)]}'
{"id":"openstack%2Ftempest~767462","triplet_id":"openstack%2Ftempest~master~I07c40119291658c3f8e823636084a8220aee70bb","project":"openstack/tempest","branch":"master","topic":"fix1218","hashtags":[],"change_id":"I07c40119291658c3f8e823636084a8220aee70bb","subject":"supplement unitest for api of resource providers","status":"ABANDONED","created":"2020-12-17 07:06:15.000000000","updated":"2022-08-26 18:53:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8392b343f5d5d7c6775db9c528dccfd136feb79e","_number":767462,"virtual_id_number":767462,"owner":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-12-18 09:37:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-17 09:56:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-21 06:08:04.000000000","updated_by":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"}],"messages":[{"id":"f643cfbe2fa96397fa0496885ec8578080dc29fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-12-17 07:06:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e94d45f23f84ab0b3490dc1c9f127926676ef2a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-17 09:56:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/61e9d5220b834983b601037d81992ac8 : SUCCESS in 59m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7322a6fa5daf489b8a6d3e3cc483c859 : SUCCESS in 1h 23m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5fdd7277f124a3792b2169f58ca1d2c : FAILURE in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f700193c28e462496eb1d32970ebe0e : FAILURE in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6d7d4a9162514a21b5f640e110728b05 : FAILURE in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bd0bbc539ec4c93ae99c6f779a50dab : FAILURE in 3m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6c685af9f714ea7a6d8a757363ad6bf : SUCCESS in 12m 23s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/eae1ce6f73574ae684800b450f4893d8 : SUCCESS in 1h 46m 12s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/bd66525a3a6d4f2f803a0e5d07729270 : SUCCESS in 1h 28m 34s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/73998f2ed1ac41abb22226a21701194c : SUCCESS in 1h 12m 33s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/a78e17fc07b3438ebd708df95730b1c2 : SUCCESS in 1h 34m 31s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/44bd7c71241e4ad1ade01d7c52d59fa4 : SUCCESS in 1h 24m 09s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/90aa2530bde845a091a5ed8240eccfd5 : SUCCESS in 1h 20m 51s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/df398903888d4ac291c9ed61481e7ff0 : SUCCESS in 1h 51m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8b2f785dcee84e249fac7051cf215560 : SUCCESS in 1h 47m 36s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/13846f3a60e8451a92b81bb39f111ed6 : SUCCESS in 49m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3ea3c76b42844d55b3325ee41b096201 : SUCCESS in 58m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/602389e8cfc74cae87d7005de9bad296 : SUCCESS in 2h 09m 25s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3b1b79f756b743b0a5c1a14709be8bc6 : SUCCESS in 50m 55s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/570a04b12fa04bb8920f34c89304e3ec : SUCCESS in 1h 18m 25s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/33f852aac64143e6ac17c32b95cf2fa1 : SUCCESS in 1h 00m 24s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/107efa6c1ed44835b6af0e46be8ea54c : SUCCESS in 1h 26m 11s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/ff642ba97a1f41de9087ba4227e6eb8a : SUCCESS in 4m 20s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ac0ea8595c3e47bb9b63ab0b39de134a : SUCCESS in 1h 42m 35s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/573df17c54d743cba0541b22e5d61a06 : FAILURE in 1h 00m 18s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d6a4049b5a5b40ac9f48579d41b0ac62 : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":1},{"id":"59717a043b4bdc2c843f014d720882661658423a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-12-18 01:29:05.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"cad1a03ac02351b3beead5deff0815511f92f2d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 04:09:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b439df816f524830a02c4130404cd9fe : SUCCESS in 1h 44m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19d4baf382c4428cb9a35275e3b8c1d8 : TIMED_OUT in 2h 10m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5c294d6dca9472cbdcfc9746e2243b8 : FAILURE in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f0efd16f5884ff69d552d1fcce82641 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f2366812e9b4843bd5baca024e30f94 : FAILURE in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49b7f12230cf4f42b0902d77da69ea7b : FAILURE in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c5c8042250e41f3b8533412242dd8a5 : SUCCESS in 12m 21s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/1658a7e4e50e45c3a87f3be5bf12a27b : TIMED_OUT in 2h 38m 13s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/b70e25b0eb944887b164d838155e984b : FAILURE in 1h 13m 21s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1af06041a9ea4a3aa12a07676f140d10 : FAILURE in 1h 08m 55s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/3e4280eddb914c7da2712a83a511f634 : SUCCESS in 1h 29m 45s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/ba5229fec81842acb7da0427682f7b83 : SUCCESS in 1h 37m 02s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/1714787e304c42e9904a04a336d41ed0 : SUCCESS in 1h 19m 14s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/2fc088f5462745eaa7499d81f0a86937 : SUCCESS in 1h 25m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dae99a7d043142fbb6934f101455be5e : SUCCESS in 1h 50m 12s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ae5877d5e4224170af3b82f260540cd5 : SUCCESS in 42m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47319edc21f64385885a55bb249ac4fc : SUCCESS in 52m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9e6c1711ec38440da66c44e9f38b2df6 : SUCCESS in 2h 00m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/db9c4e9d38544bc593e64338ac478a11 : SUCCESS in 54m 06s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/7aad7fe8ee1543068081e82e460c6cf9 : SUCCESS in 1h 04m 57s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf5b71a51bf14f43bf7c66ffe4e3e79e : SUCCESS in 1h 04m 40s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e3b8cb91a17f42f4840215dc8a049930 : FAILURE in 50m 28s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9caa1e745b784bc9acf2dd6c16245702 : SUCCESS in 2m 43s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/033f1ec387bb4e049372176c5abcae4c : FAILURE in 1h 11m 13s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/3be66f68aed544478b97245ef749e88e : SUCCESS in 1h 03m 54s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/771a8dd2d0af41378153483446dc7f7c : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":2},{"id":"c90f5b1570aef1de0d25333cb4819d3edae7bcc2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"date":"2020-12-18 06:02:51.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4062d75ca4df570fcba00bb2d128f71921ca24d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 09:37:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c420d7260990482284bb0a49a8a1a9f0 : SUCCESS in 1h 02m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a55b7cce8cf645bd941ff9d5e3b68a95 : SUCCESS in 1h 21m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b02d269e26646158f9d810c8d14a19e : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50233220fb664279936dc3b85c11b965 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f5a25d73abce4527bc5a4835c1785e48 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c640e22ce1a4b25a2882fe3d980e0cf : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba20ddedcd064f2bb39309a61614d079 : SUCCESS in 12m 29s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/877dbebb54fb4f3086b46abc5429781e : SUCCESS in 1h 51m 48s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/14cdd8a6398643a38c5f2f4f51c2e907 : SUCCESS in 1h 14m 45s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/53decdc3d0ee42038f1709d579b5a26a : SUCCESS in 1h 09m 37s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/6ecf1b3a4010457faca668c84b58a454 : SUCCESS in 1h 31m 08s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/bed395bd506f486a8eb7bc665246632b : SUCCESS in 1h 20m 26s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/5a6668aaf069472eaa2c7bc04943912a : SUCCESS in 1h 12m 05s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/7e0da18eb34942fdb4428d363632ff9e : SUCCESS in 1h 36m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/26a211a871514696b59e775c3e005d03 : SUCCESS in 1h 37m 05s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/970e4d8544cb42c2824afe658c265d32 : SUCCESS in 39m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5ee99cb65fa9444aab25cb44a50b1b58 : SUCCESS in 57m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0bda08e309c34ad780d1c2507aa1a9eb : SUCCESS in 2h 07m 24s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bb6e98a001504b4da03f34643fdf1da2 : SUCCESS in 50m 52s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0c336cfdea694de7a4872d94e77b7359 : SUCCESS in 1h 03m 59s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/89ce3ee7fbcf48699a6a80fe7856bbcb : SUCCESS in 57m 08s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f52a063a018b4f5f97b1d00079e2bed5 : SUCCESS in 1h 06m 09s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/4fa6f14e3905490cac579c4576e33f80 : SUCCESS in 2m 52s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7a0fadd08c1c40829d5fa72812097148 : SUCCESS in 1h 16m 07s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/27571c42f76349a6a9bfac42719c87f3 : SUCCESS in 1h 07m 10s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9b3608d570184276bfcd03bcb6143b3d : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":3},{"id":"8392b343f5d5d7c6775db9c528dccfd136feb79e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-08-26 18:53:37.000000000","message":"Abandoned\n\nThere is no change/response since long, we are abandoning it feel free to restore an rebase it.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7a35e569e0f7edefe4e659676fb2afaf8c55ffdf","revisions":{"57830cb8ab3564b40113a09f31ff7774340e2dc5":{"kind":"REWORK","_number":1,"created":"2020-12-17 07:06:15.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/62/767462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/62/767462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/767462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/767462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/767462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/62/767462/1"}}},"commit":{"parents":[{"commit":"0e0264636f318920e709b97c4ea6f4df1e2d27fb","subject":"Merge \"Remove slow tag from test_minbw test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e0264636f318920e709b97c4ea6f4df1e2d27fb"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-12-17 07:04:07.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-12-17 07:04:20.000000000","tz":480},"subject":"supplement unitest for api of resource providers","message":"supplement unitest for api of resource providers\n\nsupplement unitest for api of resource providers, include:\ntest_delete_resource_providers,\ntest_create_resource_provider,\ntest_update_resource_provider\n\nChange-Id: I07c40119291658c3f8e823636084a8220aee70bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/57830cb8ab3564b40113a09f31ff7774340e2dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/57830cb8ab3564b40113a09f31ff7774340e2dc5"}]},"branch":"refs/heads/master"},"1c9d0492486b2cef3e6f684b0ab44a83e284d1a4":{"kind":"REWORK","_number":2,"created":"2020-12-18 01:29:05.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/62/767462/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/62/767462/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/767462/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/767462/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/767462/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/62/767462/2"}}},"commit":{"parents":[{"commit":"0e0264636f318920e709b97c4ea6f4df1e2d27fb","subject":"Merge \"Remove slow tag from test_minbw test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e0264636f318920e709b97c4ea6f4df1e2d27fb"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-12-17 07:04:07.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-12-18 01:28:51.000000000","tz":480},"subject":"supplement unitest for api of resource providers","message":"supplement unitest for api of resource providers\n\nsupplement unitest for api of resource providers, include:\ntest_delete_resource_providers,\ntest_create_resource_provider,\ntest_update_resource_provider\n\nChange-Id: I07c40119291658c3f8e823636084a8220aee70bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1c9d0492486b2cef3e6f684b0ab44a83e284d1a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1c9d0492486b2cef3e6f684b0ab44a83e284d1a4"}]},"branch":"refs/heads/master"},"7a35e569e0f7edefe4e659676fb2afaf8c55ffdf":{"kind":"REWORK","_number":3,"created":"2020-12-18 06:02:51.000000000","uploader":{"_account_id":30643,"name":"wanglbj","email":"wanglbj@inspur.com","username":"wanglbj"},"ref":"refs/changes/62/767462/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/62/767462/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/767462/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/767462/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/767462/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/62/767462/3"}}},"commit":{"parents":[{"commit":"0e0264636f318920e709b97c4ea6f4df1e2d27fb","subject":"Merge \"Remove slow tag from test_minbw test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e0264636f318920e709b97c4ea6f4df1e2d27fb"}]}],"author":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-12-17 07:04:07.000000000","tz":480},"committer":{"name":"wanglbj","email":"wanglbj@inspur.com","date":"2020-12-18 06:02:38.000000000","tz":480},"subject":"supplement unitest for api of resource providers","message":"supplement unitest for api of resource providers\n\nsupplement unitest for api of resource providers, include:\ntest_delete_resource_providers,\ntest_create_resource_provider,\ntest_update_resource_provider\n\nChange-Id: I07c40119291658c3f8e823636084a8220aee70bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7a35e569e0f7edefe4e659676fb2afaf8c55ffdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7a35e569e0f7edefe4e659676fb2afaf8c55ffdf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
