)]}'
{"id":"openstack%2Ftempest~767632","triplet_id":"openstack%2Ftempest~master~I724e0f37d4381a493217a077492046f7a4b0a2a2","project":"openstack/tempest","branch":"master","topic":"bp/tempest-scenario-manager-stable","hashtags":[],"change_id":"I724e0f37d4381a493217a077492046f7a4b0a2a2","subject":"Pass custom args to create_port through create_server","status":"MERGED","created":"2020-12-17 20:50:22.000000000","updated":"2021-01-13 03:01:34.000000000","submitted":"2021-01-13 02:59:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"767632-bp/tempest-scenario-manager-stable","meta_rev_id":"263d52ce2cc055527dca07b56b66fd1687b67ead","_number":767632,"virtual_id_number":767632,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-13 02:59:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":2,"date":"2021-01-12 14:00:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-01-13 00:29:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-01-13 00:29:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-18 01:38:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-25 08:12:15.000000000","updated_by":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"reviewer":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"state":"REVIEWER"},{"updated":"2021-01-12 14:00:25.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2021-01-13 00:29:32.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"009d7c1804919485f65e8839e97ec4894f738472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-12-17 20:50:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a06ada3ee13a72478b58980c5dcb5a8e659a03a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-18 01:38:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c79ea9e5da3b4bffa02b373c6aedfa2a : SUCCESS in 1h 13m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/140abc53187c4c61a366143ceec2f446 : SUCCESS in 1h 29m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d70685a80b3d406eb5125dc0768cc8d0 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5db158fd08fe481b8459211f5220d944 : SUCCESS in 4m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9320b132ce8945a2b763d0ad33f5f95d : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9531231d95de4b879bc9c9081c7cf684 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecb770b2de50459dbc4721f7131c0699 : SUCCESS in 11m 19s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/eca265597a9b4e55975d7b8c3ec92a7b : SUCCESS in 1h 57m 12s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/74b5a775a2e34c1ea52d1860e229dc42 : SUCCESS in 1h 21m 00s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fcc65f784b04401485293c34a97fdc6d : SUCCESS in 1h 11m 06s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/c5ad2889575e4800a1246d2a58d54053 : SUCCESS in 1h 17m 27s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/3ae02d3c060246eba5fd3465890c889c : SUCCESS in 1h 25m 28s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/7ccc0ab75bdb4f7b98d09348f5825e72 : SUCCESS in 1h 40m 56s\n- tempest-full-stein-py3 https://zuul.opendev.org/t/openstack/build/9b235e2eeff74fa9bd8b523c648d65c3 : SUCCESS in 1h 19m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/82df14b8681c4e6798f56102833c40e9 : SUCCESS in 1h 39m 59s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b666390fa6b44272872fa199a7cf9e0d : SUCCESS in 40m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fd24c5023d1c4ab1a15fadc2b0542d52 : SUCCESS in 57m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/124d22bdd2404c0b8d208b458ab1943b : SUCCESS in 2h 13m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e704edfe934e491fac2108321076cd16 : SUCCESS in 54m 55s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/7fb0a2e6c92e49bcbfe824113a5c05e5 : SUCCESS in 1h 23m 15s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/e38e789f047746c4a0ce9e7b223c98ca : SUCCESS in 58m 02s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/0f8353f9c6e149a393f59496bfdf64eb : FAILURE in 45m 25s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/16ed2826532b4f70b59fe08a8e08d252 : SUCCESS in 3m 29s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/83bc370376264b14948745a3cf02ddef : SUCCESS in 1h 42m 07s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/a01102f4955a4390aa9647a8a6fe580e : FAILURE in 57m 20s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/be5b530ca84b403db9a0048edff97f55 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"b54763dac6406b68021067b42ba0f3ff37cd5fda","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-12-18 13:25:49.000000000","message":"Topic set to bp/tempest-scenario-manager-stable","accounts_in_message":[],"_revision_number":1},{"id":"e64f435816a3b0337aedb3523ca0615a1e3b4eb6","author":{"_account_id":20190,"name":"zhufl","email":"zhu.fanglei@zte.com.cn","username":"zhufl"},"date":"2020-12-25 08:12:15.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d32a7846dbd2a18f679af46266f2bc6a42f6b789","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-12-28 16:39:33.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3de460ba97803c7d041d11b4fad4ef4c31f2295d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2020-12-28 16:39:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d2c12a65b09a8c1859ac790be3c46e38a86a69d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-28 18:28:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3506f2d66f884f33976c60052af2d4c0 : SUCCESS in 50m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5e231aeafb154e5d809476995144452a : SUCCESS in 1h 28m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf635e528dc54686b90a16da0ad62bba : FAILURE in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08f14ff65216466abdf1ec5a9dfa6c17 : SUCCESS in 3m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b91a4e629ace429bb3969609cb4c8053 : FAILURE in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73b9f59984124bbda3b7ea68965cf01a : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fc9ab7d47e44bb79949f5bdeab1ebbd : SUCCESS in 15m 51s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/13194dfd0c484c4e98d1f76b5808e95d : FAILURE in 1h 21m 38s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/4ad64f27894d4155b90c5af1cd7db119 : SUCCESS in 1h 03m 48s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/065cb9bf9983422fb956deeb643feb3d : FAILURE in 1h 13m 44s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/986211fff34046aa9ac57ef7a16b98f2 : SUCCESS in 56m 48s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/1ca1acf632c04c5a8c2c9dfe340fe365 : SUCCESS in 1h 10m 56s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/312a90733e124d34a8707b0499b44bfe : SUCCESS in 55m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/28ef8ad2cff64d8fadab27186c054d64 : SUCCESS in 1h 44m 03s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/7fb7351822d845e09bb24b77f1a0d14b : SUCCESS in 28m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe8e84c353b1473db10bda5ccdede386 : SUCCESS in 50m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5a28c3af42134dab9438f99881db7beb : SUCCESS in 1h 36m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/20adecd1a73f4816a6c768ef5064e9e5 : SUCCESS in 41m 03s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/89844c0b1ba9426aa21f0e2d26e5280b : SUCCESS in 1h 07m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/b9b4744a427d41b8b7e9f0109b771054 : SUCCESS in 53m 17s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/6c31dcfe8bf74924943bf790c4188f01 : SUCCESS in 51m 43s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/2131af157bad4cafa26e90343f6119cb : SUCCESS in 2m 41s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c4faca1b86424769b00af5fbd7201ac6 : SUCCESS in 59m 20s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/457a1b866a2a4a6f86cc641e559bf85f : FAILURE in 54m 42s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/bc0906411967479d97411002394e5926 : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"29875a5dd9371b8fd2bc408a82834552edadab24","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-01-02 16:19:00.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"2bfa405acb393a9d01eee8fd8050181490617102","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-02 18:11:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/621e056cff7f4d5f9457f28815aba489 : SUCCESS in 55m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c97a93d218d745ac99d4895d6b32594f : SUCCESS in 1h 18m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/80ed6e3507c945a8aea8523d1c97ae3d : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37e3a3d8a54142c2b2af03b21ad69353 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0bf3a1932bd403aa63598ce064d5025 : SUCCESS in 3m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38580e8743a8485aabd05a312a43e8d2 : SUCCESS in 3m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/778519f483e7457bbd319903069d365e : SUCCESS in 11m 44s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/eea8f9875311462892ea9084fa5fa4a2 : SUCCESS in 1h 38m 16s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/935f3ad5ca504c84870b4dc958b8dbc4 : SUCCESS in 1h 10m 08s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/604f18fe3865472f87aa587ca5b69a03 : SUCCESS in 1h 16m 53s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/f0bfb73c0db445f890abd28930cfc99e : SUCCESS in 1h 30m 59s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/a8e2437596c64c5fb640e834e61d7fb7 : SUCCESS in 1h 06m 10s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/6d53940075b442b3bc8f5b89536f0018 : SUCCESS in 1h 13m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/05458c4719c145cfad7d25e69d30c7c0 : SUCCESS in 1h 40m 17s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/78dd1f2f6e23476092320651d2ff26b3 : SUCCESS in 38m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/de4ad56ee92243a787a55e9012bbe336 : SUCCESS in 49m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df10359218ab4c6fbdcb9232760e7dd0 : SUCCESS in 1h 51m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9f6221f372654bfa96e3dc4988bbc379 : SUCCESS in 50m 58s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ccb562c6d80e4439bc6379bbca5b9854 : SUCCESS in 53m 31s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9e6450388f7a4552b9b08f6c68405275 : SUCCESS in 47m 27s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/f86a386151d44f9aa3224e984c8f1724 : SUCCESS in 1h 06m 27s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f44575a530434efbaed358ad03980267 : SUCCESS in 2m 22s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/d17cf8e578e54178a24a314ac1a1960a : FAILURE in 1h 03m 00s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/58b01541b82f488f9344cbe162b6d820 : SUCCESS in 59m 00s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/35311dac04d949cb92fe0651e12ef3c4 : SUCCESS in 2m 34s","accounts_in_message":[],"_revision_number":2},{"id":"7cb033cbdb50ed16e1267e73e4bac038d2ec08da","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2021-01-12 14:00:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8c5e8d2f50e387aae12e9415ed0bf886ebd2cc27","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-01-13 00:29:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9777647c4c7caa90ef58ce1c4efbe1dc0936ea36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 00:29:46.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0f4f45c8815b8c25e8354956cad4768bfa464e7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 02:59:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/abde3bb1b30543059e73cb4e3a778ffa : SUCCESS in 53m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7970a3fa2f1e40f7a82616f3e927f6fb : SUCCESS in 1h 40m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf862a2e6c06463f96428a288355debe : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aac3543bd0fd44c0a938326732f3b12f : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7e5deaadd334ad7afabeba9c10186de : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f38cc46695049f5aff795b01ec15c1c : SUCCESS in 12m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f4f8c0f3c2dc460dbe372cddcf41cd2c : SUCCESS in 2h 23m 02s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a8102e089314e21bf2016fb90eb164e : SUCCESS in 1h 01m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/37e815bd1a5c4fc7b2aa659d817cd399 : SUCCESS in 1h 02m 39s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/aa8ec8d5967e4de2897149437ca6f06e : SUCCESS in 1h 09m 01s","accounts_in_message":[],"_revision_number":2},{"id":"d90aba0356040027ca28d9ac78472982948e5db0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 02:59:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"263d52ce2cc055527dca07b56b66fd1687b67ead","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 03:01:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc75a4e175814d7c873af3eaea3e4186 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9c874417240ba8efbd6933bfab66be09f24e1c3e","revisions":{"0758983214310effe450098afb7c8c4465ab3e33":{"kind":"REWORK","_number":1,"created":"2020-12-17 20:50:22.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/32/767632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/32/767632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/32/767632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/32/767632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/32/767632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/32/767632/1"}}},"commit":{"parents":[{"commit":"0e0264636f318920e709b97c4ea6f4df1e2d27fb","subject":"Merge \"Remove slow tag from test_minbw test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e0264636f318920e709b97c4ea6f4df1e2d27fb"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-12-17 20:43:26.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-12-17 20:49:02.000000000","tz":0},"subject":"Pass custom args to create_port through create_server","message":"Pass custom args to create_port through create_server\n\nAllow consumers of create_server to pass custom arguments to\ncreate_port method which is called from create_server one.\n\nFix namestart var usage in create_port - as the name of the var\nsuggests, it should be used to generate a name from. Therefore\nthe patch also adds rand_name call which we accidentaly removed\nrecently.\n\nImplements: blueprint tempest-scenario-manager-stable\nChange-Id: I724e0f37d4381a493217a077492046f7a4b0a2a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0758983214310effe450098afb7c8c4465ab3e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0758983214310effe450098afb7c8c4465ab3e33"}]},"branch":"refs/heads/master"},"9c874417240ba8efbd6933bfab66be09f24e1c3e":{"kind":"REWORK","_number":2,"created":"2020-12-28 16:39:40.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/32/767632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/32/767632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/32/767632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/32/767632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/32/767632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/32/767632/2"}}},"commit":{"parents":[{"commit":"0e0264636f318920e709b97c4ea6f4df1e2d27fb","subject":"Merge \"Remove slow tag from test_minbw test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0e0264636f318920e709b97c4ea6f4df1e2d27fb"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-12-17 20:43:26.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2020-12-28 16:37:53.000000000","tz":0},"subject":"Pass custom args to create_port through create_server","message":"Pass custom args to create_port through create_server\n\nAllow consumers of create_server to pass custom arguments to\ncreate_port method which is called from create_server one.\n\nFix namestart var usage in create_port - as the name of the var\nsuggests, it should be used to generate a name from. Therefore\nthe patch also adds rand_name call which we accidentaly removed\nrecently.\n\nImplements: blueprint tempest-scenario-manager-stable\nChange-Id: I724e0f37d4381a493217a077492046f7a4b0a2a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9c874417240ba8efbd6933bfab66be09f24e1c3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9c874417240ba8efbd6933bfab66be09f24e1c3e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
