)]}'
{"id":"openstack%2Ftempest~771958","triplet_id":"openstack%2Ftempest~master~Iee0d863770def63df8601222265dc22226457c0c","project":"openstack/tempest","branch":"master","topic":"bug/1912607","hashtags":[],"change_id":"Iee0d863770def63df8601222265dc22226457c0c","subject":"test_encrypted_cinder_volumes: Do not create an image for this test","status":"MERGED","created":"2021-01-22 10:14:35.000000000","updated":"2021-01-26 16:28:27.000000000","submitted":"2021-01-26 13:23:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"771958-bug/1912607","meta_rev_id":"4bb5dd6cb0d02d5ea720c53aff1dbd0a1c48d439","_number":771958,"virtual_id_number":771958,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-26 13:23:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"all":[{"value":2,"date":"2021-01-26 06:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":2,"date":"2021-01-26 09:10:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":1,"date":"2021-01-26 09:10:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-22 13:21:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 06:33:06.000000000","updated_by":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2021-01-26 09:10:13.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-01-26 16:05:43.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"}],"messages":[{"id":"32eff103a0b4098aa41a238726c0531193356e4c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-01-22 10:14:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fad65a65bd5de7925a073b919239c2c0f79354ae","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-01-22 10:16:35.000000000","message":"Topic set to bug/1912607","accounts_in_message":[],"_revision_number":1},{"id":"7e7c20a6b0452f2830080ec2b4e32280d45dc0c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 13:21:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/3e969ed5375041a6aaff87fb2f25c272 : SUCCESS in 54m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c5a5d2537ad47748919c69edfa963fd : SUCCESS in 1h 18m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/552a6462e87e4523b67a1f0c81cb0d13 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2e9c1110f54432d91a5b272aba7cabd : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4878609ecaa24adaa3a5d477df377d9f : FAILURE in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e96e7162ddf14c09816a08fdc7592cfb : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/024b95fe5930488a9a05202afb5577f2 : SUCCESS in 14m 31s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/8ef64304dee54c9fb9d87905613afacc : SUCCESS in 1h 49m 06s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/e3eaa83eefc6462ca563c71e2b640a83 : SUCCESS in 1h 28m 20s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c70b7f6dfe37422cad219b3bdf5cc928 : SUCCESS in 1h 28m 36s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/a71a06a38ea5409db47dfb81cd401aac : SUCCESS in 1h 24m 40s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/afd170d981c442b39b15a8d0af08a85b : SUCCESS in 1h 25m 22s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/484cb266e0fc4febb1c6c2eafb1d8b99 : SUCCESS in 1h 22m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6bdfa1b9ae3b4ab7b8dd6c81b818b1ad : SUCCESS in 1h 43m 31s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b48f14f0ef5a4c9a9c2cea6798ca8b4b : SUCCESS in 48m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0d7b0646d464c13a05666f7ff62942f : SUCCESS in 59m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2105bdcd39b74739b453466d14c64bd8 : SUCCESS in 2h 01m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/60bb20992b0b4347ab44d58efe02d437 : SUCCESS in 51m 10s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/eaa7d289fdb8474fbb421496b3a22e15 : SUCCESS in 1h 08m 08s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/021effbd22d4447a92309158ea565e65 : SUCCESS in 1h 24m 21s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/72e364ee7d764600ba33ee57c0672fc9 : SUCCESS in 1h 08m 54s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/073e30908d494c2d84bf91c52c0003e3 : SUCCESS in 3m 09s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a3b88b5a031f4265b5cdd6369afcde94 : SUCCESS in 1h 12m 19s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/0ef9e27d1bbb4fffb5cdec341eb85228 : SUCCESS in 1h 02m 25s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/da7b452f53ca41fa97d8c4e4c22a3ba3 : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":1},{"id":"c4adb8a12368a26228c045550afe399ecbc4064e","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-01-22 14:09:14.000000000","message":"Patch Set 1:\n\nrecheck\n\nft1.2: tempest.tests.lib.cmd.test_check_uuid.TestCLInterface.test_fix_argument_yestesttools.testresult.real._StringException: Traceback (most recent call last):\n  File \"/home/zuul/src/opendev.org/openstack/tempest/tempest/tests/lib/cmd/test_check_uuid.py\", line 60, in test_fix_argument_yes\n    check_uuid.run()\n  File \"/home/zuul/src/opendev.org/openstack/tempest/tempest/lib/cmd/check_uuid.py\", line 436, in run\n    pkg \u003d importlib.import_module(args.package)\n  File \"/usr/lib/python3.6/importlib/__init__.py\", line 126, in import_module\n    return _bootstrap._gcd_import(name[level:], package, level)\n  File \"\u003cfrozen importlib._bootstrap\u003e\", line 994, in _gcd_import\n  File \"\u003cfrozen importlib._bootstrap\u003e\", line 971, in _find_and_load\n  File \"\u003cfrozen importlib._bootstrap\u003e\", line 953, in _find_and_load_unlocked\nModuleNotFoundError: No module named \u0027tmpvmzck0zp\u0027","accounts_in_message":[],"_revision_number":1},{"id":"83d363c85979b712ffbb6639278a5ffe7c305347","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-22 16:34:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/e32faaef54924dd3978334dd942b0e41 : SUCCESS in 57m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c9e5403c815843898d74d6571646803e : SUCCESS in 1h 08m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0ae9cea6c336437c859436b7db4fedfc : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36a1c47ba784506a90f13944f2c85e1 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86e3a723c839415180cb7caedf5a835d : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c0a20d4679da4f608aae6a2c01d22887 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7858a6032573400ab0d9d8a2a027e1e0 : SUCCESS in 18m 32s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/40d8ef59e18846ef8894c0d6430bd030 : SUCCESS in 2h 18m 06s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/b86761a8f81b4c8daa574dc7991dbbce : SUCCESS in 1h 10m 36s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/903b7235526848c6bd9b6e946620fb14 : TIMED_OUT in 2h 09m 00s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/ff4bb4de17d04582b22b2cb460f96e73 : SUCCESS in 1h 27m 02s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/f020d4ddb6fd4333a95e9c30f184028c : SUCCESS in 1h 14m 31s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/fcd9cd477d9541f8b4019b8a76fb58d5 : SUCCESS in 1h 25m 04s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5fa850a5131d4cd5866d87ce0c62e5bf : SUCCESS in 1h 41m 14s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/765baa1cb7634a8a9ffb0c42d6f0b50b : SUCCESS in 39m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9b176fb546a4f1e88f8fb010f6c4148 : SUCCESS in 51m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c11765d83189436b91544e756f13350d : SUCCESS in 2h 01m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f37b71f55d62411d9d363eccf56debf3 : SUCCESS in 58m 08s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d3932eb542ed4e5f8de8fd75d1b3609e : SUCCESS in 1h 04m 28s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/7414f95c7624416aa4273e8259bc79c1 : SUCCESS in 1h 05m 22s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/e3d537e7f30c40d6afb33b0546fc5e5e : SUCCESS in 54m 29s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/05963396988843fcb061c82abea0f3a3 : SUCCESS in 2m 39s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/deb5ae9021c8493bb35a726f36cacf44 : RETRY_LIMIT in 20m 43s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/8a448b4404f140bea3b84651349db070 : SUCCESS in 1h 00m 34s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/6679c65b15f74df5894414980b6c7719 : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"1690da6e2e854973773b2fbbe52b9c4252a82065","author":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"date":"2021-01-26 06:33:06.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM. I agree with Lee. I think we don\u0027t need to create an image for this test because testing encrypted cinder volume is the main purpose of this.","accounts_in_message":[],"_revision_number":1},{"id":"7cbc757a98d6b3ae8fdcdb2117c96c4ef29c1e14","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-01-26 09:10:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nmakes sense, if the test doesn\u0027t require a unique image, better to simplify the test by reusing the creation code.","accounts_in_message":[],"_revision_number":1},{"id":"2c4ca49405a45bbc3fe3a957bd8131415594f15c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 09:10:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"53f8b76999e085eed66b7097f952aec51d501695","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 13:23:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/bc909af605384c659c07ae9647a8c85d : SUCCESS in 1h 06m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/29bcb15e8bdd4ffd9a4318f9012a2a57 : SUCCESS in 1h 22m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3806a76ff2c4c72a06c399af6ebe269 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/867e69a8e681432ba329dcdf8c783655 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c785d42c61d4466c80b8864d09aa3138 : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e50e8fe897684267a776d6b5e490c8b0 : SUCCESS in 12m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1dd0220d81884423a3860337eb8de8f3 : SUCCESS in 2h 16m 34s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/17b39baf108e47f6a225eca96ac40422 : SUCCESS in 53m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22c02a2a63b4419595f493a49dd4e5a2 : SUCCESS in 46m 56s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/aea00e6866cf4231a2263592bb519705 : SUCCESS in 1h 14m 24s","accounts_in_message":[],"_revision_number":1},{"id":"e72f843545237b77d68a10a122ba50b9bb75ab2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 13:23:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"f05c3686b70a65e55e7c63ae41dc16ad7e242fcb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 13:26:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b65e49d4bc7543e3923f4a870e141d88 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":1},{"id":"09a82b78f22f7bfe79ddb973e384ff3d699785f4","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-26 16:05:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4bb5dd6cb0d02d5ea720c53aff1dbd0a1c48d439","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2021-01-26 16:28:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26186cf964be0bdeef82267de13e7257c413bafa","revisions":{"26186cf964be0bdeef82267de13e7257c413bafa":{"kind":"REWORK","_number":1,"created":"2021-01-22 10:14:35.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/58/771958/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/58/771958/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/771958/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/771958/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/771958/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/58/771958/1"}}},"commit":{"parents":[{"commit":"cd147631f8254c2f727ccfc937566a7a200349f9","subject":"Merge \"Add system scope for admin auth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cd147631f8254c2f727ccfc937566a7a200349f9"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-01-22 10:10:32.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-01-22 10:10:32.000000000","tz":0},"subject":"test_encrypted_cinder_volumes: Do not create an image for this test","message":"test_encrypted_cinder_volumes: Do not create an image for this test\n\nAs with I50d3b874b7e0676436c434be6c059f221041560f for the\ncinder-tempest-plugin this encrypted volume test does not require a\nunique image to create an instance. The creation leading to test\nfailures in the nova-ceph-multistore job that sets\n[workarounds]/never_download_image_if_on_rbd while valdiating some edge\nusecases.\n\nThis change simply removes the creation of the image and relys\ninstead on the behaviour of tempest.common.compute.create_test_server\nthat uses CONF.compute.image_ref when an image_id is not provided [1].\n\n[1] https://opendev.org/openstack/tempest/src/branch/master/tempest/common/compute.py#L108-L109\n\nChange-Id: Iee0d863770def63df8601222265dc22226457c0c\nRelated-Bug: #1912607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/26186cf964be0bdeef82267de13e7257c413bafa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/26186cf964be0bdeef82267de13e7257c413bafa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
