)]}'
{"id":"openstack%2Ftempest~777351","triplet_id":"openstack%2Ftempest~master~Ia102acea773bd80d933a7e0a19b447f91134f81f","project":"openstack/tempest","branch":"master","topic":"bp/tempest-scenario-manager-stable","hashtags":[],"change_id":"Ia102acea773bd80d933a7e0a19b447f91134f81f","subject":"Make _get_network_by_name() non-private","status":"MERGED","created":"2021-02-24 10:18:38.000000000","updated":"2021-03-02 16:15:13.000000000","submitted":"2021-03-02 16:12:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"777351-bp/tempest-scenario-manager-stable","meta_rev_id":"4646d2cdb331a5c1eb474773b1f3d5ad8ac135a3","_number":777351,"virtual_id_number":777351,"owner":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-02 16:12:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2021-03-02 14:08:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2021-03-02 14:08:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"reviewers":{"REVIEWER":[{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-24 10:19:42.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":5689,"name":"Masayuki Igawa","email":"masayuki@igawa.io","username":"igawa"},"state":"REVIEWER"},{"updated":"2021-02-24 10:19:42.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2021-02-24 10:19:42.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":9592,"name":"wes hayutin","email":"whayutin@redhat.com","username":"weshayutin"},"state":"REVIEWER"},{"updated":"2021-02-24 10:19:42.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-02-24 10:19:42.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"REVIEWER"},{"updated":"2021-02-24 10:19:42.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-02-24 13:53:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"724176eed26c97a0eea169024ed1e79c58d34580","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-02-24 10:18:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"675b9138a2efd8e9401b6d4f47aec947c0e29948","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-24 13:53:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c13a5332a88f48d6819e3ef588bb47cc : SUCCESS in 1h 00m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d4f22be64b343bc853737066586d78d : SUCCESS in 1h 56m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25828ef880ff4587b6f17178864f19e1 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b3bed4d586e4b909984425875aee1bd : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/38c20d1514cd4d1dad4beca100229647 : SUCCESS in 8m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2084d1eed539464d9b17deccc02280da : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fbc7612ab5f4c2a8390ccfcccf8b870 : SUCCESS in 12m 49s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f5c0b3ad95ba458f9fa0e3833f63d7b7 : SUCCESS in 1h 43m 08s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/064c60c205cb4ba4a6a1bf8e8e051675 : SUCCESS in 1h 41m 38s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bce65c9c50924acdb6e017ffb7bcdb54 : FAILURE in 1h 12m 51s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/09c60ea5444141b5a79f200156958677 : SUCCESS in 1h 27m 23s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/105766028d0548aba66f4df7afc84324 : SUCCESS in 1h 34m 22s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/4af55813be524c39b78e461868d535d6 : SUCCESS in 1h 28m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/84d3c8bac22846fd9ac7336522cda9f5 : SUCCESS in 1h 47m 26s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/7fbc6ddcc83640498ef01df41a8bdc79 : SUCCESS in 49m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/357cf9cd776145f1b49940031f1b0ba9 : SUCCESS in 1h 03m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/17561a9a91b64e04a2582f2318ed410c : SUCCESS in 2h 08m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/75bac9957ca94a63854d7a455cd08fb9 : SUCCESS in 49m 55s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/baebd154aa2b475883ba3de38d9f2db5 : SUCCESS in 1h 20m 06s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/23ae5ffb7bfb48478eb2203bd40d3908 : SUCCESS in 1h 04m 38s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/94cc334af2f942048e37e4982e246d6a : SUCCESS in 1h 03m 52s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f7adc4f81d2045fe96fd7f5976c40818 : SUCCESS in 4m 51s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/021f524e5f8b48c7aaa10780466714d6 : SUCCESS in 1h 24m 47s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/a8fa2aa60c644f3db37df27720d638dc : FAILURE in 1h 05m 27s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ba8c517b9cdf491baa8f0a7c5cd67259 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"f38f207b33a7edcfdc19654640f64c5dfe702e65","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-02-24 16:17:08.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nwe do not run with CONF.shared_physical_network otherwise it could fail.","accounts_in_message":[],"_revision_number":1},{"id":"f73f4a78a2d81b151d2c8855366154a08055464f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-02-25 07:29:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"229b62fe6c12fa68981df16ade5fe8cf866036a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-25 09:34:51.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ffb2b3dc47e04fb69a597916abb2a35d : SUCCESS in 51m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/823de430a1594ccba508006230bfccbc : FAILURE in 1h 17m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f1395f8c1d749bdad5cbad71bd982d0 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/711169e4b0cc4387a33ff15f8d86bb8f : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1e8dcade3ad449f98f1d9c7a384a8c7 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d52cee6d7d049a0b38fac3fec09afd7 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8925185ac5a34d15b430a76a124c8897 : SUCCESS in 13m 44s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c921f8ac252e4bd2bce87c1ecc8120e4 : SUCCESS in 1h 32m 20s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/4d27aabac667419bab9e55531423b4b5 : SUCCESS in 1h 16m 26s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1bd3814ac7f44089856a9eb491a9f10c : SUCCESS in 1h 18m 38s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/a23c8d24ef88437ba466b07acc814b42 : SUCCESS in 1h 47m 22s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/f7713f34fce943a1aa2cdcdfeb5984fb : SUCCESS in 1h 22m 41s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/f63b6a7ff2634eb7b59ebf91a4318bbf : SUCCESS in 1h 42m 02s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d62745356aa6440899f05552c27a9dcf : SUCCESS in 1h 55m 42s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/18ebc22f17824d6bb255cd65a221ab5a : SUCCESS in 29m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/933839c87a8d4df1a328c23b03f65ff1 : SUCCESS in 46m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4215e2433db446f787537792ceef2c05 : SUCCESS in 1h 39m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6ae021d6104f418e80514139e8b98a9b : SUCCESS in 51m 25s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/3fdcdfb8e7e94f9eb4dd3407cee3de40 : SUCCESS in 1h 18m 18s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a1eae79743264d0783a004c977453a93 : SUCCESS in 58m 34s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/52a4035e10b848b9b11ff4132cbc5243 : SUCCESS in 1h 24m 23s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/57f7da25c0d74698a29b6392d64afb76 : SUCCESS in 3m 28s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ac8aca693efc479cb5ea4643010a8402 : FAILURE in 1h 36m 35s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/7d303aa6bc8842ccbe981bd2345be539 : FAILURE in 1h 06m 00s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/610b418d6355417ea84dbfc6133f9b9f : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":2},{"id":"5e61ebed231ef971f36011c8af65cccb7e01281b","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-03-01 07:48:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7d716ea5bf64fd5a483f8d3b4e93f1de4f361791","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-01 10:39:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c5cc6d62b9f24aabaa5de59cd4720162 : SUCCESS in 57m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a8376d08cf454611951c24a135cea4a4 : SUCCESS in 1h 32m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3dc759a0f8684641a2f4b6b521c10191 : SUCCESS in 7m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20e5635a068547268d21c9cc6d474514 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c9c0aef9c47246709c78583e3846ce54 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/15f2a3c1211b487d8034ddb1cb8fce58 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0f652cf9ce5456aaff63e13e2884feb : SUCCESS in 12m 36s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/055cc1cea844416dbb1bcb05c026bc84 : SUCCESS in 2h 06m 54s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/4e5d12ae00d14c679b483fce4773f9b0 : SUCCESS in 1h 06m 00s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/29cdd9e9b60041c181eaab4a5362a340 : SUCCESS in 1h 16m 57s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/b702ded394ea4ab3aa1fe503ab53f4cc : SUCCESS in 1h 42m 53s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/baaa2fe8b90945d1997fae41f2889b9f : SUCCESS in 1h 24m 52s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/ed3f0a3835d947b7b576f431fe7ea019 : SUCCESS in 1h 23m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ed3d76751de34843a68b77a42541c8c6 : SUCCESS in 2h 44m 59s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/9e6b7757a03a4b90bcb29f30c4338c8a : SUCCESS in 45m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0b76bf31fde4879ada865cf9843e900 : SUCCESS in 1h 30m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9779349447404fc29acf5b0c4574affc : SUCCESS in 2h 02m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cf154da32d904cb1858439d9d37eb11f : FAILURE in 1h 19m 17s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/64714ffee6594eb794e6e7a8da8596c9 : SUCCESS in 1h 18m 49s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/9ac5b42790aa4a118283a51aaed16e86 : SUCCESS in 55m 09s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5bc922f63b084987bb118f357aa3959c : SUCCESS in 59m 41s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/3153d78d9a1740bda5aa2f16c3b3ebc9 : SUCCESS in 3m 15s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ac03f390ef7a4710839efe7f7695eefd : FAILURE in 1h 43m 56s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/29ff6535653d487d9b4656e6a55422db : FAILURE in 1h 05m 00s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e636d2b4c0624961b170113ec7ca3f03 : SUCCESS in 4m 03s","accounts_in_message":[],"_revision_number":2},{"id":"440352a810dfa89f2d61d5c1a3eebc687a186d81","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-02 14:08:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"4b55ef6773d05d2e1bab9c9031c6e2865ed6b9d9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 14:08:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e8a463228c781db98e7e2ce537a80148070ebe48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 16:12:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/83700d35fae7482a80a38dba5945419d : SUCCESS in 56m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b64c87327b394356b949f5092ad92542 : SUCCESS in 1h 31m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3e95643d0cd498f91b3b0470bb6e13b : SUCCESS in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7f2307171ff04d72ba679814e4f806bc : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe0ba819203545a0b917dcfd047e0fbc : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/907609dea7c745d6b80bb9710016001d : SUCCESS in 13m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f8697f0806fa4d7aa4c8864b5470568f : SUCCESS in 2h 02m 53s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/dafd105d793a45ab814f37ff23a1eeda : SUCCESS in 58m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29eac2fa0a4f49cfb442cfff05103e35 : SUCCESS in 1h 11m 27s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/21631694bcac4ae49ede6b9d03073539 : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":2},{"id":"0c073957683a289cf423258b0e8a7574d3fbe502","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 16:12:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"4646d2cdb331a5c1eb474773b1f3d5ad8ac135a3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 16:15:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c51ef2a9d9204ef6b3621f626c34d0cc : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c37410fc14477a9d63eb2314a6d994cb0def764a","revisions":{"ba9fc3a4f696414f0fded2f8cbbd1ae1ed9df0e0":{"kind":"REWORK","_number":1,"created":"2021-02-24 10:18:38.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/51/777351/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/51/777351/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/777351/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/777351/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/777351/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/51/777351/1"}}},"commit":{"parents":[{"commit":"042eed72d2a237df848fe2e653616ce87f32319c","subject":"Merge \"Enable devstack parallel installation in tempest-full-parallel job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/042eed72d2a237df848fe2e653616ce87f32319c"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-02-24 09:56:27.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-02-24 09:56:27.000000000","tz":330},"subject":"Make _get_network_by_name() non-private","message":"Make _get_network_by_name() non-private\n\n_get_network_by_name() is made public by this commit as\nthe tempest.scenario.manager interface is meant to be\nconsumed by tempest plugins.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\n\nChange-Id: Ia102acea773bd80d933a7e0a19b447f91134f81f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ba9fc3a4f696414f0fded2f8cbbd1ae1ed9df0e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ba9fc3a4f696414f0fded2f8cbbd1ae1ed9df0e0"}]},"branch":"refs/heads/master"},"c37410fc14477a9d63eb2314a6d994cb0def764a":{"kind":"REWORK","_number":2,"created":"2021-02-25 07:29:31.000000000","uploader":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"ref":"refs/changes/51/777351/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/51/777351/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/777351/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/777351/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/777351/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/51/777351/2"}}},"commit":{"parents":[{"commit":"042eed72d2a237df848fe2e653616ce87f32319c","subject":"Merge \"Enable devstack parallel installation in tempest-full-parallel job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/042eed72d2a237df848fe2e653616ce87f32319c"}]}],"author":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-02-24 09:56:27.000000000","tz":330},"committer":{"name":"Soniya Vyas","email":"svyas@redhat.com","date":"2021-02-25 07:23:35.000000000","tz":330},"subject":"Make _get_network_by_name() non-private","message":"Make _get_network_by_name() non-private\n\n_get_network_by_name() is made public by this commit as\nthe tempest.scenario.manager interface is meant to be\nconsumed by tempest plugins.\n\nImplements: blueprint tempest-scenario-manager-stable\nSigned-off by: Soniya Vyas\u003csvyas@redhat.com\u003e\n\nChange-Id: Ia102acea773bd80d933a7e0a19b447f91134f81f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c37410fc14477a9d63eb2314a6d994cb0def764a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c37410fc14477a9d63eb2314a6d994cb0def764a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
