)]}'
{"id":"openstack%2Ftempest~777684","triplet_id":"openstack%2Ftempest~master~Ib301466bb0a1996997c63599918fa96059a927bd","project":"openstack/tempest","branch":"master","topic":"replace-md5-for-fips","hashtags":[],"change_id":"Ib301466bb0a1996997c63599918fa96059a927bd","subject":"Replace md5 with oslo version","status":"MERGED","created":"2021-02-25 21:14:28.000000000","updated":"2021-03-11 06:59:53.000000000","submitted":"2021-03-11 06:57:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"777684-replace-md5-for-fips","meta_rev_id":"78ae67cb8b5ddaa21209bfdfb3106394449fc3f3","_number":777684,"virtual_id_number":777684,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-03-11 06:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"recommended":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"all":[{"value":1,"date":"2021-03-04 08:50:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":2,"date":"2021-03-04 18:57:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2021-03-10 17:16:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2021-03-10 17:16:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-26 01:51:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-03 16:28:37.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2021-03-04 08:50:33.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"},{"updated":"2021-03-04 18:57:28.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-03-04 18:57:37.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"486842a493dca0f741a9bad4f6a51443da457709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-02-25 21:14:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8ac7cd784e75243ba0fc44ad138248d3e45cbdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 01:51:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/88713ebb95434988b958120645092cfd : SUCCESS in 1h 14m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9274421b1c374eada5a34e8e08cbd9b2 : FAILURE in 1h 23m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6392691cfedb4236ba392ca4ec76e7dc : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d8aa86e4d5d4d6f83a9133a773acb8e : SUCCESS in 6m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6ea1e102f7e7455b8dbf401d9f979256 : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01ffc51bb0fb4ea58ffce71c0bcd219e : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342e9ab24c354b71b373409a406c0a46 : SUCCESS in 10m 57s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/ec3fb2496ff343cd80afe2f42ccb4632 : SUCCESS in 2h 22m 28s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/94d4b71ad17e46aab45b60db6ce6328c : SUCCESS in 1h 42m 13s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8a453e392f6047e99eac1e68b548c5c1 : SUCCESS in 1h 50m 54s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/59c7a940fe0e4cda82d13d952d22ea42 : SUCCESS in 1h 23m 58s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/048669d9d1614a33bdf0268b134ab4d0 : SUCCESS in 1h 16m 00s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/1d88b13ccb45440aac424c1c5f5e51d5 : SUCCESS in 1h 22m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/44ba546b667a42958409c715e8e9d246 : SUCCESS in 1h 21m 26s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/34b8af43ab6447eb8304fd7b710b5473 : SUCCESS in 39m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6906e790cb2049e8b6e9cc6ae4184f7e : SUCCESS in 1h 05m 52s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ba9ea18ced84a78b5277ae325b03e59 : SUCCESS in 2h 03m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d19fee526203431ca1043be8d22a4dd1 : SUCCESS in 52m 57s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e9c9dec21097438188452d9c491c4489 : SUCCESS in 1h 25m 05s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/1e490e32ee924a8eb721fd0818a50bcd : SUCCESS in 1h 01m 13s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/618d10ea803041f692012576cf8fce35 : SUCCESS in 58m 01s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/5a94b01bd61d499baf26e468b3de96af : SUCCESS in 3m 16s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/34cbebee0ca742c0a2490fda343453cc : FAILURE in 1h 14m 23s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/d0fff25ff2f247dbbc1e4778459e15b6 : FAILURE in 1h 09m 50s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ce4f3f920e2847d9b92f9f96a2c68948 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0c50fe962ea12da1599da5e3caf93d641e2b1f00","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-03-03 16:28:37.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ae24c972006f48654be5e08c42574abe65ac3d2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 21:58:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7ea466ddceb5427791ab441c8cfd78d1 : SUCCESS in 1h 11m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a516cc2ce254477974050c52ffb11a8 : SUCCESS in 1h 23m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/000ea77fb1ce4c70b41a7f633ae67390 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3475e3bd01c74135b3b9aa4304892437 : SUCCESS in 6m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc800236937241cca10a2b9eb0a5c942 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e54593fd211c45f9875590f221cd646e : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d70fefbd1db47728a055b54215f7ddf : SUCCESS in 12m 46s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/70f1e9d851be428094665025f166caa7 : SUCCESS in 1h 52m 29s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/4b51a7382a1d4ef7864dd3ec2a41cb25 : SUCCESS in 1h 23m 11s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e780de4f0bbe4f9aa90f90d33185e989 : FAILURE in 1h 15m 07s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/91a43f322532458cb5637bdb391f22cd : SUCCESS in 1h 30m 26s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/d455ced7814742c0bca18d3e56d43a3e : SUCCESS in 1h 28m 20s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/e64754c83e464e85840b0899ee305ca7 : SUCCESS in 1h 24m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/438e369441984c84853e851c70908f7f : SUCCESS in 2h 07m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/fa0de4bf06e84d4497504f10aff190fd : SUCCESS in 1h 00m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/faca3a6896d04a7e9141aa6ac488339c : SUCCESS in 1h 07m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/df0be8633fff465cb47da2d3bfd1ac8d : SUCCESS in 2h 12m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6882cf6d40724374807314da556ae918 : SUCCESS in 42m 18s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d11b55170497459db98dfce843242420 : SUCCESS in 1h 14m 43s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/0a35dd2310c149ef90181565ce7260a7 : SUCCESS in 1h 05m 57s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/51e0528eb5aa4217b0e50e2469131380 : SUCCESS in 1h 04m 13s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e8b573644cb247379adf9647fe56113f : SUCCESS in 4m 24s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a8a5ffc41a5c4e12a1eeeb3a25965183 : FAILURE in 1h 02m 05s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/23d048d56b1d4cbd88ef13dae617b2a7 : FAILURE in 1h 16m 50s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7b55cb234cfb42f6843a9830c723ecc9 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":1},{"id":"93beea940a7232884bc666b122904e57c07c51d2","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-03-04 08:50:33.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks Ade! This change seems significant since md5 being a non-approved algorithm and it was still being used in tempest. Hence, there was a need for such replacement.","accounts_in_message":[],"_revision_number":1},{"id":"721020d152ff0e2488db61134ffc92b263e9d6b8","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-03-04 18:57:28.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM, as these tests are non-security context by default, nothing changes, hashlib.md5() will be called anyway however the change allows to run the tests on machines with FIPS enabled.","accounts_in_message":[],"_revision_number":1},{"id":"82bc61beb8acc3f8378fb50ab29bc4d4c99fc7fa","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-10 17:16:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2a295287854d874eecc61704853cbd9a39748074","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 17:16:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d3ab5e6360de0642939e99fe23bcb20f15fe21dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-10 21:55:11.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5369a63a89c543aeb16f5cd9448c956d : SUCCESS in 1h 04m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ce2da0362f954bbda87bebb837674deb : SUCCESS in 1h 29m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40f4538a85a04bd78e856026bb2a8f4c : SUCCESS in 5m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/58ce45ebdc8d4e54bfc05a9aebf53e89 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c37e7a9611c4ccf8c737d2b4ebfe612 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/150c44a1c1834dce8b64df8999593179 : SUCCESS in 13m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a42b6a21e6224d2bab1787370e433fb4 : SUCCESS in 2h 08m 48s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/fe4afe6f1f15487ab850886b6e922272 : SUCCESS in 59m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/82eac8fbaf8448df9031f248bd395510 : SUCCESS in 1h 00m 18s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5372dfbea9564c159aee21d8b504f75e : POST_FAILURE in 1h 19m 46s","accounts_in_message":[],"_revision_number":1},{"id":"234ac55c85184b1fd531ffda64533767288fb2a2","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-03-10 22:06:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5296167ed905c01c3605238fea9754a86eb7011d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 03:45:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/feaa8e8f38c74047b4da0efe2ef22534 : SUCCESS in 1h 05m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e121126684984363864190a20fbc3b2d : SUCCESS in 1h 25m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12556d79481d4e5192604e8c4bb82f17 : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84fc8a673dfd4e8a832bcba11947a496 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9c89ba2f640f4509aef1bd720cd1313e : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f18c11b91c634aea9999f4829e4fa12b : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90eaf5c2b8074e629a204d0de723c215 : SUCCESS in 13m 21s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/892463bfa0a64e64bed7792df0092b59 : SUCCESS in 1h 50m 28s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c0469a7997b148c48f74f98adde8d560 : SUCCESS in 1h 37m 06s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5f9436bb773042a3a88cc3bf6ea8a332 : FAILURE in 1h 28m 41s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/770672f86c4a4774b732cc31aba8f076 : SUCCESS in 1h 18m 01s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/c9b45b9c56cf4876b2a9213725bc2165 : SUCCESS in 1h 16m 47s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/0705ad85e6674d1e841546c5e1a857c9 : SUCCESS in 1h 14m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6c1a8a5df5ef4849b5043f8f4d52ab3b : SUCCESS in 2h 11m 05s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/74e7f9628d3a4390b02e5573e70f33b1 : SUCCESS in 40m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/890d9bb9b55443f2bb9f9a0b0e2582b9 : SUCCESS in 1h 05m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9bbaa2518314479b963a4fcf2808f3c9 : SUCCESS in 2h 13m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/df7d77a42f1747ec8a08d7bf014903df : FAILURE in 1h 08m 45s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/7876e6c09928495fb5dbd3e4481c7df5 : SUCCESS in 1h 05m 04s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/a3bcb1f5505146d19001cbcd954ba38c : SUCCESS in 1h 00m 43s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/eb09d3fc2d564a1b96012308713b6fb0 : FAILURE in 59m 49s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/601fa585abc742e08ccbfa433c84d67a : SUCCESS in 5m 35s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/6ac21e2510674f26822a64cae9f0f929 : FAILURE in 1h 11m 00s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/36b713eb14074fb4a9bb33d591491f0c : FAILURE in 1h 10m 41s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b35dfb2d8b41480d886f0ee99d6fdc76 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0ab04fc72af6ff0415879572e698cd7d3201cd8a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 03:46:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"47c3c1f102aabab83fe6cbd700566e43fb9c73fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 06:57:11.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3607db11022e48eda184da4272572d6c : SUCCESS in 53m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3cb7e56d8bd24cddafcd88cc4aad4036 : SUCCESS in 1h 36m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d438ada3b4fa45338de7a5b3e9a9a0f0 : SUCCESS in 5m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53baf64f139f408ea1b3c79fff474f08 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4afa876d54054216b372b01851161dda : SUCCESS in 6m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fbfc36b688a4bc1b107d67c88deb2df : SUCCESS in 12m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2c7d4b7a4c864c93b7d82659a447a1a5 : SUCCESS in 2h 06m 09s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/cb9504005e364ccfb6a560dc4a0e26b9 : SUCCESS in 58m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b286be8faf04d3490b5777fc1c17c18 : SUCCESS in 1h 01m 09s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/cea10dac21904864a581222cd859378b : SUCCESS in 1h 28m 41s","accounts_in_message":[],"_revision_number":1},{"id":"059074d8ae239b773c57a7a3cef4e46b29c6b711","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 06:57:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"78ae67cb8b5ddaa21209bfdfb3106394449fc3f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-11 06:59:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e0889ac16704ef1a9694291e99e374e : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56860ccae71114bfb32138c823ea04a50b916aa2","revisions":{"56860ccae71114bfb32138c823ea04a50b916aa2":{"kind":"REWORK","_number":1,"created":"2021-02-25 21:14:28.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/84/777684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/84/777684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/777684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/777684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/84/777684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/84/777684/1"}}},"commit":{"parents":[{"commit":"35bc9bef6a23cc8f10f1b5838e3ca99133a3b8f6","subject":"Merge \"[part1] Remove six\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/35bc9bef6a23cc8f10f1b5838e3ca99133a3b8f6"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-02-25 20:45:02.000000000","tz":-300},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-02-25 21:14:08.000000000","tz":-300},"subject":"Replace md5 with oslo version","message":"Replace md5 with oslo version\n\nmd5 is not an approved algorithm in FIPS mode, and trying to\ninstantiate a hashlib.md5() will fail when the system is running in\nFIPS mode.\n\nmd5 is allowed when in a non-security context.  There is a plan to\nadd a keyword parameter (usedforsecurity) to hashlib.md5() to annotate\nwhether or not the instance is being used in a security context.\n\nIn the case where it is not, the instantiation of md5 will be allowed.\nSee https://bugs.python.org/issue9216 for more details.\n\nSome downstream python versions already support this parameter.  To\nsupport these versions, a new encapsulation of md5() has been added to\noslo_utils.  See https://review.opendev.org/#/c/750031/\n\nThis patch is to replace the instances of hashlib.md5() with this new\nencapsulation, adding an annotation indicating whether the usage is\na security context or not.\n\nIn this case, all hashlib.md5() invocations are to run tempest tests\nso this is not a security context.  With this patch, we can run tempest\ntests from a FIPS enabled machine.\n\nChange-Id: Ib301466bb0a1996997c63599918fa96059a927bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/56860ccae71114bfb32138c823ea04a50b916aa2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/56860ccae71114bfb32138c823ea04a50b916aa2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
