)]}'
{"id":"openstack%2Ftempest~778223","triplet_id":"openstack%2Ftempest~master~I3c7d3a4993eefc6da5ec37b3c4c64fb66a3c2794","project":"openstack/tempest","branch":"master","topic":"fixed-key-flag","hashtags":[],"change_id":"I3c7d3a4993eefc6da5ec37b3c4c64fb66a3c2794","subject":"Add a [compute-feature-enabled]/fixed_key flag","status":"ABANDONED","created":"2021-03-02 17:19:59.000000000","updated":"2021-03-08 14:44:59.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5f911236ef5c0e0b3326f6ecab71db75fc7233d3","_number":778223,"virtual_id_number":778223,"owner":{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-03-02 20:14:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":1,"date":"2021-03-04 09:50:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-02 17:21:01.000000000","updated_by":{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},"reviewer":{"_account_id":31033,"name":"James Parker","email":"jparker@redhat.com","username":"jparker"},"state":"CC"},{"updated":"2021-03-02 17:21:01.000000000","updated_by":{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},"reviewer":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"state":"REVIEWER"},{"updated":"2021-03-02 17:21:01.000000000","updated_by":{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2021-03-02 17:21:01.000000000","updated_by":{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-03-02 17:21:01.000000000","updated_by":{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2021-03-02 20:14:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-03-04 09:50:47.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"}],"messages":[{"id":"189029679f479d9bcce7f8d2b68f67e242255868","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},"date":"2021-03-02 17:19:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0b02c812bd4c9dd6d3e755907ce9ff09c4edf5c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-02 20:14:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/6ac9ec3683804f3d8354dbb57826e6c2 : SUCCESS in 53m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3eb91fb1b6894587842e34753c73a567 : SUCCESS in 1h 41m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b5581d35231d41ae8a000f6b2ff88bef : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a51f765873b46c0abac9b672c95f8c3 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/27a15af611dc424e80e98cdf5f89940a : SUCCESS in 7m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d03c745d5f4541f29e2e27392484da74 : SUCCESS in 8m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0367e247c9c64b92bda410fa73448fb2 : SUCCESS in 15m 03s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/a091c6ab854c47a7a56fc8e8da0e22d1 : SUCCESS in 1h 54m 38s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/8899f3bc05ee4553bc15572c76ff7782 : SUCCESS in 1h 35m 41s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b64d2c16e42a454aa246f1457a52951d : SUCCESS in 1h 16m 49s (non-voting)\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/e1d55516e1344ab090512551a601c59e : SUCCESS in 1h 49m 59s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/dac8d1f521864608947ea59b0aeb3d60 : SUCCESS in 1h 30m 34s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/7f4b2afa6b1e429a8673fceddf62bcdf : SUCCESS in 1h 17m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c4a9c9897c63475fb4c9ee43434fe2dd : SUCCESS in 1h 44m 45s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/4302dcff90e04cd5a6b02065e1564213 : SUCCESS in 39m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ff59a5219aa476eb6c404708cfdae41 : SUCCESS in 1h 04m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d159cad96cab4dc6b75e5dc3ce2ba191 : SUCCESS in 2h 17m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b9c949084a074abfa7e3acb06da09b27 : FAILURE in 59m 15s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/4802d8bc9b234c6d9afe7897b020d869 : SUCCESS in 1h 18m 20s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/02d016869dbc4f4482d3dac083326a8f : SUCCESS in 1h 07m 14s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/475dfdec3a634706941da4e8bcf66230 : SUCCESS in 1h 11m 53s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/173ad581bcb4412dbaab7f55dda24b2a : SUCCESS in 5m 36s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3deb9f44bf7341abb1cd2e50089e118b : FAILURE in 1h 08m 02s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/17e20b1bc5a042adadf116e770399263 : FAILURE in 1h 16m 10s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7097ad7f586d436ca4b91c9ba764141d : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6181243a8042127fa6428eccb5530a50318e39e2","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-03-04 09:50:47.000000000","message":"Patch Set 1: Code-Review+1\n\nLGTM!","accounts_in_message":[],"_revision_number":1},{"id":"c80f8d94f0ce27028b1498bdb9d709062164cd6a","author":{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},"date":"2021-03-04 17:08:53.000000000","message":"Patch Set 1:\n\nAlthough this change is not necessary now that whitebox-tempest-plugin/+/778215 has merged using compute-feature-enabled.attach_encrypted_volume skip check, I\u0027m okay with adding this in case we still want to 😊","accounts_in_message":[],"_revision_number":1},{"id":"21d884727bb995f5cf392fbd5c26a42a305207f5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-03-05 15:19:27.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e Although this change is not necessary now that whitebox-tempest-plugin/+/778215 has merged using compute-feature-enabled.attach_encrypted_volume skip check, I\u0027m okay with adding this in case we still want to 😊\n\nyeah, I was checking where it is being used, we can add when we need it.","accounts_in_message":[],"_revision_number":1},{"id":"e79af9159fc0dc81de0966671bf6e74f27b82a6e","author":{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},"date":"2021-03-08 14:44:08.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e \u003e Patch Set 1:\n\u003e \u003e \n\u003e \u003e Although this change is not necessary now that whitebox-tempest-plugin/+/778215 has merged using compute-feature-enabled.attach_encrypted_volume skip check, I\u0027m okay with adding this in case we still want to 😊\n\u003e \n\u003e yeah, I was checking where it is being used, we can add when we need it.\n\nAck, abandoning this till needed","accounts_in_message":[],"_revision_number":1},{"id":"5f911236ef5c0e0b3326f6ecab71db75fc7233d3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},"date":"2021-03-08 14:44:59.000000000","message":"Abandoned\n\nWill add when we need it in future","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79f57767ecccd6a621a8a901b6ec071ba66d11fb","revisions":{"79f57767ecccd6a621a8a901b6ec071ba66d11fb":{"kind":"REWORK","_number":1,"created":"2021-03-02 17:19:59.000000000","uploader":{"_account_id":27478,"name":"Archit Modi","email":"amodi@redhat.com","username":"amodi"},"ref":"refs/changes/23/778223/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/23/778223/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/778223/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/778223/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/23/778223/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/23/778223/1"}}},"commit":{"parents":[{"commit":"12707f563321bcbe41b04c36b136dd2f9be4c3d2","subject":"Merge \"Make _get_network_by_name() non-private\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/12707f563321bcbe41b04c36b136dd2f9be4c3d2"}]}],"author":{"name":"Archit Modi","email":"amodi@redhat.com","date":"2021-03-02 17:12:57.000000000","tz":-300},"committer":{"name":"Archit Modi","email":"amodi@redhat.com","date":"2021-03-02 17:19:31.000000000","tz":-300},"subject":"Add a [compute-feature-enabled]/fixed_key flag","message":"Add a [compute-feature-enabled]/fixed_key flag\n\nFor tests requiring a fixed key or key manager service enabled\nthis can be used along with CONF.service_available.\u003ckey-manager\u003e when\nfor different deployments specifically, one not having the key manager\nenabled, we can use this flag to run the test.\n\nRelated: https://review.opendev.org/c/openstack/whitebox-tempest-plugin/+/778215\nChange-Id: I3c7d3a4993eefc6da5ec37b3c4c64fb66a3c2794\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/79f57767ecccd6a621a8a901b6ec071ba66d11fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/79f57767ecccd6a621a8a901b6ec071ba66d11fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
