)]}'
{"id":"openstack%2Ftempest~794257","triplet_id":"openstack%2Ftempest~master~I337c59abab3328ef96ed2172c43e366f23327c58","project":"openstack/tempest","branch":"master","topic":"fix/delete_snapshot","hashtags":[],"change_id":"I337c59abab3328ef96ed2172c43e366f23327c58","subject":"[DNM] Remove snapshot deletion","status":"ABANDONED","created":"2021-06-02 15:30:42.000000000","updated":"2022-10-07 09:14:42.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"65ed3f3760ce60ba904e65c7f366a2e50431ba1d","_number":794257,"virtual_id_number":794257,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2021-06-02 19:36:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-02 19:36:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d36007c7e19b2478450d8fccd935ea2419e5572c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-06-02 15:30:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"adb4e9aec06c3f931fa4424c7b98c0a456d488ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-02 19:36:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/005f546eccc4431e86cad6de78f44021 : SUCCESS in 59m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6471c60f596e47258746a9705f738dce : SUCCESS in 1h 28m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd2a7367518142ed932e11bb08286f5b : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c613dde2eb9412c8d2a03acfe3400f3 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6c785c81e9243c0963b01e82da6d46c : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f35c09b3833a428990b7a79f42f75b9b : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6383d4eb08524f6bac22acb82ad8147b : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff731d6225b04a1a91314a632a477206 : SUCCESS in 15m 05s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c2a6f80c7761489291aeae2bbefd237c : SUCCESS in 2h 02m 28s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/2b8a29b20beb48b49f7f33926aa6584d : SUCCESS in 1h 18m 54s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b1c5dabb5e3143f4a5683440a4cdbdac : SUCCESS in 1h 28m 05s (non-voting)\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/1b9ba675aed14b418c558b6765ce0328 : SUCCESS in 1h 21m 46s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/5470d8f9b3904f0fbf2665edd818f4bc : SUCCESS in 1h 27m 40s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/0412c6f5e14a4c2199331ab67c53db99 : SUCCESS in 1h 17m 08s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/a928e28be089447e867ac1828dff71c2 : SUCCESS in 1h 16m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1e8e5be20d814c62b97e96a4daeaa7cb : SUCCESS in 1h 48m 57s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0c8fc83b877941b39e361df20dcb7f5a : SUCCESS in 38m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/095de4ad7c734cd3a144f2739b26ee79 : SUCCESS in 49m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a0e904a9b62d4053ba918593ca97fd77 : SUCCESS in 2h 06m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e72eb25266554496bd8bf042b6271b4b : SUCCESS in 54m 48s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b74fac62562f40468c50b79ca3ef3191 : SUCCESS in 1h 14m 41s\n- neutron-grenade-multinode https://zuul.opendev.org/t/openstack/build/eab6c25a94d94817ac4437ec13cbdcba : SUCCESS in 1h 01m 21s\n- neutron-tempest-dvr https://zuul.opendev.org/t/openstack/build/5980a05412034e328ec8d4e15c8f4570 : SUCCESS in 56m 48s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/67c3715b8e194e2abb84b2436df49ce4 : SUCCESS in 3m 47s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2b4025943caf43b789dc80f4dc6211bb : SUCCESS in 1h 35m 21s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/3e248a2fe5274f0ba388c906f852611c : FAILURE in 1h 14m 53s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e34bd185e62a444aa93de7691a65eb4b : SUCCESS in 4m 35s","accounts_in_message":[],"_revision_number":1},{"id":"65ed3f3760ce60ba904e65c7f366a2e50431ba1d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2022-10-07 09:14:42.000000000","message":"Abandoned\n\nThe deletion is OK as the tests test the snapshot deletion itself.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1464238f59e3bc37baeb91a1dde7bc81e2e2e142","revisions":{"1464238f59e3bc37baeb91a1dde7bc81e2e2e142":{"kind":"REWORK","_number":1,"created":"2021-06-02 15:30:42.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/57/794257/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/57/794257/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/794257/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/794257/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/794257/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/57/794257/1"}}},"commit":{"parents":[{"commit":"3e127640657b294c9d4a941b6a6bb287aa496539","subject":"Update IRC network to OFTC","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3e127640657b294c9d4a941b6a6bb287aa496539"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2021-06-02 15:30:03.000000000","tz":120},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2021-06-02 15:30:03.000000000","tz":120},"subject":"[DNM] Remove snapshot deletion","message":"[DNM] Remove snapshot deletion\n\nThis patch makes sure that snapshots created within volume tests\nare deleted only once.\n\nBefore, each snapshot was deleted twice [1][2].\n\n[1] https://github.com/openstack/tempest/blob/3e127640657b294c9d4a941b6a6bb287aa496539/tempest/api/volume/test_volumes_snapshots.py#L142\n[2] https://github.com/openstack/tempest/blob/3e127640657b294c9d4a941b6a6bb287aa496539/tempest/api/volume/base.py#L142\n\nChange-Id: I337c59abab3328ef96ed2172c43e366f23327c58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1464238f59e3bc37baeb91a1dde7bc81e2e2e142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1464238f59e3bc37baeb91a1dde7bc81e2e2e142"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
