)]}'
{"id":"openstack%2Ftempest~805122","triplet_id":"openstack%2Ftempest~master~Icfca910da1b56159adc0931c537e1a0566ff4569","project":"openstack/tempest","branch":"master","topic":"master27","hashtags":[],"change_id":"Icfca910da1b56159adc0931c537e1a0566ff4569","subject":"update volume type with nonexistent type id","status":"ABANDONED","created":"2021-08-19 07:04:12.000000000","updated":"2023-02-02 21:50:44.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"394774d776f8fcb4db92c44d2fea040b5e5d371d","_number":805122,"virtual_id_number":805122,"owner":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-07 12:31:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":-1,"date":"2021-10-05 22:06:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 09:15:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 09:23:57.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-10-05 22:06:44.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"c20ce0aef4f362074cf6193e2ad95ad99e0dacc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2021-08-19 07:04:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18c95806e413f708d0ac185b4d84858eff573427","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 09:15:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/4f3efb4f00af49a6823d08cfd16f93d8 : SUCCESS in 1h 11m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/52b3919f79774c5593c9d73cba1063fe : SUCCESS in 1h 20m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0573f0db366948baa9e524f4bdddad63 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32941d8a32a844558e46fd86aa6526ef : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc660b2cd5074e7cacc483ef94b0087c : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe1a4ad8d2e448d7ae112806171dc2c7 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f7d62d7f7b0a4437b8ae85d69db885bf : SUCCESS in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50700256b9644659be8087065e572a7b : SUCCESS in 18m 18s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/07fb9dc4de4b4eaebf487ea2981e979f : SUCCESS in 1h 56m 06s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/af4265a963854fea986c7f29e3a2cf91 : SUCCESS in 1h 13m 15s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/22856a28fa0a4635a6fae49a2d56e273 : SUCCESS in 1h 20m 33s (non-voting)\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/c45c5062d8cc48d7aa497b93110be1e0 : SUCCESS in 1h 25m 08s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/4a42ee4b853241c0a17c03151d798a27 : SUCCESS in 1h 29m 23s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/593cde780aa94492ab76695955fca98e : SUCCESS in 1h 26m 18s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/0dfedaaa02c641c7903993b067b3db29 : SUCCESS in 1h 25m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ef77e69ce3234695985f87b877236004 : SUCCESS in 1h 26m 15s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/1458acef22e14714ad93c26c65bbebc7 : SUCCESS in 36m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb0773c931d9433bb98bfd3bc5ae2c0b : SUCCESS in 48m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a15bb6d82df24e1396ce318f4adf8d66 : SUCCESS in 2h 06m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f4f5d6d698414018b08320102f30c339 : SUCCESS in 56m 44s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/80e22fae83a7454ea5ae10f4ace97beb : SUCCESS in 1h 10m 42s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/afab0038b31349f2b038d5b50e41d778 : SUCCESS in 1h 02m 11s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/307a6fc7d50e4f93ad48074e3af1cdaf : SUCCESS in 55m 20s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/0628ed66b0d04b3084907a94254dfbea : SUCCESS in 4m 12s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1aed1d732c464b0da7e18c3708d11bed : FAILURE in 1h 06m 22s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/c2fc51a8c22747fbbfc903451a200845 : FAILURE in 1h 01m 21s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f173851b46c744e98afb8039854178be : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"0510d489c0ef129b3920fc0ec216c78ab83d4d88","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-08-26 09:23:57.000000000","message":"Patch Set 1: Code-Review+2\n\nLGTM, thanks for the test","accounts_in_message":[],"_revision_number":1},{"id":"ef4d3a7171774dda729e99e0ff5b8e5d7c20a8bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2021-09-07 10:10:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fafa031d674c8dfaea893cb4ee3538d3bb7aa7f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 12:31:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/71dafd7aa3a64a65aaeb4b33007a2aed : SUCCESS in 1h 01m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b0cee72b3164345be632a7ee8f9c8da : SUCCESS in 1h 26m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/710bc5c04933411f92e40e9bd12717aa : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d626c6fdf6d46b1b11219877e9e885b : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf8a4c21dded41dfb32f4c837b9f8f73 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/24fbd7dc611f4944b4edcd35b4a5f8db : SUCCESS in 5m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/791363b413034c5d8058d773997c140f : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/497fab09329d464f9d92e15316e5ab73 : SUCCESS in 20m 51s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/678556c7db444e70a005eb1c388d5d29 : SUCCESS in 1h 28m 34s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/0849adfee5df4f84af71dae8f8a733f3 : SUCCESS in 1h 14m 35s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/00918e8da96243c0906bfed6b5b0c7cf : FAILURE in 1h 07m 20s (non-voting)\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/70035e0f54ea44f3835e0efc5f689d3f : SUCCESS in 1h 36m 14s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/955b8760b30b4cbda64c75af5176a02e : SUCCESS in 1h 34m 38s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/5464dc020c514b4e94548c74494e3216 : SUCCESS in 1h 41m 31s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/e36c8a41206f4bf6ae1572bf0372d917 : SUCCESS in 1h 40m 43s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bd84ae65a84b447c92ffbc8f45cc6347 : SUCCESS in 1h 52m 34s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b096da2fee214127a54aaa07b96866aa : SUCCESS in 35m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64083de7d51547578721db91d916345d : SUCCESS in 51m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c559421d6df442f190f499e9fcd1ee85 : SUCCESS in 2h 14m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0c549c69e5034ec5b9181c5905b95159 : FAILURE in 56m 21s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/19c6f32cea544ae19ac446dfb204a067 : SUCCESS in 1h 11m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e501c3c0c3894ba7a6056d0bd799e14c : SUCCESS in 1h 01m 22s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/3149699cf0ec418592defba58cc81648 : SUCCESS in 1h 10m 39s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/a369f9f211c646b5874760150a016ed8 : SUCCESS in 5m 31s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/90a0f4b100aa44f385915fea2770155f : FAILURE in 1h 18m 31s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/8cd27b6a741b4b8d9d66f5749a1a72db : FAILURE in 1h 00m 47s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/607bff50f4204d3bb54cc5167e88dfc0 : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":2},{"id":"6148bbaee505329d20678876d4a002faebafa290","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-10-05 22:06:44.000000000","message":"Patch Set 2: Code-Review-1\n\nTempest is not best suitable place for negative API tests, we have few because interop needed it, please let us know if interop need it otherwise you can add it in cinder unit/functional tests itself -https://docs.openstack.org/tempest/latest/HACKING.html#negative-tests","accounts_in_message":[],"_revision_number":2},{"id":"394774d776f8fcb4db92c44d2fea040b5e5d371d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-02-02 21:50:44.000000000","message":"Abandoned\n\nno activities for a long time, feel free to restore it if needed.","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"096a56b32afe07b8b341ae87d6f0dc74b7a66f2e","revisions":{"be3d7a3afdb2b54fcc949e692efa052822f06c7a":{"kind":"REWORK","_number":1,"created":"2021-08-19 07:04:12.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/22/805122/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/22/805122/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/22/805122/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/22/805122/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/22/805122/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/22/805122/1"}}},"commit":{"parents":[{"commit":"663787ee794df54e7ded41e5f3e8ae246e9b4288","subject":"Merge \"Update encryption-type-create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/663787ee794df54e7ded41e5f3e8ae246e9b4288"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2021-08-19 07:02:20.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2021-08-19 07:02:34.000000000","tz":480},"subject":"update volume type with nonexistent type id","message":"update volume type with nonexistent type id\n\ntest update volume type with nonexistent type id will fail\nChange-Id: Icfca910da1b56159adc0931c537e1a0566ff4569\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/be3d7a3afdb2b54fcc949e692efa052822f06c7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/be3d7a3afdb2b54fcc949e692efa052822f06c7a"}]},"branch":"refs/heads/master"},"096a56b32afe07b8b341ae87d6f0dc74b7a66f2e":{"kind":"REWORK","_number":2,"created":"2021-09-07 10:10:46.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/22/805122/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/22/805122/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/22/805122/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/22/805122/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/22/805122/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/22/805122/2"}}},"commit":{"parents":[{"commit":"ad2c3009aed4334a0752679bf984f6c9cea49171","subject":"Expected success status for \"QoS rule update APIs\" is \"200 OK\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad2c3009aed4334a0752679bf984f6c9cea49171"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2021-08-19 07:02:20.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2021-09-07 10:09:48.000000000","tz":480},"subject":"update volume type with nonexistent type id","message":"update volume type with nonexistent type id\n\ntest update volume type with nonexistent type id will fail\n\nChange-Id: Icfca910da1b56159adc0931c537e1a0566ff4569\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/096a56b32afe07b8b341ae87d6f0dc74b7a66f2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/096a56b32afe07b8b341ae87d6f0dc74b7a66f2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
