)]}'
{"id":"openstack%2Ftempest~805125","triplet_id":"openstack%2Ftempest~master~Ib6c8cad9d12edebb11acc58a168e0f20c7d41cb4","project":"openstack/tempest","branch":"master","topic":"master30","attention_set":{},"removed_from_attention_set":{"30409":{"account":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"last_update":"2023-06-10 01:11:16.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib6c8cad9d12edebb11acc58a168e0f20c7d41cb4","subject":"update volume type with an repeat name","status":"ABANDONED","created":"2021-08-19 07:13:46.000000000","updated":"2023-06-10 01:11:16.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b37cc9976f86624cea1dab4710121a9af396893f","_number":805125,"virtual_id_number":805125,"owner":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-08-19 09:18:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"date":"2021-11-29 13:24:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":-1,"date":"2021-10-05 22:08:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-08-19 09:18:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-08-26 09:22:29.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-10-05 22:08:01.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"0809d9632311eb28fc03da8186eb057394bdf5a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2021-08-19 07:13:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58d37c952159795149983f975ceece36e39d316b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 09:18:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7f8376d23c004f32b4d7a53d206d4b65 : SUCCESS in 59m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb48d20515bf406f9d7721c9e7fccc54 : SUCCESS in 1h 15m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cbdc9216fc57453396c35393f91f7642 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d230253813d40e8bc39650a3a3680c2 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a68d270309754730bdd7432f0104b2aa : SUCCESS in 6m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5f57381590d4ffa92dc3a2239722c78 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcc8b0b55f654800a9adb14c2ba33aec : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fc2876743bd4727a5923a740a1e22bc : SUCCESS in 14m 44s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/59b7a0a48fb44ec0b75e72b430a49f9e : SUCCESS in 1h 33m 52s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/aff4829bda7b48d388820749053be89a : SUCCESS in 1h 08m 08s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/70a9b199cb3740edbca7802bae7e3eca : SUCCESS in 1h 25m 18s (non-voting)\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/7297ad114da2418eaf654961ec121cd0 : SUCCESS in 1h 34m 03s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/55adc2f32fef4b48b56e4a58ff3d2d2f : SUCCESS in 1h 21m 45s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/7e83b97ae5064aeda372fc9a4d186eb8 : SUCCESS in 1h 26m 26s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/9381baf1a0414357ac0699bcf206bedf : SUCCESS in 1h 26m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/408c2f95a5b244ffbced23cfa101f7fb : SUCCESS in 1h 28m 08s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0318467df5a04be0bc2795d4185bc8fb : SUCCESS in 39m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5cbe937bcef449628b1df817a3623b2a : SUCCESS in 48m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86c52c947d7e4f1b826dc3631e930659 : SUCCESS in 2h 02m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5ea02534c9384c5a856d6997df16ff1f : SUCCESS in 52m 29s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/842e7749aac748f3a2fb5b48bfd1eb81 : SUCCESS in 1h 09m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/48e67e2cc4a3450f81357265a28f087e : SUCCESS in 1h 05m 31s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/59fedb2b26c54c058fd2c7a75b35a717 : SUCCESS in 54m 46s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d76065dea1044d7e850c394f7b90193f : SUCCESS in 4m 38s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/433379694eea424383aaedc46f8b51bb : FAILURE in 1h 09m 21s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/3e50548bd3244711b1d68d10098918fb : FAILURE in 1h 00m 44s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/48017cf8688f4702b17ff3aff8840f58 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f6899c41de69967b25e8ea5769c82a7fe4ded09e","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-08-26 09:22:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)\n\none docstring suggestion, otherwise lgtm","accounts_in_message":[],"_revision_number":1},{"id":"725dead93d05444e109f28a1abf386f611c4759a","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-10-05 22:08:01.000000000","message":"Patch Set 1: Code-Review-1\n\nTempest is not best suitable place for negative API tests, we have few because interop needed it, please let us know if interop need it otherwise you can add it in cinder unit/functional tests itself -https://docs.openstack.org/tempest/latest/HACKING.html#negative-tests","accounts_in_message":[],"_revision_number":1},{"id":"1495c4f245ef7d18e32dcc582e5ccb4e303c8fa9","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-11-29 13:24:52.000000000","message":"Patch Set 1: -Code-Review","accounts_in_message":[],"_revision_number":1},{"id":"b37cc9976f86624cea1dab4710121a9af396893f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-06-10 01:11:16.000000000","message":"Abandoned\n\nno activity since long, feel free to restore if needed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4bbc620572c65886ca6a8d96ee55b1516fd4cb7","revisions":{"e4bbc620572c65886ca6a8d96ee55b1516fd4cb7":{"kind":"REWORK","_number":1,"created":"2021-08-19 07:13:46.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/25/805125/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/25/805125/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/25/805125/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/25/805125/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/25/805125/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/25/805125/1"}}},"commit":{"parents":[{"commit":"663787ee794df54e7ded41e5f3e8ae246e9b4288","subject":"Merge \"Update encryption-type-create\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/663787ee794df54e7ded41e5f3e8ae246e9b4288"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2021-08-19 07:12:31.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2021-08-19 07:12:39.000000000","tz":480},"subject":"update volume type with an repeat name","message":"update volume type with an repeat name\n\nTest update volume type with an repeat name will fail\n\nChange-Id: Ib6c8cad9d12edebb11acc58a168e0f20c7d41cb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e4bbc620572c65886ca6a8d96ee55b1516fd4cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e4bbc620572c65886ca6a8d96ee55b1516fd4cb7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
