)]}'
{"id":"openstack%2Ftempest~807471","triplet_id":"openstack%2Ftempest~master~If711dc682763c674b7d80721217db476ab1c485c","project":"openstack/tempest","branch":"master","topic":"add_ecc_keys","hashtags":[],"change_id":"If711dc682763c674b7d80721217db476ab1c485c","subject":"WIP Add support for ecdsa keys","status":"ABANDONED","created":"2021-09-05 05:57:12.000000000","updated":"2021-09-07 18:20:55.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e3c2292fb0a9755f77f0bc5682218b3a5378b32d","_number":807471,"virtual_id_number":807471,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-09-06 17:25:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-05 07:42:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fd6beaf8e84cdc41612df67d0123615e5c7f2c83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-09-05 05:57:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6c249e483338e131054041e691f37bf41ab7965","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 07:42:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4cf695125d9142758d0fbca00b88e588 : FAILURE in 3m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/221933e7556e473a8d7cb45bff1e295b : FAILURE in 17m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/397ab77607a641bb814de5abb8cbca5b : FAILURE in 1h 07m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c12a6a7892c4cc9b198576c58eac7f1 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10040c30a9f34e94a7db0d46ada416b4 : SUCCESS in 5m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cdd50570bb394ccc97c660c8461aa9c6 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c2ddd6f81f59417d8e8c0eb4fc1f4d65 : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a0b2753298e042a5a2dcac52ee8f6550 : SUCCESS in 5m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/703b2103f7b84d5cb376ab6462357975 : SUCCESS in 15m 14s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d3f73ab77d094abe981378fbf2141a10 : FAILURE in 1h 15m 41s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/effc221f57654fd4b1f6fc64e5228575 : FAILURE in 1h 03m 57s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f8d8d16065e44f35a3a25ab345fb5fdc : POST_FAILURE in 1h 06m 10s (non-voting)\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/de788d280c1a41a58101dea8a37f5fb8 : FAILURE in 22m 05s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/1395eb0dd171410b85941b74cc74a6f1 : FAILURE in 16m 03s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/f7138f7459a34722b220edac259ae3d6 : FAILURE in 25m 59s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/7f679d61ad6a4ee48e5785e65ee8cfc7 : FAILURE in 16m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/66b8c9c295c04c36ac8da734e71a7b0f : FAILURE in 1h 16m 59s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/606c3ee67f4e4a15b29d44e956196575 : SUCCESS in 43m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/deb4aaf7b7a54c87a0e11bae6e138103 : FAILURE in 45m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a673fa4ec47f4b779a7ec506c58006f5 : FAILURE in 1h 42m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/79a78c08e0084d4d9ce882cee3515ac8 : FAILURE in 55m 51s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0fa14c7794504a2a87e61b1791fbfa3d : FAILURE in 1h 07m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bf138e8041ed4b74bf07eabba3d03157 : FAILURE in 28m 51s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/03f0651f45c54e17bfab33193c5295cf : FAILURE in 44m 24s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/751bf933c8fe43c588195a6a0e789610 : SUCCESS in 3m 59s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/79625a4cce55492a9f6880eb26365387 : FAILURE in 1h 06m 19s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/93a4a950b4a9424590e0603c18e509b7 : FAILURE in 59m 19s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7f30dd47eeb14cb2a7e333eadae44aa4 : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":1},{"id":"3966a0eb941762682526846eaa50f22d9b6fa54a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-09-05 19:18:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"76a8fd12208d4c7ca3f785f1ba661c3eb255607b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-05 20:50:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2327e2a6a7834f2d9018b8e1d9182655 : FAILURE in 3m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/a2dd4460b5c740c99f8be57120230bb0 : FAILURE in 20m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18040e8b3e404f79912049c32e77c034 : FAILURE in 53m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7c5ee9dc67b74085a34a6677e81c6f7e : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f1edb28060c4062bf70c263a269da61 : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dab6faf7808746909e74b0b0586b85ff : FAILURE in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3aafd6a129eb4c719fa81dfe4b106661 : FAILURE in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56f64bf372cd432590889f0cd46f4901 : FAILURE in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7646809f428743a6bc16a72196852425 : SUCCESS in 20m 42s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/a8a4911cb8f24fd99dda9afefdd4c9da : FAILURE in 1h 03m 44s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c16a2c286b0f47c7a9acafbb8a0960cc : FAILURE in 50m 12s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b31856655f3a48e6af5465eb8e7a5046 : FAILURE in 1h 06m 59s (non-voting)\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/063d061a77244a099ae37773ed6be5ac : FAILURE in 18m 16s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/9bcd04bfbc9f4aa3a232b3c843ccef95 : FAILURE in 16m 55s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/b44055b0deae4e2ebf80611576fe8351 : FAILURE in 13m 36s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/1208f74f7ec641b38b525bcb9d353452 : FAILURE in 23m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e775baa59c4a4b8788d4f96f50a5663c : FAILURE in 1h 07m 55s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/908e630cf0ae49718114b519730eb829 : SUCCESS in 31m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b364ff6aada444a2af92509dbcad451c : FAILURE in 33m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dc8e662be2f243a1b39be8a21ef1e4b2 : FAILURE in 1h 26m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b419dd62b1a6408dbf5cbd286237c9dc : SUCCESS in 54m 26s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/614d4898ce104973aeb93fdcb01c1322 : FAILURE in 1h 15m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b0b95fede56749418e66c40bceb7a614 : FAILURE in 26m 17s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/187a7139339248bba3f5634316801e3d : FAILURE in 42m 49s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/5d935436b0a047b8a5e350ed090cb19d : SUCCESS in 4m 56s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ac630a99702745fab4bc9fa9d6fefa3a : FAILURE in 1h 03m 10s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/5095dd5782814c7a962225f25ca54dad : FAILURE in 51m 29s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/23a0de2311af4b50b787bedb408a4717 : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":2},{"id":"bdecf1c763e0aa46a6173c1e6d5ab4568e326a45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-09-06 15:44:06.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"09e5785e9744b65a239833e15e4673adc8efa58c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-06 17:25:41.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2adc6d92da514b6580eb0e8571f9a5e3 : FAILURE in 3m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/0fd6cd47c59e429cbbd1f712703704c0 : FAILURE in 25m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dd9e187376b4cab80c2b773eb3b7643 : FAILURE in 1h 12m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c134849817a459b90952dfb918b146f : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6175dc48b0c64555bb276c2fccab95a9 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7c9b645846934ded98442234daee13e9 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ffc5f8db57f4760bb5237b03ee6ac1f : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d18bed63b5174a079c74528675fb17c1 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e40d788615c4c2ebbfe4e57c780ff1a : SUCCESS in 13m 58s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/0af8e0dab1484222b07ccc1bbf3985a6 : FAILURE in 1h 17m 09s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3b7360dde9fa4536ab48a23354ef6cde : FAILURE in 57m 20s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/89f86938ba3a497c8f4787cb815dffbb : FAILURE in 1h 01m 33s (non-voting)\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/f5598ae3141c433b97c71ddcebcb5d44 : FAILURE in 18m 47s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/0706a248b58a4d25bad619f73d0f909d : FAILURE in 18m 03s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/f6fb76ed78ce4238b5155067f36978d3 : FAILURE in 15m 44s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/61c9e13e42c24465b141906537ed2d36 : FAILURE in 21m 49s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ed05891c0cab4538bb3f025d5a25724b : FAILURE in 1h 15m 31s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/2864f626632542e7be9848017c04c483 : SUCCESS in 37m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e49d360f6a2d4d2196ba1b8bbeabd5f9 : FAILURE in 39m 01s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c36420da215d45de96a2a67ab7e84dd7 : FAILURE in 1h 35m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3707fd562b4e4b6c8d64c368e7a65b3f : FAILURE in 57m 59s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/fa75dbffb0194eb3bee04a333f1d5f8b : FAILURE in 1h 11m 23s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e0f2faeb5e9b4894981dccb833e718b5 : FAILURE in 30m 48s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/9a7408cf126c4280bf3794d08696ac19 : FAILURE in 46m 07s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/14cac54fb993428f85a2001ff9b83d97 : SUCCESS in 4m 59s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/74c05f2864f44de8a4f528081ef59490 : FAILURE in 1h 08m 49s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/703877bb9ad84ac3bb5ca42c08a841f4 : FAILURE in 1h 08m 48s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c2a7ca6a23ac4a58988840895ba05238 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"e3c2292fb0a9755f77f0bc5682218b3a5378b32d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-09-07 18:20:55.000000000","message":"Abandoned\n\nChange actually in 807465","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"edf1abc15444898de213347be908655335077f09","revisions":{"9285866d925eefc660647644936fd4000f3199c4":{"kind":"REWORK","_number":1,"created":"2021-09-05 05:57:12.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/71/807471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/807471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/807471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/807471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/807471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/807471/1"}}},"commit":{"parents":[{"commit":"4130966f0f8bcf0f71c6eef1c2b9644f62d11c44","subject":"Merge \"test update volume type extra specs for non existent volume type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4130966f0f8bcf0f71c6eef1c2b9644f62d11c44"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-04 19:56:34.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-05 05:56:32.000000000","tz":-240},"subject":"WIP Add support for ecdsa keys","message":"WIP Add support for ecdsa keys\n\nIn FIPS mode, using RSA keys for ssh is fine as long as SHA-1 is\nnot used for the signature algorithm.  Unfortunately, the version\nof cirros used in OpenStack CI does not have a version of dropbear\nthat supports SHA-2 signatures.  So, any connections from a FIPS\nenabled machine will fail as the cirros instance will only support\nssh-rsa (SHA-1 signatures).\n\nTo get around this, we add a new option to generate a use ECDSA\nkeys instead. (validation.ssh_use_ecdsa_keys)\n\nTemporarily try adding patched paramiko.\nChange-Id: Ib59eb8432fa1a2813b3047955157d1b3d24a55f8\n\nChange-Id: If711dc682763c674b7d80721217db476ab1c485c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9285866d925eefc660647644936fd4000f3199c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9285866d925eefc660647644936fd4000f3199c4"}]},"branch":"refs/heads/master"},"e11685ba77b7a2d71d772101d6aeeafab7a74efc":{"kind":"REWORK","_number":2,"created":"2021-09-05 19:18:31.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/71/807471/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/807471/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/807471/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/807471/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/807471/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/807471/2"}}},"commit":{"parents":[{"commit":"4130966f0f8bcf0f71c6eef1c2b9644f62d11c44","subject":"Merge \"test update volume type extra specs for non existent volume type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4130966f0f8bcf0f71c6eef1c2b9644f62d11c44"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-04 19:56:34.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-05 19:18:19.000000000","tz":-240},"subject":"WIP Add support for ecdsa keys","message":"WIP Add support for ecdsa keys\n\nIn FIPS mode, using RSA keys for ssh is fine as long as SHA-1 is\nnot used for the signature algorithm.  Unfortunately, the version\nof cirros used in OpenStack CI does not have a version of dropbear\nthat supports SHA-2 signatures.  So, any connections from a FIPS\nenabled machine will fail as the cirros instance will only support\nssh-rsa (SHA-1 signatures).\n\nTo get around this, we add a new option to generate a use ECDSA\nkeys instead. (validation.ssh_use_ecdsa_keys)\n\nTemporarily try adding patched paramiko.\nChange-Id: Ib59eb8432fa1a2813b3047955157d1b3d24a55f8\n\nChange-Id: If711dc682763c674b7d80721217db476ab1c485c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e11685ba77b7a2d71d772101d6aeeafab7a74efc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e11685ba77b7a2d71d772101d6aeeafab7a74efc"}]},"branch":"refs/heads/master"},"edf1abc15444898de213347be908655335077f09":{"kind":"REWORK","_number":3,"created":"2021-09-06 15:44:06.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/71/807471/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/807471/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/807471/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/807471/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/807471/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/807471/3"}}},"commit":{"parents":[{"commit":"4130966f0f8bcf0f71c6eef1c2b9644f62d11c44","subject":"Merge \"test update volume type extra specs for non existent volume type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4130966f0f8bcf0f71c6eef1c2b9644f62d11c44"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-04 19:56:34.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-06 15:43:16.000000000","tz":-240},"subject":"WIP Add support for ecdsa keys","message":"WIP Add support for ecdsa keys\n\nIn FIPS mode, using RSA keys for ssh is fine as long as SHA-1 is\nnot used for the signature algorithm.  Unfortunately, the version\nof cirros used in OpenStack CI does not have a version of dropbear\nthat supports SHA-2 signatures.  So, any connections from a FIPS\nenabled machine will fail as the cirros instance will only support\nssh-rsa (SHA-1 signatures).\n\nTo get around this, we add a new option to specify the key type\n(validation.ssh_key_type).  This will allow the addition of other\nkey types in future if needed.\n\nTemporarily try adding patched paramiko.\nChange-Id: Ib59eb8432fa1a2813b3047955157d1b3d24a55f8\n\nChange-Id: If711dc682763c674b7d80721217db476ab1c485c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/edf1abc15444898de213347be908655335077f09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/edf1abc15444898de213347be908655335077f09"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
