)]}'
{"id":"openstack%2Ftempest~807662","triplet_id":"openstack%2Ftempest~master~Ifddfb137d40c8819d69c6ebb1bad54d8b4f44500","project":"openstack/tempest","branch":"master","topic":"master27","hashtags":[],"change_id":"Ifddfb137d40c8819d69c6ebb1bad54d8b4f44500","subject":"update volume type with nonexistent type id","status":"ABANDONED","created":"2021-09-07 01:59:35.000000000","updated":"2021-09-29 03:05:06.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b1350f76c25566456eff336374e1da751d6a5947","_number":807662,"virtual_id_number":807662,"owner":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-09-07 08:01:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"disliked":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"all":[{"value":-1,"date":"2021-09-07 09:13:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":-2,"date":"2021-09-07 08:35:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-07 04:14:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-09-07 08:35:04.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-09-07 09:13:01.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"}],"messages":[{"id":"48c1ebba6e1f88c6440441b85eb96defff7056e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2021-09-07 01:59:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0a50a0672b8b2c4850faddd1c37790289fa030d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 04:14:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2216988d78ef4ad191882b3ef7e38a6b : SUCCESS in 57m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e966851c162c4aabb916835984c513b3 : SUCCESS in 1h 32m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5da0164a57ab4675bac5a98ef5f2dc00 : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e2f53cee75f47d68b67547b4488cb55 : FAILURE in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80521eb863e942cfbf58542a158e9871 : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7dcc5bc0ff9f45c6bad59b66cdfd55f1 : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da3b451b0ba2425fa282b6f0439cb9e2 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/902a6398aac44c439e1e0fe96c41f0d6 : SUCCESS in 16m 47s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/3f7a012e8bc7456d9aaf481c5943adc3 : SUCCESS in 1h 23m 36s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/296ec8b832cf4debb8b3fb4d54e1f248 : SUCCESS in 1h 25m 06s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5c2df9f16a42456ea5778e10238b16e8 : SUCCESS in 1h 17m 11s (non-voting)\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/86e0e66400d347f1ab48e23260b507f8 : SUCCESS in 1h 47m 36s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/f31219562c2d4dd89cdd76b442358e19 : SUCCESS in 1h 31m 07s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/29071fee787e4a4681437ba668f6737d : SUCCESS in 1h 34m 34s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/431f7fee5056473ab9deca425e5e7fb2 : SUCCESS in 1h 39m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c68c74808d0545b3aea01a99a10f5d1a : SUCCESS in 1h 40m 55s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/72b3d503ec9a4e4fb20346d892dcadf4 : SUCCESS in 36m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fbb4bf5c587f42f7bae6ddab8714f638 : SUCCESS in 41m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3dcd8d052f3046f8a064c7198da08f31 : SUCCESS in 2h 11m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cba524a04033486d82b5d800dbecddfb : FAILURE in 56m 39s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ea5023805fee4acfa5ff7cfa41f52626 : SUCCESS in 1h 07m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/91ecaec473604be692383b672e9c5411 : SUCCESS in 56m 12s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/e17dc6cffe7d4d6998ebb33ed2c259ea : SUCCESS in 48m 45s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d2ca74e79f954a8eacc08703c55a4b56 : SUCCESS in 4m 17s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/b1bba2b0c0ca415cb174fe0d107ebc44 : FAILURE in 1h 15m 09s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/4d88e8fcff3d43dca959b43b5ad76b1f : FAILURE in 59m 16s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f5760cc8f4184df1a89a7420d176d73e : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":1},{"id":"90faa0eee20fe895149ac8c97c1ac1cfa927f3d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2021-09-07 05:47:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f6510022895e4d13695d57aa7213c9a672cdda03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-07 08:01:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/76c6653dd17e474384e9b16c3bc36fd4 : SUCCESS in 54m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8842a00d0daf42c89a98a3404752ace8 : SUCCESS in 1h 10m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce9c8fb37e4c49f78132df763809ccf5 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87a579da10054bccb1ebe453ce03c44c : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e1757a12af7c4a95bc858cf506ce64dc : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/880294c791a4419a917068e1846fd71f : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d14eb2e338642ae985d2e84653cd3ed : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0160426ca85349f2bd97d823193a71c7 : SUCCESS in 13m 07s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/63804f4a03294e61bd3b2bbcb3d61228 : SUCCESS in 1h 31m 28s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/cb28c8d996de4f8ca2154fc3e0a0ab2f : SUCCESS in 1h 13m 02s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/54f158a504cc4c7184558b2f1671588e : SUCCESS in 1h 13m 17s (non-voting)\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/8d2c509f0e314b0fa5eb216a78c27dbf : SUCCESS in 1h 30m 40s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/8e728c755f9b4886940bc845b6b3a1e9 : SUCCESS in 1h 36m 07s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/86de600bdb1049e4b40dcb3c4c34de1f : SUCCESS in 1h 35m 45s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/a767f2cf38d546bab645caa0ad5f851f : SUCCESS in 1h 34m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7db11525a67f49fa921457622f2f1a48 : SUCCESS in 1h 35m 24s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/95e04561c47046ea9c754c6070eb9392 : SUCCESS in 36m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3f78c6cf782476b9ff04a9f1edb6a40 : SUCCESS in 48m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/48a5ed9eca7346488361c4e3fa7ed750 : SUCCESS in 2h 08m 18s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7e4c721772c64cb8bff524f3d4069aa2 : FAILURE in 52m 14s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/7db524f377434945bc89caa57ec9b190 : SUCCESS in 1h 07m 02s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3d370d945b7b481fae70a53cb050dead : SUCCESS in 1h 02m 59s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/377a21543bf846c9adf8041afd88102d : SUCCESS in 54m 28s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b9b8ceafedf247a38eba08a46bc25e8d : SUCCESS in 4m 00s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/28a43604b8114f79a6bed990eb8bee7a : FAILURE in 1h 13m 38s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/a4c405ca01fa490693ebf194b46d90eb : FAILURE in 1h 01m 20s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/aa9671efa00c4ef2a7b0a5aed5e54673 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":2},{"id":"4ab2579326269df624259d2206bfcd2431123b05","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-09-07 08:35:04.000000000","message":"Patch Set 2: Code-Review-2\n\nthis seems like an unwanted copy of https://review.opendev.org/c/openstack/tempest/+/805122 .. please abandon this one and rather rebase the original one - https://review.opendev.org/c/openstack/tempest/+/805122","accounts_in_message":[],"_revision_number":2},{"id":"39f85839f5d8c6d4da41a8ed507a8925bdc3937b","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2021-09-07 09:13:01.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b1350f76c25566456eff336374e1da751d6a5947","tag":"autogenerated:gerrit:abandon","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2021-09-29 03:05:06.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2f3663bdfc9125dc8647e95acb54ae52a52c8fcd","revisions":{"aefea956f7b41d955ceedf340619cf9b013e9b86":{"kind":"REWORK","_number":1,"created":"2021-09-07 01:59:35.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/62/807662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/62/807662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/807662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/807662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/807662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/62/807662/1"}}},"commit":{"parents":[{"commit":"ad2c3009aed4334a0752679bf984f6c9cea49171","subject":"Expected success status for \"QoS rule update APIs\" is \"200 OK\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad2c3009aed4334a0752679bf984f6c9cea49171"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2021-08-19 07:02:20.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2021-09-07 01:59:29.000000000","tz":480},"subject":"update volume type with nonexistent type id","message":"update volume type with nonexistent type id\n\ntest update volume type with nonexistent type id will fail\nChange-Id: Icfca910da1b56159adc0931c537e1a0566ff4569\n\nChange-Id: Ifddfb137d40c8819d69c6ebb1bad54d8b4f44500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/aefea956f7b41d955ceedf340619cf9b013e9b86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/aefea956f7b41d955ceedf340619cf9b013e9b86"}]},"branch":"refs/heads/master"},"2f3663bdfc9125dc8647e95acb54ae52a52c8fcd":{"kind":"REWORK","_number":2,"created":"2021-09-07 05:47:24.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/62/807662/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/62/807662/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/807662/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/807662/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/62/807662/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/62/807662/2"}}},"commit":{"parents":[{"commit":"ad2c3009aed4334a0752679bf984f6c9cea49171","subject":"Expected success status for \"QoS rule update APIs\" is \"200 OK\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad2c3009aed4334a0752679bf984f6c9cea49171"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2021-08-19 07:02:20.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2021-09-07 05:46:39.000000000","tz":480},"subject":"update volume type with nonexistent type id","message":"update volume type with nonexistent type id\n\ntest update volume type with nonexistent type id will fail\nChange-Id: Icfca910da1b56159adc0931c537e1a0566ff4569\n\nChange-Id: Ifddfb137d40c8819d69c6ebb1bad54d8b4f44500\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2f3663bdfc9125dc8647e95acb54ae52a52c8fcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2f3663bdfc9125dc8647e95acb54ae52a52c8fcd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
