)]}'
{"id":"openstack%2Ftempest~810809","triplet_id":"openstack%2Ftempest~master~I891fd8131623a927fef637d4acac574a8fb18c10","project":"openstack/tempest","branch":"master","topic":"fips-compatibility","attention_set":{},"removed_from_attention_set":{"30002":{"account":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"last_update":"2022-01-18 16:18:24.000000000","reason":"Change was abandoned"},"9914":{"account":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"last_update":"2022-01-18 16:18:24.000000000","reason":"Change was abandoned"},"32926":{"account":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"last_update":"2022-01-18 16:18:24.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I891fd8131623a927fef637d4acac574a8fb18c10","subject":"DNM - Add temporary link to patched paramiko","status":"ABANDONED","created":"2021-09-24 03:31:12.000000000","updated":"2022-01-18 16:18:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7631bfc266bea6c93822011604a279a3ce96448e","_number":810809,"virtual_id_number":810809,"owner":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"actions":{},"labels":{"Verified":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-09-24 05:17:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"cd2fa14251707d7f8eb1ba227d40fd93978ac276","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-09-24 03:31:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a9b2245dad8a048762d500f10546914054452036","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-24 05:17:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/72de106267fb49dcb10ab89215d02365 : FAILURE in 3m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/90338080f61046e09d13eab56311175b : FAILURE in 21m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cb427cd1b42425db764064022ec7947 : FAILURE in 1h 03m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66ab8d8a61a147b9be9d490d87f9cd41 : SUCCESS in 6m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c80d5b2c7f0b44048a171dbde48421e9 : SUCCESS in 4m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b89e1cbdf7314514ad218dd6246c9b0e : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a3e26dc7b9354d42909a0b1793845781 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8594a23bf54a43979d5ae045b24cc09b : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57bed303fd804ee0b317452f97c1650d : SUCCESS in 11m 39s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d99a3db6689042dd9507a2efccd217f0 : FAILURE in 1h 18m 59s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/27830ba268ac45b68cefec19010bbadb : FAILURE in 58m 26s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/372aaac7189e4cbbb53e9f286fea4c1d : FAILURE in 1h 02m 14s (non-voting)\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/2e149e83da4a4a23a8532f168a81322c : FAILURE in 21m 19s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/76930e32dc8d4163818b26de7f832a5f : FAILURE in 20m 40s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/78092ee6edaf418e92fc8d74f924a566 : FAILURE in 21m 15s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/cdf14a74e5d04f2da9ccd09c83253fef : FAILURE in 21m 07s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0246f914c4e945f89ccb20dd3b8d5e96 : FAILURE in 1h 15m 59s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/fb5c2a0240414ac1a5719494a776ef62 : SUCCESS in 30m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/00ae515dae664edb8744aa49907a3134 : FAILURE in 41m 54s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/844e3d07418f4dd19874406cdb396555 : FAILURE in 1h 41m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c046f8b625424673b0e22472b549a698 : FAILURE in 52m 03s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9181c00554194f99bae89df87c33daea : FAILURE in 1h 05m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ff73d9555144738bb0211eced1fbc6e : FAILURE in 31m 02s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/51c5828be7d444548f4627313779d8e4 : FAILURE in 49m 26s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/603a57667c1f4fd897a19837b34b1de3 : SUCCESS in 4m 46s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9043c1fe7b99418089594b809f6099bc : FAILURE in 1h 09m 01s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/9e83e3cdafd149c39cb09e3c782256d4 : FAILURE in 57m 20s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/15f205ead41b4df6afd259e937e99865 : SUCCESS in 3m 45s","accounts_in_message":[],"_revision_number":1},{"id":"4c318d3fa04617f66d7e1fd387e615ec1e95f2eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-10-01 15:24:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"e56da30c79b53cc4c1ebadb5f91b3f23d54cf25a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 17:39:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/4a5fd621052448c68999288731bd96ce : FAILURE in 3m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/5055d0b71c504578b7b9eb182a502305 : FAILURE in 32m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac623cecc34c477b82df7e38222f9888 : FAILURE in 1h 00m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f7e97efc2d4b45ee81044543be81bb2f : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/181810f84cf34a4e996cb7c7bbfa2c48 : SUCCESS in 8m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf6a0551369c415dae8f2ced98b8bfe8 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5f43cd7a8d1c493b84deb991ba431fcd : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9d97e2b33a624b8c9a2c4b0c62a2d645 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c31e385f0924def8cb03d5b75644c14 : SUCCESS in 15m 29s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/459d4853d1a44a549e0a2b93c1cad03c : FAILURE in 1h 27m 41s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/ed8f91fb86594804a6c4694ba3a6480d : FAILURE in 1h 01m 47s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ee49c3bdbadd4519945dc91d28cf7a91 : FAILURE in 1h 13m 53s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/add3759a5986480ca7b2e74b224ebbfe : FAILURE in 1h 00m 57s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/b8a7558b261f40639a675c56676fc364 : FAILURE in 22m 49s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/929eeb64aad043098a2553c0272ec508 : FAILURE in 15m 24s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/f883b7df1b3e4c2487e33d5952098d68 : FAILURE in 16m 52s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/807e3e4da55a4849ba6dc9f0aa2951e6 : FAILURE in 19m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/544ea583ab7b4f008c9aace6e792a002 : FAILURE in 1h 13m 51s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/55846751204e445c8c263f4dcac906bd : SUCCESS in 32m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6311e3b6a61c41f88f7735b56f6cc4c9 : FAILURE in 43m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3b01babe8d094e14846cb603f99cbdb0 : FAILURE in 1h 44m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4f94def1876a40b4bdc005721f1160f1 : SUCCESS in 51m 14s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6a08d1eed56548dd84048200e67fa949 : FAILURE in 1h 13m 12s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/384a00a7728d43868fba5b7aec39290e : FAILURE in 45m 02s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/0623e5d653734404b667845761cfb0d4 : FAILURE in 40m 59s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/4f199d9d994a4dad8be483b4fc8441dd : SUCCESS in 5m 24s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/bf4fd1d4f33845ed883a64ba5dd21eeb : FAILURE in 1h 00m 28s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/5f55a6ff387e4cdab497ecea179e27da : FAILURE in 1h 01m 00s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/bac7404ec20b43fe979e4e279030d669 : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":2},{"id":"fd851f38cdb22b22e9a8951fc17fd85db0eb1d3e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-10-04 20:11:58.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"e62555a3a2c8a280da59d27d1852dd0de9c7afb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-04 22:11:28.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0aee096e6e8e4241861fbe3b0b66ba48 : FAILURE in 3m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/746146d7509d4e2bb17bd678317c457f : FAILURE in 32m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1a105a696ab8469483ed148578132a29 : FAILURE in 1h 12m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1af97ce15b8d4ef284caeb6e1eb33abf : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/688a5473300f4407b054de482f100ba6 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d1ac9080cbdc4ce9a18105d63b27848d : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a1f1aebc45d4d469e093bce5872fc6b : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55644368217a464b97d7f6e129ed13a3 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42665b86dbc34f258bc7f6c4781a6991 : SUCCESS in 13m 12s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/a2da89e68b3949a1a3ec6f2c7b872fcb : FAILURE in 1h 54m 30s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/5589eb1544704c60bb9e94d60fc3d715 : FAILURE in 54m 20s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/808ddf60d04e4b6793c0142622413116 : FAILURE in 58m 13s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/ad96041693484632954fc2fda390ed0a : FAILURE in 1h 00m 59s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/6494acdb030343ca9268aef7ff029946 : FAILURE in 16m 40s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/d99535dc2f954489bff47979bfa95a2c : FAILURE in 18m 03s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/4f1f6afd8e2c4a71b82d2efb83e7e496 : FAILURE in 12m 52s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/882b6bb447e84b1a9f887b3518fbfa7f : FAILURE in 17m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7e9379fd9ed24201bfc6529d69055611 : FAILURE in 1h 20m 53s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/429abf36e0844ae8b53aa6e79621fda1 : SUCCESS in 28m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/499ecd567a21420dabc1342803f773e5 : FAILURE in 42m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/20e5803ea1464dc89c138f2606fc860d : FAILURE in 1h 35m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c53a6b34121d44e399aa921956226d1b : SUCCESS in 50m 17s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b97ff0dae72e4569a09f5c8c24261a65 : FAILURE in 1h 16m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/803c906457e54bd7a4c2b8942ce6ad05 : FAILURE in 44m 35s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/e254ce7d575d4a1da7df8e5b07bcd231 : FAILURE in 30m 18s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/aae2cb93f6de402b8c63f384f79906e8 : SUCCESS in 6m 00s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/bb4c7db1e09540dfbdd4ce887c2a08f7 : FAILURE in 52m 50s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/627151db5ab94bb692b8c413d3b772db : FAILURE in 55m 54s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5a4dfdeef80d4d6d9b1b19f1a563fcba : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":3},{"id":"3f4b5de9e054d89723ccbd0b883b6131dff4535f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-10-15 15:42:51.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"515d7a5de61d62ea8f7e0f14e952795ce940f83e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 17:31:42.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a4b75014a08146bba236b6743576ded7 : FAILURE in 3m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/ba243f80cecd45b092f5daed78ebaea3 : FAILURE in 45m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b3d4d52d07e4699acb7626d2342641a : FAILURE in 1h 07m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9deb2aa173b4fa49296ee29bac6afd9 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e18c27b220472da069affdd601f5b4 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7a696cbd5794901abba63c48ce9cf52 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5e6aefa00e940ecb7c6f8f19d740f0d : FAILURE in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d43b2b291f448038582b13301623cee : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/100fdba3adf34ec2bcd0b1336cd1eae0 : SUCCESS in 13m 55s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d97be0be812545e7a85b3a594c541cf5 : FAILURE in 1h 16m 47s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/ac15d691d7254a7ab35929a9f766bd25 : FAILURE in 1h 00m 08s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e188b87b902942f0a3a83094ccb38cc1 : FAILURE in 1h 12m 38s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/db9f27530df049c29f9a1b92c01447eb : FAILURE in 1h 01m 42s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/ea087ee1b7bb47c191111e8f38fc0cc6 : FAILURE in 21m 52s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/afbdd5405fc9491ebac16299ec58884d : FAILURE in 23m 39s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/7c8c3cd7eafe4c758b67e71424b6c800 : FAILURE in 20m 20s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/f9c0fa24d3ee4dd19e721fdd372e4a31 : FAILURE in 21m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a4646c1db3a94eb4a95a315b3ed42ce1 : FAILURE in 1h 27m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a43709d6a5a146ebb45e9ee8f54ecd95 : SUCCESS in 31m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9224aea4d06742e290545ef588289447 : FAILURE in 39m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b43a4a52f409429599e71730bd0679f1 : FAILURE in 1h 45m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2382ef1cd5c84ff09ad31470894f45f1 : SUCCESS in 53m 00s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/78e2c11264c143b78629a437f5ce1c12 : FAILURE in 1h 21m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a77a1c729364828ba9951e157b32352 : FAILURE in 42m 53s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/9c4d8b620d0249a7a3c62bef48ed033e : FAILURE in 47m 48s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/4a87cc7db0c54eb598659da72af05bb8 : SUCCESS in 6m 37s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/649df4c3fed44976a4989f1d0bbf37c7 : FAILURE in 1h 08m 12s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/60ddd775909b423ebdcc2cc53f252ed6 : FAILURE in 53m 54s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/6ce4af872664447a89c5d091698164ef : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":4},{"id":"483d45a1855395d71fd0b7e19e66eca839f9cd2d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-11-04 19:41:16.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"1c2fe0869f229451ce87f2f7e78ce9d9e9ba027e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 21:28:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d9aaa285ec7b4fe99c009a7a798b944e : FAILURE in 3m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/bfc457fd73494a3aa4c05b1eb4c6a13d : FAILURE in 27m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1124ab0122d84f7baf039ee3ff67f231 : FAILURE in 1h 07m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7055a03d04e54e08a8bd247ea4f02e94 : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbd127e735af46e3ab9440f52dd88761 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9988f1ed6467440f9aeb152ecb4db08c : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1ab9da3c6ed41ba95f975945b076dfb : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bf80239877f45e7badf386a611b0b27 : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c07d8134f8f34902ab09ffad70591b76 : SUCCESS in 12m 56s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/174ba3bbde33447889f64e2b38cf7d05 : FAILURE in 1h 26m 23s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/88acf66bdf784b6aa54321bb4d06c2c8 : FAILURE in 1h 06m 25s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8d0f34fc508f489facdc940b6eac8622 : FAILURE in 1h 05m 07s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/2382b4b0cd4044ceb7bce86208107694 : FAILURE in 1h 03m 18s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/8355fbd3ca5b4b698f335a03f8697d6f : FAILURE in 29m 29s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/ab5cdb13fa644c53935708a3b197c205 : FAILURE in 25m 54s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/5ce07ecfe2e94601b6612d5aa2ae9bdc : FAILURE in 26m 59s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/df9c3c6e99ab4f288230b8ac82b9f491 : FAILURE in 22m 49s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fbb385f67a7041e28f72bc74b0ba9a81 : FAILURE in 1h 26m 31s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/6415fece9e61401da2355d1004e6ee9d : SUCCESS in 33m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/286a7ee265594359a4f730f8c5765f98 : FAILURE in 43m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/78e510c351c54c27b052622e887dbd48 : FAILURE in 1h 45m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6fca28cae03e4c0ead789a72bba276eb : SUCCESS in 41m 07s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/66d50e9f3df04c02b2e7feedc1717bc4 : FAILURE in 1h 10m 52s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6aa019c6c3b4802888979c5e230acc5 : FAILURE in 38m 49s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/4dc39203da064e758d8154d65e9d180f : FAILURE in 44m 59s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8a514be0d803466da3ad3c5c962e893b : SUCCESS in 5m 11s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5cea46e44f184446b087c7146c449b91 : FAILURE in 1h 03m 07s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/2e2e56527ccf49799717e965d1112056 : FAILURE in 1h 08m 39s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ca6de75c518b4a9587e12bc4ecca588d : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":5},{"id":"c7821db29dd20595b3b08e895a5099c2adb22242","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-11-05 16:46:58.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"e0ef1c70d449449f52276d3020dca24d0e57fd73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 17:44:51.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d3b93049c033415b9fbc3a826a41b0da : FAILURE in 3m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/ce0701b09dd248fcb81955feb6e99440 : FAILURE in 32m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ffcf0f23fb1943f2867f6407b6c41056 : FAILURE in 34m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb2ce14df27744a58c216dbd1f8d2e1a : FAILURE in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d6af3fc4bde4c62a25725389a66f4e3 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/796246822e7741f68d26edc9f2146dc5 : FAILURE in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80f1e8a1f7654fe7926c527fd8d6edc2 : FAILURE in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28f1a6bce06b402ea55dd724f6b86761 : FAILURE in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b6031d381141b79ad6b82bd308dd98 : SUCCESS in 13m 03s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/db301d6d50e64005b4b9aa4bdf848004 : FAILURE in 32m 26s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/cf12be789e984133b651998908585c6b : FAILURE in 31m 12s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0a8c75adfa65455bb89a499524ebd5c9 : FAILURE in 37m 58s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/01d0cdaa33da441f8176b4bea13ce8d2 : FAILURE in 35m 55s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/6bd0a7a0902044caaf87a4c1457b3504 : FAILURE in 22m 58s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/028a24e2b715475fa85f7c40a89be28c : FAILURE in 16m 14s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/1b0537832a4b4760aad2e690cbf2d80f : FAILURE in 23m 00s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/44b961903ff7406d99a6a45825195e8c : FAILURE in 25m 04s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9384df94cce14317a44f4c2856559359 : FAILURE in 41m 59s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/43558ab9948d48e18cf1a19b89c1441a : SUCCESS in 31m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7a4b69d10cb406cb2f67c0641cf27e3 : FAILURE in 29m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/cdf58ba085e2407587273fd5b2bf42d3 : FAILURE in 39m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5695c2c90c7e4d579e7f412cd6c60ef8 : FAILURE in 50m 33s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/80732b0657394cacad20731c1987241e : FAILURE in 36m 10s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/299d408b1fa445c6a5cc5c73f603c48d : FAILURE in 38m 13s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/0532da738c7e4372a5af28845e183853 : FAILURE in 28m 12s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/6de32cf2eea44e3b9e24d66ff69f6efa : SUCCESS in 4m 50s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/816acb03e19c4ddbb75fef0dd449b511 : FAILURE in 36m 38s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/2be36ca9323747e5bf9796940e3f5d6c : FAILURE in 33m 41s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/06fb0de70a7d4889a50d3fdc353f8ba4 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":6},{"id":"f07b20ee8ee9b08bf9ee6d457393b8609c1b6eb4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-11-05 20:37:09.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"cc7d88d6c01cacaed222e3bf5a2525f179d4b23b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-05 21:48:44.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/97a55cca232549569b5c88d4b8b84cef : FAILURE in 4m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/5607f581fb214aa29f8912aa61ebbd23 : FAILURE in 47m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/76ad33ddb01241f391838332ddf00bce : FAILURE in 1h 02m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4635c9c165114460a392725f65aec78e : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35c119f876d64613aaaf41ee7a9a0aa5 : SUCCESS in 5m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e90b6eb0a9d5458aa5f9ba6a4bf6b1eb : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dde5ebf6dd44b9eafabee7b3d7ad2a0 : FAILURE in 6m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7447f4bed31b446c826c5192a89a5a67 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a98f25497e543429bee8728d8906baa : SUCCESS in 13m 38s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f5af82a928fb41f185b735a9e97e85dd : FAILURE in 1h 06m 52s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/462a530d94014c82a5d9bc76a5aa3638 : FAILURE in 56m 07s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ce33c0c230c4471dadaad0cb1fd664d9 : FAILURE in 1h 00m 12s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/984e1524c1b547a7b101c35abdbbd1bc : FAILURE in 1h 05m 07s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/be9e7bad03104b4c9ef658e8f56eadc6 : FAILURE in 14m 27s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/c503685c756f4ddc8862caadbfd7c494 : FAILURE in 15m 06s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/969c0759c23b4cd78b7c7fdfd34749c6 : FAILURE in 14m 59s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/54c9846df61f47bebd3de1dfc89401d0 : FAILURE in 14m 33s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9fc5eb4de4ad40e292784c518ca5a99e : FAILURE in 1h 06m 50s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/11742dba2e504646afd8fcc5fce3ce32 : SUCCESS in 29m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f636cf2f044b4573a707e6eb7f2a7d29 : FAILURE in 36m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b1e5aef923f34b96b4376b77ee6c7e00 : FAILURE in 15m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ffe502e0d9f7447891c63b92a498b700 : SUCCESS in 40m 08s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c57a14e286484cc89a29c559b949a463 : FAILURE in 1h 02m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d67ee2d945bc40f9bd3da03442a6ab62 : FAILURE in 33m 29s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/a1719311240a44f18910bcd9416c6fe9 : FAILURE in 48m 39s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/c3a53896b92b4b638f38be33d885e2a6 : SUCCESS in 5m 34s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9feac9bbcd8e49ff9349c968cdff65cf : FAILURE in 55m 14s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/85b86c1d448546efaef49ceb4acb5d3e : FAILURE in 53m 08s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a4c1a462cfdb415fbe9d8bbf9df6fed7 : SUCCESS in 4m 23s","accounts_in_message":[],"_revision_number":7},{"id":"07d037140ed177ff43955735be36526ddb6906af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-11-12 19:13:12.000000000","message":"Patch Set 8: Patch Set 7 was rebased","accounts_in_message":[],"_revision_number":8},{"id":"0a0777da9851bb57ee87adee58de15bb71d47cb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 21:36:29.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0fb55a208ebb4dc0add223c2e7bb8a95 : FAILURE in 3m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/995eb1c64c41445388aa8babc4844af6 : SUCCESS in 1h 01m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/59f8851bf50145f3ad8a311a156872c1 : SUCCESS in 1h 26m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d602c2341e4402e85fcc037f58a0baa : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b56091fd7de54ca19e418fa20ac59869 : SUCCESS in 6m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b1c8209d04ef41a793d961384e03527d : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aeea1523889a4129b583fe73f6717cab : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7364e3765416475fa3b6b95fbd2d2a84 : SUCCESS in 6m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/832376def282448fb5d471ae943dd18c : SUCCESS in 13m 16s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f049e81fb2e24dd6a9827976cf4aa15a : SUCCESS in 1h 41m 53s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/bf2a87afe43344b68b012598348d196d : SUCCESS in 1h 10m 46s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1f1baf718a814f1b8b642907fe84193f : FAILURE in 1h 13m 29s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/5a445b409a2f46cdb3d23cb56d7a731a : SUCCESS in 1h 26m 04s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/dfc5eff7cb9f46b2a0105b4bb4cb234f : FAILURE in 22m 25s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/ed988dc9fd884aefa331e8faefd5a770 : FAILURE in 26m 04s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/f3e336da91184518a9221166b9d8d26b : FAILURE in 24m 59s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/09641ef0f002405482f65aedc81aa2a9 : FAILURE in 22m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/007ca29e2c9b4140bd28b1e214edc247 : SUCCESS in 1h 32m 27s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a172035be5f74b33b32f2287fbd9009b : SUCCESS in 32m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45d9abc386f64aa690bb6ad770efd866 : SUCCESS in 50m 08s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ca784d2d87ca4c19b381b0336d50416b : SUCCESS in 2h 14m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8b9be3912d34454a92e1df0e9a88f9ad : SUCCESS in 49m 41s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/81135c9caec3483a93f27af58424751a : SUCCESS in 1h 08m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d4fdb407b8f94c9e86316a3b4483a493 : SUCCESS in 1h 05m 26s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/c6971c11feb64957a9f3d51a71ab0969 : SUCCESS in 1h 16m 14s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/32d7a94890dd41a0bcf4a5e9698dbeaa : SUCCESS in 5m 40s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/85a5c3d831b348a4b303f518056189f1 : FAILURE in 1h 26m 55s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/befdddc8a9654b178e845734fe64d312 : FAILURE in 1h 01m 29s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3129ea08ef00436ab0d6c25f68c0425f : SUCCESS in 4m 24s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/02dc7efd0ac4444daedc8628f8a13ba7 : SUCCESS in 1h 26m 23s","accounts_in_message":[],"_revision_number":8},{"id":"5cda8d4bc5c88b7885a1d649cc6928fea41394bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-11-17 14:52:45.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"2334c948f005f21117707502e2cb93ff02914d7b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-17 17:35:00.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a61500a02f7942c59304d5dc9783cc46 : FAILURE in 4m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/7001b9d67ddb4a18a144d3ceba4467eb : SUCCESS in 57m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3a91ce5678a340288999b10219efc091 : SUCCESS in 1h 43m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da2aa69e70f34a6a956daf929d8c65c9 : SUCCESS in 8m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff0fabb4911b46d8b0689b09d4c2aff3 : SUCCESS in 7m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f22e495bc7da4c39a9fbd724703d8895 : SUCCESS in 7m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3c524c827de846259dcbb59bf8ec4478 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eebfa6d2f00f462e94909d1df7e1fa30 : SUCCESS in 7m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97bff0641b23445783cb907498c70176 : SUCCESS in 15m 35s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/87ee50dc251c4c10b427c6a402693c87 : SUCCESS in 1h 42m 35s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3038f5559fea4ad2923eada0ee17fe9e : SUCCESS in 1h 13m 22s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3a0effd83ea54a0dae0fde7ec6b6f9a6 : SUCCESS in 1h 22m 39s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/19524e676b9b43dcb0fce237416fcadc : SUCCESS in 1h 19m 44s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/09ac42ec89f1481e872d5b845cf20c93 : FAILURE in 25m 22s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/6ff5f57b468d4444811d62d2e57aba55 : FAILURE in 24m 39s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/d4d71de7ff004f56ab9a75e0f4dcd8d6 : FAILURE in 22m 18s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/d124d10d8c084754a5cb7e3b8cb86781 : FAILURE in 24m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d9c07198e0884d519cfc827bfb16e7d3 : SUCCESS in 1h 33m 35s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3e0f614f2589464e86d79640c4ef64a9 : SUCCESS in 35m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8c46c7047df456d9be8d31ef7be0247 : SUCCESS in 55m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2430e721ee6d46bea7ff67a43b8a6095 : SUCCESS in 2h 11m 30s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/19871d3954be4f50b10f004d7ab3a76a : SUCCESS in 53m 26s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/38491188a8354a11b84a0824043b7740 : SUCCESS in 1h 20m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6dbc2c9773244de6bf26ba6a657c3c0d : SUCCESS in 58m 28s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/b25b957f51584fe58233aaa8f29fb38f : SUCCESS in 58m 47s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/184205e7d6184f139fcc59562c07e625 : SUCCESS in 6m 20s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7474a57effd94f509e991143419a5af1 : FAILURE in 1h 32m 30s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/9029a2df17584499af4d481132f6b6eb : FAILURE in 1h 00m 22s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4d41335f89204944a2a8c27ea65cb576 : SUCCESS in 6m 00s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/c1505e789b424ed6bfddc9a1efcbb2c9 : FAILURE in 1h 21m 50s","accounts_in_message":[],"_revision_number":9},{"id":"7397c5ea8304dcbe4d04aca2405d71a33aec65b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"date":"2021-11-29 18:45:12.000000000","message":"Patch Set 10: Patch Set 9 was rebased","accounts_in_message":[],"_revision_number":10},{"id":"2768572e3a86c52a028a64ed71ddfd8cd55ed4fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 20:59:26.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/466a9b8553ad43978233c4428e8e15c5 : FAILURE in 3m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/13a9731669d9453d8d187a0526e706fe : SUCCESS in 54m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/461445a7b69f4304903a657c0a0ecd66 : SUCCESS in 1h 17m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b26272bc4e154f5fb7373bc99872c2de : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12cadb73d73a4752a817be6b5a9f9f66 : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0b487776dc442aaa90b763d972eb002 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5285f6c03f6f42c3b918629b8e19126f : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8f4aafc9b9fc448eb453e0781616f9ee : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19fcc89c0d904ab1983dac494fb522a5 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5d523eb7daed45cf8166ba720fdd5770 : SUCCESS in 6m 24s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/36b906e4b0144eaeafb47c134fe37ff1 : SUCCESS in 1h 29m 38s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/9d07bf70005f4facba80fa51c4733a83 : SUCCESS in 1h 00m 02s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3bef7008abad4f14b30c572dd0593481 : FAILURE in 58m 17s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/a277d905765a45fb856d7182bda466cf : SUCCESS in 1h 18m 14s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/40ce03d34a4e4f0a8bf9242fae40279c : FAILURE in 24m 17s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/4b821a92d363432ea105924f654acb44 : FAILURE in 25m 09s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/7affa14b8714453cab8edbe091dc8b46 : FAILURE in 23m 58s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/f63b0cf155c54dddbf5aa7231ee7186c : FAILURE in 15m 01s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a4fd374e69ac4bb88012019a34465b3c : SUCCESS in 1h 28m 15s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c130ee0443014233b8b145785682722b : SUCCESS in 36m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3d50f42eef1b424f9dc03ea23dc7610a : SUCCESS in 50m 21s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b9c36c7d79b4026a82665c5c80e35a6 : SUCCESS in 2h 08m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2d618d156d2846bf885b4fc837b90056 : SUCCESS in 51m 32s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1bc96a9136f748578cfa472000eb7749 : SUCCESS in 1h 27m 17s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0f51268db7fd44649fe93988f0b801ba : SUCCESS in 1h 05m 52s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/9417d18e71ed40a8828fd6215edd1928 : SUCCESS in 52m 07s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b0e1a5a1085b4741923a466ed2db5731 : SUCCESS in 5m 42s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/36600f17b9e644459256540612e487cb : FAILURE in 1h 12m 02s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/d44034e917fa423dab33042e471d9ef8 : FAILURE in 1h 04m 55s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/927b0fb8b08842a582659363263b01fa : SUCCESS in 3m 48s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/8c81f6acb0884e50a88587c20fadb62d : SUCCESS in 1h 52m 24s","accounts_in_message":[],"_revision_number":10},{"id":"27dd7c8a422272eb37d97045fbdce247d6c389de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-11-29 22:11:35.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"0cffc03307e098bcf0f26aa087eddb5e5ca45acb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 00:33:58.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/86e3ebb37aa44574a944e724991ae6ad : FAILURE in 2m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/dd41a7101d46468891d80f965d84254b : SUCCESS in 1h 06m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f74fd4825ff64e41851b7b1a95e93e06 : SUCCESS in 1h 25m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b953b5afe7674e67bcf2efef5e5c26a6 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31b4c1bb834d46b6a30cc7b983a917be : SUCCESS in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8288d6aaf4f64b20b8c8ff1f5f13d47f : SUCCESS in 5m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98932a2a64564c429a8ae3d395e28b80 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bca1f869b244187a1bcc98e81ff10f5 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d66364df2ba7419cae388216bc8b8dea : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03e33f6f1ba448b7bb3a8eb4f0a034bf : SUCCESS in 5m 49s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/de89d99290104f37b33dbb9b1f03d5c3 : SUCCESS in 1h 47m 11s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/7252f7d2d0004235aa4b208748fd8ec1 : SUCCESS in 1h 26m 48s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c12ae5eec3544513888c7dd27b7ce8d4 : SUCCESS in 1h 12m 31s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/54ab09ef15194a85a9e4d5882d08d92c : SUCCESS in 1h 17m 43s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/63daec04a1cf495f8fee72adf545e6f8 : FAILURE in 16m 04s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/c6b1734130ff41bc9038c73dedffaa20 : FAILURE in 15m 13s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/5dba29d7fd67475591a25beaeba5b059 : FAILURE in 16m 00s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/fede36b331814b779d61e6d12f85463e : FAILURE in 16m 07s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bbea6710d48141e1b4ca9060dc360d37 : SUCCESS in 1h 36m 52s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f06a3f47512f4685bd1c8f5983fea439 : SUCCESS in 24m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e536a04b11bb404d9af941ba02688ba5 : SUCCESS in 45m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4d7897c1df724cedb6a5a1e198fa6e87 : SUCCESS in 2h 15m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/59bed0d9be0c4ed98bff94a5669a2520 : SUCCESS in 44m 47s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d3ff186124914b3ca31b7dcdc2108e12 : FAILURE in 1h 50m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c1ccc4781a1c44cdbcf6d6ddd6437eee : SUCCESS in 53m 51s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/7ceb010011b54fa3bcf20767b1c83290 : SUCCESS in 52m 08s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/14eb779cbd6b4e8684f3f4b83b72396d : SUCCESS in 5m 20s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/171562af178e44caa83d50a312dfb1ac : FAILURE in 1h 04m 53s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/6fb4da8dd7a34b5d9d1ce20946211ca5 : FAILURE in 58m 19s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5d35b74e607747f6bd8d204ec679b63e : SUCCESS in 3m 47s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/d4041c51be3d4846814f1a79ffd02da0 : SUCCESS in 1h 25m 36s","accounts_in_message":[],"_revision_number":11},{"id":"0cefbdeec19fd2308e481a7a1b18e454e42665eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-11-30 04:08:54.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"387b671a0c50591a1a0ba5bf5b27ba241eb1ea67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 06:27:52.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7820d8c1f86f424d97737054703ffb6e : FAILURE in 3m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/5160ef2343e44a30849d96fe02ddce6d : SUCCESS in 45m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7b44e5cc65b64904b133c775c361c56e : SUCCESS in 1h 19m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7766972e53664f15a9aa8a6bd0086720 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97c86c493a2a4978af664aebdd14a38c : SUCCESS in 18m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6d242ccf90d4931b6c57dbeea5b6dbc : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf57ab01360649bebe35b2917f4e2bd2 : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/33b764724d7646aa99767c6c365785de : FAILURE in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cc7f2177d97429ca9095e049b779c45 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5388b8dc4c394acb80760eb26d0ee736 : SUCCESS in 6m 45s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/eda4fbc2cfa74d4b8c77f7186cafa8f8 : SUCCESS in 1h 35m 52s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/afc1ceb8ccac4562a0c61821ac745de9 : SUCCESS in 58m 41s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/95c8717e4a1d48a6bf620634e04fc525 : SUCCESS in 1h 25m 22s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/69c35295929d474696bb5b1e52b6071d : SUCCESS in 1h 17m 12s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/53d777ad3c3146959c81998c71196b0c : FAILURE in 17m 00s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/89edf8cb95df4b4cb57b82f8d87e46d0 : FAILURE in 16m 26s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/d9d34a5996ac4d7382119f6a7d6beb17 : FAILURE in 15m 17s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/2268c5cc2381448ea5e8e30395ac2dbc : FAILURE in 19m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c15f63ed64804eb3867aac689fed33be : SUCCESS in 1h 34m 11s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0e97906096d0463bbdd15415c5765627 : SUCCESS in 27m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29df0a50222b49e8aea7eece1a020ee4 : SUCCESS in 48m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/508813dfd49b42bda29e004757f07863 : SUCCESS in 2h 16m 59s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/664b2ae52cb045769e025a7aa24c5289 : SUCCESS in 45m 34s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c14e9520b9b748a5956990d700a1fcc5 : SUCCESS in 1h 22m 50s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ef1ec7763c64b57bc0ee6a95c38b732 : SUCCESS in 1h 04m 44s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/d6a6553a99b342c598da795274b5ea93 : SUCCESS in 1h 10m 22s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/01f6c8e6ed4a48bd837c15aef27be9ce : SUCCESS in 4m 28s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ce5276e8b13543f7b140c40f8e4cc898 : FAILURE in 1h 07m 33s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/95748ceeccbe4091a1583edc57194dcb : FAILURE in 56m 09s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/63a2b2d65ff94cb1923f322ea6cb2237 : SUCCESS in 3m 46s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/5a78998256ee4772818e48cb67148169 : SUCCESS in 1h 31m 49s","accounts_in_message":[],"_revision_number":12},{"id":"68e3ce450971c295103d531bcbba26de82c392e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-12-02 22:41:39.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.","accounts_in_message":[],"_revision_number":13},{"id":"2c80d118ed111cd704ca43737650f80e8d29861f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 23:26:59.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0a3179ad09694ac3aa44ac239c5e5d1b : FAILURE in 3m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/b7b3f7b83f04439492784355c13c0393 : FAILURE in 18m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d1d529df234644c8a0ca52e071dc9a96 : FAILURE in 25m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6fcc51114b664bbb88707f7488d560fd : FAILURE in 3m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46fcb09a4ba449b08945e1f8077b3448 : SUCCESS in 13m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8fe9ad9d0f3417998206b8860d32f8e : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2af3ba5d1084114bb8ee9b16d6c862a : FAILURE in 3m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0f30e82bd8584a3b9aca0e4a44b8e95c : FAILURE in 3m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df36f32248de4b0d97951cc2dbb7f55e : FAILURE in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/78be0555c0e24f92ba72034031334e77 : FAILURE in 3m 33s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/a6bb8ccce36f468fa06eae4971fc0f41 : FAILURE in 35m 12s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/2fe180dc929243499d781ceb0fe652b4 : FAILURE in 30m 11s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1da16542fe854214bab3415e4fccbca3 : FAILURE in 33m 39s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/6f72f298521242a98d2852eb16db4d83 : FAILURE in 32m 33s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/ea066b2f7fcd41629aa0c484ff272ec3 : FAILURE in 26m 48s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/910dce8189734ede98ec3e3cabfa4892 : FAILURE in 22m 18s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/db276af93cab449d9daf408a64009960 : FAILURE in 21m 55s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/b4b59b660ca64d7bb5a5d793b197812c : FAILURE in 26m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b64cdda7730c41b2b7dc3fe9c986dd47 : FAILURE in 36m 20s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/40147ae6edc942189826b885a510dca1 : FAILURE in 2m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65fdf53195be4206b42474c2fbebb5d6 : FAILURE in 29m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dc85ea8784794d93bd27a2c1cc260ce6 : FAILURE in 37m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/651c2ad9d51e4d609b96062825aeeefb : POST_FAILURE in 37m 09s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/21f908c754664197a921cb5de82e078d : FAILURE in 30m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dcb387435a4f432b85e20dbe581a6c55 : FAILURE in 35m 26s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/0b7bf35f6e1b410f876cd9676132152a : FAILURE in 27m 33s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f9216e068a2c4cc28e67f6571a38a489 : SUCCESS in 4m 09s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/955bb7f9997a4e629129813e203d7907 : FAILURE in 31m 11s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/8fe851c9ae7a48258455c400197cf212 : FAILURE in 31m 16s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9b84ccb4a60c49d5b39c067dc26b7ad8 : SUCCESS in 3m 21s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2abef6ab9ed4ae3bffd1bdcfd852e4e : FAILURE in 37m 57s","accounts_in_message":[],"_revision_number":13},{"id":"fed314edd8ab0082c01cb2f1ccaadd760c60d649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2021-12-08 22:24:17.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"122c54d909c8370b04dd0075f7b412b0490d260f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 00:45:34.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/b252f5999be14f36bfeb37b62c7bebe5 : FAILURE in 3m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/06e25772ba554b3db04ea398320d7113 : SUCCESS in 1h 06m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80893d761954447db843f8f101f7f640 : SUCCESS in 1h 20m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ada51ff60904892ad50a13d15e2440c : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8df7d1ca6fb42279be28105c051acd4 : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df5dbd14cd274ab49f629b96acee6cbf : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97fd79626b514b8393251d7419758611 : SUCCESS in 5m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4193229fd73c42468fd40edc7a04c374 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/556eaa9dce3f43a2be905f5f62279a7e : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/101f528a0a6543fda4b796ec1a53fdc8 : SUCCESS in 5m 17s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/fc741ca74c3b4e2d8ce2fcae8e13ae79 : SUCCESS in 1h 40m 44s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/21b713d0ff3e4ad58634b05225d0136a : SUCCESS in 1h 10m 35s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d782a427a39e42809a9d697abeafcf7c : SUCCESS in 1h 13m 18s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/bef5b803198e464babe5329257489201 : SUCCESS in 1h 16m 26s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/b1daa36aa3a646fd84aed633ed15ef9a : FAILURE in 22m 08s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/d754148e99e144d5b511844a9ee0d00b : FAILURE in 16m 36s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/05a595a686994c5a8570bb2b4b012f46 : FAILURE in 13m 46s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/52a8def7d6534e04986a0b325474d836 : FAILURE in 20m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a41ab257a5cb4fd893f8dc824093fa30 : SUCCESS in 1h 31m 47s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f51a3512874f49e593d0e1cf869d6398 : SUCCESS in 30m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2a1fe730340444d8123bbfa5bdd38dd : SUCCESS in 44m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/61905199aafe441db9016dab7bcff811 : SUCCESS in 2h 17m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/040c14fd312a4bf0af8626ce8f852cac : SUCCESS in 45m 45s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a188b39014484ddea3ac33136f4b66a7 : SUCCESS in 1h 08m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/35e24b9d65c846a1ac5f950a3fbf8e89 : SUCCESS in 54m 20s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/77a6306b9d2c4f3fa3015b4fd29ae299 : SUCCESS in 52m 51s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/aeea461fce814ec9b6be3412f22b84a0 : SUCCESS in 4m 46s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a81d8e42165a4b69bc3aeda04f7228b5 : FAILURE in 1h 03m 51s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/faa29c780739484f87b2dfe65e76d4a6 : FAILURE in 58m 51s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9861f1b1e96e436686a184e6ca14335b : SUCCESS in 3m 50s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e0ee76d97654e5c9b849610eb409f39 : SUCCESS in 1h 32m 18s","accounts_in_message":[],"_revision_number":14},{"id":"5a309e4b703cd04686ece48dbce4f7af70593e85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-01-03 07:28:51.000000000","message":"Patch Set 15: Patch Set 14 was rebased","accounts_in_message":[],"_revision_number":15},{"id":"871d77a81d4851a4843cac6690d310da8c201068","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-03 09:53:30.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2a46f3fcfe364c8fb06c76aeb69edcbe : FAILURE in 3m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/eb5e2fe68fd44638b2edab407a416f36 : SUCCESS in 57m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/65493ff8657c45baaf4d8cacdcb258b9 : SUCCESS in 1h 35m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf0a07fa328646c6a04c5b578afe9cee : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9d260ee622749f783f38286e479cb13 : SUCCESS in 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acd9dac64aa241ccb05046282a855c49 : SUCCESS in 5m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e6aea5af85c346bd81de63abee0c8777 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/878b618962bd4a6290bf84e6a168071a : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c083b353d29c41edbcc39b9d17304fee : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8ab97d709db404a878034af6e52af22 : SUCCESS in 5m 45s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/e09340034f4d402fbe2b5004dbb1997d : SUCCESS in 1h 37m 24s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/69e39507611e42ee8ec2df3a7fbfcc3c : SUCCESS in 1h 21m 24s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/77189042872c4ed8a24d03aa4dac1cab : SUCCESS in 1h 31m 51s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/fd795f2d83ca455c8177e655d2292f7f : SUCCESS in 1h 19m 10s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/3b4d94ed62a14efdaab7debd40e253bc : FAILURE in 16m 05s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/0b7c76271a6a48549a79ed4fb7bff960 : FAILURE in 16m 01s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/32ae7ebc107d459c8c618cc8a277c499 : FAILURE in 16m 07s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/7fe53f736b6f4b52a510370ac9998e33 : FAILURE in 20m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/719c6252deec4377b87b68921b0a3769 : SUCCESS in 1h 40m 09s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a530f7e6497c49849bbd6f54ff6b3a95 : SUCCESS in 31m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f212db79822408a9e51c21f7ba2d407 : SUCCESS in 46m 47s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f2b50d8ac30f43ab94056324c038afbb : SUCCESS in 2h 19m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1cd2e5e945ea4956b3f9910cf74d574c : SUCCESS in 48m 47s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/dd3b7de9adc148b68fdb17ea19bbbc12 : SUCCESS in 1h 22m 00s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c8a255319e794eef97d44ca75fd70db0 : SUCCESS in 1h 03m 31s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/ef1a1e576de743eaabdfc5669defe69e : SUCCESS in 1h 00m 43s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/87a9ee3f8bc24d90b9dd361cde435d86 : SUCCESS in 5m 27s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/713bc4749bc34b818b66b76c1bc52a38 : FAILURE in 1h 19m 33s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/35453b72098148e6934fc65cff95ddcc : FAILURE in 1h 04m 41s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/60e3dfe055064ec6ac6f2b7b7387362e : SUCCESS in 4m 13s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/379f9ea67ed24e66a0c08f86e99e3f71 : SUCCESS in 1h 51m 45s","accounts_in_message":[],"_revision_number":15},{"id":"aaceb24daa93b7ff79170d38919d48eba6794a4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-01-04 11:39:03.000000000","message":"Patch Set 16: Patch Set 15 was rebased","accounts_in_message":[],"_revision_number":16},{"id":"f31283c16a7330f539fd71b1609112996ecbb3f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-04 14:08:45.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/536a397f5f2a4c14bc5c41536953829c : FAILURE in 4m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/f1d9c62933ec4e83942e20b57a59eb0f : SUCCESS in 58m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0d7bbe7a929641368934d0efd6ef491a : SUCCESS in 1h 23m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/95500d3f98c244bdbfe9a8ced3a641ba : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbb18f28bf32430499d90d9d4a5748fb : SUCCESS in 14m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/094bfcd660904a6eac17f94ace64ee32 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a9345ec694724bf68518ab3bc4d32f69 : SUCCESS in 6m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3cf02f3489a140e894a7767b502ef2c8 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/21261de0282740c9910b88a4327a961e : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9346c8cca6a14b7fb0d6b20c88d52306 : SUCCESS in 6m 33s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/53fec5782e28453380f804c2eea5da09 : SUCCESS in 1h 47m 25s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/6ff0c73c7c8f401c9071b68761f4f61b : SUCCESS in 1h 17m 32s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c4117191acde4228a40f500161819a56 : SUCCESS in 1h 36m 14s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/432ca24e0abc41f68ab4c7aafe1f7232 : SUCCESS in 1h 20m 43s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/0d35933d1c1c4f6d875a8aa27c4d2a75 : FAILURE in 24m 42s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/a52eabf795ef47868c1aa894b463a421 : FAILURE in 21m 04s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/bfdab7b09187476589f9cf9a353b7648 : FAILURE in 22m 48s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/deec178fcdd6402cac13138a57c8aec5 : FAILURE in 24m 51s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9a9fc59ad966472bb2a36f3281e02dcc : SUCCESS in 1h 34m 14s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/16b53e7a708a4a34a038b4c899854c86 : SUCCESS in 30m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b417920d856f46328115900d9c52bc5f : SUCCESS in 55m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a206f0225e364a8eb65b8c3839a12d9f : SUCCESS in 2h 20m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1f51c9ce050a44c28e67d1999c798735 : SUCCESS in 52m 39s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/54a9c5cb5ee346329018446cd5117420 : SUCCESS in 1h 12m 27s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c90357d64ca04cada6fc3e4b16319e5a : SUCCESS in 58m 17s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/fcd9bf63816c4111affe06449f64a152 : SUCCESS in 58m 30s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/70cbed5f0d6c4e94b6e402f3d11788b4 : SUCCESS in 6m 52s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/14d221281370427c8acd7e81ba253c11 : FAILURE in 1h 26m 50s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/b28cf746b4bf4729abc4ef099b6362d5 : FAILURE in 1h 04m 54s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/6d629f47c2ea4625adbf5d412874a240 : SUCCESS in 5m 10s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/5b345527db36448aa03b084b971a33a7 : SUCCESS in 1h 33m 48s","accounts_in_message":[],"_revision_number":16},{"id":"38fcd5537a81f5377292715a014eb6f38c6c0f19","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-17 20:24:27.000000000","message":"Topic fips_changes removed","accounts_in_message":[],"_revision_number":16},{"id":"49b0c4873a59e4aa91b52d2e2c53f5dc48415545","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-17 20:24:34.000000000","message":"Topic set to fips-compatibility","accounts_in_message":[],"_revision_number":16},{"id":"0bbac307386de9b76f9b332ba6f0f06ca4c12c6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"date":"2022-01-18 14:37:38.000000000","message":"Patch Set 17: Patch Set 16 was rebased","accounts_in_message":[],"_revision_number":17},{"id":"7631bfc266bea6c93822011604a279a3ce96448e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2022-01-18 16:18:24.000000000","message":"Abandoned\n\nThis is being abandoned pending https://review.opendev.org/c/openstack/tempest/+/822560","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"51e6a5423f34fd6e7f3a325cae5a920afe763395","revisions":{"e7f688635c92bff09b4fb443a434c81eaa113948":{"kind":"REWORK","_number":1,"created":"2021-09-24 03:31:12.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/09/810809/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/1"}}},"commit":{"parents":[{"commit":"3949141ace9f0a72daa483f320fc19880fb641e1","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3949141ace9f0a72daa483f320fc19880fb641e1"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-24 03:30:46.000000000","tz":-240},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nUntil we can replace paramiko and to see real failures instead\nof test infrastructure failures, set paramiko to a patches version\nthat will allow the md5 calls to succeed on a centos/fedora system.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e7f688635c92bff09b4fb443a434c81eaa113948"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e7f688635c92bff09b4fb443a434c81eaa113948"}]},"branch":"refs/heads/master"},"182727d8bca874c856f7f1e7a3fd9ffebfeb4c02":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2021-10-01 15:24:53.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/09/810809/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/2"}}},"commit":{"parents":[{"commit":"2135d825fb7ba917f151ee3fbe451eb3af488267","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2135d825fb7ba917f151ee3fbe451eb3af488267"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-10-01 15:24:53.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nUntil we can replace paramiko and to see real failures instead\nof test infrastructure failures, set paramiko to a patches version\nthat will allow the md5 calls to succeed on a centos/fedora system.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/182727d8bca874c856f7f1e7a3fd9ffebfeb4c02"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/182727d8bca874c856f7f1e7a3fd9ffebfeb4c02"}]},"branch":"refs/heads/master","description":"Rebase"},"1241844677042e3a3766a746c512174e98ea5025":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-10-04 20:11:58.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/09/810809/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/3"}}},"commit":{"parents":[{"commit":"7b3662b06c1a0cca7df516b5526f633f6f013858","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7b3662b06c1a0cca7df516b5526f633f6f013858"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-10-04 20:11:58.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nUntil we can replace paramiko and to see real failures instead\nof test infrastructure failures, set paramiko to a patches version\nthat will allow the md5 calls to succeed on a centos/fedora system.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1241844677042e3a3766a746c512174e98ea5025"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1241844677042e3a3766a746c512174e98ea5025"}]},"branch":"refs/heads/master","description":"Rebase"},"65aae32226687825ec0dbcc2151a7f5cfd2bb731":{"kind":"NO_CHANGE","_number":4,"created":"2021-10-15 15:42:51.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/09/810809/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/4"}}},"commit":{"parents":[{"commit":"f8aca8680271afbe69d0ad705c03e040957eb12f","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f8aca8680271afbe69d0ad705c03e040957eb12f"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-10-15 15:42:51.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nUntil we can replace paramiko and to see real failures instead\nof test infrastructure failures, set paramiko to a patches version\nthat will allow the md5 calls to succeed on a centos/fedora system.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/65aae32226687825ec0dbcc2151a7f5cfd2bb731"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/65aae32226687825ec0dbcc2151a7f5cfd2bb731"}]},"branch":"refs/heads/master","description":"Rebase"},"7ae2c5d72e8da8bc63c0d0435550f11733c93d97":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2021-11-04 19:41:16.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/09/810809/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/5"}}},"commit":{"parents":[{"commit":"586b1e5c4bae6714988636a5f945e747524c81fc","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/586b1e5c4bae6714988636a5f945e747524c81fc"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-11-04 19:41:16.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nUntil we can replace paramiko and to see real failures instead\nof test infrastructure failures, set paramiko to a patches version\nthat will allow the md5 calls to succeed on a centos/fedora system.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7ae2c5d72e8da8bc63c0d0435550f11733c93d97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7ae2c5d72e8da8bc63c0d0435550f11733c93d97"}]},"branch":"refs/heads/master","description":"Rebase"},"ecfd93a14867fa600be48741450c6adb5cfa244a":{"kind":"REWORK","_number":6,"created":"2021-11-05 16:46:58.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/09/810809/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/6"}}},"commit":{"parents":[{"commit":"586b1e5c4bae6714988636a5f945e747524c81fc","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/586b1e5c4bae6714988636a5f945e747524c81fc"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-11-05 16:46:58.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nWe\u0027re going to have to replace paramiko eventually because it won\u0027t\nbe FIPS compliant, but until then we might be able to achieve\ncompatibility by patching paramiko.\n\nIn particular, the patched paramiko will allow md5 calls in a case\nwhere they are allowed (fingerprints used in log messages with no\nsecret data).\n\nParamiko will still fail where we read in encrypted private keys\nthat are not generated in the new (default) OpenSSH format, because it\ntries to do a key derivation there using md5().  That is definitely\ndisallowed under fips.  We should find those errors and replace the\nkeys.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ecfd93a14867fa600be48741450c6adb5cfa244a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ecfd93a14867fa600be48741450c6adb5cfa244a"}]},"branch":"refs/heads/master"},"12b0b841be9b64be6a0ab114901bef2647885548":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-11-05 20:37:09.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/09/810809/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/7"}}},"commit":{"parents":[{"commit":"36b0b2db8f4f464c1d6910a53b06fbd269c8bd31","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/36b0b2db8f4f464c1d6910a53b06fbd269c8bd31"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-11-05 20:37:09.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nWe\u0027re going to have to replace paramiko eventually because it won\u0027t\nbe FIPS compliant, but until then we might be able to achieve\ncompatibility by patching paramiko.\n\nIn particular, the patched paramiko will allow md5 calls in a case\nwhere they are allowed (fingerprints used in log messages with no\nsecret data).\n\nParamiko will still fail where we read in encrypted private keys\nthat are not generated in the new (default) OpenSSH format, because it\ntries to do a key derivation there using md5().  That is definitely\ndisallowed under fips.  We should find those errors and replace the\nkeys.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/12b0b841be9b64be6a0ab114901bef2647885548"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/12b0b841be9b64be6a0ab114901bef2647885548"}]},"branch":"refs/heads/master","description":"Rebase"},"5c913afe822d0ebf1b6b4b1822b272cd8d63b9ee":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2021-11-12 19:13:12.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/09/810809/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/8"}}},"commit":{"parents":[{"commit":"01689ed7313c0965ce30c09405be81fc83543651","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/01689ed7313c0965ce30c09405be81fc83543651"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-11-12 19:13:12.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nWe\u0027re going to have to replace paramiko eventually because it won\u0027t\nbe FIPS compliant, but until then we might be able to achieve\ncompatibility by patching paramiko.\n\nIn particular, the patched paramiko will allow md5 calls in a case\nwhere they are allowed (fingerprints used in log messages with no\nsecret data).\n\nParamiko will still fail where we read in encrypted private keys\nthat are not generated in the new (default) OpenSSH format, because it\ntries to do a key derivation there using md5().  That is definitely\ndisallowed under fips.  We should find those errors and replace the\nkeys.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5c913afe822d0ebf1b6b4b1822b272cd8d63b9ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5c913afe822d0ebf1b6b4b1822b272cd8d63b9ee"}]},"branch":"refs/heads/master","description":"Rebase"},"009512d485af5924fe734057d3802ac9352e3c0e":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-11-17 14:52:45.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/09/810809/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/9"}}},"commit":{"parents":[{"commit":"e2ce6edd80a77f7d8bb6ccc1c6b72e94aee7bc39","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e2ce6edd80a77f7d8bb6ccc1c6b72e94aee7bc39"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-11-17 14:52:45.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nWe\u0027re going to have to replace paramiko eventually because it won\u0027t\nbe FIPS compliant, but until then we might be able to achieve\ncompatibility by patching paramiko.\n\nIn particular, the patched paramiko will allow md5 calls in a case\nwhere they are allowed (fingerprints used in log messages with no\nsecret data).\n\nParamiko will still fail where we read in encrypted private keys\nthat are not generated in the new (default) OpenSSH format, because it\ntries to do a key derivation there using md5().  That is definitely\ndisallowed under fips.  We should find those errors and replace the\nkeys.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/009512d485af5924fe734057d3802ac9352e3c0e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/009512d485af5924fe734057d3802ac9352e3c0e"}]},"branch":"refs/heads/master","description":"Rebase"},"1647c136476e47720c3b18af2de4afcae1500749":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-11-29 18:45:12.000000000","uploader":{"_account_id":30002,"name":"Douglas Viroel","email":"viroel@gmail.com","username":"dviroel"},"ref":"refs/changes/09/810809/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/10"}}},"commit":{"parents":[{"commit":"1d8bcfcc6afeb4a2737b152a2032a20cbdaeaad9","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1d8bcfcc6afeb4a2737b152a2032a20cbdaeaad9"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Douglas Viroel","email":"viroel@gmail.com","date":"2021-11-29 18:45:12.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nWe\u0027re going to have to replace paramiko eventually because it won\u0027t\nbe FIPS compliant, but until then we might be able to achieve\ncompatibility by patching paramiko.\n\nIn particular, the patched paramiko will allow md5 calls in a case\nwhere they are allowed (fingerprints used in log messages with no\nsecret data).\n\nParamiko will still fail where we read in encrypted private keys\nthat are not generated in the new (default) OpenSSH format, because it\ntries to do a key derivation there using md5().  That is definitely\ndisallowed under fips.  We should find those errors and replace the\nkeys.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1647c136476e47720c3b18af2de4afcae1500749"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1647c136476e47720c3b18af2de4afcae1500749"}]},"branch":"refs/heads/master","description":"Rebase"},"3f1cba45628ef94294dfc780daa2cc495bc4c254":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-11-29 22:11:35.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/09/810809/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/11"}}},"commit":{"parents":[{"commit":"b0c3bd7bd0462a25ebca8f6c3ff9f7eb3faff5b2","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b0c3bd7bd0462a25ebca8f6c3ff9f7eb3faff5b2"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-11-29 22:11:02.000000000","tz":-300},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nWe\u0027re going to have to replace paramiko eventually because it won\u0027t\nbe FIPS compliant, but until then we might be able to achieve\ncompatibility by patching paramiko.\n\nIn particular, the patched paramiko will allow md5 calls in a case\nwhere they are allowed (fingerprints used in log messages with no\nsecret data).\n\nParamiko will still fail where we read in encrypted private keys\nthat are not generated in the new (default) OpenSSH format, because it\ntries to do a key derivation there using md5().  That is definitely\ndisallowed under fips.  We should find those errors and replace the\nkeys.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3f1cba45628ef94294dfc780daa2cc495bc4c254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3f1cba45628ef94294dfc780daa2cc495bc4c254"}]},"branch":"refs/heads/master"},"3827b2c46b00e1c5ae8e3e4f5eeec0c6f361aab6":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-11-30 04:08:54.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/09/810809/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/12"}}},"commit":{"parents":[{"commit":"fb1196583b9a892efa410ef5fab224b6a6583ad3","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fb1196583b9a892efa410ef5fab224b6a6583ad3"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-11-30 04:08:54.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nWe\u0027re going to have to replace paramiko eventually because it won\u0027t\nbe FIPS compliant, but until then we might be able to achieve\ncompatibility by patching paramiko.\n\nIn particular, the patched paramiko will allow md5 calls in a case\nwhere they are allowed (fingerprints used in log messages with no\nsecret data).\n\nParamiko will still fail where we read in encrypted private keys\nthat are not generated in the new (default) OpenSSH format, because it\ntries to do a key derivation there using md5().  That is definitely\ndisallowed under fips.  We should find those errors and replace the\nkeys.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3827b2c46b00e1c5ae8e3e4f5eeec0c6f361aab6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3827b2c46b00e1c5ae8e3e4f5eeec0c6f361aab6"}]},"branch":"refs/heads/master","description":"Rebase"},"de91864695e68edb4ab1ebf28ca1f1db941e8ff0":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-12-02 22:41:39.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/09/810809/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/13"}}},"commit":{"parents":[{"commit":"6c8b2d26986c809342d5df6eb8465cd4ce118d03","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6c8b2d26986c809342d5df6eb8465cd4ce118d03"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-12-02 22:35:48.000000000","tz":-300},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nWe\u0027re going to have to replace paramiko eventually because it won\u0027t\nbe FIPS compliant, but until then we might be able to achieve\ncompatibility by patching paramiko.\n\nIn particular, the patched paramiko will allow md5 calls in a case\nwhere they are allowed (fingerprints used in log messages with no\nsecret data).\n\nParamiko will still fail where we read in encrypted private keys\nthat are not generated in the new (default) OpenSSH format, because it\ntries to do a key derivation there using md5().  That is definitely\ndisallowed under fips.  We should find those errors and replace the\nkeys.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/de91864695e68edb4ab1ebf28ca1f1db941e8ff0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/de91864695e68edb4ab1ebf28ca1f1db941e8ff0"}]},"branch":"refs/heads/master"},"d080ce938a62bcf550c57c9f54290366d38e6e29":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-12-08 22:24:17.000000000","uploader":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"ref":"refs/changes/09/810809/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/14"}}},"commit":{"parents":[{"commit":"1ddd0578c3d2aa4a500965e06fdc28ebc8eb8960","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1ddd0578c3d2aa4a500965e06fdc28ebc8eb8960"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-12-08 22:24:17.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nWe\u0027re going to have to replace paramiko eventually because it won\u0027t\nbe FIPS compliant, but until then we might be able to achieve\ncompatibility by patching paramiko.\n\nIn particular, the patched paramiko will allow md5 calls in a case\nwhere they are allowed (fingerprints used in log messages with no\nsecret data).\n\nParamiko will still fail where we read in encrypted private keys\nthat are not generated in the new (default) OpenSSH format, because it\ntries to do a key derivation there using md5().  That is definitely\ndisallowed under fips.  We should find those errors and replace the\nkeys.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d080ce938a62bcf550c57c9f54290366d38e6e29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d080ce938a62bcf550c57c9f54290366d38e6e29"}]},"branch":"refs/heads/master","description":"Rebase"},"64c867443f456f742049df8a1eb3c4e1434397ae":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2022-01-03 07:28:51.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/09/810809/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/15"}}},"commit":{"parents":[{"commit":"69425461c1f28d50d0b6ae9095534f37f0bc7a0f","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/69425461c1f28d50d0b6ae9095534f37f0bc7a0f"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-01-03 07:28:51.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nWe\u0027re going to have to replace paramiko eventually because it won\u0027t\nbe FIPS compliant, but until then we might be able to achieve\ncompatibility by patching paramiko.\n\nIn particular, the patched paramiko will allow md5 calls in a case\nwhere they are allowed (fingerprints used in log messages with no\nsecret data).\n\nParamiko will still fail where we read in encrypted private keys\nthat are not generated in the new (default) OpenSSH format, because it\ntries to do a key derivation there using md5().  That is definitely\ndisallowed under fips.  We should find those errors and replace the\nkeys.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/64c867443f456f742049df8a1eb3c4e1434397ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/64c867443f456f742049df8a1eb3c4e1434397ae"}]},"branch":"refs/heads/master","description":"Rebase"},"b93c732ac541993c1ad7bb57a07d6a27f1f73a7b":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2022-01-04 11:39:03.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/09/810809/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/16"}}},"commit":{"parents":[{"commit":"99339e3dfd305466f2084aca27e930f0a6303163","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/99339e3dfd305466f2084aca27e930f0a6303163"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-01-04 11:39:03.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nWe\u0027re going to have to replace paramiko eventually because it won\u0027t\nbe FIPS compliant, but until then we might be able to achieve\ncompatibility by patching paramiko.\n\nIn particular, the patched paramiko will allow md5 calls in a case\nwhere they are allowed (fingerprints used in log messages with no\nsecret data).\n\nParamiko will still fail where we read in encrypted private keys\nthat are not generated in the new (default) OpenSSH format, because it\ntries to do a key derivation there using md5().  That is definitely\ndisallowed under fips.  We should find those errors and replace the\nkeys.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b93c732ac541993c1ad7bb57a07d6a27f1f73a7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b93c732ac541993c1ad7bb57a07d6a27f1f73a7b"}]},"branch":"refs/heads/master","description":"Rebase"},"51e6a5423f34fd6e7f3a325cae5a920afe763395":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2022-01-18 14:37:38.000000000","uploader":{"_account_id":32926,"name":"Jiri Podivin","display_name":"jpodivin","email":"jpodivin@redhat.com","username":"jpodivin"},"ref":"refs/changes/09/810809/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/810809/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/810809/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/810809/17"}}},"commit":{"parents":[{"commit":"6675d7b1edd1e2b74801739101bbc9f05fe35265","subject":"Add skip for tests that wont work under fips","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6675d7b1edd1e2b74801739101bbc9f05fe35265"}]}],"author":{"name":"Ade Lee","email":"alee@redhat.com","date":"2021-09-16 21:03:22.000000000","tz":-240},"committer":{"name":"Jiri Podivin","email":"jpodivin@redhat.com","date":"2022-01-18 14:37:38.000000000","tz":0},"subject":"DNM - Add temporary link to patched paramiko","message":"DNM - Add temporary link to patched paramiko\n\nparamiko has some calls to md5() which will break under fips.\nWe\u0027re going to have to replace paramiko eventually because it won\u0027t\nbe FIPS compliant, but until then we might be able to achieve\ncompatibility by patching paramiko.\n\nIn particular, the patched paramiko will allow md5 calls in a case\nwhere they are allowed (fingerprints used in log messages with no\nsecret data).\n\nParamiko will still fail where we read in encrypted private keys\nthat are not generated in the new (default) OpenSSH format, because it\ntries to do a key derivation there using md5().  That is definitely\ndisallowed under fips.  We should find those errors and replace the\nkeys.\n\nChange-Id: I891fd8131623a927fef637d4acac574a8fb18c10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/51e6a5423f34fd6e7f3a325cae5a920afe763395"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/51e6a5423f34fd6e7f3a325cae5a920afe763395"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
