)]}'
{"id":"openstack%2Ftempest~814085","triplet_id":"openstack%2Ftempest~master~Ic99ba61d87dfc24a684263d41e0caa4f54d1d263","project":"openstack/tempest","branch":"master","topic":"bug/1945495","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2022-01-28 20:23:38.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-01-28 20:23:38.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-01-28 16:18:13.000000000","reason":"Ghanshyam replied on the change"}},"hashtags":[],"change_id":"Ic99ba61d87dfc24a684263d41e0caa4f54d1d263","subject":"Fix test_rebuild_server test by waiting for floating ip","status":"MERGED","created":"2021-10-15 00:07:05.000000000","updated":"2022-02-08 17:18:02.000000000","submitted":"2022-01-28 20:23:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"814085-bug/1945495","meta_rev_id":"dfa671d9641bafb50f7fe49d437c632d7dd70925","_number":814085,"virtual_id_number":814085,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-28 20:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-01-28 16:18:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-01-28 16:18:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}],"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-15 02:42:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-19 08:54:06.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"state":"CC"},{"updated":"2021-10-19 12:22:35.000000000","updated_by":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"reviewer":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"state":"REVIEWER"},{"updated":"2022-01-28 13:00:42.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-02-04 14:47:18.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"941b1ecd46b03a076985d7036939a6f0c97eacf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-15 00:07:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59025e51a3773cdc7fdacf62ef145f5d9130f488","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 02:42:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/7bea52fcb1a84350a32fe21d1459ceb7 : SUCCESS in 1h 03m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/435d9563198343ea886e491cfd7295b0 : SUCCESS in 1h 13m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d43b32e423494dde96fe21306898cc68 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a281c7283a1845a3ab92b8e261d2fc2b : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/010f22ee653d4046956f21c7509ad0ba : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ddc1c00e2dce4a1d8f7bcfdd96995ddc : SUCCESS in 5m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35beb50d848d47af949018a16d78f96e : SUCCESS in 6m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/558d7a243b73428d977314f2e59c7edc : SUCCESS in 11m 00s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/cb3f52785c0b4088b2b951379b6b2c70 : SUCCESS in 1h 32m 34s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/34dfb1e93e564f12b2c54f99512bec25 : SUCCESS in 1h 17m 22s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/77377a4815ff4776b252559b602b3cbd : SUCCESS in 1h 30m 49s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/a4c339cf4b524ff7b0c40e4803d9c9e0 : SUCCESS in 1h 18m 08s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/12e951cde360491f9b6f4bd5c2d72e0e : SUCCESS in 1h 24m 42s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/790d7f92803040b59f7860428f3b9387 : SUCCESS in 1h 38m 17s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/98d556e0efce47c499d3a28da590ba08 : SUCCESS in 1h 29m 50s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/2fe42eb8a2c8472eaa5576619e39ea4d : SUCCESS in 1h 28m 17s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a91615f708d44f75ad56843d6bbd9ec9 : SUCCESS in 1h 29m 35s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c14f597ce4de4b3aa1d49efc8077c328 : SUCCESS in 35m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a647f56cf4b4c0aafd50acb1f12b918 : SUCCESS in 51m 43s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dcdaa069ce4a4814b09362f992970a9d : SUCCESS in 2h 29m 43s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5426a5b288cd42348892f446c5fa7d2a : SUCCESS in 48m 52s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/7452af091bd34e0d878c4670c4ed29cd : FAILURE in 1h 31m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a1b913e507d431f913c59c60c956988 : SUCCESS in 59m 39s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/43703b57fd2e4095b4f9e6d417b9e430 : SUCCESS in 59m 06s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b431c05afce747fe8c9f5ae84e8901ed : SUCCESS in 4m 43s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/62502349184e4955bb2cfc4825fdf81e : FAILURE in 1h 10m 12s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/9ba51a7192f54f0faca97e0adf947ed9 : FAILURE in 1h 07m 05s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/25a9712b91fb43bb8b7604d2f87e84a6 : SUCCESS in 3m 39s","accounts_in_message":[],"_revision_number":1},{"id":"c41895a9f9c7333f5fbf48042735a23411f358b9","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-15 06:51:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"280c998460b85b81d0460e03d3ffaaa65f47a610","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-15 10:05:14.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/97d9bc641fc9489a9c974a49a6630004 : SUCCESS in 56m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/82240e8c34aa42f499d6a813fd2d0b36 : SUCCESS in 1h 18m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6ee6f71662b42839fdb388a25ada0c3 : SUCCESS in 7m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c549c49ebdb41a2ad515565334d5c97 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34726a5e8e444bac965fe12383772e82 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b7a312b3a8894428863329be41cc7d48 : FAILURE in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2acce1fdc414417ba7e27ca3bc15a372 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0301d1d266ff4b318cf1a322b60c0de4 : SUCCESS in 12m 59s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/3c9257b2ac8246bf9f9948272290316d : SUCCESS in 1h 47m 49s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/c1999017e63d47fb9bad59d8fcacc34c : SUCCESS in 1h 23m 51s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2960002fc05c49a99d9107125257ea32 : SUCCESS in 1h 20m 01s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/a602d7f8de554c2f97ffdff4e28ccfaa : SUCCESS in 1h 16m 24s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/ddfcce5ec1df4556a489b779633be212 : SUCCESS in 1h 23m 05s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/4059b08801a34cdf8a87961de516a95b : SUCCESS in 1h 25m 59s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/4838129af58f433a8ed54bc1369c3590 : SUCCESS in 1h 28m 00s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/174997f4abf140fb96895d4a48aac0f3 : SUCCESS in 1h 25m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/af806963653246408801aa9047538548 : SUCCESS in 1h 30m 17s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/60f76873a19442ce8996744a7735764b : SUCCESS in 28m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9c3b9dfa8754f62a3e69ed1ea8535ef : SUCCESS in 53m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3cf69ee2e5f442028c530556b1ca150b : SUCCESS in 2h 19m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bede5d2ff4414384bf38f52352fdcd68 : SUCCESS in 48m 43s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c90932ddca3347a1b79f952e427d19bc : FAILURE in 1h 17m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8a3003bbf6e34808acfa3e885f6dff6c : SUCCESS in 55m 55s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/0690729a67814611864608d51b137e8a : SUCCESS in 52m 30s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/0eb5cd727505418886a8abd1a82a194a : SUCCESS in 6m 10s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/b0d12dbaba6b44e2b7ee69b9e7eb8817 : FAILURE in 1h 08m 26s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/0d9c494e0a074401add1eb5dfa300d52 : FAILURE in 49m 40s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1a86861521124ac981309bacfdab73a2 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"79b496416309b065e3d8e88f428918f447070788","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-19 08:54:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b824aae6501fcf3994f7ddb4f2d5f1c808a8849a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 10:57:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/149620d2b8f8446ca89f44371578e76c : SUCCESS in 1h 00m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/278ed9248a054575a5005fe9508ef1cb : SUCCESS in 1h 34m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cdbf99be1a53419096bf8d22f997e09d : SUCCESS in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd2cbb6cf28342dc95860a18234939fc : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4dc33ae3f7a64ad791782e69dd0dc23a : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52bb381768b74235b387aa8c944f1164 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db4f6355456543ce992b2b83aa089ce1 : SUCCESS in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b36fa3851b8d493183d0c858a7e77923 : SUCCESS in 12m 44s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b31dd1d2921c48cea300de8e0af3bd88 : SUCCESS in 1h 33m 24s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/e706cc4ce1ce43b7affff0999b3e38e2 : SUCCESS in 1h 16m 09s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/44922dac8a76434497152095232babe8 : SUCCESS in 1h 16m 45s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/684a32fd76e342f0a4b2903b2c5f54c6 : SUCCESS in 1h 13m 44s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/bcca4e25a1df414ba09205a42ab62996 : SUCCESS in 1h 37m 35s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/c8eb5422d5ff4c1a9eae6552cdd6246d : SUCCESS in 1h 25m 20s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/d53269c802c5494f9e3e3291da326476 : SUCCESS in 1h 30m 22s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/56c71406e6214364ab65c04bcf03fc59 : SUCCESS in 1h 35m 04s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2432fab832ad44d3a4a887b8f7a43685 : SUCCESS in 1h 23m 12s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/69b4d06a048240fcbbc0034c466dec4d : SUCCESS in 29m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44f7ba92a6f9423298ca116168603e8d : SUCCESS in 49m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/177d11f29ce5435aba1054e58e32b837 : SUCCESS in 2h 01m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/73edcb490ac34abc9efd6983d7ecba12 : SUCCESS in 48m 01s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/53bcec3a9e3f4d2ca6727a534a3bfc1d : FAILURE in 1h 00m 24s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bdcc8c45da6d4f0fb7dae0785c5e2819 : SUCCESS in 46m 39s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/a7e87c83889f4c2dba92b91761e608d3 : SUCCESS in 48m 50s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5b70f16e3e0d4da982bac481086cb15c : FAILURE in 1h 06m 29s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/80c46570a25544f4a74e9f47b0eea131 : FAILURE in 1h 08m 19s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/76e865dd50984045aba15add9d48ccb8 : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"b8886a683fd69d7e84aea51d73df8f090106a058","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-19 12:22:35.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d76dff5b59e9b41e56e7d014dcab8fb5e3b37476","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-19 12:22:50.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a181b1ed060baf1de4a0592854c99b7bc71d5a12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-19 14:50:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/1c3c1bf0e8a445f3a3efba14caa92a71 : SUCCESS in 45m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e215d4d795a4a2986aba77fd98b7055 : SUCCESS in 1h 19m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00fb31cfe86743f0959f54adf0c2dff3 : FAILURE in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd171daadb12497a9c9758cf096e64ca : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4c36d724170341639bc50140b233fef9 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c143ed8c787f4e25a648628d0d8576da : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b140a636cfcb4035b8d7e2e005e06778 : FAILURE in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a4e38f8b37c490a884565e489b569cf : SUCCESS in 17m 12s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/db89b64726d349fa8df039a75e473561 : SUCCESS in 1h 30m 24s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/2c9cee3220874a3eb78bd864e2f3c48b : SUCCESS in 1h 04m 35s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/441c691a6ea140b4ac5923d564b68dc9 : SUCCESS in 1h 18m 10s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/04a70ab2787244678413df92361c2ceb : SUCCESS in 1h 18m 14s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/e0c47d00802540ea917a8dddf5bbee0e : SUCCESS in 1h 18m 14s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/c4dff28072064c8b9ab4c6ba3b577d06 : SUCCESS in 1h 15m 05s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/0bb9c374ad4240e3b4c08e3b3dbc8977 : SUCCESS in 1h 18m 29s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/90a8493f61274de99ba52e1b108cd8b0 : SUCCESS in 1h 16m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7b10f0c2747e42fea4d95ffda1ffd985 : SUCCESS in 1h 23m 02s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/86323fdc80e94d888fb2d059c287f8d7 : SUCCESS in 35m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8077ecdf9b474fddba13401be02e3c6e : SUCCESS in 38m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/063e3ed552254faba99e50ca044b9803 : SUCCESS in 2h 20m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9773e34cf367414f8adf8ceb61950673 : SUCCESS in 48m 20s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/cf3684d8fcfb48ebaab2d9633a3f0652 : FAILURE in 1h 23m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/01880ce4bcd949f18964fd2b28e6d261 : SUCCESS in 56m 46s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/b1384d0a88e042589cfbe87e3df0989b : SUCCESS in 40m 11s (non-voting)\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/54440f77eee044129ded9705edc56616 : FAILURE in 1h 30m 15s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/74237b7bbb9d4c7789f51b85547ea5ff : FAILURE in 1h 18m 00s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7b160a5ac8bf4ab898c27a319cd7fe5d : SUCCESS in 3m 47s","accounts_in_message":[],"_revision_number":1},{"id":"617fe1912c8b4b0a866e58c2de095b9ed9964dfc","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-10-21 21:19:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5b7d7bfafb63f5b268448901e9b70753aabe3860","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2021-10-22 09:36:22.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ed0be5d9b97b5a693706d4f32db39b45d48c96d5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-01-26 20:27:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"35a6b6b50ad6e5fd4f660b694e98290c76968bb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-26 23:04:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/66ceefd4ea9e4db2b0b682960ddece3e : SUCCESS in 1h 02m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/97eace75ae3d41069088e1f551e52f0c : SUCCESS in 1h 32m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7db135e377844539390039e40140135 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa4b1c9824a34e9482a919da2b18f05e : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e48ecf63a83541bdb132b509f9ec4a61 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ac96d32ed2b742a984c182b6a61b8ef1 : SUCCESS in 5m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/458ed59f1e1e491d832d90c23a066439 : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7daf44289a84e339058b4277be3ac53 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7e6a54a64fd4adfb83946fb664aa48c : SUCCESS in 6m 03s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b8f1f1304af348bab786758ab84960d3 : SUCCESS in 1h 27m 47s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3cacac2c4c614f9f9760675a725c4237 : SUCCESS in 1h 20m 41s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/061f714ca94a4c3c9f4860a689272df1 : SUCCESS in 1h 42m 49s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/27b14200402e4122b801b3e8fb3eeb22 : SUCCESS in 1h 18m 00s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/f408e49127a440aa85d693522db9385f : SUCCESS in 1h 26m 30s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/1ca15251beb74e0394a6091935feee9b : SUCCESS in 1h 45m 33s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/6de7d5433a9344e08d8bfb33b1560b05 : SUCCESS in 1h 27m 59s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/86897fe366da47319191d045e4fdf47a : SUCCESS in 1h 37m 55s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/9c62f8ff3b774c5587c53cbd5f69a3dd : SUCCESS in 30m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da77fe4637f34e0386392a3d8cea2294 : SUCCESS in 53m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b1e30f8d0db474e9aeaccfe1953b570 : SUCCESS in 2h 09m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/743c1f909bb3491c8467680dbd3f9938 : SUCCESS in 54m 04s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/7e4743948b2345c7a957c4047f57b779 : FAILURE in 1h 23m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1184413bec7349388e0812137c7bbffe : SUCCESS in 59m 27s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/aa4ab82b948744cabda34d4ae43b86da : SUCCESS in 1h 01m 32s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/5cb3988f93ac47d99d43e66092b51779 : SUCCESS in 4m 48s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8bae89f8dcfd4290bb9b5cceb39a418c : FAILURE in 1h 14m 03s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/b64446e633fc4498bf23ad3e502dd5b0 : FAILURE in 1h 10m 01s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9b7685fc63ee43e193bce62fe5b53cd7 : SUCCESS in 3m 52s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/262a0b08337d49cdad82ef118f7aa476 : SUCCESS in 1h 36m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b7d6ff3ecffb84cb4e5a70b7817967921efe7fb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-01-27 15:40:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"3bc2b6000849cfbeb3fc6c7c65f38b3d81bb9256","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-01-27 15:40:54.000000000","message":"Patch Set 3:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":3},{"id":"c03a0e330df268525b65084f256aa8288f907578","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 18:07:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a4c24c26f82645788923fa28356ea4ce : SUCCESS in 57m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/54ae332433d24e1a9de509f83f204064 : SUCCESS in 1h 29m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9b8ee22396d48328f6a16d43c0d26c0 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3acae2cdf4a64d0baed385c5fdb37d3c : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67a1a24b63964ba39532da13066ab025 : FAILURE in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c1c24a9826f436da17e88e0cc75c2da : SUCCESS in 6m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/324177db9f8c418db218d190d99b7d46 : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/298380fdda394e13a0593ccba539b37f : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e2d2b864768421ea392b8e66568d3b7 : SUCCESS in 5m 28s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/a1428cc53ecc451cb7d63550014ffd7c : SUCCESS in 2h 02m 05s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/077cbfde25384f53bee972de9088b368 : SUCCESS in 1h 20m 13s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5123bbc691214940b147122bd80d812f : SUCCESS in 1h 49m 39s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/ffed1eb455ef4b63b1248bd20d09b55f : SUCCESS in 1h 19m 06s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/43e971c3740045d8b9ef435b515e64a7 : SUCCESS in 1h 37m 08s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/20cf15c1d7744b37ab8a726cc80eb9bf : SUCCESS in 1h 42m 33s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/4f5f4edc229f47c4ada3874fd10d0bed : SUCCESS in 1h 29m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e06edc95bb044be99e0e992f40c1df12 : SUCCESS in 1h 39m 58s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/5b347f9613144766a6d76659f517d788 : SUCCESS in 26m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f74598d220e43e19f7df0a736430f95 : SUCCESS in 50m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/187bb409f797469597e75aefecedfef3 : SUCCESS in 2h 22m 48s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b16034d1ba6a47be8cb322668359f987 : SUCCESS in 49m 05s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9dc1a7e9f7d143569b7750d2708cf339 : SUCCESS in 1h 10m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9eae9324b92441bc96436f021f5be30e : SUCCESS in 1h 02m 58s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/e1ddb273005e437e9ff34dbce4f010bc : SUCCESS in 1h 01m 39s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/033bab680c72492c8f70bbbde94681b1 : SUCCESS in 5m 01s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/caeaa54c1d5e4ac1921f5c0f6abd8530 : FAILURE in 1h 14m 23s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/5d73a59e813f4063ab2e0838adca3ff9 : FAILURE in 1h 01m 40s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/49ef87b7874e4021aa9f5b8a9bdc570d : SUCCESS in 4m 05s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/d0fc6191796548739a00e7c966b31c12 : SUCCESS in 1h 36m 11s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d9f3d9d97133cf49ebb586f8b905da8918ef2394","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-01-28 08:02:44.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0b8904e6b6dbcd41f2345377b0f6985a4b59496f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 10:45:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/40b28ef35f384340be045f79b6f6d5c9 : SUCCESS in 1h 05m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/935c7fb73d51462a96e8620a1522e294 : SUCCESS in 1h 30m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c77b86cfd3048d096e4d8ddbd79b726 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47efc98b63e8440da4369423a8e4533a : SUCCESS in 11m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5dec91859f6401bb877ced37c0279a4 : SUCCESS in 6m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d04afff389e14f339a43dd0d4e441358 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6a1dcea3af94421a81171d72b4a58730 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7503f2f81cde4e07bfc6f1ea0719ce73 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6ebeb9f811324f5db07986a0905a6c38 : SUCCESS in 4m 52s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/55dfa461c7ac4acda8f6668dfa293f12 : SUCCESS in 2h 14m 11s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d81895f7565148e09d893e2490c36271 : SUCCESS in 1h 23m 39s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cd04873753e8484eb642d22f1fa01597 : SUCCESS in 1h 31m 40s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/61146c0994284a8daacaea0642bca218 : SUCCESS in 1h 18m 07s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/4e94bc9128514bc894f270ea29aefc8a : SUCCESS in 1h 35m 31s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/d0bd93ede17849f3aea58a68493bc8a8 : SUCCESS in 1h 33m 57s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/6c741b9c666b4d2dbc06d0cc3ee5a488 : SUCCESS in 1h 29m 04s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7b0ed3bc053940aea68cffb41774ec73 : SUCCESS in 1h 44m 30s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a338cc6a88124bb3abb672eee85bdac8 : SUCCESS in 31m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce70cd87bd5d47b380310f6d06e47144 : SUCCESS in 57m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0e1604df606046ba8c3db1db1bae0270 : SUCCESS in 2h 34m 00s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d819298d175e4b858aeddeef67ccc013 : SUCCESS in 53m 12s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/903fbe05eac34d53b8751d38bf83a1c7 : SUCCESS in 1h 10m 25s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3b01007262b74c63a592fd646ccfc8cf : SUCCESS in 49m 22s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/30087211f10843b08a6594d1c4628972 : SUCCESS in 1h 04m 21s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/ab6de7d3c4ea41b5be882569e24cbe3b : SUCCESS in 4m 17s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7fe8d8b0d25f4a6b9675ccd0d2e368d4 : FAILURE in 1h 08m 15s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/49fda59fdbe34c099dedf256a945aa26 : FAILURE in 1h 02m 42s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2d95b141af754ce8ab5d0546e165274c : SUCCESS in 3m 38s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/3fafdd36fa6a42a8936361a4313f7176 : SUCCESS in 1h 24m 17s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"641e0b8726c28db4745a7cd042252e43671b25b3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-01-28 16:18:13.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"842cfe117d2fd85bfd6455e8d40d1da402c02e7e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 16:20:07.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"87d85e18cafcc55da0f12238ad9abf84c327f415","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 20:23:37.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/c579180b37394d59a3cd6778296e0ec1 : SUCCESS in 51m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/18a5c6bcfce24892be074ef4e989b3e4 : SUCCESS in 1h 14m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfbcd20acff34a9d8a2f88c9fba09edb : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00df0304653443599ce4df26db61de84 : SUCCESS in 5m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42219aaf957e4c72b19630eedf5345f0 : SUCCESS in 6m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/88cb7bfd6d6846a18303691d03162dbe : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73bbcb3a72574186b7c0203ffa82f83f : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9895b2a826714ddf9e5ee107689ed206 : SUCCESS in 5m 45s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/681baa9c620942c1b398930f94e22ba9 : SUCCESS in 2h 25m 57s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/db7dbed3435c4200bfa9f8ee5036d9a3 : SUCCESS in 51m 50s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/b10d7f1d716e4c9697ae970b64e349b4 : FAILURE in 1h 21m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/52d0dc7163624b048862c1a6d410a110 : SUCCESS in 43m 44s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0a9c5a1c08d74d219119cac1b32d9564 : SUCCESS in 1h 12m 32s","accounts_in_message":[],"_revision_number":4},{"id":"e28ecf19f23b022fb9a715fe06cc3d75d3cede53","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 20:23:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"bc5cbcf9b01e08d887877a4e9c96adc14443224a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 20:27:08.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f164b7c34f624b31861ab6e25a356f7e : SUCCESS in 1m 35s","accounts_in_message":[],"_revision_number":4},{"id":"267bd31fe328a52c816ecb06926a46bcef1ff09e","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2022-02-04 14:47:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"dfa671d9641bafb50f7fe49d437c632d7dd70925","tag":"autogenerated:gerrit:revert","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-02-08 17:18:02.000000000","message":"Created a revert of this change as Iddfa4ef874328fdee5418de5a8b1112007541883","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d4c9cc89f199899b20cfaea975ecd15ee66d3523","revisions":{"25c92bb03eeaebcf84dee5c5df484ac15aaa55b5":{"kind":"REWORK","_number":1,"created":"2021-10-15 00:07:05.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/85/814085/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/85/814085/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/85/814085/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/85/814085/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/85/814085/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/85/814085/1"}}},"commit":{"parents":[{"commit":"a47c8afc0eb3b14a7482b884fc75dd701712bc1b","subject":"Get rid of jsonschema.compat usage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a47c8afc0eb3b14a7482b884fc75dd701712bc1b"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-15 00:01:36.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-15 00:05:35.000000000","tz":0},"subject":"Fix test_rebuild_server by waiting for floating ip","message":"Fix test_rebuild_server by waiting for floating ip\n\nThe test could fail on an adresses mismatch when an ip wasn\u0027t\nassigned to either original or rebuilt server within a short\nperiod of time.\nThe commit implements a waiter for floating ip attachment which\nis used every time for original and rebuilt system if\nconnect_method \u003d\u003d floating.\n\nCloses-Bug: #1945495\nChange-Id: Ic99ba61d87dfc24a684263d41e0caa4f54d1d263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/25c92bb03eeaebcf84dee5c5df484ac15aaa55b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/25c92bb03eeaebcf84dee5c5df484ac15aaa55b5"}]},"branch":"refs/heads/master"},"e3c2cd6ee97d9230729d7a61b2d8e8b7de0a04c8":{"kind":"REWORK","_number":2,"created":"2022-01-26 20:27:59.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/85/814085/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/85/814085/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/85/814085/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/85/814085/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/85/814085/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/85/814085/2"}}},"commit":{"parents":[{"commit":"86db21ea6afb6c26a21fc143a7d061e947c31a93","subject":"Patch paramiko","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/86db21ea6afb6c26a21fc143a7d061e947c31a93"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-15 00:01:36.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-01-26 20:26:51.000000000","tz":0},"subject":"Fix test_rebuild_server by waiting for floating ip","message":"Fix test_rebuild_server by waiting for floating ip\n\nThe test could fail on an adresses mismatch when an ip wasn\u0027t\nassigned to either original or rebuilt server within a short\nperiod of time.\nThe commit implements a waiter for floating ip attachment which\nis used every time for original and rebuilt system if\nconnect_method \u003d\u003d floating.\n\nCloses-Bug: #1945495\nChange-Id: Ic99ba61d87dfc24a684263d41e0caa4f54d1d263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e3c2cd6ee97d9230729d7a61b2d8e8b7de0a04c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e3c2cd6ee97d9230729d7a61b2d8e8b7de0a04c8"}]},"branch":"refs/heads/master"},"3d44a6300c5c112e129e402291c2799b8fd2368a":{"kind":"REWORK","_number":3,"created":"2022-01-27 15:40:45.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/85/814085/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/85/814085/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/85/814085/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/85/814085/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/85/814085/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/85/814085/3"}}},"commit":{"parents":[{"commit":"86db21ea6afb6c26a21fc143a7d061e947c31a93","subject":"Patch paramiko","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/86db21ea6afb6c26a21fc143a7d061e947c31a93"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-15 00:01:36.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-01-27 15:35:59.000000000","tz":0},"subject":"Fix test_rebuild_server test by waiting for floating ip","message":"Fix test_rebuild_server test by waiting for floating ip\n\nThe test could fail on an adresses mismatch when the test server\ndidn\u0027t have floating ip assigned yet at the beginning of the test.\n\nCloses-Bug: #1945495\nChange-Id: Ic99ba61d87dfc24a684263d41e0caa4f54d1d263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3d44a6300c5c112e129e402291c2799b8fd2368a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3d44a6300c5c112e129e402291c2799b8fd2368a"}]},"branch":"refs/heads/master"},"d4c9cc89f199899b20cfaea975ecd15ee66d3523":{"kind":"REWORK","_number":4,"created":"2022-01-28 08:02:44.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/85/814085/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/85/814085/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/85/814085/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/85/814085/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/85/814085/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/85/814085/4"}}},"commit":{"parents":[{"commit":"86db21ea6afb6c26a21fc143a7d061e947c31a93","subject":"Patch paramiko","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/86db21ea6afb6c26a21fc143a7d061e947c31a93"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2021-10-15 00:01:36.000000000","tz":0},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2022-01-28 08:00:50.000000000","tz":0},"subject":"Fix test_rebuild_server test by waiting for floating ip","message":"Fix test_rebuild_server test by waiting for floating ip\n\nThe test could fail on an adresses mismatch when the test server\ndidn\u0027t have floating ip assigned yet at the beginning of the test.\n\nCloses-Bug: #1945495\nChange-Id: Ic99ba61d87dfc24a684263d41e0caa4f54d1d263\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d4c9cc89f199899b20cfaea975ecd15ee66d3523"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d4c9cc89f199899b20cfaea975ecd15ee66d3523"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
