)]}'
{"id":"openstack%2Ftempest~816315","triplet_id":"openstack%2Ftempest~master~Ib817e26ac87eeb9e7313a1473ce69e48199ce5f6","project":"openstack/tempest","branch":"master","topic":"fix_alloc_check","attention_set":{},"removed_from_attention_set":{"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2021-11-03 18:50:28.000000000","reason":"Change was abandoned"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2021-11-03 18:50:28.000000000","reason":"Change was abandoned"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2021-11-02 18:39:07.000000000","reason":"Ghanshyam replied on the change"}},"hashtags":[],"change_id":"Ib817e26ac87eeb9e7313a1473ce69e48199ce5f6","subject":"Add check for port-resource-request-groups","status":"ABANDONED","created":"2021-11-02 10:23:52.000000000","updated":"2021-11-03 18:50:28.000000000","total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"c877d7fdb9ff18036e6b3f53a5adfffd64541dff","_number":816315,"virtual_id_number":816315,"owner":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-02 12:58:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"disliked":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":1,"date":"2021-11-02 10:30:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":-1,"date":"2021-11-02 18:39:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2021-11-02 10:42:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-02 10:30:13.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2021-11-02 10:42:47.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-11-02 10:43:07.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-11-02 12:58:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2d12023b90025f1abf287f85dd9e8a6d95d208b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-02 10:23:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"77ac8641fa0fe42ea122b28f81e5407771bdf6e3","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2021-11-02 10:30:13.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0b0dc37bdd3b2754d45e7071b4821d63f1a7f7fe","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-02 10:42:47.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c0881a441b9ce9fc27c0342cb3152b29f0bf37c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-02 12:58:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/57344cb290524961a3cccd0af742ae11 : SUCCESS in 49m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5b6d46c1c6904b8681b571063b4b9f6a : SUCCESS in 1h 21m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/559ff3f00bbf426d8dc515e56e41175b : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4b73b9ed6f64f3aa3962405a7b968fd : SUCCESS in 5m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/daf871cf9d954f7d95de6e56a82551fd : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0aebcbfcb2344bcda8b4899fa51073e2 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a65712f9eac44169173d46a452a72e3 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d9b1d1e8e8c41d0b6d45d731677136d : SUCCESS in 14m 17s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/4f78bd000a234aeb9ae48b036080403e : SUCCESS in 1h 44m 28s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/31f71698b11543b39f36624d9293e2b7 : SUCCESS in 1h 12m 50s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6fd15eb687414c2b8236cb4d7e900282 : SUCCESS in 1h 20m 33s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/64e56d0e6ebe439e92f3965d2ce15b9b : SUCCESS in 1h 11m 41s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/730b669c7a3143808ef2dda18b387ebc : SUCCESS in 1h 24m 02s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/424e68f571104002bc12d48985370861 : SUCCESS in 1h 30m 20s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/6b1f87df28f443468ebe7cd789b4830d : SUCCESS in 1h 17m 56s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/50d17d4412254a89901f5e4a7c33ddc6 : SUCCESS in 1h 19m 15s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c9d685a1c6674c4e8eedc1eec7a13919 : SUCCESS in 1h 41m 21s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/cc7c3521eb8b4da2ac5309ffa20b1d3a : SUCCESS in 27m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8209a434495d425383268fef64393353 : SUCCESS in 46m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d5603c6c6c994ecc8aa7db1a74f7851c : FAILURE in 2h 32m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f4f5877ac50c4b518ca0906cd7938e78 : SUCCESS in 50m 10s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/62c2261213754e528030333fd84569f1 : SUCCESS in 1h 33m 06s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/94aaf5f9bffd4f08b2e6a39f51ed2509 : SUCCESS in 45m 34s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/1becaf642dee4062b1bcc2b5100c8b3b : SUCCESS in 58m 46s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/87e4a9fc16f94fed9ce869ff8f419191 : SUCCESS in 4m 35s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a96ad173843f4a2cb0c4fc9d204a298c : FAILURE in 1h 28m 08s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/483c6f59d2414bf19730866c71b81e94 : FAILURE in 57m 27s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/99accfd322b4485d8819e2b0151f04e0 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":1},{"id":"acfbbe1549d858333acbfd02fb46e2b69cd179f9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-02 18:39:07.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c877d7fdb9ff18036e6b3f53a5adfffd64541dff","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2021-11-03 18:50:28.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03e5d4452c385391a34cda868c00a99f8c60f3d7","revisions":{"03e5d4452c385391a34cda868c00a99f8c60f3d7":{"kind":"REWORK","_number":1,"created":"2021-11-02 10:23:52.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/15/816315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/15/816315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/816315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/816315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/15/816315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/15/816315/1"}}},"commit":{"parents":[{"commit":"8cd09500bea38bfd9478814be95fddf3df56fbff","subject":"Merge \"Adding Bandwidth Limit QoS rule client to Manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8cd09500bea38bfd9478814be95fddf3df56fbff"}]}],"author":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-11-02 10:18:49.000000000","tz":60},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2021-11-02 10:18:49.000000000","tz":60},"subject":"Add check for port-resource-request-groups","message":"Add check for port-resource-request-groups\n\nport-resource-request-groups Neutron extension changes the format of\nbinding:profile allocation to be a dictionary. It seems that in some\njobs we have \"all\" as network_api_extensions and perhaps this is the\nreason why the allocation returned for the port is in the old format.\nTo avoid such failure add an extra check to be sure that allocation is\na dictionary.\n\nChange-Id: Ib817e26ac87eeb9e7313a1473ce69e48199ce5f6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/03e5d4452c385391a34cda868c00a99f8c60f3d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/03e5d4452c385391a34cda868c00a99f8c60f3d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
