)]}'
{"id":"openstack%2Ftempest~816553","triplet_id":"openstack%2Ftempest~master~I422da8da06509e529f4091e21cdc3d392d3cf910","project":"openstack/tempest","branch":"master","topic":"py39-voting","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2021-11-25 00:11:19.000000000","reason":"Ghanshyam replied on the change"}},"hashtags":[],"change_id":"I422da8da06509e529f4091e21cdc3d392d3cf910","subject":"Test all supported python version for stable branch","status":"MERGED","created":"2021-11-03 20:05:13.000000000","updated":"2021-11-25 05:38:06.000000000","submitted":"2021-11-25 05:33:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"816553-py39-voting","meta_rev_id":"4bf9cac935bb0734d18d75effdee39dc42cf5e78","_number":816553,"virtual_id_number":816553,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-25 05:33:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2021-11-04 09:35:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2021-11-04 09:35:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-03 22:39:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-04 09:35:38.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"93ac762077cbd2772c2f3523bce73c3ad8659437","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-03 20:05:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d68e0030964b0e32c895947ee082e3e0bd9f8fd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-03 20:09:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c13aa795648bbff48451845be63e2150cdae2c52","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-03 20:39:02.000000000","message":"Topic testing-non-system removed","accounts_in_message":[],"_revision_number":2},{"id":"1bb03dbacbf1a5bd8f92630ff3dd8d94c0966391","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-03 20:39:04.000000000","message":"Topic set to py39-testing","accounts_in_message":[],"_revision_number":2},{"id":"25003f0f2d7c2cfe0e1a7a45e85db452f42a5e90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-03 22:39:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/97a921790ee644048d9030cab7099d37 : SUCCESS in 7m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/1b185a5921a148dabb909b5332905894 : SUCCESS in 47m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a1f911a19f6140f2afe187fbae37b0e5 : SUCCESS in 1h 23m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65717b603ce04d9bbae732b98f5ef9c3 : SUCCESS in 6m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96d904c7a606423199547ffbf02e656f : SUCCESS in 13m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9969f63f974b4b3fa9fecb5b560c044d : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d572160292f9486dbdb39891a29be879 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8f3e3e5d7c84435b61b89893f81b9ba : SUCCESS in 5m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b1893d45214f42d6b6a252b33bd0382d : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b64351d3475540a286e5f2bdd195cebf : SUCCESS in 5m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc615da74b8640b2bc75d37273be3a55 : SUCCESS in 5m 44s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/dab5a69d6a4a4eb1b98cdc6fb9d4a74e : SUCCESS in 1h 40m 23s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d31959aa9d8e47f58303ae4dcf399e3c : SUCCESS in 1h 28m 03s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7fb6fad4613a4967a3a7205446f15213 : FAILURE in 59m 00s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/1c0e69b0118c4abcade1e3a6e1c18717 : SUCCESS in 1h 20m 37s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/845c3dacc5a840d587cd74bf3b0073af : SUCCESS in 1h 26m 18s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/1885a47006f04dacacecf5a41bf9c0d8 : SUCCESS in 1h 27m 41s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/6972f55364fe4306886104d30ffcefc0 : SUCCESS in 1h 44m 55s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/50f7f9575f61423582e8ab02f5fdb3ad : SUCCESS in 1h 34m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f8bb2f20fa4146a5bdc2f36ba266ad0a : SUCCESS in 1h 38m 57s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ee9b3618a60645bbb022dbc37c8698cd : SUCCESS in 31m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08b1bb8619954b578ba32fefd5af06b1 : SUCCESS in 1h 04m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4d2bd21a22f94d41b78105360848ea37 : SUCCESS in 2h 21m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2c88e122eb6948f089d678bb465665d1 : SUCCESS in 43m 45s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0b6c9c7c40724378a7371e16a72b9d4f : SUCCESS in 1h 24m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/da807f64ea934984af41b768ecd53e37 : SUCCESS in 1h 01m 21s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/bd75e30e92944ecf87b0d2d8262e8447 : SUCCESS in 52m 41s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d58e5e0a399f47ea956dd098f88b100e : SUCCESS in 7m 51s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2b6aba8b91c94999a07cdcf3191aacee : FAILURE in 1h 01m 15s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/f6b4c0ccc4c1452bb7b253265ef4f844 : FAILURE in 1h 11m 39s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/493891a8e4b94986baadb3fc0c694799 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":2},{"id":"6f632e7f9243c5c1b8fe3756995cdffff6c970d2","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-04 00:33:32.000000000","message":"Topic py39-testing removed","accounts_in_message":[],"_revision_number":2},{"id":"0d9d5cbb3c965fd09788570d842929aec2c75b73","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-04 00:33:34.000000000","message":"Topic set to py39-voting","accounts_in_message":[],"_revision_number":2},{"id":"2aff018dd9cef70bb10de088632fc903b10ae4fa","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-11-04 09:35:38.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e0941f9596184e89650321eb87f35dbc1ef959ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 09:36:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"352df6af62615d85f19611fc5d743918971e244e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 11:53:38.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4b19c495600245beb0df478efec242b6 : SUCCESS in 4m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/016aa97e167a4121b722ce918b4c77ca : SUCCESS in 1h 23m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9589d744656146a2b6e41650a0966a0c : SUCCESS in 1h 41m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897c37b8b4574ccc88e6b1c21aa0eb11 : SUCCESS in 13m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b0f880e40fd4c7da4f2c5efe3b753b7 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10575577df8b49b1b75b12d399b7aec4 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f3397090d4b048aba80fe3bd3a209c44 : SUCCESS in 7m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/24b781d6b98f40a0837f084766fe3a29 : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de23b9ff561140b98205e2e8f219205a : FAILURE in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebf34e5a33b3402494c4a6c6f530b67c : SUCCESS in 6m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/55b97a9efb4a4e189d58feb8331e8c21 : SUCCESS in 2h 09m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c13d448476ca46819851e2681b244865 : SUCCESS in 53m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31f87a0285294dc58d2dc5626df6b7f8 : SUCCESS in 49m 09s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9eae8f49ff61454e9ff4839918cf6140 : SUCCESS in 1h 17m 01s","accounts_in_message":[],"_revision_number":2},{"id":"55e6383081e01c88e424732d52cede4766922867","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-11-04 13:05:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a07fbf83f709f7a4ed9e725fa535b7d05b9a5321","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-04 16:29:47.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/07c4788f43394eaaac9da4a26b168839 : SUCCESS in 3m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/421be421f11e43ccbb4c0ecb6ea1d8c9 : SUCCESS in 58m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5703862f279e4abe871378ff67e82e6b : SUCCESS in 1h 19m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/615168d5b3ed447196b7b2944a54d4b3 : SUCCESS in 6m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0fefb7573e74850a45d13886b1965eb : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35e1ffcb94cd4622b5cc421a3dee52ec : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21dd4f4fa77644cda6f399613569d2d9 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25e6d4637e674ce5a3f639081a48c455 : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e193dcb85bf440b4a69509dfc84885f6 : SUCCESS in 7m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae2c87bf5ba44d4299cd00e6e9c90a1e : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/779d027b89044779a722619eaa7a57bd : FAILURE in 6m 33s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d3f4df242eb44d32a3f2cce08b4f0c46 : SUCCESS in 2h 00m 02s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/9f65af60d9ae4deabe713748f47f76fe : SUCCESS in 1h 37m 10s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a28eba0f965d46e3bb7346fe35e03602 : SUCCESS in 1h 16m 29s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/492df68407dc4d9a97b6722f2ddef215 : SUCCESS in 1h 17m 49s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/fe2f84a3480744c59aef889904db47fa : SUCCESS in 1h 36m 43s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/742ac1c3b7ee405f992144f0d2b7660f : SUCCESS in 1h 32m 54s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/db3fab4d687b4ac7882d740a1954ad08 : SUCCESS in 1h 45m 38s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/25a4e736c7894a3c9403ce0f2acd62ac : SUCCESS in 1h 33m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c2be333faf5f4e608c50e45222d13d47 : SUCCESS in 1h 38m 29s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/41c092dadcd84c12929ef6efdcee86c5 : SUCCESS in 33m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6131f8127914ddab47d25a724f48612 : SUCCESS in 1h 11m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf3abeeafcec4a639917773a54c21ca5 : SUCCESS in 2h 19m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a10faa4df77f4333810fe50e3738db11 : TIMED_OUT in 3h 06m 48s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/24eba7acd15241a9b7b0f693c627f72b : SUCCESS in 1h 21m 33s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c7ef2d6495d4171b4781ca1b2e29f14 : SUCCESS in 50m 10s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/1efd215d1b674e27bdf42a45e6355706 : SUCCESS in 55m 44s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b317f47574884eb3908c96b26c5bd5a9 : SUCCESS in 6m 01s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/507d274e1f594ebd93446b8ddabe9c1b : FAILURE in 1h 26m 24s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ce81c2a7f8874e6a9c09a17fbca50dbd : FAILURE in 1h 07m 38s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a27103a47b154f39b71ee06311682b2c : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":2},{"id":"7edf27c90e3e2872fdd150de457cc2f93f9b3e13","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-08 18:38:26.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6d1624c9ae45d679db81abd9941a8c45eece85b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 18:42:58.000000000","message":"Patch Set 2:\n\nUnknown configuration error","accounts_in_message":[],"_revision_number":2},{"id":"752a94396e59b20b83af737094c07b16c4578ea6","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-08 19:40:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"60e329c47b16f419356d5c3febbd4a0cf2566f64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-08 22:26:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5b179c15b63f40b0b35b572f29305af9 : SUCCESS in 2m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/9387cfd547f64215a56ac2db50957f1a : SUCCESS in 56m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c8c85dd2288c4001a53529331c42c0b6 : SUCCESS in 1h 36m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c35ecd775e24dd6b14ce71a59bc9701 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24d66a817afb4ef98900409b39596cd0 : SUCCESS in 13m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/446e132fc20344478d71059c064684fd : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c508dbb3a8c44c189dc53a5baa6a814 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/094269e144e74833809984ef6361a436 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cd9f6649dba3477e8cc379c8bd517dc2 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d12cbb64dc1b457fbad1c507b264906c : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28a3ca7046ef4a42ab564c2be87ed4e0 : FAILURE in 7m 12s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/5e2fae64f6ee4a558fe9b099fb1899f9 : SUCCESS in 1h 32m 09s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/5a94a4811cf74f2a9f15925b0d68c739 : SUCCESS in 1h 13m 54s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0be7ab174fb04160a6b186be9c9ce65e : SUCCESS in 1h 31m 33s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/00db38ce8b084120b879dbf2000bbd01 : SUCCESS in 1h 30m 41s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/c2dcff166bee498a9adbc761575aac9e : SUCCESS in 1h 09m 15s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/b50c81064fea459dbefb618a197c3922 : SUCCESS in 1h 38m 46s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/f4db65cfec9f4b39ac675c078658032b : SUCCESS in 1h 42m 04s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/20c002c329574cd996ab0a0d69f75a2a : SUCCESS in 1h 28m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/935dcfe7bcff4f51b8f7913de7bac506 : SUCCESS in 1h 31m 40s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/46a540ad31a446f09bcdde211575868b : SUCCESS in 27m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c8bd09a36f54a059ed7c0d2548ba672 : SUCCESS in 51m 54s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/014091111d8e4225bd409fc9333f8b80 : SUCCESS in 2h 13m 47s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6115c25f882e406e938ca1fcdb970052 : SUCCESS in 51m 55s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0ae45fd62b764f1fb917e152c53cf174 : SUCCESS in 1h 37m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/409d2837481a438d8dc871f3bea80e9f : SUCCESS in 53m 46s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/f6c6004da1f344249c62e5079c422700 : SUCCESS in 1h 09m 52s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8cb4f3d29e6a4e72830cbfa90959d3f6 : SUCCESS in 4m 40s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/b5fdc4928f6544108984225a51bdd856 : FAILURE in 1h 12m 51s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/3d15b6268bdb4f44868cc1c849b4c4e9 : FAILURE in 1h 07m 50s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/197615e50f3c477e9e06b8d87b02a76a : SUCCESS in 4m 43s","accounts_in_message":[],"_revision_number":2},{"id":"85da69516880ff83385f2db0e2857b9a2fa47288","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-11 19:09:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"37cc596135c610bf633b8dc1565f146e421cbddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 21:30:32.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/44320bdc31254788961e4402643a138e : SUCCESS in 3m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/15971c03d7b94056825d3e33494fff9a : FAILURE in 36m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3056cf1b6e7d49d3b2d895e716e4d3aa : SUCCESS in 1h 20m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/27b04c9e338849028ffa75bfc2c77913 : SUCCESS in 7m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4059d9296274b0dbf6d1bfb7b4e4fa3 : SUCCESS in 17m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/565fa92a31bb4694982a96e83b6a3361 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcf122667ae7498cb4b177e542195369 : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f86aad3f6f345deb13a109c4bdf6ac7 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0323f9fb2f284e138a0ff7c846f9bfed : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07627ad5079d4c81805c8733771e4ec9 : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c1f1d750d584d3fb647a4deb1b5fa0d : SUCCESS in 5m 55s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/4ff933290b5f4363b1a1483eab3ab16e : SUCCESS in 2h 17m 51s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/6bcfb9ff6d8a4327864cbab2eac1bc84 : SUCCESS in 1h 22m 37s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/139c440c2c494ae3a4ac413f44e71a54 : FAILURE in 1h 19m 36s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/88b70ca44a0a4ac6aabc52c80f9e7e3e : FAILURE in 1h 13m 42s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/2db92e789778425c837fdfbc9ddf2c3c : SUCCESS in 1h 34m 10s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/4eaccd2796d141a7be4622f66bed256d : SUCCESS in 1h 36m 12s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/88e9c69af7bb470f8b10a3f703d0b6fe : SUCCESS in 1h 36m 27s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/5db30fbd98864174920ed0c40685c744 : SUCCESS in 1h 23m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9c23520fc0a4459fb57652d0a1205072 : SUCCESS in 1h 34m 37s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/19d7ad38eda443329972c2e0aaffafa0 : SUCCESS in 33m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78d8cad1684242fbb8b7fce03f7c0a5c : SUCCESS in 56m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f2ca7b8f155c43fabc818398d8a8dbaf : FAILURE in 19m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d27347ad3c4e4cf0b1eb65b18509782d : SUCCESS in 46m 33s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ee9cd78206dd49cf943f027e22299d0d : SUCCESS in 1h 26m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8f28c9813d274273bf540199026028da : SUCCESS in 59m 47s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/f7f576a1cb7c4775956642380e72be80 : SUCCESS in 1h 05m 41s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/4678f854abf74738b5de4a0f92573de0 : SUCCESS in 7m 31s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2a05261a37b9416d92b6bb18ad6b1b15 : FAILURE in 1h 30m 07s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/71628b2b2ff94c7784a3319eda63f36a : FAILURE in 1h 00m 06s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/cfaf71e1e3eb447593731de8817d5c3a : SUCCESS in 3m 54s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/3f560437885946a6a89f22fcfcb4c914 : SUCCESS in 1h 44m 18s","accounts_in_message":[],"_revision_number":2},{"id":"03efb7f34359baf191547c96d9158263b3f01ff3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-11 21:42:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"016c2668e87251027b3556dca308eb8af3fc74cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 00:23:19.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/54f83501704d4ccdace970461c52aa37 : SUCCESS in 2m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/15a4d3f5553d4cefab4345e9701a71a1 : SUCCESS in 1h 05m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/37076b12f61b4138b56d5921fa5a7e51 : SUCCESS in 1h 14m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b3e9b42c7514470a9cd0934c2b0b1ad : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33eeecbf186464da49b02c130906fd7 : SUCCESS in 20m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c0ee151c61e4d708c32d0e9416bb9d5 : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf1a0649113144a4aa83dfe804cab510 : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/45c58d105a044bcdb754cf56c71053a2 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fb89bf835a38472384c808693d263888 : SUCCESS in 8m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9eef0075f184eb09add9d0e73a77dc4 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9bf859ccfe74485499b2a20602479f25 : SUCCESS in 6m 31s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/1f7e357a8abf4d76b3163f197835e691 : SUCCESS in 1h 35m 18s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/bb8680389c1842df82e296f74ada1736 : SUCCESS in 1h 09m 45s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f7f063089166486b87efc9d147bd8567 : SUCCESS in 1h 12m 52s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/20f4578550f644658fb5bc7868d6a2ce : SUCCESS in 1h 27m 53s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/8950405d97d74368b27e711c8019f5bb : SUCCESS in 1h 08m 06s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/095416367ba94b30abc99d6846d6d398 : SUCCESS in 1h 28m 25s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/6f98b2fddde94caea95892eb44beab5e : SUCCESS in 1h 32m 06s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/9a2aee73be224e479f0260e7ebf776ee : SUCCESS in 1h 39m 14s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8ee8b764fc7e4d7d93ca5d070cf2d999 : SUCCESS in 1h 43m 05s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/aa890c97d7b048a38b39e1b29966a4ba : SUCCESS in 27m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ce79b32416a49daaf5affb283536bc6 : SUCCESS in 43m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3d20b2a6f1684e0b8ddc51d0c0df907d : SUCCESS in 2h 31m 39s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/59bd91f9e775467095b7160214f93537 : SUCCESS in 40m 54s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/cd76492a7ab74c29a9359b531c075c09 : SUCCESS in 1h 11m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b6443b4d005944408106a9390b8f1df3 : SUCCESS in 49m 03s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/08b3e7aa0f524902807bb565b5399e7b : SUCCESS in 44m 03s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/746d25a5f8ce465f9002e50a2f184d03 : SUCCESS in 4m 47s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/e1e8baab229342f8a89fc0972c5ecc8f : FAILURE in 1h 23m 23s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/6e6244eb89104ed5b897a1838dd8e9ea : FAILURE in 57m 14s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/54a25409092348b1bb1468790044a6b0 : SUCCESS in 4m 04s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/1599f476b01942e78d3557124a1000a6 : SUCCESS in 1h 44m 23s","accounts_in_message":[],"_revision_number":2},{"id":"ecf80bc11370fb1de73a679d837bcd3d99038f5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 00:24:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"7ea59713bf78be44b549e6cc486c3403256b542a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 02:55:12.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1e6675c7a80b45818c5dc1adb4e1f6cc : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/1b3552bb2bd947d492cd8cbe3f3decbf : SUCCESS in 1h 02m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/717e17f50a78472d9d4d79161d8d3e37 : SUCCESS in 2h 04m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/177c9a5a883741aeb8bfa2bb0aaf628b : SUCCESS in 21m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c35f307aeca24cad8bb511511161be80 : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/324be67a79034799b6791fc0c1e83f59 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f62855fec2fc4fa18b46a616869a3046 : SUCCESS in 9m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c3074a19007045d3beccce50dbd1a2c7 : SUCCESS in 8m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e150fe32f8544f6d9c9afaaa94455b1c : SUCCESS in 6m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/151699ea4f804e6bb729d25934ca6d45 : SUCCESS in 9m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e958350e9c974003ad76ef87ac2c0391 : SUCCESS in 2h 14m 32s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2c69367231c84a75abd8fe6ac164ac3b : SUCCESS in 56m 25s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/0578f1f16847488ca93cb4881210a8c8 : SUCCESS in 1h 44m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf126c207dec4a2d9f51fe67383baccc : SUCCESS in 55m 30s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/acaafbedfc93478e953e38612162b4cc : FAILURE in 1h 02m 56s","accounts_in_message":[],"_revision_number":2},{"id":"6591ac7fed09868475506bd70bd0d01abb9bdd93","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-11-12 07:58:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"faa6fcec7d4b38cf77ad2f0237ca6af8070adcc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-12 10:49:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ab31196a1380414ea068a415223f8de2 : SUCCESS in 3m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/0f1e239c626d44acb9f082dcfbbfcc8b : SUCCESS in 1h 00m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68900f5d4f5747799864fa022d210e2e : SUCCESS in 1h 16m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d514631c7ee4e0d9ddc21f17f380cab : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a951c64505455c9cdca3ba3f08b8b5 : SUCCESS in 18m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0228c2fd605410e893defdf3826e9da : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98d324d0b0b044339bd93f0406484821 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af2895b6ccc840318a3a43d902c3b82c : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6ecbb52706c64eb190bf5c98939263b4 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d463a7f2cfd4a0a98cf86254b019548 : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f7a6fa05efd45cfb0eb98b8473cdbcb : SUCCESS in 6m 47s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/0c5f8ccd93a94233a13bca35fd0ff780 : SUCCESS in 1h 31m 33s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3372e50a10e74d879417393a25bcd1b0 : SUCCESS in 1h 09m 51s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7b12b620f431420ab945e4d7b317a5a2 : SUCCESS in 1h 33m 42s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/e3292907de364e5dbd263b3d577bef50 : SUCCESS in 1h 13m 01s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/0cc976042e91485da89758e90d2a6be1 : SUCCESS in 1h 28m 08s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/e9f358332db64027a75eb879383bb0cf : SUCCESS in 1h 25m 29s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/16cee03ec2144c4a8c8d12fe08e771b6 : SUCCESS in 1h 33m 56s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/c03011371e824b8fb9cb42d18cefcaf3 : TIMED_OUT in 2h 09m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9600ec8af3814038906425cf4e97f6a3 : SUCCESS in 1h 38m 06s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/49b3480cb22d47e5b47b3ed23693aa08 : SUCCESS in 33m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5df4d0f8d21e4d5cb74f1b53bfd0dd44 : SUCCESS in 57m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8311ac7e9e3346dba09cdabc904b2eb0 : SUCCESS in 2h 42m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4776d5e998db4c66a83e3172f48be3f2 : SUCCESS in 1h 04m 10s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/32d09fa5285342d3a11ded57ec574814 : SUCCESS in 1h 19m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/28bced821359427880586914ad303476 : SUCCESS in 1h 04m 42s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/50156d42c009455e9fd3ebae70d29502 : SUCCESS in 1h 02m 18s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/84ca1a2f66b34aac840196fac2812123 : SUCCESS in 5m 05s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c1f6fea089fd4830b0f166bb48885677 : FAILURE in 1h 09m 54s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/aea9642344ed49b993b02e6a949d26b1 : FAILURE in 1h 00m 36s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/cf6572aa847645a99c90805e7943ff42 : SUCCESS in 4m 10s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/2c201c745f424a62aa76bd1118086989 : FAILURE in 1h 28m 19s","accounts_in_message":[],"_revision_number":2},{"id":"65d6b684f10707acc20cdbde9ee0757239eb8f66","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-11-13 22:27:52.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8d58482c4940d5c0a85697388110beb4dc529aff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-14 00:27:17.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8f797e61b21b4b3eb999291ec1dbfa18 : SUCCESS in 2m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/88ddc512597b474fa7e9dbaf2d3b5924 : SUCCESS in 1h 00m 22s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3533d211c0a84c53af8ef47be2df5ead : SUCCESS in 1h 30m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4accfff981d8448fbc89164888d992fa : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64013e21b09d4567b7827282c047a523 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3990f036274f47b0810e47a1989f557d : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31b92d2242de495f9af8d32a4ce90850 : SUCCESS in 4m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/921bda316e4249dfab727f32dd243771 : SUCCESS in 5m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d84378a6d9004b1484afb388e56afe17 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96da68b76db547a88c1075ad269b8539 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2859df748f3c4d318a96aa2f4a35fecb : SUCCESS in 5m 48s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/1793acdad6b34794944051818a880267 : SUCCESS in 1h 28m 01s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/81a72ba357bd4551b5729956ef2882e5 : SUCCESS in 1h 19m 22s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cdc2c4734e2546cb85e69f47b3e718c0 : SUCCESS in 1h 27m 41s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/69b5b424826347b4ac7340adfbef4580 : FAILURE in 1h 18m 10s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/ff4b020881d44fcd95748b52360a3693 : SUCCESS in 1h 28m 15s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/196969a8fee3468aa50d835839dfe497 : SUCCESS in 1h 41m 37s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/2ffc322b20ac49e1988e6cda13a62a73 : SUCCESS in 1h 46m 38s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/86f6ee7b6bf64b94835f64fcd2c90d7e : SUCCESS in 1h 33m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5d10865c6e1448c1894a093d74be2887 : SUCCESS in 1h 28m 29s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a69409cb7d1847e08e9c5aa4ac5ace39 : SUCCESS in 34m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7122a688dc0341a3ab24da87fac32aeb : SUCCESS in 1h 13m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8947b39b6f14452a9fe78b1bf0f8ef62 : SUCCESS in 1h 46m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e0373869c72e407d8f2de898fa4212ea : SUCCESS in 48m 49s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9776f0a1e4374c3b93fdc06d4a824d93 : SUCCESS in 1h 17m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3fb557e566a642d2ac9bd8c8b37f72cd : SUCCESS in 1h 02m 12s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/4a0d9b1a36e8428a8e1bbafe5ec7bce9 : SUCCESS in 55m 38s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/afc12d2646114e80bf201f067266427e : SUCCESS in 4m 49s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9357130d9b40430e93d791a8729b6c22 : FAILURE in 1h 12m 52s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/d282b16343ef4397b3c1f090c12af2bd : FAILURE in 1h 07m 24s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5a6c4d2946fd4b568cf91903b6e1b375 : SUCCESS in 3m 39s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/050f925c247243768af3c6d01044e456 : SUCCESS in 1h 58m 19s","accounts_in_message":[],"_revision_number":2},{"id":"01eee69182d5608c572d15e95076c48a6d00eaf0","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-11-15 11:27:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"53eb10c5b03942de2d26dde17560ba7a04c3c30f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 13:28:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/09b9a4a7fe6645829024168a19a0be07 : SUCCESS in 3m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/0b9f96283c9a446caff8876e5acea98e : SUCCESS in 1h 03m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8915456918f340aab62d31db51976f13 : SUCCESS in 1h 26m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea7f3e6aa5444706a0b25425f0df971a : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/51bd12017bbb4ccaab40c07a4d2ee5d7 : SUCCESS in 12m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90df80aba1294ce0aca9a51309d6a2ef : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8134c95789b044af9b36cb3dd91bc594 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/76d32e8fb9834cbaafb49a046b3cc6ca : SUCCESS in 5m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/632bb1b568a24b69a9004195a4fff663 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27a50114c38845839aa6a0306bc852f8 : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5b34a750b6340898eaadb8823cab553 : SUCCESS in 5m 04s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d6b2264710f24cabad9321188c28f6ea : SUCCESS in 1h 57m 33s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/439323a242c2490680ee98f2745d9fd6 : SUCCESS in 1h 16m 36s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d7acd37f634448e4b006698df9208a27 : SUCCESS in 1h 16m 33s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/3133b15444b5431c92cf87e1d9f0b210 : SUCCESS in 1h 14m 10s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/d690d16fa5fa4063ab7875ac05c1511a : SUCCESS in 1h 17m 58s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/d483154980694cf9acf38d54ce255e8a : SUCCESS in 1h 26m 02s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/319de1d8a1b04e2d8b0691ade68cb890 : SUCCESS in 1h 42m 06s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/ecd6d9fba6ee46ebb891411e4e698b85 : SUCCESS in 1h 17m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b1d0dc2d8ffb406a8001964548013516 : SUCCESS in 1h 41m 54s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/4c23534ec46343ec81dd0f00701dd503 : SUCCESS in 25m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93f773d0da934cb097f5f3a6d7d2b1a0 : SUCCESS in 43m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ebd742bfd24848499ed31f9766e3306e : SUCCESS in 1h 54m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/370bb444c8a14ef09719ced205868b07 : SUCCESS in 49m 13s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/21fc3c78bc014d7b812a0bd85989f650 : SUCCESS in 1h 25m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/525effed1d894f3486914abda8e8ed93 : SUCCESS in 52m 51s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/c1bd5fcfcda848df80e5fde98f83fd60 : SUCCESS in 56m 16s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/c72100c425d947ffa022e7989ce2da64 : SUCCESS in 4m 37s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/19add8d8a537489a859fae017ee962e2 : FAILURE in 1h 11m 32s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/5444157fd46c42c298a822feac08b1e1 : FAILURE in 1h 00m 48s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/dec8b7e448024464a54c3939b1a803db : SUCCESS in 4m 54s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/601a391e6fc54cd2958ba868ec31c7a0 : SUCCESS in 1h 27m 11s","accounts_in_message":[],"_revision_number":2},{"id":"3692447e4bf6362e1f0b99294c33936c1dc5ae82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 13:30:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b98dbb1a41db8aa676b796e0f36d2f03130af3c4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-15 18:50:50.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3c4b17c68320466e848ff1faeb5cd605 : SUCCESS in 3m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/b5e16c89cebf4e2fa7e480e2fdce19e0 : SUCCESS in 57m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9d8659fd5a0f4ee3914bbc3fc1c55967 : SUCCESS in 1h 22m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1a394f713fc4922a3788713ea4b9a5b : SUCCESS in 13m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc5302eb430a4d18b04b02bdb015f13a : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a24517058f2441d496c990706ceb26f2 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86f6df353c934b07b030a6b98d034cbf : SUCCESS in 6m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/face21f98c694ccebed0f415d1ea33fa : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4aed0a57605f486686cbc599269a2f75 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d1945b08f95248c8ae233d15f0160698 : SUCCESS in 6m 40s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d0ccf22f57f6442990d5d8d08bb66aa2 : SUCCESS in 1h 48m 35s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4f29bd5c9e41453caf4faf2a2747424e : SUCCESS in 1h 04m 59s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/6bf139d232e543ed8f0b6b7bee32a1f1 : POST_FAILURE in 1h 41m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/75e850f649c542389245b7433eee2a17 : SUCCESS in 47m 36s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/51230698077246579a424e1a6cd3d820 : SUCCESS in 1h 05m 35s","accounts_in_message":[],"_revision_number":2},{"id":"5a834e09e008b3d36f50e4716c20938fb92ed3dc","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-11-16 11:55:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2751e9c68e24a4ee57bfdff9d3eba8a9675ce745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-16 14:01:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5d3f905b04d6452da52cdbbb0c7c059c : SUCCESS in 3m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/e543f9dda2684c529611d408b11bca91 : SUCCESS in 55m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d06a1d75264046c7adb93f8200ea1d67 : SUCCESS in 1h 25m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ca5064973704781a1b4560cb8fc744e : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56280c9a83b04d068d24ac9c2e8dbe5a : SUCCESS in 15m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/211dc9595b8a431fa3dc452a2ac6a6c1 : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f85d5f04cd984e07b7bee23e7c3ee3f8 : SUCCESS in 5m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0a69314dd6e246df9e4e815e5bee3b01 : SUCCESS in 6m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e18eee81822e49dc8c4f4458c8f66e7d : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08629027245c4a21a2591c589c84e9b6 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/23d2bddd9f564cfb9ddf495cb2fd197f : SUCCESS in 7m 12s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/4afbfff1524345068c039ec7361f8aa3 : SUCCESS in 1h 35m 33s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/6a03520bcbbe4fa3ae12d93e4767ed53 : SUCCESS in 1h 11m 41s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e2ef82e44c3743dbac08d0e2b189a425 : SUCCESS in 1h 10m 38s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/3586109d8cd347d19fa94196ff42b424 : SUCCESS in 1h 44m 00s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/632a5bd3d78945708f6c983b040388e1 : SUCCESS in 1h 25m 25s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/0c6e87eb0f5447618ddd7badeb4290e5 : SUCCESS in 1h 33m 52s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/27b831dbeb834d0d98384d314b9805e0 : SUCCESS in 1h 28m 13s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/27784538a287489098579d31607e9933 : SUCCESS in 1h 59m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6449e087c431407287202dc0da46198e : SUCCESS in 1h 31m 21s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/4c5d0f3c979046a4ab55d920a796c044 : SUCCESS in 33m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/596390090dec40c0994e5f7688e57678 : SUCCESS in 51m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d28fe3632eb24ca5bbbffbacb258e690 : FAILURE in 23m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cdbc5df3c6054aa98c9f91ac44863604 : SUCCESS in 51m 18s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d7c8852eb2ad43ceb893ac39ca3cef53 : SUCCESS in 1h 08m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/94adaba4f96f4e91a67f870eb37e9933 : SUCCESS in 1h 00m 53s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/99b7dac2beb54245ba67f0e7b0a11579 : FAILURE in 48m 28s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/4deba6d0c4164b5cb6722ff3c73eefd2 : SUCCESS in 5m 33s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7eb075e0a8174e89a916e4fddcf728b3 : FAILURE in 18m 58s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/713f084ec9ef452c8af0b83f6178a479 : FAILURE in 20m 42s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/48af3f86193f4acf933d12897dc1c048 : SUCCESS in 4m 43s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/f4932eb9794d47469bc45a586af2f6b6 : SUCCESS in 1h 30m 22s","accounts_in_message":[],"_revision_number":2},{"id":"573d941198a9cb1a4ed65ec7555f5d9f97c26f2e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-22 18:04:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"498aba510616834af72ab82d7219a5eb1c35083c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-22 20:19:44.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/09eb97c789a7418fa3d4d2873eaa0339 : SUCCESS in 3m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/ffb32b0227804117be3189a032f3231d : SUCCESS in 53m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b598a0040b9d4dd3b404d2b1764e4b61 : SUCCESS in 1h 15m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce5e76e784874e19b66aaff011febf16 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/561a62ed68ee41d48fe52679b794d782 : SUCCESS in 21m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab4f31aaf6d449e9ab20a4a410194879 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcdeb8bd2b6f4eaba78df433725b776a : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/698a6599235c472b9364c7ed9d4e27c3 : SUCCESS in 4m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8c2f0e1768dc443ba209c5e0be514934 : SUCCESS in 7m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5704559b2a3b416d9fe4f3b25d35b758 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27a57814d93a427fb60ff5765b21d57f : SUCCESS in 4m 56s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/31cd8ca768d348a6ac31d3c0d63c0ada : SUCCESS in 1h 42m 07s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/9b49cd2871c04a9ab496aa5b65031b77 : SUCCESS in 1h 30m 03s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c6ca5b281c214a3799be6e019ed8bed6 : SUCCESS in 1h 33m 30s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/8f6416ce499b4139b54ba84ad2208a6d : SUCCESS in 1h 32m 03s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/17288884239b4c1e973072c68fb5c905 : SUCCESS in 1h 36m 25s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/e55cfe9664c5416eb15f8a50af0b2bb4 : SUCCESS in 1h 31m 55s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/1cd384f2e63b4a23934abc07329da97a : SUCCESS in 1h 24m 51s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/2d2fa82791cb4492bb459058babd3c73 : SUCCESS in 1h 26m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/aa7e23534fbf4049b22aa28747eda6df : SUCCESS in 1h 33m 22s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e01b40baca8141ce930dbea23e743128 : SUCCESS in 25m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76a453cce573458380fd0b5e9a4fbfec : SUCCESS in 44m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7ad1bd309af047ce8963486d4e481e41 : FAILURE in 2h 11m 52s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/48fc4b25afc84b34b6e8d793276a2a7b : SUCCESS in 50m 53s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/50324b1bf8f844c8b3b0fa034887a6af : SUCCESS in 1h 28m 46s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/75a1c88ea0164e698f896ad48b530948 : SUCCESS in 1h 04m 42s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/5ba51bcbd1b84c55840578e40ee60128 : SUCCESS in 53m 49s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/ba4d15d5e91640b09971531e30f0f77c : SUCCESS in 4m 56s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/31933e2351b242a5831432b34c898892 : FAILURE in 59m 05s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/33382ca0d61a4859a228d108a07a0452 : FAILURE in 1h 11m 05s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/34cf7c97c2f444569e000794469ebe9d : SUCCESS in 3m 51s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/26cfa42384e44e88bcfbde1788a958a6 : SUCCESS in 1h 29m 58s","accounts_in_message":[],"_revision_number":2},{"id":"ba719673ebfb228ba48678f9e960b228a81af311","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-25 00:11:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e297118325785324c8ed6c9ddbcb74f5ba0087bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 02:25:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/18848b0d0b584708a5fbfde3d1fd4131 : SUCCESS in 3m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/cb825085ab9f4e468a14d3baef239640 : SUCCESS in 50m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a6d54a5e4df4123a3b1bf2a1e9873e1 : SUCCESS in 1h 19m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fa6536d2fb1470e94a86a88cc7279ca : SUCCESS in 5m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c405dbd8ba684ccdbf5ed5140f6d3ea8 : SUCCESS in 13m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/360860da23ee47a6834390915f8693ca : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c12e18f759df4b0c916e9816cf35c45e : SUCCESS in 4m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b950e4c18e6c46f2984405ba0373a739 : SUCCESS in 5m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19cdc8ed8c15405fb9612d4beaa69033 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8151de7f4b814ca89db4720f2c1e46fb : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/172e9ee35cfc46869319596769493f82 : SUCCESS in 5m 40s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/7d22d1cf2ba447db9a474b36ba0adbb6 : SUCCESS in 1h 29m 41s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/be2af95a196e40188f43b998bfe60c8a : SUCCESS in 58m 06s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/58f2cc5cd4424641aef2c1346bf9631d : SUCCESS in 1h 13m 51s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/1c905390845f40e59d3e24ed5d89d069 : SUCCESS in 1h 14m 03s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/80ad6ef778944309b62d7862242be47b : SUCCESS in 1h 24m 05s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/ac42888fea0a48dca3b0e4da811ab94e : SUCCESS in 1h 31m 24s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/d2679f736df2492eace0bcc0ad5aa767 : SUCCESS in 1h 18m 51s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/03b26c5b4dff4ef7b08a37ae60392bf8 : SUCCESS in 1h 24m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7d38cd00ff37435ebfda94cc2d301ceb : SUCCESS in 1h 44m 39s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b484d73f2a6e4201a1ac4d7533846c62 : SUCCESS in 30m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ea60fb7c76b49229e7f4fb4ff54ec2a : SUCCESS in 51m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/149123af6895434399414b1dde64d7cf : SUCCESS in 2h 10m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1c6631718e8f48e5b53f2ad7302ead95 : SUCCESS in 49m 49s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/03f8d8d758ac40649fc8df3d4793625e : SUCCESS in 1h 13m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/39af3c6d36974c898312839c6086d427 : SUCCESS in 43m 46s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/c485e31d7d75467ab0645a3cf49058fd : SUCCESS in 57m 21s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/17ce2987c26a4480bb52e0f41d34d1b6 : SUCCESS in 3m 58s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/b54d24c8bb704462ba5349477105b3a5 : POST_FAILURE in 55m 51s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/b24b595e0a2d4ec5b149c0891f1fde6f : FAILURE in 54m 38s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3a1b983806724f8c907d311216a80ce7 : SUCCESS in 4m 19s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b752b745d2e4b06ae9cb9573780b36d : SUCCESS in 1h 09m 00s","accounts_in_message":[],"_revision_number":2},{"id":"3a9518dc5ad1ff8583207f596233591cbd8de472","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 02:36:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2dda9c5d6a8e26176a247555e6e37da90ef86269","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 05:33:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd24e4fe033b49b4bff95fd2f468cc5b : SUCCESS in 2m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/8a6a20f5c2874884ba2918cc06e2f8f2 : SUCCESS in 56m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d8a5a6fcc6144eb8a9e304b8a7369c0 : SUCCESS in 1h 20m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40b91e99827f48f29b6aba35da27f325 : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d73990b385874d7d9b1db26d35b0a549 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac66a7010e74aa2abaf151159883888 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99ff14d75d4f4d88bcbf4115507a2223 : SUCCESS in 5m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ebed2cc940d54100993d254387f3c9f4 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/257f3dbc958541d8b52785615e5bf5ac : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e6f098f54704e92838de2d9d0b7f53d : SUCCESS in 5m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b3fc77219b4f4ed4b43b1aaebd0c6d96 : SUCCESS in 1h 59m 21s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b5cc705ec1ae4e5ba0eef5a47def495a : SUCCESS in 57m 41s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/d7b73902e1f54633aceeb2351b4eb7a2 : SUCCESS in 1h 16m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d25f8ad2f064bbca40f27fe28789370 : SUCCESS in 48m 55s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/77132553f9ec4033b15e717d5fda3dab : SUCCESS in 1h 16m 51s","accounts_in_message":[],"_revision_number":2},{"id":"ff6d0e824921681845778bb7955971c81c73560a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 05:33:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4bf9cac935bb0734d18d75effdee39dc42cf5e78","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-25 05:38:06.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6068cca92804a2c8b8f99a66b86f9b0 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba28546ca6414ac39cfdfff5161bbcf4 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1aac05c7b2f1fe4699a7f08183c93d6f92ce1efe","revisions":{"21dbdcfa07f2c692810b26662e143df9f300d793":{"kind":"REWORK","_number":1,"created":"2021-11-03 20:05:13.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/53/816553/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/53/816553/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/816553/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/816553/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/816553/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/53/816553/1"}}},"commit":{"parents":[{"commit":"6753c266d74275eb1e818f1952bdd55f67674059","subject":"DNM: Test if system creds request pass if enforce scope is disabled","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6753c266d74275eb1e818f1952bdd55f67674059"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-03 19:56:11.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-03 19:56:11.000000000","tz":-300},"subject":"Test all supported python version for stable branch","message":"Test all supported python version for stable branch\n\nAs Tempest is branchless and needs to support older\nsupported stable branches, let\u0027s test it with all the\npython version supported in those supported stable branches.\nFor that, we need to explicitly dd the py jobs instead of using\nthe template. This way, we will explicitly know that we are testing\nall supported python versions when the new release template bumps the\nmin python version in the template.\n\nIn Yoga testing, we are making py3.9 job to run as voting, so\nadding py3.9 job as voting.\n\nChange-Id: I422da8da06509e529f4091e21cdc3d392d3cf910\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/21dbdcfa07f2c692810b26662e143df9f300d793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/21dbdcfa07f2c692810b26662e143df9f300d793"}]},"branch":"refs/heads/master"},"1aac05c7b2f1fe4699a7f08183c93d6f92ce1efe":{"kind":"REWORK","_number":2,"created":"2021-11-03 20:09:11.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/53/816553/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/53/816553/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/816553/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/816553/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/53/816553/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/53/816553/2"}}},"commit":{"parents":[{"commit":"8cd09500bea38bfd9478814be95fddf3df56fbff","subject":"Merge \"Adding Bandwidth Limit QoS rule client to Manager\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8cd09500bea38bfd9478814be95fddf3df56fbff"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-03 20:08:44.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2021-11-03 20:08:44.000000000","tz":-300},"subject":"Test all supported python version for stable branch","message":"Test all supported python version for stable branch\n\nAs Tempest is branchless and needs to support older\nsupported stable branches, let\u0027s test it with all the\npython version supported in those supported stable branches.\nFor that, we need to explicitly dd the py jobs instead of using\nthe template. This way, we will explicitly know that we are testing\nall supported python versions when the new release template bumps the\nmin python version in the template.\n\nIn Yoga testing, we are making py3.9 job to run as voting, so\nadding py3.9 job as voting.\n\nChange-Id: I422da8da06509e529f4091e21cdc3d392d3cf910\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1aac05c7b2f1fe4699a7f08183c93d6f92ce1efe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1aac05c7b2f1fe4699a7f08183c93d6f92ce1efe"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
