)]}'
{"id":"openstack%2Ftempest~817179","triplet_id":"openstack%2Ftempest~master~I802c0cdb3c8875e33819aed335877f1caef39fef","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2021-11-09 19:28:04.000000000","reason":"Change was submitted"},"10135":{"account":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"last_update":"2021-11-09 19:28:04.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2021-11-09 19:28:04.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2021-11-09 16:24:06.000000000","reason":"Ghanshyam replied on the change"}},"hashtags":[],"change_id":"I802c0cdb3c8875e33819aed335877f1caef39fef","subject":"zuul: Limit tempest-integrated-compute-centos-8-stream to \u003e\u003d stable/wallaby","status":"MERGED","created":"2021-11-09 10:24:28.000000000","updated":"2021-11-09 19:33:04.000000000","submitted":"2021-11-09 19:28:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"817179","meta_rev_id":"33eb932ca4225e9452b95da41938bf7a3535872f","_number":817179,"virtual_id_number":817179,"owner":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-09 19:28:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"all":[{"value":1,"date":"2021-11-09 14:33:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":2,"date":"2021-11-09 16:24:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":1,"date":"2021-11-09 16:24:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"reviewers":{"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-09 10:25:05.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-11-09 12:52:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-09 14:33:34.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"state":"REVIEWER"},{"updated":"2021-11-09 14:35:06.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2021-11-09 14:35:06.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-11-09 14:57:57.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"972c34927c266c8bb113f8ea049f29d6249ace6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"date":"2021-11-09 10:24:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61906300098a132d082ce1de42b1917644c6b4b6","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2021-11-09 10:25:05.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/tempest on branch master.  The error was:\n\n  The nodeset \"devstack-single-node-centos-8-stream\" was not found.\n\nThe error appears in the following job stanza:\n\n  job:\n      name: tempest-integrated-compute-centos-8-stream\n      parent: tempest-integrated-compute\n      nodeset: devstack-single-node-centos-8-stream\n      branches: ^(?!stable/(ocata|pike|queens|rocky|stein|train|ussuri|victoria)).*$\n      description: |\n        This job runs integration tests for compute. This is\n        subset of \u0027tempest-full-py3\u0027 job and run Nova, Neutron, Cinder (except backup tests)\n        and Glance related tests. This is meant to be run on Nova gate only.\n        This version of the job also uses CentOS 8 stream.\n      vars:\n        # Required until bug/1949606 is resolved when using libvirt and QEMU\n        # \u003e\u003d5.0.0 with a [libvirt]virt_type of qemu (TCG).\n        configure_swap_size: 4096\n\n  in \"openstack/tempest/zuul.d/integrated-gate.yaml@master\", line 131, column 3","accounts_in_message":[],"_revision_number":1},{"id":"72be5319f28bf4563198c251a59f756e304aa6a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 12:52:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/de7d81b3cf574a74af5601445a56cb2e : SUCCESS in 44m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d491b6be8e6f4d028fa9fc7b4382835f : SUCCESS in 1h 27m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e618e1bb41c4463d9bf5c7f97df1424a : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7681848034b0409fab24288d7267864c : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a7f90ca9d2274188a10d072e4570ba50 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68cea75b0b254cd29cfb9c81d960ee15 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ebd8684614da4d85b17e37a923966a33 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bc8c7e7c12641cbbe4ce4e95f2b313f : SUCCESS in 13m 07s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/317215a82ab04ce0a76b45034df3d312 : SUCCESS in 1h 34m 31s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/1ad33f9fea75449aabca48bf19951fa4 : SUCCESS in 1h 05m 38s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/51b3864577264d5887435266c7a77e08 : SUCCESS in 1h 25m 52s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/cc4306218a98479a86bf10b021b08e49 : SUCCESS in 1h 24m 51s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/a7119c018bdc4772a3af5513a402e1c2 : SUCCESS in 1h 17m 27s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/3391804f02ba4a119e0da604711911fc : SUCCESS in 1h 34m 57s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/1d580b890f244ac483170f6d3efdfa32 : SUCCESS in 1h 15m 16s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/441182dcf3384dada2b9dbaced3a0781 : SUCCESS in 1h 40m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/674fce79877b4630aae5ca91284d4ed5 : SUCCESS in 1h 37m 21s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/1f4b65cd63b84a96aafa9bde8d59375d : SUCCESS in 33m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/67e478cc5d9b487593bc9b4c4bf53eee : SUCCESS in 48m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a30360e1a648486d8100e235a4cfb739 : SUCCESS in 2h 17m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fc310f0873d645eb929751bcff7fc2c0 : SUCCESS in 54m 44s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/84f516bc797f4329a74e7141164e58f8 : SUCCESS in 1h 30m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9950ddd73d044d1ba65270d45216472b : SUCCESS in 57m 03s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/0dd29d72c36441a3bcdc81ba708a2b90 : SUCCESS in 58m 54s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e9a724f937d84b06b80f297043094255 : SUCCESS in 4m 53s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9e8b3bc059824d738962f76b1563478a : FAILURE in 1h 13m 34s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/b2a8d4e5922f4939bdfaeb534da33285 : FAILURE in 1h 03m 00s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e47c7607b314421380c399902280cdc9 : SUCCESS in 4m 32s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/710eed64d51d4368940bdfb5461f246a : SUCCESS in 1h 33m 02s","accounts_in_message":[],"_revision_number":1},{"id":"83228926c3daa23ccd88181a62a8a73cb6cb85aa","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2021-11-09 14:33:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b4e98b6e397ad1c972712f04ddd190ee77e0888","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2021-11-09 14:57:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5ce31cf874b910f784dd824cd6460f6cbcb05a17","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-11-09 16:24:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae0dc8e9e059fc3759c1444b0a1c61808eb07134","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 16:25:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4ee0fc2ccc2d0460a0d49141f28535144a360f8f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 19:28:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/a65f9d63f63f4bf0a140ff4dc8b9766c : SUCCESS in 1h 00m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3142a215b98447ec8dda3ba79224011c : SUCCESS in 1h 34m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bbdc6cd33b314ff883f4231b037a4462 : SUCCESS in 5m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/74b81cac91c74b0495bb8ddf163cb49a : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63ec533fa73c4977a011941904a6c3e0 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af526bb1fbd746d2b9e9438e7321705f : SUCCESS in 12m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3eaf3b57254d4e208e7ed0c80c6ae5d9 : SUCCESS in 2h 16m 03s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/92ce7230555047c0804372df8bf0d2f0 : SUCCESS in 58m 25s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/2a5b0842a2a84ab9b92b74ed3ac2f44f : SUCCESS in 1h 27m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/34591422c35242bdb2caac626cc5453f : SUCCESS in 48m 42s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bbae0d544490402b8dbde98af539adbc : SUCCESS in 1h 18m 01s","accounts_in_message":[],"_revision_number":1},{"id":"d8f2bb9e973f3f76aad59749e5a9618e4b1ea272","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 19:28:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"33eb932ca4225e9452b95da41938bf7a3535872f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-09 19:33:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74b1d3a17efb4208ae7b27f996b7bfd0 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ed89c77222917235290c8cc51974835528ed4cfa","revisions":{"ed89c77222917235290c8cc51974835528ed4cfa":{"kind":"REWORK","_number":1,"created":"2021-11-09 10:24:28.000000000","uploader":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"ref":"refs/changes/79/817179/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/79/817179/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/817179/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/817179/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/817179/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/79/817179/1"}}},"commit":{"parents":[{"commit":"d27a543fc936ab223279d240b16936c2c1d5787a","subject":"Merge \"Reactivate image before deletion\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d27a543fc936ab223279d240b16936c2c1d5787a"}]}],"author":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-11-09 10:12:38.000000000","tz":0},"committer":{"name":"Lee Yarwood","email":"lyarwood@redhat.com","date":"2021-11-09 10:23:24.000000000","tz":0},"subject":"zuul: Limit tempest-integrated-compute-centos-8-stream to \u003e\u003d stable/wallaby","message":"zuul: Limit tempest-integrated-compute-centos-8-stream to \u003e\u003d stable/wallaby\n\nFull support for CentOS8Stream in devstack didn\u0027t land until\nI39ccefbd06f46adf5077f8d8001f37d3b190f040 fixed is_fedora to include the\nnewly introduced name.\n\nAs such the recently introduced\ntempest-integrated-compute-centos-8-stream job within tempest that forms\npart of the integrated-gate-compute template needs to be restricted to\nbranches \u003e\u003d stable/wallaby.\n\nCloses-Bug: 1950296\nChange-Id: I802c0cdb3c8875e33819aed335877f1caef39fef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ed89c77222917235290c8cc51974835528ed4cfa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ed89c77222917235290c8cc51974835528ed4cfa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
