)]}'
{"id":"openstack%2Ftempest~81834","triplet_id":"openstack%2Ftempest~master~I499a11a128b1caa0f92f2392e2f7a511583bf3c9","project":"openstack/tempest","branch":"master","topic":"bp/ssh-auth-strategy","hashtags":[],"change_id":"I499a11a128b1caa0f92f2392e2f7a511583bf3c9","subject":"Enable instance validation with the server_actions v2","status":"ABANDONED","created":"2014-03-20 15:32:42.000000000","updated":"2014-12-05 15:04:55.000000000","total_comment_count":39,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"dc09aa34056fef39d963435a774cbf1fe404f24f","_number":81834,"virtual_id_number":81834,"owner":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":-1,"date":"2014-09-22 15:46:38.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"date":"2014-05-23 20:44:24.000000000","_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},{"value":1,"date":"2014-05-27 13:34:54.000000000","_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5174,"name":"Mauro Sergio Martins Rodrigues","email":"maurosr@linux.vnet.ibm.com","username":"maurosr"},{"_account_id":8824,"name":"Sergey Murashov","email":"smurashov@mirantis.com","username":"SergeyMurashov"},{"_account_id":7227,"name":"Timur Nurlygayanov","email":"tnurlygayanov@mirantis.com","username":"tnurlygayanov"},{"date":"2014-12-05 15:04:55.000000000","_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"all":[{"value":1,"date":"2014-05-27 12:52:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":-1,"date":"2014-05-27 15:41:19.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5174,"name":"Mauro Sergio Martins Rodrigues","email":"maurosr@linux.vnet.ibm.com","username":"maurosr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8824,"name":"Sergey Murashov","email":"smurashov@mirantis.com","username":"SergeyMurashov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7227,"name":"Timur Nurlygayanov","email":"tnurlygayanov@mirantis.com","username":"tnurlygayanov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"date":"2014-05-27 14:24:43.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"_account_id":5174,"name":"Mauro Sergio Martins Rodrigues","email":"maurosr@linux.vnet.ibm.com","username":"maurosr"},{"_account_id":8824,"name":"Sergey Murashov","email":"smurashov@mirantis.com","username":"SergeyMurashov"},{"_account_id":7227,"name":"Timur Nurlygayanov","email":"tnurlygayanov@mirantis.com","username":"tnurlygayanov"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5174,"name":"Mauro Sergio Martins Rodrigues","email":"maurosr@linux.vnet.ibm.com","username":"maurosr"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8824,"name":"Sergey Murashov","email":"smurashov@mirantis.com","username":"SergeyMurashov"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":7227,"name":"Timur Nurlygayanov","email":"tnurlygayanov@mirantis.com","username":"tnurlygayanov"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},{"_account_id":5174,"name":"Mauro Sergio Martins Rodrigues","email":"maurosr@linux.vnet.ibm.com","username":"maurosr"},{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},{"_account_id":7227,"name":"Timur Nurlygayanov","email":"tnurlygayanov@mirantis.com","username":"tnurlygayanov"},{"_account_id":8824,"name":"Sergey Murashov","email":"smurashov@mirantis.com","username":"SergeyMurashov"},{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-04-07 20:15:34.000000000","updated_by":{"_account_id":5174,"name":"Mauro Sergio Martins Rodrigues","email":"maurosr@linux.vnet.ibm.com","username":"maurosr"},"reviewer":{"_account_id":5174,"name":"Mauro Sergio Martins Rodrigues","email":"maurosr@linux.vnet.ibm.com","username":"maurosr"},"state":"REVIEWER"},{"updated":"2014-04-21 06:54:17.000000000","updated_by":{"_account_id":8824,"name":"Sergey Murashov","email":"smurashov@mirantis.com","username":"SergeyMurashov"},"reviewer":{"_account_id":8824,"name":"Sergey Murashov","email":"smurashov@mirantis.com","username":"SergeyMurashov"},"state":"REVIEWER"},{"updated":"2014-04-23 17:20:35.000000000","updated_by":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"reviewer":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"state":"REVIEWER"},{"updated":"2014-04-24 08:04:06.000000000","updated_by":{"_account_id":7227,"name":"Timur Nurlygayanov","email":"tnurlygayanov@mirantis.com","username":"tnurlygayanov"},"reviewer":{"_account_id":7227,"name":"Timur Nurlygayanov","email":"tnurlygayanov@mirantis.com","username":"tnurlygayanov"},"state":"REVIEWER"},{"updated":"2014-05-21 17:15:46.000000000","updated_by":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"reviewer":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"state":"REVIEWER"},{"updated":"2014-05-21 22:09:35.000000000","updated_by":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"reviewer":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"state":"REVIEWER"},{"updated":"2014-05-23 20:44:24.000000000","updated_by":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-05-27 12:52:36.000000000","updated_by":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"reviewer":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"state":"REVIEWER"},{"updated":"2014-05-27 13:34:54.000000000","updated_by":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2014-05-27 15:41:19.000000000","updated_by":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"reviewer":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"state":"REVIEWER"},{"updated":"2014-09-22 15:46:38.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2014-12-05 15:04:55.000000000","updated_by":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"reviewer":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"state":"REVIEWER"}],"messages":[{"id":"3dabfd6297d59b3bdb6b3e1cfccf022e28b1020b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-20 15:32:43.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"df713ce9080b75460b6e628f5c0563d0867a3025","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-03-20 16:05:29.000000000","message":"Patch Set 1:\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://70dcd673047f0ebed6a7-dda4e480e005aaa13ec303551d2d8155.r49.cf5.rackcdn.com/34/81834/1/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":1},{"id":"370726475a4bbc20ebb1b897282b3759719d2e68","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-20 16:25:19.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/34/81834/1/check/check-grenade-dsvm/b4d5828 : FAILURE in 28m 47s\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/1/check/check-tempest-dsvm-full/9edf1ae : FAILURE in 31m 58s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/1/check/check-tempest-dsvm-neutron/b3e3385 : FAILURE in 43m 04s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/1/check/check-tempest-dsvm-neutron-heat-slow/b97153f : SUCCESS in 24m 34s\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/1/check/gate-tempest-pep8/9190ef1 : SUCCESS in 4m 48s\n- gate-tempest-docs http://logs.openstack.org/34/81834/1/check/gate-tempest-docs/8452dd5 : SUCCESS in 4m 14s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/1/check/gate-tempest-python27/2ccacd7 : SUCCESS in 4m 23s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/1/check/check-tempest-dsvm-postgres-full/2faf291 : FAILURE in 33m 15s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/1/check/check-tempest-dsvm-neutron-full/8b55dcc : FAILURE in 46m 52s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/34/81834/1/check/check-tempest-dsvm-cells-full/24a26a6 : FAILURE in 35m 59s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/1/check/gate-tempest-dsvm-large-ops/1a8bbc1 : SUCCESS in 15m 00s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/1/check/gate-tempest-dsvm-neutron-large-ops/733edee : SUCCESS in 15m 28s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/1/check/check-grenade-dsvm-partial-ncpu/4ffca39 : FAILURE in 28m 34s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/1/check/check-tempest-dsvm-ironic-nv/12b7da9 : FAILURE in 30m 24s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/1/check/check-tempest-dsvm-ironic-postgres-nv/81f7326 : FAILURE in 33m 55s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/34/81834/1/check/check-tempest-dsvm-sahara-full-nv/ebe3b75 : FAILURE in 24m 25s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/34/81834/1/check/check-tempest-dsvm-sahara-postgres-nv/b4b4e1a : FAILURE in 22m 04s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/34/81834/1/check/check-tempest-dsvm-sahara-neutron-nv/31f136d : FAILURE in 35m 41s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"5b6ad0a90443ff1f414ee607dbb763c0ad0caceb","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-03-20 16:27:33.000000000","message":"Patch Set 1:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-ironic-postgres-nv: https://bugs.launchpad.net/bugs/1253896\n- check-tempest-dsvm-postgres-full: https://bugs.launchpad.net/bugs/1253896\n- check-tempest-dsvm-cells-full: https://bugs.launchpad.net/bugs/1253896\n- check-tempest-dsvm-sahara-neutron-nv: https://bugs.launchpad.net/bugs/1283522\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1253896\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":1},{"id":"07fece8949e8e013a0b40325bce001a80a699fdf","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-03-20 16:56:15.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2aa84d19afa422f4511f1b6449aff73b1da8422a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-20 16:57:57.000000000","message":"Patch Set 2:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"81c4165b911502c8943b41b005e3728d03705653","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-03-20 18:07:23.000000000","message":"Patch Set 2:\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://70dcd673047f0ebed6a7-dda4e480e005aaa13ec303551d2d8155.r49.cf5.rackcdn.com/34/81834/2/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":2},{"id":"a4d5476611d20f33853e90b9448354826503add2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-20 18:57:35.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/34/81834/2/check/check-grenade-dsvm/0db9aa7 : FAILURE in 38m 27s\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-full/09151ec : FAILURE in 57m 16s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-neutron/eac38ed : SUCCESS in 41m 48s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-neutron-heat-slow/0ec6020 : SUCCESS in 23m 18s\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/2/check/gate-tempest-pep8/7ec531e : SUCCESS in 3m 12s\n- gate-tempest-docs http://logs.openstack.org/34/81834/2/check/gate-tempest-docs/b60f06f : SUCCESS in 3m 38s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/2/check/gate-tempest-python27/4766d9a : SUCCESS in 3m 20s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-postgres-full/f6c1aa2 : SUCCESS in 56m 14s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-neutron-full/20569cf : SUCCESS in 59m 07s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-cells-full/b2e4439 : SUCCESS in 58m 07s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/2/check/gate-tempest-dsvm-large-ops/764d41b : SUCCESS in 14m 46s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/2/check/gate-tempest-dsvm-neutron-large-ops/77e12b2 : SUCCESS in 15m 04s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/2/check/check-grenade-dsvm-partial-ncpu/37153fe : SUCCESS in 37m 31s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-ironic-nv/8cb442b : SUCCESS in 37m 15s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-ironic-postgres-nv/ec7c591 : FAILURE in 38m 42s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-sahara-full-nv/f59cf5b : SUCCESS in 59m 40s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-sahara-postgres-nv/647a971 : FAILURE in 27m 44s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-sahara-neutron-nv/442f147 : SUCCESS in 1h 08m 45s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"487978b7a7a88abb7a78c6ae09e5616fb91da28e","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-03-20 18:58:19.000000000","message":"Patch Set 2:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-sahara-postgres-nv: https://bugs.launchpad.net/bugs/1253896\n- check-grenade-dsvm: https://bugs.launchpad.net/bugs/1253896\n- check-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1253896\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1253896\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":2},{"id":"36a2bd09713d95f78a7740639f23c12f879ca748","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-03-20 18:58:19.000000000","message":"Patch Set 2:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-sahara-postgres-nv: https://bugs.launchpad.net/bugs/1253896\n- check-grenade-dsvm: https://bugs.launchpad.net/bugs/1253896\n- check-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1253896\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1253896\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":2},{"id":"52fa080cc222aac45a0c2584975198e2af52ae92","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-03-20 21:38:46.000000000","message":"Patch Set 2:\n\nThe failures are with nova-net after the HARD reboot action.\nLook like the instance obtained the IP address (based on the logs from the dnsmasq), but the ssh connection is refused for an unknown reason.","accounts_in_message":[],"_revision_number":2},{"id":"321de9f93c60d8b8188c447557f681033d85fb63","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-03-20 21:53:07.000000000","message":"Patch Set 2:\n\nrecheck no bug","accounts_in_message":[],"_revision_number":2},{"id":"a13532e8080822ee5fbcfa211b6483769f4befb6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-20 21:53:08.000000000","message":"Patch Set 2: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":2},{"id":"1aa28e0891f2a01e3177d57ae88874125cd3235f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-20 23:06:18.000000000","message":"Patch Set 2: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/34/81834/2/check/check-grenade-dsvm/c1c4514 : SUCCESS in 30m 25s\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-full/60397b9 : SUCCESS in 44m 11s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-neutron/969bd09 : SUCCESS in 40m 00s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-neutron-heat-slow/6b78816 : SUCCESS in 21m 34s\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/2/check/gate-tempest-pep8/81decf4 : SUCCESS in 3m 51s\n- gate-tempest-docs http://logs.openstack.org/34/81834/2/check/gate-tempest-docs/e626216 : SUCCESS in 3m 44s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/2/check/gate-tempest-python27/71f1572 : SUCCESS in 3m 42s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-postgres-full/fe99450 : SUCCESS in 49m 29s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-neutron-full/0159c2e : SUCCESS in 43m 15s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-cells-full/15fcd64 : SUCCESS in 49m 16s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/2/check/gate-tempest-dsvm-large-ops/08aad4b : SUCCESS in 15m 31s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/2/check/gate-tempest-dsvm-neutron-large-ops/07aff6b : SUCCESS in 20m 08s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/2/check/check-grenade-dsvm-partial-ncpu/f67d2e8 : FAILURE in 33m 03s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-ironic-nv/c9b0355 : SUCCESS in 30m 58s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-ironic-postgres-nv/5cc5d1b : SUCCESS in 30m 32s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-sahara-full-nv/1a0b665 : SUCCESS in 48m 17s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-sahara-postgres-nv/bb4ed05 : SUCCESS in 25m 02s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/34/81834/2/check/check-tempest-dsvm-sahara-neutron-nv/b15876d : SUCCESS in 41m 45s (non-voting)\n","accounts_in_message":[],"_revision_number":2},{"id":"3887e58e6fb6e70dd5adcd878f0684af4d2e70aa","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-03-20 23:06:36.000000000","message":"Patch Set 2:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-grenade-dsvm-partial-ncpu: https://bugs.launchpad.net/bugs/1253896\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1253896\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":2},{"id":"2198394b7cef294482d291bebdb23d09b0d9dbf2","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-03-24 16:03:41.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"9457f8be95d7f03e2f701c231f735b0530c2b81f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-24 16:03:47.000000000","message":"Patch Set 3:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"c53c3c7ca0eb1001a926452e9ff570d9c4aa8765","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-03-24 17:04:20.000000000","message":"Patch Set 3:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://70dcd673047f0ebed6a7-dda4e480e005aaa13ec303551d2d8155.r49.cf5.rackcdn.com/34/81834/3/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":3},{"id":"7cc481f6cb63ed20660c11c0e45b543ba5ca4fd6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-24 17:26:28.000000000","message":"Patch Set 3: Works for me\n\nBuild succeeded.\n\n- check-grenade-dsvm http://logs.openstack.org/34/81834/3/check/check-grenade-dsvm/3a35c74 : SUCCESS in 35m 49s\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/3/check/check-tempest-dsvm-full/eeedbce : SUCCESS in 45m 56s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/3/check/check-tempest-dsvm-neutron/1e48b6f : SUCCESS in 51m 32s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/3/check/check-tempest-dsvm-neutron-heat-slow/87230f7 : SUCCESS in 23m 42s\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/3/check/gate-tempest-pep8/268b133 : SUCCESS in 3m 54s\n- gate-tempest-docs http://logs.openstack.org/34/81834/3/check/gate-tempest-docs/58d4ec3 : SUCCESS in 3m 26s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/3/check/gate-tempest-python27/6c0908c : SUCCESS in 3m 47s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/3/check/check-tempest-dsvm-postgres-full/a5ea234 : SUCCESS in 1h 12m 49s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/3/check/check-tempest-dsvm-neutron-full/3195346 : SUCCESS in 1h 14m 14s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/34/81834/3/check/check-tempest-dsvm-cells-full/76f86e7 : SUCCESS in 1h 04m 34s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/3/check/gate-tempest-dsvm-large-ops/c0864c6 : SUCCESS in 16m 01s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/3/check/gate-tempest-dsvm-neutron-large-ops/4c21218 : SUCCESS in 17m 11s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/3/check/check-grenade-dsvm-partial-ncpu/db9c384 : SUCCESS in 31m 35s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/3/check/check-tempest-dsvm-ironic-nv/cd09b98 : SUCCESS in 38m 32s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/3/check/check-tempest-dsvm-ironic-postgres-nv/fd1a250 : SUCCESS in 37m 01s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/34/81834/3/check/check-tempest-dsvm-sahara-full-nv/b1954e5 : SUCCESS in 1h 04m 32s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/34/81834/3/check/check-tempest-dsvm-sahara-postgres-nv/87a45bf : SUCCESS in 33m 45s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/34/81834/3/check/check-tempest-dsvm-sahara-neutron-nv/b53c2b8 : SUCCESS in 47m 23s (non-voting)\n\n","accounts_in_message":[],"_revision_number":3},{"id":"2a4f83723ff88ea9f004c4100e17f28d92776407","author":{"_account_id":7227,"name":"Timur Nurlygayanov","email":"tnurlygayanov@mirantis.com","username":"tnurlygayanov"},"date":"2014-03-24 19:17:07.000000000","message":"Patch Set 3: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":3},{"id":"de76625ed23ae5f0590e35b3d830bb01bacfea07","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-04-02 09:01:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0a704e3ca2e783464e5b35ac1ace3542413ad6c2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-02 09:01:44.000000000","message":"Patch Set 4:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":4},{"id":"2619b1a07fa5a3a54dcfd887995b268ad3df3a1b","author":{"_account_id":5494,"name":"Trivial Rebase","email":"trivial-rebase@review.openstack.org","username":"trivial-rebase"},"date":"2014-04-02 09:02:01.000000000","message":"Patch Set 4:\n\nNew patchset patch-id matches previous patchset, but whitespace content has changed.","accounts_in_message":[],"_revision_number":4},{"id":"a8c0a8271e08faf52b1520a2e560f4713a1a17dd","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-02 10:13:55.000000000","message":"Patch Set 4: Works for me\n\nBuild succeeded.\n\n- check-grenade-dsvm http://logs.openstack.org/34/81834/4/check/check-grenade-dsvm/c6e782d : SUCCESS in 36m 06s\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/4/check/check-tempest-dsvm-full/7a414d2 : SUCCESS in 1h 04m 10s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/4/check/check-tempest-dsvm-neutron/1b09c85 : SUCCESS in 57m 10s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/4/check/check-tempest-dsvm-neutron-heat-slow/e916ec0 : SUCCESS in 29m 26s\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/4/check/gate-tempest-pep8/1f5c983 : SUCCESS in 3m 08s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/4/check/gate-tempest-docs/1debc7c/doc/build/html/ : SUCCESS in 2m 47s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/4/check/gate-tempest-python27/2450479 : SUCCESS in 3m 21s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/4/check/check-tempest-dsvm-postgres-full/e235e8d : SUCCESS in 56m 25s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/4/check/check-tempest-dsvm-neutron-full/48c228c : SUCCESS in 1h 09m 05s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/34/81834/4/check/check-tempest-dsvm-cells-full/a91f70c : FAILURE in 28m 26s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/4/check/gate-tempest-dsvm-large-ops/f954963 : SUCCESS in 15m 19s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/4/check/gate-tempest-dsvm-neutron-large-ops/cc7836e : SUCCESS in 15m 26s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/4/check/check-grenade-dsvm-partial-ncpu/4899d93 : SUCCESS in 39m 31s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/4/check/check-tempest-dsvm-ironic-nv/6b4431e : SUCCESS in 38m 25s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/4/check/check-tempest-dsvm-ironic-postgres-nv/d7f52ea : SUCCESS in 37m 32s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/34/81834/4/check/check-tempest-dsvm-sahara-full-nv/b55138f : SUCCESS in 57m 53s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/34/81834/4/check/check-tempest-dsvm-sahara-postgres-nv/033e145 : SUCCESS in 24m 48s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/34/81834/4/check/check-tempest-dsvm-sahara-neutron-nv/61f6047 : SUCCESS in 55m 05s (non-voting)\n\n","accounts_in_message":[],"_revision_number":4},{"id":"bdd5753ea52ee2e2258960d69ab2e9043f88abc2","author":{"_account_id":8824,"name":"Sergey Murashov","email":"smurashov@mirantis.com","username":"SergeyMurashov"},"date":"2014-04-02 12:14:15.000000000","message":"Patch Set 4: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\n","accounts_in_message":[],"_revision_number":4},{"id":"8231292bfb2c785280c514956d0059586e688bbd","author":{"_account_id":7227,"name":"Timur Nurlygayanov","email":"tnurlygayanov@mirantis.com","username":"tnurlygayanov"},"date":"2014-04-02 12:19:47.000000000","message":"Patch Set 4: (1 inline comment)\n\n","accounts_in_message":[],"_revision_number":4},{"id":"0551d9fe2afa547cb237b6ecc08f88cea79da62d","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-04-02 17:31:03.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d002258d898487864f5ea144201d85b1cec96662","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-02 17:31:04.000000000","message":"Patch Set 5:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":5},{"id":"c11010996d2176676f403bdce5ca856aa3732d97","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-02 18:56:08.000000000","message":"Patch Set 5: Works for me\n\nBuild succeeded.\n\n- check-grenade-dsvm http://logs.openstack.org/34/81834/5/check/check-grenade-dsvm/972bead : SUCCESS in 39m 35s\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/5/check/check-tempest-dsvm-full/086c0d1 : SUCCESS in 50m 15s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/5/check/check-tempest-dsvm-neutron/fb6e7a5 : SUCCESS in 53m 56s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/5/check/check-tempest-dsvm-neutron-heat-slow/35e8f7e : SUCCESS in 22m 19s\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/5/check/gate-tempest-pep8/bcec65a : SUCCESS in 4m 01s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/5/check/gate-tempest-docs/2be362a/doc/build/html/ : SUCCESS in 3m 36s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/5/check/gate-tempest-python27/058ba39 : SUCCESS in 3m 58s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/5/check/check-tempest-dsvm-postgres-full/13a89e9 : SUCCESS in 58m 04s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/5/check/check-tempest-dsvm-neutron-full/c6b93e0 : SUCCESS in 1h 10m 19s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/34/81834/5/check/check-tempest-dsvm-cells-full/b91cf10 : FAILURE in 35m 33s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/5/check/gate-tempest-dsvm-large-ops/9f8e824 : SUCCESS in 16m 28s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/5/check/gate-tempest-dsvm-neutron-large-ops/98c3f9b : SUCCESS in 16m 42s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/5/check/check-grenade-dsvm-partial-ncpu/7ec3c12 : SUCCESS in 39m 32s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/5/check/check-tempest-dsvm-ironic-nv/8766315 : SUCCESS in 32m 20s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/5/check/check-tempest-dsvm-ironic-postgres-nv/c67e5ee : SUCCESS in 30m 54s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/34/81834/5/check/check-tempest-dsvm-sahara-full-nv/f2a4967 : SUCCESS in 41m 10s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/34/81834/5/check/check-tempest-dsvm-sahara-postgres-nv/b4ce998 : SUCCESS in 20m 51s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/34/81834/5/check/check-tempest-dsvm-sahara-neutron-nv/f433dea : SUCCESS in 47m 30s (non-voting)\n- check-tempest-dsvm-full-havana http://logs.openstack.org/34/81834/5/check/check-tempest-dsvm-full-havana/c2279a2 : SUCCESS in 45m 17s (non-voting)\n- check-tempest-dsvm-neutron-havana http://logs.openstack.org/34/81834/5/check/check-tempest-dsvm-neutron-havana/b9828fa : SUCCESS in 32m 10s (non-voting)\n\n","accounts_in_message":[],"_revision_number":5},{"id":"ba37804d7d5b737da5b36c51570fa7773756a280","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-04-02 21:44:03.000000000","message":"Patch Set 5:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/34/81834/5/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":5},{"id":"76648f2ce689163db5cd85974e1a59751e7a28ff","author":{"_account_id":8824,"name":"Sergey Murashov","email":"smurashov@mirantis.com","username":"SergeyMurashov"},"date":"2014-04-03 06:53:53.000000000","message":"Patch Set 5: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":5},{"id":"8ca613b434cfab32dddd0a66765e0af9c0539ab1","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-04-07 08:07:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ef2bac8bb5044a685aaeeaff75de3ef982459e6f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-07 08:07:51.000000000","message":"Patch Set 6:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":6},{"id":"3671f2d353041179baba14b9fb316dfc103a04c8","author":{"_account_id":8824,"name":"Sergey Murashov","email":"smurashov@mirantis.com","username":"SergeyMurashov"},"date":"2014-04-07 08:08:07.000000000","message":"Patch Set 6: Looks good to me, but someone else must approve\n\nAutomatically re-added by Gerrit trivial rebase detection script.","accounts_in_message":[],"_revision_number":6},{"id":"ae07469ce5d19e0df9ce6509a437176db8d7a7e7","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-07 09:17:05.000000000","message":"Patch Set 6: Works for me\n\nBuild succeeded.\n\n- check-grenade-dsvm http://logs.openstack.org/34/81834/6/check/check-grenade-dsvm/7ae30fb : SUCCESS in 41m 10s\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/6/check/check-tempest-dsvm-full/6be10ee : SUCCESS in 1h 08m 51s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/6/check/check-tempest-dsvm-neutron/f4d9494 : SUCCESS in 50m 33s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/6/check/check-tempest-dsvm-neutron-heat-slow/85cae42 : SUCCESS in 24m 23s\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/6/check/gate-tempest-pep8/a521859 : SUCCESS in 3m 41s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/6/check/gate-tempest-docs/7827437/doc/build/html/ : SUCCESS in 3m 46s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/6/check/gate-tempest-python27/cce0295 : SUCCESS in 3m 29s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/6/check/check-tempest-dsvm-postgres-full/66f8760 : SUCCESS in 1h 09m 07s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/6/check/check-tempest-dsvm-neutron-full/6e2ec8e : SUCCESS in 1h 01m 18s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/34/81834/6/check/check-tempest-dsvm-cells-full/fc3ff21 : FAILURE in 36m 02s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/6/check/gate-tempest-dsvm-large-ops/e70aa55 : SUCCESS in 17m 14s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/6/check/gate-tempest-dsvm-neutron-large-ops/28e6042 : SUCCESS in 14m 34s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/6/check/check-grenade-dsvm-partial-ncpu/6899d55 : SUCCESS in 48m 27s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/6/check/check-tempest-dsvm-ironic-nv/5f018ab : SUCCESS in 32m 00s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/6/check/check-tempest-dsvm-ironic-postgres-nv/3829c29 : SUCCESS in 31m 44s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/34/81834/6/check/check-tempest-dsvm-sahara-full-nv/29c76ec : SUCCESS in 52m 13s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/34/81834/6/check/check-tempest-dsvm-sahara-postgres-nv/b6d93d2 : SUCCESS in 28m 47s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/34/81834/6/check/check-tempest-dsvm-sahara-neutron-nv/0c66ffb : SUCCESS in 50m 42s (non-voting)\n- check-tempest-master-dsvm-full-havana http://logs.openstack.org/34/81834/6/check/check-tempest-master-dsvm-full-havana/6cd23a1 : SUCCESS in 27m 55s (non-voting)\n- check-tempest-master-dsvm-neutron-havana http://logs.openstack.org/34/81834/6/check/check-tempest-master-dsvm-neutron-havana/563c2ff : SUCCESS in 32m 56s (non-voting)\n\n","accounts_in_message":[],"_revision_number":6},{"id":"79ebc7a33b0fb8c33ca116b1d08da9bb1e77b9ae","author":{"_account_id":7227,"name":"Timur Nurlygayanov","email":"tnurlygayanov@mirantis.com","username":"tnurlygayanov"},"date":"2014-04-07 09:22:00.000000000","message":"Patch Set 6: I would prefer that you didn\u0027t merge this\n\n(1 inline comment)\n\nLooks good, but need to fix one small issue.","accounts_in_message":[],"_revision_number":6},{"id":"23151c0051db5b20c8a6a0af3121d231380faa08","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-04-07 09:40:57.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"512e18835f84fcb9fb2e2c20f20255fd471af892","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-07 09:40:58.000000000","message":"Patch Set 7:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":7},{"id":"f108342adde67f2fd9a4070451c090fd7ee3473f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-07 10:54:44.000000000","message":"Patch Set 7: Works for me\n\nBuild succeeded.\n\n- check-grenade-dsvm http://logs.openstack.org/34/81834/7/check/check-grenade-dsvm/5825c9c : SUCCESS in 42m 22s\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/7/check/check-tempest-dsvm-full/adf26f9 : SUCCESS in 55m 41s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/7/check/check-tempest-dsvm-neutron/b342716 : SUCCESS in 49m 08s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/7/check/check-tempest-dsvm-neutron-heat-slow/79b3b91 : SUCCESS in 19m 48s\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/7/check/gate-tempest-pep8/c529715 : SUCCESS in 3m 33s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/7/check/gate-tempest-docs/d97919e/doc/build/html/ : SUCCESS in 3m 16s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/7/check/gate-tempest-python27/0a5d42b : SUCCESS in 3m 32s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/7/check/check-tempest-dsvm-postgres-full/be6a00c : SUCCESS in 51m 41s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/7/check/check-tempest-dsvm-neutron-full/739fb13 : SUCCESS in 1h 13m 39s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/34/81834/7/check/check-tempest-dsvm-cells-full/20f38ac : FAILURE in 31m 16s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/7/check/gate-tempest-dsvm-large-ops/9482f50 : SUCCESS in 15m 29s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/7/check/gate-tempest-dsvm-neutron-large-ops/3a4f26a : SUCCESS in 16m 18s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/7/check/check-grenade-dsvm-partial-ncpu/3973e31 : SUCCESS in 43m 28s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/7/check/check-tempest-dsvm-ironic-nv/746c4f3 : SUCCESS in 36m 29s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/7/check/check-tempest-dsvm-ironic-postgres-nv/dd60bb6 : SUCCESS in 33m 24s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/34/81834/7/check/check-tempest-dsvm-sahara-full-nv/0c0bbf2 : SUCCESS in 59m 11s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/34/81834/7/check/check-tempest-dsvm-sahara-postgres-nv/d44510c : SUCCESS in 27m 22s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/34/81834/7/check/check-tempest-dsvm-sahara-neutron-nv/3bff0bb : SUCCESS in 50m 22s (non-voting)\n- check-tempest-master-dsvm-full-havana http://logs.openstack.org/34/81834/7/check/check-tempest-master-dsvm-full-havana/15a1d3c : SUCCESS in 39m 09s (non-voting)\n- check-tempest-master-dsvm-neutron-havana http://logs.openstack.org/34/81834/7/check/check-tempest-master-dsvm-neutron-havana/74094f3 : SUCCESS in 40m 12s (non-voting)\n\n","accounts_in_message":[],"_revision_number":7},{"id":"bb2a45e717460150db4c3842d6848a9cea6f6424","author":{"_account_id":7227,"name":"Timur Nurlygayanov","email":"tnurlygayanov@mirantis.com","username":"tnurlygayanov"},"date":"2014-04-07 11:22:39.000000000","message":"Patch Set 7: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":7},{"id":"4f05619253c38dac06999ba700f23aadb17f8fca","author":{"_account_id":8824,"name":"Sergey Murashov","email":"smurashov@mirantis.com","username":"SergeyMurashov"},"date":"2014-04-07 11:45:56.000000000","message":"Patch Set 7: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":7},{"id":"2a21668a1f1e3fadea47d362831fda9295fc4a29","author":{"_account_id":5174,"name":"Mauro Sergio Martins Rodrigues","email":"maurosr@linux.vnet.ibm.com","username":"maurosr"},"date":"2014-04-07 20:15:34.000000000","message":"Patch Set 7: I would prefer that you didn\u0027t merge this\n\n(2 inline comments)\n\na typo and some suggestions","accounts_in_message":[],"_revision_number":7},{"id":"fe0793e8d1704a42709a496ad65404971bf98b8b","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-04-08 13:11:40.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"c02828c9f972bf04d3f2c5974eb6c2bb6804d779","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-08 13:11:43.000000000","message":"Patch Set 8:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":8},{"id":"b94d7a4fc4f5385c2619196080492d5703a7e399","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-08 14:26:30.000000000","message":"Patch Set 8: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/34/81834/8/check/check-grenade-dsvm/188c5e2 : SUCCESS in 39m 47s\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-full/85b2feb : SUCCESS in 43m 03s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-neutron/5544ba0 : SUCCESS in 1h 05m 32s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-neutron-heat-slow/0976a2d : FAILURE in 20m 20s\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/8/check/gate-tempest-pep8/4e8b7eb : SUCCESS in 3m 29s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/8/check/gate-tempest-docs/41673e6/doc/build/html/ : SUCCESS in 3m 02s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/8/check/gate-tempest-python27/3f9d204 : SUCCESS in 3m 41s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-postgres-full/f2cecd2 : SUCCESS in 59m 59s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-neutron-full/9c16ccb : FAILURE in 1h 14m 23s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-cells-full/c91202e : FAILURE in 34m 40s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/8/check/gate-tempest-dsvm-large-ops/c46ca7d : SUCCESS in 17m 06s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/8/check/gate-tempest-dsvm-neutron-large-ops/9cfbf2b : SUCCESS in 14m 53s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/8/check/check-grenade-dsvm-partial-ncpu/49ab6f5 : SUCCESS in 39m 52s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-ironic-nv/56677c1 : SUCCESS in 33m 34s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-ironic-postgres-nv/41a135f : SUCCESS in 33m 09s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-sahara-full-nv/4ecd11c : SUCCESS in 43m 04s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-sahara-postgres-nv/35212e2 : SUCCESS in 28m 28s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-sahara-neutron-nv/1bf986c : SUCCESS in 53m 49s (non-voting)\n- check-tempest-master-dsvm-full-havana http://logs.openstack.org/34/81834/8/check/check-tempest-master-dsvm-full-havana/a414d74 : SUCCESS in 28m 42s (non-voting)\n- check-tempest-master-dsvm-neutron-havana http://logs.openstack.org/34/81834/8/check/check-tempest-master-dsvm-neutron-havana/f554650 : SUCCESS in 29m 41s (non-voting)\n","accounts_in_message":[],"_revision_number":8},{"id":"6e83e582b03aed5c1a0b5a255d48bd65b19dcf82","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-04-08 19:43:35.000000000","message":"Patch Set 8:\n\nrecheck bug 1285833","accounts_in_message":[],"_revision_number":8},{"id":"c05ff24c1271cadf892a7ebdbea83333df6be049","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-08 19:43:36.000000000","message":"Patch Set 8: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":8},{"id":"b866f56fad56a79a1ac4dc20d2cd7230fa36a381","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-08 20:52:50.000000000","message":"Patch Set 8: Works for me\n\nBuild succeeded.\n\n- check-grenade-dsvm http://logs.openstack.org/34/81834/8/check/check-grenade-dsvm/bd79922 : SUCCESS in 45m 02s\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-full/1e989e1 : SUCCESS in 58m 54s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-neutron/7bab1e4 : SUCCESS in 56m 27s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-neutron-heat-slow/fce8624 : SUCCESS in 26m 10s\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/8/check/gate-tempest-pep8/1fb3a53 : SUCCESS in 3m 48s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/8/check/gate-tempest-docs/343f645/doc/build/html/ : SUCCESS in 2m 48s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/8/check/gate-tempest-python27/03dc25a : SUCCESS in 3m 28s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-postgres-full/15517b4 : SUCCESS in 1h 03m 00s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-neutron-full/cf7838a : SUCCESS in 59m 40s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-cells-full/5595b18 : FAILURE in 26m 21s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/8/check/gate-tempest-dsvm-large-ops/e98d326 : SUCCESS in 15m 16s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/8/check/gate-tempest-dsvm-neutron-large-ops/aca94f3 : SUCCESS in 16m 27s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/8/check/check-grenade-dsvm-partial-ncpu/aefd1db : SUCCESS in 45m 09s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-ironic-nv/34b12a5 : SUCCESS in 35m 46s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-ironic-postgres-nv/38fe3d6 : SUCCESS in 36m 47s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-sahara-full-nv/f17ada1 : SUCCESS in 56m 12s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-sahara-postgres-nv/8caa876 : SUCCESS in 27m 56s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/34/81834/8/check/check-tempest-dsvm-sahara-neutron-nv/8ee2ab5 : SUCCESS in 49m 09s (non-voting)\n- check-tempest-master-dsvm-full-havana http://logs.openstack.org/34/81834/8/check/check-tempest-master-dsvm-full-havana/88dd6cc : SUCCESS in 32m 48s (non-voting)\n- check-tempest-master-dsvm-neutron-havana http://logs.openstack.org/34/81834/8/check/check-tempest-master-dsvm-neutron-havana/24de2d2 : SUCCESS in 27m 57s (non-voting)\n","accounts_in_message":[],"_revision_number":8},{"id":"f46f45029f1b42bc3164d993ab248c83fbf9e723","author":{"_account_id":8824,"name":"Sergey Murashov","email":"smurashov@mirantis.com","username":"SergeyMurashov"},"date":"2014-04-09 13:03:13.000000000","message":"Patch Set 8: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":8},{"id":"c0b09634a4cd4556e16a6643a0a7e8df31e0d2a3","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-04-19 12:54:29.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4638c5abd5c98c5eb9aac6b5bcf12857dd0a8979","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-19 12:54:31.000000000","message":"Patch Set 9:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":9},{"id":"7d1717582d5a7cedc05d0d7b627d8fcef96bd255","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-04-19 13:53:39.000000000","message":"Patch Set 9: Works for me\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/34/81834/9/5306/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":9},{"id":"023220ffb7321353f3cf8e824ae977abd6d01507","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-19 14:53:04.000000000","message":"Patch Set 9: Works for me\n\nBuild succeeded.\n\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-full/bc9aece : SUCCESS in 52m 20s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron/796a4f6 : SUCCESS in 57m 15s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-heat-slow/0ab4986 : SUCCESS in 21m 16s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm/97020df : SUCCESS in 32m 05s\n- check-grenade-dsvm-neutron http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-neutron/f546e67 : FAILURE in 24m 10s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/9/check/gate-tempest-pep8/7d34eaf : SUCCESS in 4m 00s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/9/check/gate-tempest-docs/4a7c34e/doc/build/html/ : SUCCESS in 3m 52s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/9/check/gate-tempest-python27/7102dd9 : SUCCESS in 3m 11s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-postgres-full/0eba4b5 : SUCCESS in 59m 32s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-full/63ac4a6 : SUCCESS in 49m 11s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-cells-full/0f75e38 : FAILURE in 32m 09s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-large-ops/c26e1f3 : SUCCESS in 14m 14s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-neutron-large-ops/47125a4 : SUCCESS in 16m 21s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-partial-ncpu/2234bd6 : SUCCESS in 33m 47s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-ironic-nv/93f4c03 : SUCCESS in 50m 29s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-ironic-postgres-nv/49f0d1d : SUCCESS in 34m 55s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-full-icehouse/6bf1962 : SUCCESS in 49m 24s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-icehouse/d0c6dfc : SUCCESS in 55m 06s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-heat-slow-icehouse/f9e1bdd : SUCCESS in 21m 02s (non-voting)\n- check-grenade-dsvm-icehouse http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-icehouse/cd42c41 : SUCCESS in 41m 40s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-neutron-icehouse/c6e079f : FAILURE in 44m 39s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-postgres-full-icehouse/c06f415 : SUCCESS in 1h 01m 51s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-full-icehouse/7cf8f7e : SUCCESS in 1h 07m 12s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-large-ops-icehouse/18d2d16 : SUCCESS in 15m 40s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-neutron-large-ops-icehouse/8ea856b : SUCCESS in 17m 41s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-partial-ncpu-icehouse/ae645cd : SUCCESS in 36m 12s\n- check-tempest-master-dsvm-full-havana http://logs.openstack.org/34/81834/9/check/check-tempest-master-dsvm-full-havana/b77a798 : FAILURE in 1h 56m 20s (non-voting)\n- check-tempest-master-dsvm-neutron-havana http://logs.openstack.org/34/81834/9/check/check-tempest-master-dsvm-neutron-havana/45ea9c2 : FAILURE in 1h 38m 51s (non-voting)\n","accounts_in_message":[],"_revision_number":9},{"id":"aa4cb718dc14c3fa959d21dc534dc5b84d0dce09","author":{"_account_id":8824,"name":"Sergey Murashov","email":"smurashov@mirantis.com","username":"SergeyMurashov"},"date":"2014-04-21 06:54:17.000000000","message":"Patch Set 9: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":9},{"id":"8b37ccf55020a56a63930e096f89c4ff9623c19a","author":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"date":"2014-04-23 13:08:14.000000000","message":"Patch Set 9: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":9},{"id":"a55417a1ea39685da8629f6e76c42b2b2ea0e254","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-23 13:08:15.000000000","message":"Patch Set 9: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":9},{"id":"e9a1a902b514ef76b8b1ff3a44cb8f71bf794a76","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-23 15:38:23.000000000","message":"Patch Set 9: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-full/e034d93 : SUCCESS in 1h 02m 06s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron/3b633eb : FAILURE in 59m 06s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-heat-slow/77856a7 : SUCCESS in 25m 28s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm/54400dd : SUCCESS in 41m 56s\n- check-grenade-dsvm-neutron http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-neutron/80a54f5 : FAILURE in 27m 57s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/9/check/gate-tempest-pep8/4719f16 : SUCCESS in 4m 33s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/9/check/gate-tempest-docs/39e3f20/doc/build/html/ : SUCCESS in 3m 18s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/9/check/gate-tempest-python27/430da54 : SUCCESS in 4m 20s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-postgres-full/d622ba8 : SUCCESS in 55m 34s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-full/698f93c : SUCCESS in 1h 09m 34s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-large-ops/ff30cab : SUCCESS in 17m 19s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-neutron-large-ops/980c874 : SUCCESS in 15m 26s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-partial-ncpu/e301af3 : SUCCESS in 32m 32s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-ironic-nv/276641b : SUCCESS in 32m 17s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-ironic-postgres-nv/63be5f0 : SUCCESS in 35m 00s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-full-icehouse/2ab5667 : SUCCESS in 44m 16s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-icehouse/dc08098 : SUCCESS in 1h 04m 56s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-heat-slow-icehouse/81187f1 : SUCCESS in 24m 35s (non-voting)\n- check-grenade-dsvm-icehouse http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-icehouse/0d63b97 : SUCCESS in 38m 58s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-neutron-icehouse/cae1717 : FAILURE in 47m 04s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-postgres-full-icehouse/775bec3 : SUCCESS in 58m 07s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-full-icehouse/8b6698d : FAILURE in 44m 40s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-large-ops-icehouse/7744f37 : SUCCESS in 14m 03s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-neutron-large-ops-icehouse/520ec10 : SUCCESS in 16m 48s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-partial-ncpu-icehouse/d36b810 : SUCCESS in 37m 48s\n- check-tempest-master-dsvm-full-havana http://logs.openstack.org/34/81834/9/check/check-tempest-master-dsvm-full-havana/29f150b : FAILURE in 1h 56m 03s (non-voting)\n- check-tempest-master-dsvm-neutron-havana http://logs.openstack.org/34/81834/9/check/check-tempest-master-dsvm-neutron-havana/de3bfc8 : FAILURE in 1h 31m 48s (non-voting)\n","accounts_in_message":[],"_revision_number":9},{"id":"abb8ae2ba05bc86025977afcf346f2493a691530","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-04-23 15:39:40.000000000","message":"Patch Set 9:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-grenade-dsvm-neutron: https://bugs.launchpad.net/bugs/1307344\n- check-grenade-dsvm-neutron-icehouse: unrecognized error\n- check-tempest-master-dsvm-neutron-havana: https://bugs.launchpad.net/bugs/1253896 https://bugs.launchpad.net/bugs/1224001 https://bugs.launchpad.net/bugs/1283522\n- check-tempest-dsvm-neutron: unrecognized error\n- check-tempest-dsvm-neutron-full-icehouse: https://bugs.launchpad.net/bugs/1253896\n- check-tempest-master-dsvm-full-havana: unrecognized error\n\nYou have some unrecognized errors.\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":9},{"id":"f0b068d78ef2c6e8db21897b400567fb2660e484","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-04-23 17:20:35.000000000","message":"Patch Set 9:\n\nrecheck bug 1309831","accounts_in_message":[],"_revision_number":9},{"id":"7f46bc52a16f4ff6833941e407529836e0065653","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-23 17:20:40.000000000","message":"Patch Set 9: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":9},{"id":"9825e13274340fcadcb662bfad57f733b04fe844","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-04-23 19:32:43.000000000","message":"Patch Set 9:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/34/81834/9/5861/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":9},{"id":"0692caac64abe5f2ea877ebec2da468e8a64ccfc","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-23 19:49:12.000000000","message":"Patch Set 9: Works for me\n\nBuild succeeded.\n\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-full/9825629 : SUCCESS in 1h 06m 36s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron/77102fd : SUCCESS in 55m 27s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-heat-slow/923ff84 : FAILURE in 31m 22s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm/9291d63 : SUCCESS in 46m 16s\n- check-grenade-dsvm-neutron http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-neutron/ab51d42 : FAILURE in 33m 37s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/9/check/gate-tempest-pep8/46904c6 : SUCCESS in 4m 17s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/9/check/gate-tempest-docs/ee745e9/doc/build/html/ : SUCCESS in 2m 37s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/9/check/gate-tempest-python27/8a8d6f4 : SUCCESS in 4m 45s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-postgres-full/b0496cf : SUCCESS in 1h 12m 57s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-full/1e8b7ca : FAILURE in 1h 22m 36s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-large-ops/3e4b417 : SUCCESS in 16m 05s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-neutron-large-ops/a07ec68 : SUCCESS in 18m 22s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-partial-ncpu/5851b8a : SUCCESS in 46m 35s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-ironic-nv/a30f87a : SUCCESS in 37m 26s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-ironic-postgres-nv/d229bec : SUCCESS in 34m 22s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-full-icehouse/25e192a : SUCCESS in 1h 07m 31s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-icehouse/d280f2b : SUCCESS in 1h 05m 19s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-heat-slow-icehouse/ecbcee7 : SUCCESS in 17m 14s (non-voting)\n- check-grenade-dsvm-icehouse http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-icehouse/3c689b8 : SUCCESS in 36m 27s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-neutron-icehouse/4fbab06 : FAILURE in 31m 42s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-postgres-full-icehouse/2c794f3 : SUCCESS in 59m 07s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-full-icehouse/991e306 : SUCCESS in 44m 47s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-large-ops-icehouse/7c8356d : SUCCESS in 17m 01s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-neutron-large-ops-icehouse/6419306 : SUCCESS in 16m 21s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-partial-ncpu-icehouse/653a6a9 : SUCCESS in 40m 22s\n- check-tempest-master-dsvm-full-havana http://logs.openstack.org/34/81834/9/check/check-tempest-master-dsvm-full-havana/e3d471a : FAILURE in 1h 54m 44s (non-voting)\n- check-tempest-master-dsvm-neutron-havana http://logs.openstack.org/34/81834/9/check/check-tempest-master-dsvm-neutron-havana/52e801e : FAILURE in 14m 19s (non-voting)\n","accounts_in_message":[],"_revision_number":9},{"id":"41ba20d875234e2087c3913fd6679833a9854170","author":{"_account_id":7227,"name":"Timur Nurlygayanov","email":"tnurlygayanov@mirantis.com","username":"tnurlygayanov"},"date":"2014-04-24 08:04:06.000000000","message":"Patch Set 9: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":9},{"id":"24e67ca6fbdcc3156322d42b1597fffc04528879","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-04-30 19:54:33.000000000","message":"Patch Set 9: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"9669acccfb0a4ab6105d6428cec9c72cc64bc12a","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2014-05-21 13:55:55.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)\n\nThe idea is very good, thanks. Could you address David\u0027s comments and mine, and add some more docstring to explain how this is used.","accounts_in_message":[],"_revision_number":9},{"id":"b40d5889f9493f750d0a125b42fd7dcfa9c0ecd8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-21 13:55:58.000000000","message":"Patch Set 9: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":9},{"id":"a89c204b14182a3144b9bb5f833e524967a72a66","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-21 15:07:51.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-full/05f1d3d : SUCCESS in 40m 18s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron/597dd16 : SUCCESS in 1h 04m 05s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-heat-slow/cfb017f : SUCCESS in 20m 11s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm/690aed1 : SUCCESS in 41m 09s\n- check-grenade-dsvm-neutron http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-neutron/674f0d2 : FAILURE in 36m 59s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/9/check/gate-tempest-pep8/6b9f3b4 : SUCCESS in 3m 38s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/9/check/gate-tempest-docs/d2fe21f/doc/build/html/ : SUCCESS in 3m 47s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/9/check/gate-tempest-python27/1b525d2 : SUCCESS in 3m 34s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-postgres-full/0413f74 : SUCCESS in 53m 09s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-full/2a1b135 : SUCCESS in 59m 45s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-large-ops/8d08089 : SUCCESS in 22m 00s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-neutron-large-ops/2bbb1c5 : SUCCESS in 20m 50s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-partial-ncpu/5154810 : SUCCESS in 47m 07s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-ironic-nv/91eff90 : SUCCESS in 36m 13s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-ironic-postgres-nv/f79953c : SUCCESS in 38m 57s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-full-icehouse/a5f4a83 : SUCCESS in 58m 10s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-icehouse/2ebe15e : SUCCESS in 57m 22s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-heat-slow-icehouse/1140d90 : FAILURE in 24m 39s (non-voting)\n- check-grenade-dsvm-icehouse http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-icehouse/1cc3aa5 : FAILURE in 39m 59s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-neutron-icehouse/37a7145 : FAILURE in 52m 34s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-postgres-full-icehouse/2fdeeae : SUCCESS in 59m 05s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-neutron-full-icehouse/3531146 : SUCCESS in 1h 07m 46s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/34/81834/9/check/check-tempest-dsvm-virtual-ironic-nv/565f37b : SUCCESS in 30m 26s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-large-ops-icehouse/c7c4922 : SUCCESS in 20m 32s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/34/81834/9/check/gate-tempest-dsvm-neutron-large-ops-icehouse/096bf26 : SUCCESS in 21m 31s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/34/81834/9/check/check-grenade-dsvm-partial-ncpu-icehouse/a9348ed : SUCCESS in 40m 40s","accounts_in_message":[],"_revision_number":9},{"id":"ff94f3a4107f910ac21557f831c20e73bb97e90a","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-05-21 15:08:22.000000000","message":"Patch Set 9:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-grenade-dsvm-neutron-icehouse: https://bugs.launchpad.net/bugs/1298355 https://bugs.launchpad.net/bugs/1283522\n- check-grenade-dsvm-neutron: unrecognized error\n- check-tempest-dsvm-neutron-heat-slow-icehouse: unrecognized error\n- check-grenade-dsvm-icehouse: https://bugs.launchpad.net/bugs/1306029\n\nYou have some unrecognized errors.\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":9},{"id":"691f74fe68cce4c4432f543ce3d66fea606593e9","author":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"date":"2014-05-21 16:11:12.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"aba730807aa4eca892ef9ceafcbb1e2ebe18e667","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-21 16:11:13.000000000","message":"Patch Set 10:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":10},{"id":"d54cc37b76f74fc7559c6fbacf9f8702138dba07","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-21 17:14:55.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-full/9b1ab23 : FAILURE in 49m 48s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-neutron/98228d1 : SUCCESS in 51m 49s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-neutron-heat-slow/05bbcdd : SUCCESS in 17m 03s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/34/81834/10/check/check-grenade-dsvm/3292518 : FAILURE in 23m 31s\n- check-grenade-dsvm-neutron http://logs.openstack.org/34/81834/10/check/check-grenade-dsvm-neutron/ae1e1c2 : FAILURE in 38m 10s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/10/check/gate-tempest-pep8/0150043 : SUCCESS in 3m 48s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/10/check/gate-tempest-docs/393c598/doc/build/html/ : SUCCESS in 4m 20s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/10/check/gate-tempest-python27/93bf3ae : SUCCESS in 3m 43s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-postgres-full/655b45b : FAILURE in 40m 49s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-neutron-full/7ea6283 : FAILURE in 52m 36s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/10/check/gate-tempest-dsvm-large-ops/248a5a8 : SUCCESS in 20m 51s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/10/check/gate-tempest-dsvm-neutron-large-ops/3ae4c78 : SUCCESS in 20m 38s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/10/check/check-grenade-dsvm-partial-ncpu/2ba85c9 : SUCCESS in 38m 07s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-ironic-nv/aed5008 : SUCCESS in 41m 19s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-ironic-postgres-nv/0ed33a0 : SUCCESS in 33m 49s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-full-icehouse/ce0a56f : FAILURE in 54m 06s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-neutron-icehouse/c987228 : SUCCESS in 59m 28s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-neutron-heat-slow-icehouse/5672794 : FAILURE in 21m 53s (non-voting)\n- check-grenade-dsvm-icehouse http://logs.openstack.org/34/81834/10/check/check-grenade-dsvm-icehouse/b18439c : SUCCESS in 1h 00m 54s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/10/check/check-grenade-dsvm-neutron-icehouse/b3de397 : FAILURE in 32m 27s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-postgres-full-icehouse/573880d : FAILURE in 56m 45s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-neutron-full-icehouse/fce6738 : FAILURE in 57m 30s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-virtual-ironic-nv/931f6b8 : SUCCESS in 29m 30s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/34/81834/10/check/gate-tempest-dsvm-large-ops-icehouse/05a5aa5 : SUCCESS in 20m 24s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/34/81834/10/check/gate-tempest-dsvm-neutron-large-ops-icehouse/d1abfeb : SUCCESS in 21m 26s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/34/81834/10/check/check-grenade-dsvm-partial-ncpu-icehouse/487debc : SUCCESS in 34m 07s","accounts_in_message":[],"_revision_number":10},{"id":"e5d8a12ccdb508baf489728a1bd3bcc793ae502b","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-05-21 17:15:46.000000000","message":"Patch Set 10:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-full: unrecognized error\n- check-grenade-dsvm-neutron: unrecognized error\n- check-tempest-dsvm-postgres-full: unrecognized error\n- check-tempest-dsvm-neutron-full: unrecognized error\n- check-tempest-dsvm-full-icehouse: unrecognized error\n- check-grenade-dsvm-neutron-icehouse: unrecognized error\n- check-tempest-dsvm-postgres-full-icehouse: unrecognized error\n- check-grenade-dsvm: https://bugs.launchpad.net/bugs/1262566\n- check-tempest-dsvm-neutron-full-icehouse: unrecognized error\n- check-tempest-dsvm-neutron-heat-slow-icehouse: unrecognized error\n\nYou have some unrecognized errors.\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":10},{"id":"e540623b059d21fe995e6e8e549581f3262bff64","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-05-21 17:21:52.000000000","message":"Patch Set 10: Verified-1\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/34/81834/10/8596/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":10},{"id":"5f3a76c33d0ebaec1d68d6a65038275507079b4e","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-05-21 18:12:50.000000000","message":"Patch Set 10:\n\nBuild succeeded.\n\n- gate-ibm-tempest-python27 http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/81834/10/gate-ibm-tempest-python27/47d70f3 : SUCCESS in 6m 50s\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/81834/10/check-ibm-tempest-dsvm-full/350d9cc : SUCCESS in 2h 01m 26s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/81834/10/check-ibm-tempest-dsvm-postgres-full/c3d46e4 : SUCCESS in 38m 34s","accounts_in_message":[],"_revision_number":10},{"id":"01db86bd8e25abcb643766dc62e7aaadc083805a","author":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"date":"2014-05-21 22:09:35.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"30e67755080a5b5add48a7a07a4b7b5331b699ec","author":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"date":"2014-05-21 22:18:52.000000000","message":"Patch Set 10:\n\nrecheck bug 1285833","accounts_in_message":[],"_revision_number":10},{"id":"c843807c03022c55b30d6df1b47911f0f35b2fe4","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-21 22:18:53.000000000","message":"Patch Set 10: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":10},{"id":"71d11abb0e8cdd3653fd6d2ff4a78f08cc792d3f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-21 23:26:42.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-full/f227ffc : FAILURE in 1h 07m 28s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-neutron/b717a00 : SUCCESS in 58m 21s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-neutron-heat-slow/93538ad : SUCCESS in 18m 35s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/34/81834/10/check/check-grenade-dsvm/15aa1e0 : SUCCESS in 35m 55s\n- check-grenade-dsvm-neutron http://logs.openstack.org/34/81834/10/check/check-grenade-dsvm-neutron/f774c69 : FAILURE in 40m 52s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/10/check/gate-tempest-pep8/362a03e : SUCCESS in 3m 30s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/10/check/gate-tempest-docs/ea73668/doc/build/html/ : SUCCESS in 3m 24s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/10/check/gate-tempest-python27/5288aa9 : SUCCESS in 3m 17s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-postgres-full/d466586 : FAILURE in 1h 06m 43s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-neutron-full/44bcbc3 : FAILURE in 1h 00m 23s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/10/check/gate-tempest-dsvm-large-ops/88bb36d : SUCCESS in 19m 52s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/10/check/gate-tempest-dsvm-neutron-large-ops/98fe6a8 : SUCCESS in 19m 12s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/10/check/check-grenade-dsvm-partial-ncpu/9076b45 : SUCCESS in 41m 43s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-ironic-nv/36afc14 : SUCCESS in 35m 23s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-ironic-postgres-nv/8a92b16 : SUCCESS in 35m 33s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-full-icehouse/8f41b78 : FAILURE in 55m 09s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-neutron-icehouse/4a428bf : SUCCESS in 55m 05s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-neutron-heat-slow-icehouse/aa4ad72 : FAILURE in 21m 04s (non-voting)\n- check-grenade-dsvm-icehouse http://logs.openstack.org/34/81834/10/check/check-grenade-dsvm-icehouse/28ebeca : FAILURE in 37m 22s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/10/check/check-grenade-dsvm-neutron-icehouse/bf44d3e : FAILURE in 46m 29s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-postgres-full-icehouse/4949794 : FAILURE in 53m 32s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-neutron-full-icehouse/3311571 : FAILURE in 1h 02m 55s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/34/81834/10/check/check-tempest-dsvm-virtual-ironic-nv/95f65a0 : SUCCESS in 25m 31s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/34/81834/10/check/gate-tempest-dsvm-large-ops-icehouse/042626d : SUCCESS in 18m 49s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/34/81834/10/check/gate-tempest-dsvm-neutron-large-ops-icehouse/a795a2f : SUCCESS in 19m 55s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/34/81834/10/check/check-grenade-dsvm-partial-ncpu-icehouse/6c17a38 : FAILURE in 37m 35s","accounts_in_message":[],"_revision_number":10},{"id":"7e644a6c5229659de910babbad882239ce7b5010","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-05-21 23:28:10.000000000","message":"Patch Set 10:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-grenade-dsvm-partial-ncpu-icehouse: https://bugs.launchpad.net/bugs/1320670\n- check-tempest-dsvm-full: unrecognized error\n- check-grenade-dsvm-neutron: unrecognized error\n- check-tempest-dsvm-postgres-full: unrecognized error\n- check-tempest-dsvm-neutron-full: unrecognized error\n- check-tempest-dsvm-full-icehouse: unrecognized error\n- check-grenade-dsvm-icehouse: https://bugs.launchpad.net/bugs/1320670\n- check-tempest-dsvm-postgres-full-icehouse: unrecognized error\n- check-grenade-dsvm-neutron-icehouse: https://bugs.launchpad.net/bugs/1320670\n- check-tempest-dsvm-neutron-full-icehouse: unrecognized error\n- check-tempest-dsvm-neutron-heat-slow-icehouse: unrecognized error\n\nYou have some unrecognized errors.\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":10},{"id":"7721af6dbfb03b0ef555dc3c3fe15e1dda1ca3f6","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-05-21 23:33:01.000000000","message":"Patch Set 10:\n\nFailed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/34/81834/10/8646/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":10},{"id":"e19bc3e500d01f2935cd21631406465efabb0e72","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2014-05-22 07:07:17.000000000","message":"Patch Set 10: Code-Review-1\n\n(10 comments)\n\nThanks for progressing on this!\nThere are still a couple of issues to be addressed, else it looks good to me.","accounts_in_message":[],"_revision_number":10},{"id":"a6424d21f8eb4f77ce372b44805e35bd866fc1d5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-23 13:12:24.000000000","message":"Patch Set 11:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":11},{"id":"43a1465fb3b7163cd1f30e2dfbfc12d7e7422396","author":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"date":"2014-05-23 13:12:24.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"9f138a9461a9a0659049ce7a267afafb38c3a4fe","author":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"date":"2014-05-23 14:23:52.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"b1f613eace1df9d3176504d2645ed45c29d4b88b","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-23 14:24:12.000000000","message":"Patch Set 12:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":12},{"id":"d8b46a90991b05197f80f86c63937c43fc6d3422","author":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"date":"2014-05-23 14:45:08.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"e828f24fe92e9c33cf4d0ec95d2c34489e0322c5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-23 14:45:10.000000000","message":"Patch Set 13:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":13},{"id":"222a92524ee003d3b7e0b8238ebde1e193d76a08","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-23 15:47:17.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/13/check/check-tempest-dsvm-full/000d8d0 : FAILURE in 51m 19s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/13/check/check-tempest-dsvm-neutron/5b07cdc : SUCCESS in 1h 01m 20s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/13/check/check-tempest-dsvm-neutron-heat-slow/4bce483 : SUCCESS in 24m 06s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/34/81834/13/check/check-grenade-dsvm/bb35596 : SUCCESS in 40m 52s\n- check-grenade-dsvm-neutron http://logs.openstack.org/34/81834/13/check/check-grenade-dsvm-neutron/0fbdf8b : FAILURE in 46m 58s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/13/check/gate-tempest-pep8/a3748b3 : SUCCESS in 3m 35s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/13/check/gate-tempest-docs/8975c86/doc/build/html/ : SUCCESS in 3m 16s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/13/check/gate-tempest-python27/ae50566 : SUCCESS in 2m 59s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/13/check/check-tempest-dsvm-postgres-full/21f75ed : FAILURE in 56m 56s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/13/check/check-tempest-dsvm-neutron-full/5faa35e : SUCCESS in 1h 01m 56s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/13/check/gate-tempest-dsvm-large-ops/ef6fe14 : SUCCESS in 22m 58s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/13/check/gate-tempest-dsvm-neutron-large-ops/4a4c531 : SUCCESS in 19m 57s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/13/check/check-grenade-dsvm-partial-ncpu/c478045 : SUCCESS in 43m 42s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/13/check/check-tempest-dsvm-ironic-nv/edec8e2 : SUCCESS in 42m 59s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/13/check/check-tempest-dsvm-ironic-postgres-nv/6581241 : SUCCESS in 37m 18s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/34/81834/13/check/check-tempest-dsvm-full-icehouse/24fbb1b : FAILURE in 59m 48s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/13/check/check-tempest-dsvm-neutron-icehouse/d746bce : SUCCESS in 58m 40s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/34/81834/13/check/check-tempest-dsvm-neutron-heat-slow-icehouse/d7f89a6 : FAILURE in 24m 28s (non-voting)\n- check-grenade-dsvm-icehouse http://logs.openstack.org/34/81834/13/check/check-grenade-dsvm-icehouse/774b656 : SUCCESS in 41m 54s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/13/check/check-grenade-dsvm-neutron-icehouse/6ea333e : FAILURE in 49m 24s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/34/81834/13/check/check-tempest-dsvm-postgres-full-icehouse/431bdff : FAILURE in 52m 56s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/34/81834/13/check/check-tempest-dsvm-neutron-full-icehouse/b28be6b : FAILURE in 51m 28s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/34/81834/13/check/check-tempest-dsvm-virtual-ironic-nv/396cb55 : SUCCESS in 30m 45s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/34/81834/13/check/gate-tempest-dsvm-large-ops-icehouse/a0a9009 : SUCCESS in 21m 57s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/34/81834/13/check/gate-tempest-dsvm-neutron-large-ops-icehouse/8c64c5d : SUCCESS in 20m 06s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/34/81834/13/check/check-grenade-dsvm-partial-ncpu-icehouse/99460c1 : SUCCESS in 43m 51s","accounts_in_message":[],"_revision_number":13},{"id":"32a70194349d60e609da632d1002cbda42e19936","author":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"date":"2014-05-23 15:47:22.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"c78901bb795bd3c395a40b647acaa3d291de4ab2","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-23 15:47:23.000000000","message":"Patch Set 14:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":14},{"id":"f2efb46ae99571b2f388d97a4c16e79bdd810236","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-23 17:16:43.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-full/4d33e1b : SUCCESS in 56m 10s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-neutron/193870b : SUCCESS in 57m 29s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-neutron-heat-slow/a4b620b : SUCCESS in 20m 57s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/34/81834/14/check/check-grenade-dsvm/2c74935 : SUCCESS in 44m 37s\n- check-grenade-dsvm-neutron http://logs.openstack.org/34/81834/14/check/check-grenade-dsvm-neutron/7b73808 : FAILURE in 44m 53s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/14/check/gate-tempest-pep8/82fe613 : SUCCESS in 3m 38s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/14/check/gate-tempest-docs/e3ed31d/doc/build/html/ : SUCCESS in 4m 13s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/14/check/gate-tempest-python27/fe1960a : SUCCESS in 3m 34s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-postgres-full/8988683 : SUCCESS in 55m 50s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-neutron-full/ea09f97 : FAILURE in 1h 29m 03s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/14/check/gate-tempest-dsvm-large-ops/f88e3ce : SUCCESS in 22m 10s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/14/check/gate-tempest-dsvm-neutron-large-ops/56279d0 : SUCCESS in 20m 48s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/14/check/check-grenade-dsvm-partial-ncpu/58ac64f : SUCCESS in 44m 42s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-ironic-nv/7bd2ff5 : FAILURE in 36m 24s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-ironic-postgres-nv/9da9668 : SUCCESS in 36m 48s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-full-icehouse/5a94a02 : SUCCESS in 54m 05s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-neutron-icehouse/d6f9f76 : SUCCESS in 55m 37s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-neutron-heat-slow-icehouse/bb5067c : FAILURE in 23m 31s (non-voting)\n- check-grenade-dsvm-icehouse http://logs.openstack.org/34/81834/14/check/check-grenade-dsvm-icehouse/7a23be7 : SUCCESS in 37m 47s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/14/check/check-grenade-dsvm-neutron-icehouse/03a89dd : FAILURE in 50m 16s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-postgres-full-icehouse/61255a1 : SUCCESS in 55m 43s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-neutron-full-icehouse/2b90ec1 : FAILURE in 1h 05m 22s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-virtual-ironic-nv/61c6d52 : SUCCESS in 33m 07s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/34/81834/14/check/gate-tempest-dsvm-large-ops-icehouse/8492f8a : SUCCESS in 19m 12s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/34/81834/14/check/gate-tempest-dsvm-neutron-large-ops-icehouse/c2534f5 : SUCCESS in 20m 04s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/34/81834/14/check/check-grenade-dsvm-partial-ncpu-icehouse/0988703 : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":14},{"id":"a1b21bf10df1581e27d993ba68c6e05d26937703","author":{"_account_id":10118,"name":"IBM PowerKVM CI","email":"kvmpower@linux.vnet.ibm.com","username":"powerkvm","tags":["SERVICE_USER"]},"date":"2014-05-23 20:44:24.000000000","message":"Patch Set 14:\n\nBuild succeeded.\n\n- gate-ibm-tempest-python27 http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/81834/14/gate-ibm-tempest-python27/32e7a06 : SUCCESS in 11m 08s\n- check-ibm-tempest-dsvm-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/81834/14/check-ibm-tempest-dsvm-full/73a5d67 : SUCCESS in 43m 45s (non-voting)\n- check-ibm-tempest-dsvm-postgres-full http://dal05.objectstorage.softlayer.net/v1/AUTH_3d8e6ecb-f597-448c-8ec2-164e9f710dd6/pkvmci/81834/14/check-ibm-tempest-dsvm-postgres-full/7fb8bb2 : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":14},{"id":"8d22530671626ad431c84f319cac524eaed2f433","author":{"_account_id":1921,"name":"Andrea Frittoli","email":"andrea.frittoli@gmail.com","username":"andrea-frittoli"},"date":"2014-05-27 12:52:36.000000000","message":"Patch Set 14: Code-Review+1\n\n(1 comment)\n\nIt looks good to me, thank you!\n\nJust a nit re a docstring, but I wouldn\u0027t make a new patchset just for that.","accounts_in_message":[],"_revision_number":14},{"id":"795f91a353a21a7a8ac98d91fae7a0ed4259ad05","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-27 12:52:37.000000000","message":"Patch Set 14: -Verified\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":14},{"id":"640ed878c256082755e78b6d633099a2020ba458","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-05-27 13:34:54.000000000","message":"Patch Set 14: Verified+1\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/34/81834/14/8996/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":14},{"id":"ce2aea7a041bd7955eb5796f5b4d51efccb77e00","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-05-27 13:59:54.000000000","message":"Patch Set 14: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":14},{"id":"cdbbb0eb795b23c4c8a7257f4337f3851b1f51e3","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-27 14:01:11.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded.\n\n- check-tempest-dsvm-full http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-full/f5c5b8f : SUCCESS in 53m 34s\n- check-tempest-dsvm-neutron http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-neutron/cda6385 : SUCCESS in 57m 57s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-neutron-heat-slow/fa5bb77 : SUCCESS in 24m 39s (non-voting)\n- check-grenade-dsvm http://logs.openstack.org/34/81834/14/check/check-grenade-dsvm/1230325 : SUCCESS in 42m 01s\n- check-grenade-dsvm-neutron http://logs.openstack.org/34/81834/14/check/check-grenade-dsvm-neutron/dc86fa4 : FAILURE in 34m 37s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/14/check/gate-tempest-pep8/1883938 : SUCCESS in 3m 49s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/14/check/gate-tempest-docs/330bc1a/doc/build/html/ : SUCCESS in 4m 23s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/14/check/gate-tempest-python27/c497761 : SUCCESS in 3m 16s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-postgres-full/f1616b5 : SUCCESS in 54m 26s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-neutron-full/fa17674 : SUCCESS in 1h 01m 44s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/14/check/gate-tempest-dsvm-large-ops/714d659 : SUCCESS in 21m 38s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/14/check/gate-tempest-dsvm-neutron-large-ops/4278875 : SUCCESS in 19m 08s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/14/check/check-grenade-dsvm-partial-ncpu/8eb3642 : SUCCESS in 39m 41s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-ironic-nv/d263a5e : SUCCESS in 35m 40s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-ironic-postgres-nv/d80b9dd : SUCCESS in 37m 25s (non-voting)\n- check-tempest-dsvm-full-icehouse http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-full-icehouse/d46b338 : SUCCESS in 50m 00s\n- check-tempest-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-neutron-icehouse/cf14a26 : SUCCESS in 54m 57s\n- check-tempest-dsvm-neutron-heat-slow-icehouse http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-neutron-heat-slow-icehouse/e56e514 : FAILURE in 26m 29s (non-voting)\n- check-grenade-dsvm-icehouse http://logs.openstack.org/34/81834/14/check/check-grenade-dsvm-icehouse/46f4781 : SUCCESS in 42m 07s\n- check-grenade-dsvm-neutron-icehouse http://logs.openstack.org/34/81834/14/check/check-grenade-dsvm-neutron-icehouse/fcffa4a : SUCCESS in 1h 00m 58s (non-voting)\n- check-tempest-dsvm-postgres-full-icehouse http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-postgres-full-icehouse/3a40201 : SUCCESS in 56m 37s\n- check-tempest-dsvm-neutron-full-icehouse http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-neutron-full-icehouse/c0bad9b : FAILURE in 1h 05m 00s (non-voting)\n- check-tempest-dsvm-virtual-ironic-nv http://logs.openstack.org/34/81834/14/check/check-tempest-dsvm-virtual-ironic-nv/eef7781 : SUCCESS in 30m 48s (non-voting)\n- gate-tempest-dsvm-large-ops-icehouse http://logs.openstack.org/34/81834/14/check/gate-tempest-dsvm-large-ops-icehouse/b33207a : SUCCESS in 24m 13s\n- gate-tempest-dsvm-neutron-large-ops-icehouse http://logs.openstack.org/34/81834/14/check/gate-tempest-dsvm-neutron-large-ops-icehouse/cf00e41 : SUCCESS in 21m 44s\n- check-grenade-dsvm-partial-ncpu-icehouse http://logs.openstack.org/34/81834/14/check/check-grenade-dsvm-partial-ncpu-icehouse/b8991af : SUCCESS in 41m 50s","accounts_in_message":[],"_revision_number":14},{"id":"7c78c07f82777360cd05dac53fe67ea84c1513af","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-27 14:01:18.000000000","message":"Patch Set 14: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":14},{"id":"2a7ec90eb78f51e8b25ea79af62d6ae87d5c4426","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-05-27 14:24:43.000000000","message":"Patch Set 14: -Workflow -Code-Review\n\naccidentally +A -ed. Removing my vote for a while.","accounts_in_message":[],"_revision_number":14},{"id":"72aaf4f324e6751dff3646b27b533f2de14b4051","author":{"_account_id":5196,"name":"Matthew Treinish","email":"mtreinish@kortar.org","username":"treinish"},"date":"2014-05-27 15:41:19.000000000","message":"Patch Set 14: Code-Review-1\n\n(15 comments)\n\nSo I still need to do the spec review for this bp, at a quick glance I feel we should put this in a more common location. Like the isolated creds stuff. I also think you should wait until the spec review is finished so we can iron out some of the design decisions in the spec. I have some feedback on how I feel this should be done, but I will move it to the spec review because the whole point of design specs is to figure out how we should be implementing something.","accounts_in_message":[],"_revision_number":14},{"id":"58fc5b5782c574ecb28893bc5e0bcaa08c361bd1","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-05-27 16:04:41.000000000","message":"Patch Set 14: Verified+2\n\nBuild succeeded.\n\n- gate-tempest-dsvm-full http://logs.openstack.org/34/81834/14/gate/gate-tempest-dsvm-full/7454c2f : SUCCESS in 54m 57s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/34/81834/14/gate/gate-tempest-dsvm-neutron/2a2cf79 : SUCCESS in 56m 01s\n- gate-grenade-dsvm http://logs.openstack.org/34/81834/14/gate/gate-grenade-dsvm/8c2133b : SUCCESS in 45m 49s\n- gate-tempest-pep8 http://logs.openstack.org/34/81834/14/gate/gate-tempest-pep8/c5a140a : SUCCESS in 3m 40s\n- gate-tempest-docs http://docs-draft.openstack.org/34/81834/14/gate/gate-tempest-docs/32965cd/doc/build/html/ : SUCCESS in 3m 20s\n- gate-tempest-python27 http://logs.openstack.org/34/81834/14/gate/gate-tempest-python27/0b3d496 : SUCCESS in 3m 17s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/34/81834/14/gate/gate-tempest-dsvm-postgres-full/8130971 : SUCCESS in 1h 00m 43s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/34/81834/14/gate/gate-tempest-dsvm-large-ops/540df30 : SUCCESS in 22m 17s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/34/81834/14/gate/gate-tempest-dsvm-neutron-large-ops/cb3aab0 : SUCCESS in 21m 56s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/34/81834/14/gate/gate-grenade-dsvm-partial-ncpu/62b06d6 : SUCCESS in 42m 21s","accounts_in_message":[],"_revision_number":14},{"id":"02fb564fc5e22b5d90011ee4995b747939c15e01","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2014-05-27 20:50:53.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"6aa983bb52fd32837768958d13b51614287d4dc8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-09-22 15:46:38.000000000","message":"Patch Set 14: Verified-1\n\nMerge Failed.\n\nThis change was unable to be automatically merged with the current state of the repository. Please rebase your change and upload a new patchset.","accounts_in_message":[],"_revision_number":14},{"id":"f75d0f7d4ceefdc0cb90ca8f3ddf22c872f3b27d","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2014-12-05 15:04:55.000000000","message":"Patch Set 14:\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":14},{"id":"9576bb3e5033f7e0d82822ebd6b9ec839b048dbb","author":{"_account_id":2750,"name":"Sean Dague","email":"sean@dague.net","username":"sdague"},"date":"2014-12-05 15:04:55.000000000","message":"Abandoned\n\nThis review is \u003e 4 weeks without comment, and failed Jenkins the last time it was checked. We are abandoning this for now. Feel free to reactivate the review by pressing the restore button and leaving a \u0027recheck\u0027 comment to get fresh test results.","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"60da146bac2df925cd8e6777fe625bc90247c863","revisions":{"9446220a678eee14e8d833807d9b06c6fd9ecc2b":{"kind":"REWORK","_number":1,"created":"2014-03-20 15:32:42.000000000","uploader":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"ref":"refs/changes/34/81834/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/1"}}},"commit":{"parents":[{"commit":"9a2207c882964c8397fd1b355b380df0cda27bf8","subject":"ssh instance validation add options for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9a2207c882964c8397fd1b355b380df0cda27bf8"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 15:32:01.000000000","tz":60},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\n fixed and floating ip.\n\nThe after the init_instance_validation method call the\n create_test_server method will help in setting up an\n ssh able server.\nIf the init_instance_validation not called the create_test_server,\n behaves as before.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9446220a678eee14e8d833807d9b06c6fd9ecc2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9446220a678eee14e8d833807d9b06c6fd9ecc2b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9a2207c882964c8397fd1b355b380df0cda27bf8","is_merged_in_target_branch":false,"change_id":"Ic5fc9bd1f7407d3430fcd33b03a226deed696d57","change_number":80620,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e9de09a223376948f4b5929aff34d0e1cd73344c":{"kind":"REWORK","_number":2,"created":"2014-03-20 16:56:15.000000000","uploader":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"ref":"refs/changes/34/81834/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/2"}}},"commit":{"parents":[{"commit":"9a2207c882964c8397fd1b355b380df0cda27bf8","subject":"ssh instance validation add options for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9a2207c882964c8397fd1b355b380df0cda27bf8"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 16:54:57.000000000","tz":60},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\n fixed and floating ip.\n\nThe after the init_instance_validation method call the\n create_test_server method will help in setting up an\n ssh able server.\nIf the init_instance_validation not called the create_test_server,\n behaves as before.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e9de09a223376948f4b5929aff34d0e1cd73344c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e9de09a223376948f4b5929aff34d0e1cd73344c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9a2207c882964c8397fd1b355b380df0cda27bf8","is_merged_in_target_branch":false,"change_id":"Ic5fc9bd1f7407d3430fcd33b03a226deed696d57","change_number":80620,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"2ea10fa18e10cd298403432c44efef607fcac872":{"kind":"REWORK","_number":3,"created":"2014-03-24 16:03:41.000000000","uploader":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"ref":"refs/changes/34/81834/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/3"}}},"commit":{"parents":[{"commit":"c7e6086d4a4d59a57cdd05a1c4eab95d09732fca","subject":"ssh instance validation add options for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c7e6086d4a4d59a57cdd05a1c4eab95d09732fca"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-24 15:58:20.000000000","tz":60},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\n fixed and floating ip.\n\nThe after the init_instance_validation method call the\n create_test_server method will help in setting up an\n ssh able server.\nIf the init_instance_validation not called the create_test_server,\n behaves as before.\n\nNow just the test_rebuild_server will run with instance validation\nas a demo.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2ea10fa18e10cd298403432c44efef607fcac872"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2ea10fa18e10cd298403432c44efef607fcac872"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c7e6086d4a4d59a57cdd05a1c4eab95d09732fca","is_merged_in_target_branch":false,"change_id":"Ic5fc9bd1f7407d3430fcd33b03a226deed696d57","change_number":80620,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c3bfd3ece1e483b7ff47936ec7bac2238d50bb88":{"kind":"REWORK","_number":4,"created":"2014-04-02 09:01:42.000000000","uploader":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"ref":"refs/changes/34/81834/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/4"}}},"commit":{"parents":[{"commit":"1f66c7fa0886f9f41fce53842fca826d7fc82655","subject":"ssh instance validation add options for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1f66c7fa0886f9f41fce53842fca826d7fc82655"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-04-02 09:01:08.000000000","tz":120},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\n fixed and floating ip.\n\nThe after the init_instance_validation method call the\n create_test_server method will help in setting up an\n ssh able server.\nIf the init_instance_validation not called the create_test_server,\n behaves as before.\n\nNow just the test_rebuild_server will run with instance validation\nas a demo.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c3bfd3ece1e483b7ff47936ec7bac2238d50bb88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c3bfd3ece1e483b7ff47936ec7bac2238d50bb88"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1f66c7fa0886f9f41fce53842fca826d7fc82655","is_merged_in_target_branch":false,"change_id":"Ic5fc9bd1f7407d3430fcd33b03a226deed696d57","change_number":80620,"patch_set_number":3,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0405caca21759c7a65c38eec815f1b3a030c0086":{"kind":"REWORK","_number":5,"created":"2014-04-02 17:31:03.000000000","uploader":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"ref":"refs/changes/34/81834/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/5"}}},"commit":{"parents":[{"commit":"35f17d18d5eb6bff6d2d69c0a4638f6cbe404f5b","subject":"ssh instance validation add options for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/35f17d18d5eb6bff6d2d69c0a4638f6cbe404f5b"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-04-02 17:30:10.000000000","tz":120},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\n fixed and floating ip.\n\nThe after the init_instance_validation method call the\n create_test_server method will help in setting up an\n ssh able server.\nIf the init_instance_validation not called the create_test_server,\n behaves as before.\n\nNow just the test_rebuild_server will run with instance validation\nas a demo.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0405caca21759c7a65c38eec815f1b3a030c0086"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0405caca21759c7a65c38eec815f1b3a030c0086"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"35f17d18d5eb6bff6d2d69c0a4638f6cbe404f5b","is_merged_in_target_branch":false,"change_id":"Ic5fc9bd1f7407d3430fcd33b03a226deed696d57","change_number":80620,"patch_set_number":4,"change_status":"MERGED"}],"branch":"refs/heads/master"},"02babc5bbb795995ab09f3208dc7ca402b0fe94a":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2014-04-07 08:07:49.000000000","uploader":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"ref":"refs/changes/34/81834/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/6"}}},"commit":{"parents":[{"commit":"9ec223b5fa9147dffe7542be39c6501782c6a517","subject":"ssh instance validation add options for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9ec223b5fa9147dffe7542be39c6501782c6a517"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-04-07 08:07:13.000000000","tz":120},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\n fixed and floating ip.\n\nThe after the init_instance_validation method call the\n create_test_server method will help in setting up an\n ssh able server.\nIf the init_instance_validation not called the create_test_server,\n behaves as before.\n\nNow just the test_rebuild_server will run with instance validation\nas a demo.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/02babc5bbb795995ab09f3208dc7ca402b0fe94a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/02babc5bbb795995ab09f3208dc7ca402b0fe94a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9ec223b5fa9147dffe7542be39c6501782c6a517","is_merged_in_target_branch":false,"change_id":"Ic5fc9bd1f7407d3430fcd33b03a226deed696d57","change_number":80620,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"0a1d75ce2fedffe2f402c4e5c6cd1c8a65cf976c":{"kind":"REWORK","_number":7,"created":"2014-04-07 09:40:57.000000000","uploader":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"ref":"refs/changes/34/81834/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/7"}}},"commit":{"parents":[{"commit":"9ec223b5fa9147dffe7542be39c6501782c6a517","subject":"ssh instance validation add options for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9ec223b5fa9147dffe7542be39c6501782c6a517"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-04-07 09:37:07.000000000","tz":120},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\n fixed and floating ip.\n\nThe after the init_instance_validation method call the\n create_test_server method will help in setting up an\n ssh able server.\nIf the init_instance_validation not called the create_test_server,\n behaves as before.\n\nNow just the test_rebuild_server will run with instance validation\nas a demo.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0a1d75ce2fedffe2f402c4e5c6cd1c8a65cf976c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0a1d75ce2fedffe2f402c4e5c6cd1c8a65cf976c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"9ec223b5fa9147dffe7542be39c6501782c6a517","is_merged_in_target_branch":false,"change_id":"Ic5fc9bd1f7407d3430fcd33b03a226deed696d57","change_number":80620,"patch_set_number":5,"change_status":"MERGED"}],"branch":"refs/heads/master"},"4d8ecb82453adf0d0b865d6250e90907496d76c1":{"kind":"REWORK","_number":8,"created":"2014-04-08 13:11:40.000000000","uploader":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"ref":"refs/changes/34/81834/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/8"}}},"commit":{"parents":[{"commit":"b1c323d978c7d110886cbc648536f3cccec19ecd","subject":"ssh instance validation add options for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b1c323d978c7d110886cbc648536f3cccec19ecd"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-04-08 13:06:28.000000000","tz":120},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\n fixed and floating ip.\n\nThe after the init_instance_validation method call the\n create_test_server method will help in setting up an\n ssh able server.\nIf the init_instance_validation not called the create_test_server,\n behaves as before.\n\nNow just the test_rebuild_server will run with instance validation\nas a demo.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4d8ecb82453adf0d0b865d6250e90907496d76c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4d8ecb82453adf0d0b865d6250e90907496d76c1"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b1c323d978c7d110886cbc648536f3cccec19ecd","is_merged_in_target_branch":false,"change_id":"Ic5fc9bd1f7407d3430fcd33b03a226deed696d57","change_number":80620,"patch_set_number":6,"change_status":"MERGED"}],"branch":"refs/heads/master"},"318fdb0d9e60ef00e171df095f5470474a1ec5da":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2014-04-19 12:54:29.000000000","uploader":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"ref":"refs/changes/34/81834/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/9"}}},"commit":{"parents":[{"commit":"423834d26b6d75aa3b8e32aaf6ed63f566644939","subject":"ssh instance validation add options for Neutron","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/423834d26b6d75aa3b8e32aaf6ed63f566644939"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-04-19 12:53:37.000000000","tz":120},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\n fixed and floating ip.\n\nThe after the init_instance_validation method call the\n create_test_server method will help in setting up an\n ssh able server.\nIf the init_instance_validation not called the create_test_server,\n behaves as before.\n\nNow just the test_rebuild_server will run with instance validation\nas a demo.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/318fdb0d9e60ef00e171df095f5470474a1ec5da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/318fdb0d9e60ef00e171df095f5470474a1ec5da"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"423834d26b6d75aa3b8e32aaf6ed63f566644939","is_merged_in_target_branch":true,"change_id":"Ic5fc9bd1f7407d3430fcd33b03a226deed696d57","change_number":80620,"patch_set_number":7,"change_status":"MERGED"}],"branch":"refs/heads/master"},"800dbd5f52c9fe96803bc7982401b15025c51142":{"kind":"REWORK","_number":10,"created":"2014-05-21 16:11:12.000000000","uploader":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"ref":"refs/changes/34/81834/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/10"}}},"commit":{"parents":[{"commit":"6c1a8f08f3a232bca55286c64021c0487e7441c8","subject":"Merge \"Neutron Allowed Address Pair API test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6c1a8f08f3a232bca55286c64021c0487e7441c8"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","date":"2014-05-21 16:11:11.000000000","tz":0},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\n fixed and floating ip.\n\nThe after the init_instance_validation method call the\n create_test_server method will help in setting up an\n ssh able server.\nIf the init_instance_validation not called the create_test_server,\n behaves as before.\n\nNow just the test_rebuild_server will run with instance validation\nas a demo.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/800dbd5f52c9fe96803bc7982401b15025c51142"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/800dbd5f52c9fe96803bc7982401b15025c51142"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"6c1a8f08f3a232bca55286c64021c0487e7441c8","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"3ca0fd9f41c2f352196d879bdf0cfc2578c8e788":{"kind":"REWORK","_number":11,"created":"2014-05-23 13:12:24.000000000","uploader":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"ref":"refs/changes/34/81834/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/11"}}},"commit":{"parents":[{"commit":"fc137b5dd743f6d031f3d7b9ce32e6eafb367db6","subject":"Merge \"Verify more information in floating ip tests(part2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fc137b5dd743f6d031f3d7b9ce32e6eafb367db6"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","date":"2014-05-23 13:11:45.000000000","tz":0},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\nfixed and floating ip.\n\nAfter the init_instance_validation method, call to the\ncreate_test_server method will help in setting up an\nssh able server. If the init_instance_validation is not\ncalled the create_test_server, behaves as before.\n\nNow just the test_rebuild_server and test_change_server_password\nwill run with instance validation as a demo.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3ca0fd9f41c2f352196d879bdf0cfc2578c8e788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3ca0fd9f41c2f352196d879bdf0cfc2578c8e788"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fc137b5dd743f6d031f3d7b9ce32e6eafb367db6","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"af683ff1577fd70a73652dee1b56e9bd3639300a":{"kind":"REWORK","_number":12,"created":"2014-05-23 14:23:52.000000000","uploader":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"ref":"refs/changes/34/81834/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/12"}}},"commit":{"parents":[{"commit":"fc137b5dd743f6d031f3d7b9ce32e6eafb367db6","subject":"Merge \"Verify more information in floating ip tests(part2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fc137b5dd743f6d031f3d7b9ce32e6eafb367db6"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","date":"2014-05-23 14:23:43.000000000","tz":0},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\nfixed and floating ip.\n\nAfter the init_instance_validation method, call to the\ncreate_test_server method will help in setting up an\nssh able server. If the init_instance_validation is not\ncalled the create_test_server, behaves as before.\n\nNow just the test_rebuild_server and test_change_server_password\nwill run with instance validation as a demo.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/af683ff1577fd70a73652dee1b56e9bd3639300a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/af683ff1577fd70a73652dee1b56e9bd3639300a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fc137b5dd743f6d031f3d7b9ce32e6eafb367db6","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"b92849029d5a70c694cd88e676cd1f6006410fdb":{"kind":"REWORK","_number":13,"created":"2014-05-23 14:45:08.000000000","uploader":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"ref":"refs/changes/34/81834/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/13"}}},"commit":{"parents":[{"commit":"fc137b5dd743f6d031f3d7b9ce32e6eafb367db6","subject":"Merge \"Verify more information in floating ip tests(part2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fc137b5dd743f6d031f3d7b9ce32e6eafb367db6"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","date":"2014-05-23 14:44:56.000000000","tz":0},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\nfixed and floating ip.\n\nAfter the init_instance_validation method, call to the\ncreate_test_server method will help in setting up an\nssh able server. If the init_instance_validation is not\ncalled the create_test_server, behaves as before.\n\nNow just the test_rebuild_server and test_change_server_password\nwill run with instance validation as a demo.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b92849029d5a70c694cd88e676cd1f6006410fdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b92849029d5a70c694cd88e676cd1f6006410fdb"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fc137b5dd743f6d031f3d7b9ce32e6eafb367db6","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"60da146bac2df925cd8e6777fe625bc90247c863":{"kind":"REWORK","_number":14,"created":"2014-05-23 15:47:22.000000000","uploader":{"_account_id":1839,"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","username":"nithya-ganesan"},"ref":"refs/changes/34/81834/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/34/81834/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/34/81834/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/34/81834/14"}}},"commit":{"parents":[{"commit":"fc137b5dd743f6d031f3d7b9ce32e6eafb367db6","subject":"Merge \"Verify more information in floating ip tests(part2)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fc137b5dd743f6d031f3d7b9ce32e6eafb367db6"}]}],"author":{"name":"Attila Fazekas","email":"afazekas@redhat.com","date":"2014-03-20 14:12:29.000000000","tz":60},"committer":{"name":"nithya-ganesan","email":"nithya.ganesan@hp.com","date":"2014-05-23 15:47:14.000000000","tz":0},"subject":"Enable instance validation with the server_actions v2","message":"Enable instance validation with the server_actions v2\n\nAdding the base methods for instance validation on both\nfixed and floating ip.\n\nAfter the init_instance_validation method, call to the\ncreate_test_server method will help in setting up an\nssh able server. If the init_instance_validation is not\ncalled the create_test_server, behaves as before.\n\nNow just the test_rebuild_server and test_change_server_password\nwill run with instance validation as a demo.\n\nChange-Id: I499a11a128b1caa0f92f2392e2f7a511583bf3c9\nImplements: blueprint ssh-auth-strategy\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/60da146bac2df925cd8e6777fe625bc90247c863"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/60da146bac2df925cd8e6777fe625bc90247c863"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"fc137b5dd743f6d031f3d7b9ce32e6eafb367db6","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
