)]}'
{"id":"openstack%2Ftempest~818861","triplet_id":"openstack%2Ftempest~master~Ide9836afba4a12b2ca780cf447c747797023651c","project":"openstack/tempest","branch":"master","topic":"tempest05","attention_set":{},"removed_from_attention_set":{"31280":{"account":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"last_update":"2022-12-24 05:47:51.000000000","reason":"Change was abandoned"},"11075":{"account":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"last_update":"2021-12-16 09:25:33.000000000","reason":"Benny Kopilov replied on the change"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-12-24 05:47:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ide9836afba4a12b2ca780cf447c747797023651c","subject":"Test multiattach volume booted server does not delete multi attachments","status":"ABANDONED","created":"2021-11-23 08:01:42.000000000","updated":"2022-12-24 05:47:51.000000000","total_comment_count":11,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"fd16f0653c3791b9915912b683890629ae55e721","_number":818861,"virtual_id_number":818861,"owner":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-12-03 03:23:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2021-12-16 09:25:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-23 10:01:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-01 22:42:44.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-12-14 11:08:36.000000000","updated_by":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"reviewer":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"state":"REVIEWER"}],"messages":[{"id":"c9b5d788843fa80dd13de222fa5971beaa779684","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-11-23 08:01:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e70d275dc754f8e90c9c4ddcca5db788abf5affe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-11-23 08:09:16.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0eac0616350a9ccb4bd799b83858e15a4004d1b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 10:01:14.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/dd0e337207814bdeac4c1c3c44424c38 : SUCCESS in 53m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/30674e9b92524fd7ad2721ab014316c6 : FAILURE in 1h 41m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4482dd5656f54090883ce6e3f1c36358 : SUCCESS in 5m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9523456cbcb3436a91306a3ebbfc54bc : FAILURE in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18f802d742c744149cf9e8e294fa7fc9 : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8ff907d67d024439a1341386e38e720f : SUCCESS in 5m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f21be9b306004d9ca729be73ee9a1b0f : SUCCESS in 5m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef8751f6d9e4c8cb681482cdaec4fc5 : SUCCESS in 16m 55s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/e44175c2e3a248ec80d3f7ef001dea57 : FAILURE in 1h 32m 42s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3a8e32814f2845208efab2f03a1bec39 : SUCCESS in 1h 43m 27s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8945328dc2ec44589952ec17872df5d2 : FAILURE in 53m 59s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/ba64cd8db309482cb8ceafc92cde7709 : FAILURE in 1h 08m 56s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/6648d03683b5458188aa20e5a10b55a5 : FAILURE in 1h 10m 22s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/b324889d4f2944f29b466dbc072896e8 : FAILURE in 1h 14m 04s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/5c4eb06a58c5423d9f857f4c8c18fcd6 : FAILURE in 1h 22m 59s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/c0a7b108307344349ed0b91b28569572 : FAILURE in 1h 01m 28s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e08801d961fa4044aa304ce5f732c2e9 : SUCCESS in 1h 34m 04s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e68d63c5d71e4ce0be7b68b69fe68051 : SUCCESS in 27m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/766dccb3a23a4aebae20ec2a492f6a96 : SUCCESS in 1h 16m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/28aedd5d67174416b845570e6822ed96 : SUCCESS in 1h 49m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3dbd901bda5f480d948e7304ac450509 : SUCCESS in 44m 59s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c9f564120caf48798a3e1b09c6cac1e8 : FAILURE in 59m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/16324a0f00944175a65eceed4d45f801 : SUCCESS in 57m 44s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/198b782d50d348099650dd55e4342c06 : SUCCESS in 1h 04m 37s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9736b29f092644bead33b197c4a53889 : SUCCESS in 5m 02s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2c86602a1b204754a24da74592244f71 : FAILURE in 1h 08m 25s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/47bf79a9653543698a039d4b116d73e8 : FAILURE in 56m 40s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7e76891e992043d0a30faa892a6288b0 : SUCCESS in 4m 30s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/764d16785210497497d5ed14cf6f095f : FAILURE in 1h 08m 39s","accounts_in_message":[],"_revision_number":2},{"id":"b34645ddd0b4f0a6954f5c774d32052a27a3cbbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-11-24 00:44:30.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4a72f67d76d0695d809cd074a3f92d9988dacb1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 03:17:26.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5617b5b7008e4a47be7f16ce4b21e109 : SUCCESS in 53m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/402fa10eada74feba834772ba6acbd0a : FAILURE in 1h 10m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bc7a4784ab6473992f708a8a25ae886 : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a7887fa3757466f8bddb1226cc524c5 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/122831aa9b60454e9f35d179f4b96a89 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/997160ca108c42b090c4ef0dbb1dccb8 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d19e1b4ce5641be9d8022ee62fe4fe1 : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecefc5b81ea340a2b203f8f52c61967d : SUCCESS in 15m 12s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d7922940209347adbd6d60b24f2fbad9 : FAILURE in 1h 44m 03s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/a4eff210ceb849c9b97ed4983f28f4ec : SUCCESS in 1h 28m 09s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e50245f46da84520832dd084a3a7d78e : FAILURE in 1h 13m 43s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/0e7a1647a07b4d109185dad1908b132a : FAILURE in 1h 18m 49s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/f7e5bc06375c45e6bca66ce86ee01c4b : FAILURE in 1h 08m 54s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/f1870f6a1e32415fbc4bc90017a3ddac : FAILURE in 1h 25m 13s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/ec34fbf13f6a4c5a95e68d7414d0afd4 : FAILURE in 1h 08m 55s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/638bf4697a3849359592b226de512a03 : FAILURE in 1h 05m 48s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/719304288ca045188838a20b524465e0 : SUCCESS in 1h 33m 18s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b5a9dab5504c413ba24defc0dc6313d7 : SUCCESS in 27m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2cf80a6871604ee58221d93c4dea0ace : SUCCESS in 46m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f891913b1459423495d67cfb0f9ee6f8 : FAILURE in 2h 30m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/eee8275188004ae3b35b2a3cd631a011 : SUCCESS in 45m 53s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e2be8df17e504e89ab75f67d36dbd8ab : FAILURE in 1h 02m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/02bcd1d218344715b73838fd7f413e06 : SUCCESS in 50m 53s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/a1f95a9013f84158a3af1a7a93707e5c : SUCCESS in 54m 54s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/626fc5522b5947d59c1a04291eb99fd4 : SUCCESS in 4m 25s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/424c10d657d442b4a736819e250017dc : FAILURE in 1h 00m 07s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/b19c18bd87dd491ca344f6a80e171c45 : FAILURE in 55m 54s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/6977046641944cfc918fc8f1f08772d8 : SUCCESS in 4m 14s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/0ee913ba8c47457083eba0dbb25a4d21 : FAILURE in 50m 50s","accounts_in_message":[],"_revision_number":3},{"id":"2dea0e4a9dda10393059602bf0e88f9f48e770b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-11-24 06:23:54.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"86ad01cc6258d36a2e3f50284e986edb02495f94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-24 08:28:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/7afbe448f3a847c18a2a32d9669c7fea : SUCCESS in 51m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/593e78cd7988442cb4f8b5eab67022cd : SUCCESS in 1h 17m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b7fbf4abd40e416f9e78b2c9210a8117 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e273227bdca04836b3b801a44bb74be5 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a1e010a8d89484ca34ea7bfcf476f35 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b66b38680c9141828bd705008aef3211 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/444fdde6f99447668debae4f0770d9c4 : SUCCESS in 4m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb1a748725747fca6f199940efb0bf5 : SUCCESS in 12m 57s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/31a7ba1e6c8f42a49b524730331ab7f1 : SUCCESS in 1h 50m 27s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/5b53133eb5864313afb5de4d1a097e73 : SUCCESS in 1h 11m 04s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4785cca6fa2a41f9936d8d20fb6505dd : SUCCESS in 1h 24m 27s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/8837bd9723c847e497a6cf5a384f010e : SUCCESS in 1h 13m 51s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/b48f7d9e719845eca7598034a70b39df : SUCCESS in 1h 52m 53s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/42afd34c8c7e4e159b10954ceb528eed : SUCCESS in 1h 30m 52s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/c253aa7d636c4486b2440a2119a021a6 : SUCCESS in 1h 28m 05s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/94819874f13a4131b201de28a0251820 : SUCCESS in 1h 10m 55s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e7f44a7b4303493e8e653c2ad94d26f6 : SUCCESS in 1h 29m 10s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e12fe3878f274508be67ef32cfc6b0a0 : SUCCESS in 36m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86d7fdad86d24174acdacbe66d8aba76 : SUCCESS in 1h 03m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/17af2138e7cd4976a19d3edc17bec97f : SUCCESS in 2h 02m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b55d3de7aaf34af38615b3d48bec5d06 : SUCCESS in 53m 34s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/4992b59891544ed099770b56f8f1e9ab : SUCCESS in 1h 08m 15s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bb53d743ed6c42f585215d2403a0a5e7 : SUCCESS in 54m 25s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/54f34ebbed9c43fcb790845bedb81d07 : SUCCESS in 53m 27s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/edf43d38935644ad96a3f971c04d0903 : SUCCESS in 4m 52s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5a5687c4762743d4abf3b373be453055 : FAILURE in 1h 24m 50s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/da2c45cbfe3c47db979adb7bb9e4440b : FAILURE in 57m 34s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/56b26f5a08eb4f65b743b81525eba9a5 : SUCCESS in 3m 21s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/066773c6caf549ef98b2e91290ff7eea : SUCCESS in 1h 50m 42s","accounts_in_message":[],"_revision_number":4},{"id":"d1c8f5700f425bc10e384198f12d3fb814e33c91","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-12-01 22:42:44.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"77a3afb922a8f394308dd2a97eaf2fbb864f2e20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-12-02 00:40:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"78b72fbbcc2706e309c6b382822927473e45107a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 03:21:11.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/d489fc33dee144d6a9b6e1486c4efeb0 : SUCCESS in 1h 08m 56s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fdf74992fd574398897962c70775eafc : SUCCESS in 1h 31m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cf78a65ca554add9e7b68ee6dfa6136 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d8db565d70c48db898bf0063dafbce8 : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b4c3e2cfb774b69b6c02d7c0d934b22 : FAILURE in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8db31e82351f441da53623ae65178372 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6d648cb58b542e2a29aea184ed87cb2 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6373f69096349fca3fcb11328eefa5e : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2cf5f99d5db45d59607d2740c3fe790 : SUCCESS in 4m 40s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/4a55a1442d8b43ffa7f64a6dca14d4f3 : SUCCESS in 2h 02m 33s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/9c623bf6d49944b587172e76d562bd1a : SUCCESS in 1h 16m 07s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0f80b1a0f5f1440bae0fc0c05ff21640 : SUCCESS in 1h 24m 01s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/e0f2d30c79024eee88a4c6f2401df8b7 : SUCCESS in 1h 17m 11s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/cfec9000d0bc447b8cb89a5e7fa8ff0c : SUCCESS in 1h 49m 31s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/5a498496e8d24c709ac96ff3ba919ab1 : SUCCESS in 1h 43m 40s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/2dfff1a6f48a412f840e941662738113 : SUCCESS in 1h 23m 45s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/6f101c04ff224b4bad087f6be9bc5845 : SUCCESS in 1h 28m 49s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/2dcafc8c62744ef89c62314040eaaf07 : SUCCESS in 1h 37m 11s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/81dd6b7c7240479cb6530c4087624d91 : SUCCESS in 34m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/27470b1ba38349a9849696f5f1549fc7 : SUCCESS in 52m 21s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2292fc54f99e47d2bdba5c8f3d1d79a9 : SUCCESS in 2h 32m 41s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/416b578510fe49a29883ef7e2d041b0e : SUCCESS in 50m 53s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9bf22d50215a4851a300051596a9f530 : TIMED_OUT in 2h 16m 08s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/62fd7ab7771c413992ff4459038812e0 : SUCCESS in 1h 03m 03s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/667927b5550e4b2184f7ffa625b2f745 : SUCCESS in 58m 36s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e62c29bbbd354f38b305faeb08c29ec7 : SUCCESS in 4m 03s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c54a5cf61a10447aa2991063b0a2d75b : FAILURE in 1h 24m 11s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/d4ded9e3cf1644eb90dd19bdd57bd718 : FAILURE in 59m 59s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2450ef51df1b4b4698736d5bd116f027 : SUCCESS in 4m 19s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/febbfed04a044aa6bbd4d8e61809b0b6 : SUCCESS in 1h 33m 56s","accounts_in_message":[],"_revision_number":5},{"id":"f960e0d14cea040b45ebe9d8342bec9d8666119e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-12-02 03:56:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"7d7d4e6f59a75dd96aa1b8daaa1a4b453a2fedcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 06:31:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a34099acd4d84ec2bd1ccb7b927d1819 : SUCCESS in 56m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c7c377347e13408a8fae39e79f3c3fe1 : SUCCESS in 1h 28m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cf4af094d8c4f1c90479fdd6ee69268 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ece831bd0c3243659925f330ea67c477 : SUCCESS in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc894c4857e040a9b41f4fb9c5da48a2 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cda62c4328244aa68aee4d5517c3f5a8 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e6a466df874a42d382e8f3d0be6469e8 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0b11268b09f64f0489d5a3785691d350 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67b8c2744dc94964a43678839e26228b : SUCCESS in 4m 45s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d64f2419694040809c721edb85752a64 : SUCCESS in 1h 49m 37s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/5b080061a7b244c58c7558c01d182aa2 : SUCCESS in 1h 08m 39s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ae0234d637ff472a98ea07e64144fd0b : SUCCESS in 1h 25m 55s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/64438a0d404f4517b420fd40882013da : SUCCESS in 1h 16m 56s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/af3b4c3ac9dd47b79a9675693c493f87 : SUCCESS in 1h 24m 49s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/74c6e1ccb8674183b6652845b890d768 : SUCCESS in 1h 24m 22s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/7d20d60ce9074a5bacbccb8bb75416aa : SUCCESS in 1h 41m 18s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/9fbb0e244ed34e24b9ad9574093765ec : SUCCESS in 1h 35m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/550e3b7a37bb4929ab5da6092068017e : SUCCESS in 1h 41m 24s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ead35c386d75477da452f7bec81a6c1e : SUCCESS in 30m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d486f6b28d749109becad9bf5347a0a : SUCCESS in 49m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/385cfaca72054d618966f112f263e7ac : SUCCESS in 2h 33m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fa310cdbdb1241118df9754b913d03b3 : SUCCESS in 37m 39s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9b135feb7c204a90a0b79939687ed1da : SUCCESS in 1h 01m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/db15e0f810934ef8822110ff816bd017 : SUCCESS in 53m 52s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/5258efbff0fa4d4ebafe60a3e78d14f6 : SUCCESS in 52m 38s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/86eed15adc5d406e8f9d0c63d18c9f03 : SUCCESS in 4m 32s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3579d3079c5f42e3980c63ae157b7159 : FAILURE in 54m 00s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/9b5c37a9432d494488a763437a7d1f21 : FAILURE in 1h 00m 07s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8f3d1e88c69b4dba8e7d254a8484722a : SUCCESS in 3m 59s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/235b9ff607bf44299d4741255bef2e28 : FAILURE in 1h 16m 56s","accounts_in_message":[],"_revision_number":6},{"id":"90462c815dbd555142abd8caa0e4903b9598d2f4","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-12-02 08:59:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7138cd6ac6e80f19d75ec67330541452e562d606","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-02 11:19:59.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/f7cc6d8e9fe64a439b393da95600aedf : SUCCESS in 57m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4bfe60a5404a4b2c8a3d59908a276d52 : SUCCESS in 1h 26m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/839070d8788b45e9b1423738f0bc906d : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3950796c908484cb6abdb84e02749dd : SUCCESS in 13m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/381b0390a00647ef83435d6f9569d8b8 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7a116c63183e407e93859d1f3c32e26a : SUCCESS in 4m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/7d89c51d2ed343d68a518a2c71eb375f : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4376fad4390c4aa193c46eb21eaf2be9 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/85492c64c2b94f37950c8b8f7b2601b0 : SUCCESS in 4m 59s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/0138b408bd5f430abf604b57413d587f : SUCCESS in 1h 46m 09s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/64544c444aa443008d3e28c0f07c8d07 : SUCCESS in 1h 15m 29s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/88caf04968b845e4b0906690a52ec84e : FAILURE in 1h 24m 22s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/afa1ed07c64a48288de7d621fe6f425b : SUCCESS in 1h 12m 16s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/6e516ae08c224d73979f9ee2dbe89d49 : SUCCESS in 1h 21m 44s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/5739d15c9a764df08b7f1d43e1195ef0 : SUCCESS in 1h 45m 03s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/932847d8cd804b329d7415f214d2a9ff : SUCCESS in 1h 15m 17s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/55a3015c042d4c80ac61feff404f6451 : SUCCESS in 1h 29m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0992e0e249144a37945aeed3d1ee4709 : SUCCESS in 1h 41m 51s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/31bf3856621a42979eeabaab8e9e825c : SUCCESS in 30m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b566dfddda66474ab77653bb403586d5 : SUCCESS in 48m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ed43aa36fd584117816351f9995d22f5 : SUCCESS in 2h 17m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/32993713ca0a469e8fcec27b4a33d553 : SUCCESS in 49m 40s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/e0ebfc8d748944c0b5922c6c019b5d23 : SUCCESS in 1h 12m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/172d94435ded4e7a8a55c7feb5ded740 : SUCCESS in 52m 54s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/025fd02c4ab1460db715e201dfafe026 : SUCCESS in 1h 00m 24s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/90040af7de7a40c7bc132c3af7bc7f53 : SUCCESS in 4m 50s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f5e227ebf96149e1a5509d5d05f906c2 : FAILURE in 1h 02m 16s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/1070ecb389844941919c9e0959098a02 : FAILURE in 1h 01m 40s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/65f45cc212194f12aaee9b68fc58ab83 : SUCCESS in 3m 46s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/df947a37cfe14324ad69747ec03126b7 : SUCCESS in 1h 42m 24s","accounts_in_message":[],"_revision_number":6},{"id":"3ec75e079f9379eccad374676aef2757f5383d83","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-12-02 22:31:21.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2b74740af5f2d420c8357d55bfd7f3eccffdda82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-12-03 00:43:21.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"5386f331f355ef621a3a1915bda70a288d8387ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-03 03:23:30.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/ab7466288c6048aca4b0e506dba719b8 : SUCCESS in 39m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7ddac331570c47998bbe888875310637 : SUCCESS in 1h 07m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d2cb941a9f848b5b148fa6a1ac35a12 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3808dc186664b6eb78b7495c1125872 : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c0040e6e8dd47caaa5442d6ede30652 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0fc94b974d5e457597c6955f944f02f4 : SUCCESS in 5m 20s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ded00c42ea8b44aba96e026be298b6c8 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49ed2a9c8a604b0e872aa25a8deac87c : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c09f09f70c7b4f42a35e893f0cbb09a1 : SUCCESS in 4m 56s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/d3719c2189744ccfb83c199a2542164c : SUCCESS in 2h 03m 19s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/889ed9be247d4e0e8327678270f5db34 : SUCCESS in 1h 28m 18s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/95e7367158f64424b7eaa7b305dc2ced : SUCCESS in 1h 28m 22s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/cfe0f5d02a08412893b975d74e0368df : SUCCESS in 1h 16m 14s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/4bd340edbff448d4b633129dfddf29c9 : SUCCESS in 1h 31m 02s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/8170dec2c1b34e048b2ae6055a36219a : SUCCESS in 1h 52m 25s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/b685fbb356e549738bea498910bad915 : SUCCESS in 1h 37m 53s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/ba96add9d78e4cb3abb7fbb11604f4fb : SUCCESS in 1h 32m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/83a74dcc767b43a0be2fbc5a254d0370 : SUCCESS in 2h 14m 39s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/d73f170980bd48eca765b311179bb9a2 : SUCCESS in 35m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/317db8d07c9f4ce088a5f2fb575818df : SUCCESS in 1h 01m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c732e7ba0b2a40edad51af251d3f4070 : SUCCESS in 2h 34m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/15f8dfe90f7742db8997aa020a1596d9 : SUCCESS in 42m 55s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/13e043c78c404a519b2faa0ea48d3be4 : SUCCESS in 1h 33m 29s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/57b5ad3dab7343cbb90e04f234dcf496 : SUCCESS in 48m 24s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/8ee707f3da3b429eb011824d3a61ab2a : SUCCESS in 1h 10m 21s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/824e59fa8069487d9b59499e72cadfe5 : SUCCESS in 4m 30s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/46c135f20b6b41a184b52ea57bf19af3 : FAILURE in 1h 06m 03s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/d342fb094a594ef489c0acbf328df1b6 : FAILURE in 57m 09s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/bbc4a9f05c45400a98d5a13a2e1e1760 : SUCCESS in 3m 23s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/d10d602da5a84bbca56fdb459d70e976 : SUCCESS in 1h 23m 48s","accounts_in_message":[],"_revision_number":7},{"id":"99806c222021a101bb07959788cf6fe47a2f7ce3","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-12-14 10:37:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1a8f77950a532a571b71fc847660841ca9eb7f51","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2021-12-14 11:08:36.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"72b34f6b57aafc872e91bde373c9b6ca3e692fda","author":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"date":"2021-12-16 09:10:13.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bad46bad3308324dee718282d642fbe19b540746","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2021-12-16 09:25:33.000000000","message":"Patch Set 7: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"fd16f0653c3791b9915912b683890629ae55e721","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-24 05:47:51.000000000","message":"Abandoned\n\nNo activities on this since long time, feel free to restore it if needed.","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6d38fb282c047d95338a83bae535ae1e73470f1e","revisions":{"395ddf8537f8d9bd63d5921b6411f1e1f7596a28":{"kind":"REWORK","_number":1,"created":"2021-11-23 08:01:42.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/61/818861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/61/818861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/61/818861/1"}}},"commit":{"parents":[{"commit":"ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead","subject":"Merge \"Remove race from wait_for_interface_detach waiter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-23 07:58:24.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-23 07:58:24.000000000","tz":480},"subject":"Test delete server with multiattached bootable volume","message":"Test delete server with multiattached bootable volume\n\nChange-Id: Ide9836afba4a12b2ca780cf447c747797023651c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/395ddf8537f8d9bd63d5921b6411f1e1f7596a28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/395ddf8537f8d9bd63d5921b6411f1e1f7596a28"}]},"branch":"refs/heads/master"},"d6ad70a4f23e02340e5f9bf5c34a43a0f4e1b3c4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-23 08:09:16.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/61/818861/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/61/818861/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/61/818861/2"}}},"commit":{"parents":[{"commit":"ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead","subject":"Merge \"Remove race from wait_for_interface_detach waiter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-23 07:58:24.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-23 08:02:01.000000000","tz":480},"subject":"Test delete server with multiattached bootable volume","message":"Test delete server with multiattached bootable volume\n\nCreate a bootable multiattach volume with delete_on_termination\u003dTrue, create server1 from the volume, then attach it to server2, and then delete server1 in which case the volume won\u0027t be deleted because it\u0027s still attached to server2 and make sure the volume is still attached to server2.\n\nChange-Id: Ide9836afba4a12b2ca780cf447c747797023651c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d6ad70a4f23e02340e5f9bf5c34a43a0f4e1b3c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d6ad70a4f23e02340e5f9bf5c34a43a0f4e1b3c4"}]},"branch":"refs/heads/master"},"e76719dc52196379d7c32ce214add70d8e2d3822":{"kind":"REWORK","_number":3,"created":"2021-11-24 00:44:30.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/61/818861/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/61/818861/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/61/818861/3"}}},"commit":{"parents":[{"commit":"ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead","subject":"Merge \"Remove race from wait_for_interface_detach waiter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-23 07:58:24.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-24 00:43:49.000000000","tz":480},"subject":"Test delete server with multiattached bootable volume","message":"Test delete server with multiattached bootable volume\n\nCreate a bootable multiattach volume with delete_on_termination\u003dTrue, create server1 from the volume, then attach it to server2, and then delete server1 in which case the volume won\u0027t be deleted because it\u0027s still attached to server2 and make sure the volume is still attached to server2.\n\nChange-Id: Ide9836afba4a12b2ca780cf447c747797023651c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e76719dc52196379d7c32ce214add70d8e2d3822"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e76719dc52196379d7c32ce214add70d8e2d3822"}]},"branch":"refs/heads/master"},"916a189f4c6c7dba1673cd1395d5512d0028566e":{"kind":"REWORK","_number":4,"created":"2021-11-24 06:23:54.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/61/818861/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/61/818861/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/61/818861/4"}}},"commit":{"parents":[{"commit":"ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead","subject":"Merge \"Remove race from wait_for_interface_detach waiter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-23 07:58:24.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-24 06:23:16.000000000","tz":480},"subject":"Test delete server with multiattached bootable volume","message":"Test delete server with multiattached bootable volume\n\nCreate a bootable multiattach volume with delete_on_termination\u003dTrue, create server1 from the volume, then attach it to server2, and then delete server1 in which case the volume won\u0027t be deleted because it\u0027s still attached to server2 and make sure the volume is still attached to server2.\n\nChange-Id: Ide9836afba4a12b2ca780cf447c747797023651c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/916a189f4c6c7dba1673cd1395d5512d0028566e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/916a189f4c6c7dba1673cd1395d5512d0028566e"}]},"branch":"refs/heads/master"},"18e416ac8ab79fc1cf7424c2f5a51948061bbf3f":{"kind":"REWORK","_number":5,"created":"2021-12-02 00:40:57.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/61/818861/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/61/818861/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/61/818861/5"}}},"commit":{"parents":[{"commit":"ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead","subject":"Merge \"Remove race from wait_for_interface_detach waiter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-23 07:58:24.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-12-02 00:37:45.000000000","tz":480},"subject":"Test multiattach volume booted server does not delete multi attachments","message":"Test multiattach volume booted server does not delete multi attachments\n\nCreate a bootable multiattach volume, create server1 from the volume, then attach it to server2, and then delete server1 in which case the volume won\u0027t be deleted because it\u0027s still attached to server2 and make sure the volume is still attached to server2.\n\nChange-Id: Ide9836afba4a12b2ca780cf447c747797023651c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/18e416ac8ab79fc1cf7424c2f5a51948061bbf3f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/18e416ac8ab79fc1cf7424c2f5a51948061bbf3f"}]},"branch":"refs/heads/master"},"f8fca90e5cda4fdcbc2084458d0f1681b2b4f896":{"kind":"REWORK","_number":6,"created":"2021-12-02 03:56:17.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/61/818861/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/61/818861/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/61/818861/6"}}},"commit":{"parents":[{"commit":"ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead","subject":"Merge \"Remove race from wait_for_interface_detach waiter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-23 07:58:24.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-12-02 03:54:51.000000000","tz":480},"subject":"Test multiattach volume booted server does not delete multi attachments","message":"Test multiattach volume booted server does not delete multi attachments\n\nCreate a bootable multiattach volume, create server1 from the volume, then attach it to server2, and then delete server1 in which case the volume won\u0027t be deleted because it\u0027s still attached to server2 and make sure the volume is still attached to server2.\n\nChange-Id: Ide9836afba4a12b2ca780cf447c747797023651c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f8fca90e5cda4fdcbc2084458d0f1681b2b4f896"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f8fca90e5cda4fdcbc2084458d0f1681b2b4f896"}]},"branch":"refs/heads/master"},"6d38fb282c047d95338a83bae535ae1e73470f1e":{"kind":"REWORK","_number":7,"created":"2021-12-03 00:43:21.000000000","uploader":{"_account_id":31280,"name":"lkk","email":"likangkang01@inspur.com","username":"likangkang"},"ref":"refs/changes/61/818861/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/61/818861/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/61/818861/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/61/818861/7"}}},"commit":{"parents":[{"commit":"ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead","subject":"Merge \"Remove race from wait_for_interface_detach waiter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad8f599b32e875c438bd49b8d81bfcd9d4eb8ead"}]}],"author":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-11-23 07:58:24.000000000","tz":480},"committer":{"name":"likangkang01","email":"likangkang01@inspur.com","date":"2021-12-03 00:41:27.000000000","tz":480},"subject":"Test multiattach volume booted server does not delete multi attachments","message":"Test multiattach volume booted server does not delete multi attachments\n\nCreate a bootable multiattach volume, create server1 from the volume, then attach it to server2, and then delete server1 in which case the volume won\u0027t be deleted because it\u0027s still attached to server2 and make sure the volume is still attached to server2.\n\nChange-Id: Ide9836afba4a12b2ca780cf447c747797023651c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6d38fb282c047d95338a83bae535ae1e73470f1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6d38fb282c047d95338a83bae535ae1e73470f1e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
