)]}'
{"id":"openstack%2Ftempest~819496","triplet_id":"openstack%2Ftempest~master~I9891b43948aba1acaa0596585246825599ead1e7","project":"openstack/tempest","branch":"master","topic":"bug/1951623","attention_set":{},"removed_from_attention_set":{"8864":{"account":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"last_update":"2023-07-18 01:49:48.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I9891b43948aba1acaa0596585246825599ead1e7","subject":"Assert Nova fix for bug 1951623 works","status":"ABANDONED","created":"2021-11-26 21:15:05.000000000","updated":"2023-07-18 01:49:48.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e876d74e80b6ef82e599edb23dde85326d8e6d52","_number":819496,"virtual_id_number":819496,"owner":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-11-27 14:58:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-27 00:03:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd179c6c7fa3f950f5d4c6943532fbc11330bab0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-11-26 21:15:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c456feb4f800e104fe219e06a17ef6b23305109d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-27 00:03:49.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/5134cd9413be4b2a84af3edc47ee026d : FAILURE in 53m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7107df24d8644bb3bf99055080cc4e57 : FAILURE in 1h 18m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94e0c4146a3f430984d5af4cbabf5697 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ca7c749380b49e4a5591cbbfa4e9833 : SUCCESS in 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38b30ceabd3748f7bb3815c8a234989f : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d6ce67fdf69409db5fc9e830b6648e5 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6e8fbaa054d24187babdaa2f2cc47c45 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68077f7c3e524f7dab632bf6d632dd85 : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fc6f2801ab24b9699f954b60f2a2217 : FAILURE in 5m 37s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/bf7c0a471ad74e189a8b265a3be45554 : FAILURE in 1h 25m 17s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/8d9bc1762eef4c5f94dc9cea68d1223f : FAILURE in 1h 28m 34s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6c197d2aede8439fb0bd87ab2f70095b : FAILURE in 1h 55m 10s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/27c548b9828f4224a35eebd51f7c544d : SUCCESS in 1h 18m 16s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/8a8aa42a335f4dae89d75b68bd25822f : SUCCESS in 1h 23m 00s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/7638c1e638574bf88b85a18141235099 : SUCCESS in 1h 31m 32s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/f6f4e7ef088f4e9bb7df767aa5631519 : SUCCESS in 1h 10m 33s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/474f188e4e0142a69b183625b2eb0525 : SUCCESS in 1h 20m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8df92fe0db3b481d8726613f6475fea8 : FAILURE in 1h 39m 14s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/1f05a37667794d7a8704c821f0102c62 : SUCCESS in 26m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae0b6177ca7148fd84ca3eda235498f7 : FAILURE in 35m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8dbb4f2cd4d5493989b90c92e4891e1a : FAILURE in 2h 41m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5e4ff13054c840f49ab67f702ef9f305 : SUCCESS in 47m 29s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bf8fb3caa47640c69b35be96b39fd740 : FAILURE in 1h 30m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8dbdd96ff9e244b78a92f0755eea64f8 : SUCCESS in 53m 05s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/6c9e8730e77245378b6ddeaa1d889223 : FAILURE in 58m 32s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b9dae11478504f079cae12af5c0afd0a : SUCCESS in 4m 16s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/696b72f288f8459aa09b2727ebd19c2e : FAILURE in 1h 06m 14s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/d8d9a5bd74bd485eba4a778a2c6de2f4 : FAILURE in 1h 12m 37s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/62eefc21611a4f6caba4f882b54981c6 : SUCCESS in 3m 57s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/0827965347764ea4b6c660e8973d2d69 : FAILURE in 1h 27m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9da6b043dd57e4152d35b202b6244335ee40396d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-11-27 03:46:07.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"530c9a3ad5e0c7cc009937658b5bec67f3358014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-27 05:46:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/2f9d311d3dbe4091a0d44cf7808be847 : SUCCESS in 57m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f0a3ddab6654eacbf4a41a3f58325d3 : SUCCESS in 1h 12m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93cd08c5c61944049885882f898b1ace : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed0e73d5dae74345a86bffcdb5a14e8a : SUCCESS in 13m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d5dd3eab7a34302a474f4e732f4910d : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8b494ca960494cf8a82670d4e348e64c : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a5fb3deb6c6545ca8528c371e141a51e : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d25335e72cf2420d8e77f546ed8601d2 : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6082bf49cdb74bebb523e8f7461d1b29 : SUCCESS in 5m 03s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/300a521ad3c9427b817bc6199090c823 : SUCCESS in 1h 26m 59s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/bbc54154445e45039245ffc20062a749 : SUCCESS in 1h 15m 53s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/44eabdbdccef4d6dab7890f6386506f9 : SUCCESS in 1h 16m 16s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/3fe306a469734e03bb25be4816846908 : SUCCESS in 1h 12m 21s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/0aae86c40ce84fd188fa24c053b08eb5 : SUCCESS in 1h 30m 56s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/358c0f6e3e634702b873fc77176cd7af : SUCCESS in 1h 19m 32s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/da99a1a91a3044268afc65b479874192 : SUCCESS in 1h 17m 57s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/9806289485974ca48349148a9b72e9a3 : SUCCESS in 1h 22m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e33a9fa502cb468aa6d75b12b46904ca : FAILURE in 1h 17m 31s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b8b84ffdf6e942a681ecc4380a527a7d : SUCCESS in 24m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e400175eb27b4022bb4d9c027001fe84 : SUCCESS in 51m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eef509987e664c7f9e7834b580fcad2f : SUCCESS in 1h 57m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6070f4ff2ffb4ced856bb0294e2987bc : FAILURE in 42m 19s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/26353da437384825bcefe58f8e72d0b8 : SUCCESS in 1h 11m 39s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/234b6b781b5a4beabbc9f393a400f721 : SUCCESS in 58m 32s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/fe0054fc94e44c449cdd2c8e7b7851fb : SUCCESS in 56m 56s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e46c86027a264117b505189278cfce1e : SUCCESS in 5m 02s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/14e88c78a4cb4c74ad6b61ff3c2eae07 : FAILURE in 1h 28m 03s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/8a8e9d0db6294e7ca6523246cd875b32 : FAILURE in 1h 04m 12s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/147462d4246342aca8fe638c7f3be418 : SUCCESS in 3m 59s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/3e6d7cc5930a469facf277ebfa4b1469 : SUCCESS in 1h 24m 07s","accounts_in_message":[],"_revision_number":2},{"id":"f2e884868afe74cc97a5011afa90245809c97cf4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"date":"2021-11-27 14:58:01.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"fefab7f22d888a64dae9e37b7658afe94e3d21e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-27 14:58:19.000000000","message":"Patch Set 3: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":3},{"id":"e876d74e80b6ef82e599edb23dde85326d8e6d52","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-18 01:49:48.000000000","message":"Abandoned\n\nThanks. No activity on this since long, feel free to restore it if it is still needed.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"375e9297b93a8962e3a2b4bd5549dddd8351b7a0","revisions":{"5fbcbee05e8e7d1d22ee2640dfa5cd13ece5d49a":{"kind":"REWORK","_number":1,"created":"2021-11-26 21:15:05.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/819496/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/96/819496/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/819496/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/819496/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/819496/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/96/819496/1"}}},"commit":{"parents":[{"commit":"5eba9435dd0855f755096046d9240bf19369c7f7","subject":"Regression test for bug 1951623","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5eba9435dd0855f755096046d9240bf19369c7f7"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-11-26 21:07:44.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-11-26 21:13:46.000000000","tz":-300},"subject":"Assert Nova fix for bug 1951623 works","message":"Assert Nova fix for bug 1951623 works\n\nNow that Nova bug 1951623 is fixed, fix the Tempest test.\n\nChange-Id: I9891b43948aba1acaa0596585246825599ead1e7\nRelated-bug: 1951623\nDepends-on: https://review.opendev.org/c/openstack/nova/+/819494\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5fbcbee05e8e7d1d22ee2640dfa5cd13ece5d49a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5fbcbee05e8e7d1d22ee2640dfa5cd13ece5d49a"}]},"branch":"refs/heads/master"},"b3461c29c09ba98faf1509e88e87ac3a1a1140e1":{"kind":"REWORK","_number":2,"created":"2021-11-27 03:46:07.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/819496/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/96/819496/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/819496/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/819496/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/819496/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/96/819496/2"}}},"commit":{"parents":[{"commit":"e3a1e9e3cefa2b3b5d0badfc9f9eff0b3a1b97f3","subject":"Regression test for bug 1951623","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e3a1e9e3cefa2b3b5d0badfc9f9eff0b3a1b97f3"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-11-26 21:07:44.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-11-27 03:45:09.000000000","tz":-300},"subject":"Assert Nova fix for bug 1951623 works","message":"Assert Nova fix for bug 1951623 works\n\nNow that Nova bug 1951623 is fixed, fix the Tempest test.\n\nChange-Id: I9891b43948aba1acaa0596585246825599ead1e7\nRelated-bug: 1951623\nDepends-on: https://review.opendev.org/c/openstack/nova/+/819494\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b3461c29c09ba98faf1509e88e87ac3a1a1140e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b3461c29c09ba98faf1509e88e87ac3a1a1140e1"}]},"branch":"refs/heads/master"},"375e9297b93a8962e3a2b4bd5549dddd8351b7a0":{"kind":"REWORK","_number":3,"created":"2021-11-27 14:58:01.000000000","uploader":{"_account_id":8864,"name":"Artom Lifshitz","email":"notartom@gmail.com","username":"artom"},"ref":"refs/changes/96/819496/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/96/819496/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/819496/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/819496/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/96/819496/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/96/819496/3"}}},"commit":{"parents":[{"commit":"6cf0228f9ec6ae494e3b4f32f9a911ff1df3aefc","subject":"Regression test for bug 1951623","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6cf0228f9ec6ae494e3b4f32f9a911ff1df3aefc"}]}],"author":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-11-26 21:07:44.000000000","tz":-300},"committer":{"name":"Artom Lifshitz","email":"alifshit@redhat.com","date":"2021-11-27 14:57:54.000000000","tz":-300},"subject":"Assert Nova fix for bug 1951623 works","message":"Assert Nova fix for bug 1951623 works\n\nNow that Nova bug 1951623 is fixed, fix the Tempest test.\n\nChange-Id: I9891b43948aba1acaa0596585246825599ead1e7\nRelated-bug: 1951623\nDepends-on: https://review.opendev.org/c/openstack/nova/+/819494\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/375e9297b93a8962e3a2b4bd5549dddd8351b7a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/375e9297b93a8962e3a2b4bd5549dddd8351b7a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
