)]}'
{"id":"openstack%2Ftempest~819594","triplet_id":"openstack%2Ftempest~master~I43e2269d4b1c425119c86919eb869941d621c6de","project":"openstack/tempest","branch":"master","topic":"log-warn","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2022-03-02 13:15:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I43e2269d4b1c425119c86919eb869941d621c6de","subject":"Use LOG.warning instead of deprecated LOG.warn","status":"MERGED","created":"2021-11-29 06:48:37.000000000","updated":"2022-03-02 13:16:55.000000000","submitted":"2022-03-02 13:15:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819594-log-warn","meta_rev_id":"77019bbb8a593a1ac9505e8d79d1d807a2e2bc99","_number":819594,"virtual_id_number":819594,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 13:15:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2022-03-02 10:51:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2022-02-14 02:15:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2022-03-02 10:51:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-29 11:39:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-11 15:20:43.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2021-12-22 06:48:18.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2022-03-02 10:51:24.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"25ba1562d772c34c5cc494156e6f2b0e9a05b6bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 06:48:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"793eaffaf58ceba4e6b897f11360a7837856364f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 06:50:25.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"52b555a6259b787672de4d1dbefa95b9270003d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 09:24:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6452178df69c0012ba7613d8585a55cb8386b5bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 11:39:00.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b4f9636d84334c76a0f069d2f9ec2934 : SUCCESS in 51m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/56c55b2199074ad8b609211b1b920af6 : SUCCESS in 1h 17m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f7f91d951a544888c841d199d7744b9 : FAILURE in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bb6db7bb9d746aa8e2f6894155f43f9 : SUCCESS in 13m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b09d1497d4494b45bb5dfde0e267bb9f : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49f1d55938a94541888e259ad6fe4dfa : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87fbef966ee24bdda0982334b3200022 : FAILURE in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/dbcfd07b94004392a2a21e7a4f1bffa5 : FAILURE in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91b93ced4fa2444ba69322e692434872 : FAILURE in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/09b49db8b3ca41509f8360f9d7d7be9b : FAILURE in 4m 39s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/743acea977754ca1bc6f8e6d1a7724ae : SUCCESS in 1h 41m 51s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/e81b00f36f574b089130bf063d0c87b8 : SUCCESS in 1h 18m 03s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ae30dd7940fa4c2b9348a86fd67383bb : SUCCESS in 1h 36m 24s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/9b14e3c3676b486e88ffb2fad9680491 : SUCCESS in 1h 25m 57s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/824761f178fb49f4ba6b75f981ebee8b : SUCCESS in 1h 25m 49s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/c46b6bf24daa481bbd73f9a457ae5aa5 : SUCCESS in 1h 23m 23s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/c232166f90474bcbbcbd53238a464263 : SUCCESS in 1h 15m 05s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/8db9a1b945584cb0ad772dfa77427b33 : SUCCESS in 1h 17m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f08a44a4c39e456faf358c04bbb6d58d : SUCCESS in 1h 37m 29s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/2331476c88cd4fbbb5bcadd01ca298af : SUCCESS in 27m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a16d6bb0dd474e1b908e7ad8764e4168 : SUCCESS in 43m 35s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/45f7215a14c14008b675b8285014bfc0 : SUCCESS in 2h 10m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e8985eb975634faab0a2a7aa66207f7c : SUCCESS in 47m 10s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0c639e222b234714bb5341024fe82cb3 : SUCCESS in 1h 16m 30s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6d058799fe445bdaeaec9fceed68e12 : SUCCESS in 1h 07m 30s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/45bd3ebf2b894d55be0943912d958576 : SUCCESS in 56m 28s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d5c3303bca1443459939e450f33b68ec : SUCCESS in 4m 47s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0396a6bbc4954571a351ffc726a22bc9 : FAILURE in 1h 16m 20s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/45edf5bc8ae744dfb2a2f43853a2e16e : FAILURE in 1h 01m 13s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/526a947de0c244eeb7c95b019eff78db : SUCCESS in 3m 48s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/3a68d761f7474b0b8c31741b3a1f71dc : SUCCESS in 1h 41m 27s","accounts_in_message":[],"_revision_number":3},{"id":"0b8ede538dc7237b7e21d90dce3465833733b0a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-29 12:05:20.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"bc883bbe8e1184927e9725cd0dafc91f5456e5c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-29 14:32:28.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/b1f97f6a36874f5bb9b02dd1ac0eced0 : SUCCESS in 52m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4927c3b935264ce5aad61da96addfdfb : SUCCESS in 1h 22m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec7b142fbc66440580e876a9723572f6 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61958cdbf56148468ca15b8aee2e7241 : SUCCESS in 16m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e276bf88a2f6470eaaa8d404d38bec98 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34025c20c9e5442eb1c99e1e09c716c1 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5aff5f216b44c9e9129c51b5ac90c22 : FAILURE in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/eb53e7bba5674aee9ce5150076cb8cbc : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f986ef3c7fc4f0ba5d80b9b5e33c048 : FAILURE in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ec9db2bad2a947d2b67dc1c5445c7039 : SUCCESS in 4m 29s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/946f985677304971b924378bd847fe00 : SUCCESS in 1h 28m 13s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/bf6cd72404cd4b89962fb43c39b6fff9 : SUCCESS in 1h 30m 37s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6d47451d506247ffba9f6c7fee7b8f2f : SUCCESS in 1h 10m 42s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/4ccf9182e0624d9dbf853c451a073a7a : SUCCESS in 1h 23m 20s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/4a1b07a837c6402fb6b1a80f74f82f83 : SUCCESS in 1h 05m 36s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/4b7a2891f5f04b589268264beae3ef72 : SUCCESS in 1h 19m 22s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/276a84cb578a4bfaa99c796240e0d95c : SUCCESS in 1h 27m 25s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/d1aba0411bbb4d08b694bb865dcac435 : SUCCESS in 1h 11m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/0364f07ac24b4f798e838ee5a0c3367e : SUCCESS in 1h 18m 58s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c17c789b59ec452584cb1e5004147822 : SUCCESS in 31m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65a5dc58046c4d9aaed2a934f0b17348 : SUCCESS in 52m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4390077b0e8043e7971d42daa6085283 : SUCCESS in 2h 19m 42s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3877bb990b3e4c698dce5f14b7c5c6f9 : SUCCESS in 59m 58s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c845ae09d9874ff299fd9f5c98d7ccbe : SUCCESS in 1h 29m 45s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f59f1f4eedc742fdb47e52a1b4632a06 : SUCCESS in 1h 01m 37s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/26c03a91564e4adf92fbeb1e8ab5a226 : SUCCESS in 1h 00m 05s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/19fdedc30a4b415c9f604ef9ec5e6c8d : SUCCESS in 4m 24s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3b2241ef335a4c3994c39d946a7e0d19 : FAILURE in 1h 21m 24s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/7005ff858dba4fb989ba8fbb1ef16818 : FAILURE in 1h 14m 09s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b15c97789f714b3287652a7582ddb440 : SUCCESS in 3m 20s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/54b045df46354c1e8bd48692ac50bfc3 : SUCCESS in 1h 26m 11s","accounts_in_message":[],"_revision_number":4},{"id":"b826c250d4b421cc4c62c3efb301b348ad4247c2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-06 14:29:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3cd6d847a1481cdcd8b97b27cf3812eb96c36cf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 16:45:36.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/76209d151ada4e4a9f6e026da3e9cc65 : SUCCESS in 56m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/72541725ba4945c6b0d4d26211c92aaf : SUCCESS in 1h 31m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f16d5551dfb435db84179bbd8c27950 : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/333e125af41048bea414c281963cdbf4 : SUCCESS in 15m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d9667e12a8743a3a47f8d6f6e70ba49 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3f2e79bdecb49f68bda5048260b2e90 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07be5d4eaec840f2b890f4f0db8f2ffe : SUCCESS in 4m 57s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/85d6ed80f4f54a7395db7613152d4091 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0498f21a263b48c3b31e746863af1755 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb6664ddae8b4dd4ab95d42a0a117dae : SUCCESS in 5m 29s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/2ad14bb379d24ff98d5b3a18b83e05b4 : SUCCESS in 1h 39m 52s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/d7434a1c2fae4a9ab54e860c0b9cd680 : SUCCESS in 1h 13m 56s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/31ed0c8582fd4081a0295ffa09976747 : SUCCESS in 1h 19m 05s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/ee1aea6ab7d04ea5a342733c128d33f9 : SUCCESS in 1h 20m 31s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/13a78ad15e124dfa8e7278fcfcc56830 : SUCCESS in 1h 25m 25s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/debd80994bd84b3e8c63924d3f93587d : SUCCESS in 1h 26m 20s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/77f23b0786fb4b14a3907fe886fd76f7 : TIMED_OUT in 2h 08m 43s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/fddf21f124164642964fd62e6f744679 : SUCCESS in 1h 07m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/91f024317e7e405d9aa0350a5d6b0f32 : SUCCESS in 1h 41m 26s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ebecea4abdcb424b90097676233f605a : SUCCESS in 33m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8c39e578ecc5422f847b81459d640cf3 : SUCCESS in 44m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/50a1ce3ab89f4bee9c280f26dd2bc627 : SUCCESS in 1h 56m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1c5f16830eed4262b44b95518d131d5b : SUCCESS in 39m 14s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/da4f3f7b40f94b4c82aecec083bd338b : SUCCESS in 1h 21m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a6d1847fe46943e184581adb4a639256 : SUCCESS in 1h 02m 08s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/1d9958b9aa4e4c708f9717d8421b1664 : SUCCESS in 45m 24s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/facb418862ac4fc2af2f46695222680d : SUCCESS in 4m 30s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a016a38503464ebd8a7170c821b9dd30 : FAILURE in 1h 15m 48s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/b3e8004b350d4007a086e909adf7d73d : FAILURE in 51m 51s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d82971206f6847198ecaa8dd982aa9d8 : SUCCESS in 5m 13s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/921f81aa8c5d4a65a5111999405386f5 : SUCCESS in 1h 33m 39s","accounts_in_message":[],"_revision_number":4},{"id":"bc4da61b067df99fcf4c8a245d25e2283bbaea8d","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2021-12-11 15:20:43.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"f30e657167585a525414abb704488969636df7b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-11 17:16:40.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ee9f9f7b1ecd4938b33e793951ace33a : SUCCESS in 49m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3176336968614ba28594a93ce1f408c1 : SUCCESS in 1h 41m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5df13e14c904a0cab39266438321cf9 : SUCCESS in 6m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2b2d5dacf9a4e0c83b60751a0e538df : SUCCESS in 12m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cae2d71b90f640198794c4b465e6be14 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d73660d54a04b36a57b406b9235d673 : SUCCESS in 4m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1dc9f70bfddc4243a2a4d5994f322146 : SUCCESS in 4m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e3cf15a7c3b34217a823ea8f1561f24b : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf61a90e198e4dc0ad517ec2575aaa3a : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b8ba5866d5e4da3a3211f4abf14d4b0 : FAILURE in 4m 21s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/6ce2757dbd3f42f8bacd9b8cf805d4f5 : SUCCESS in 1h 46m 47s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/60194c2f7124407c94a2c4bb75f8d190 : SUCCESS in 1h 13m 17s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/754dd6b784c64b8a8a6e8ab50c47cf0e : FAILURE in 1h 03m 13s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/6b77053c4070441594d566dd4a18e871 : SUCCESS in 1h 05m 13s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/8b82d710948548ebafb00f025786474c : SUCCESS in 1h 44m 24s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/c6f3d5f7e6604a188eff75870861a0fa : SUCCESS in 1h 24m 02s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/bca6cf35b10e4c25a2e4da69fdcc87df : SUCCESS in 1h 13m 46s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/e26f43c9f6f141508a32afdfbf9da8ba : SUCCESS in 1h 21m 07s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f1300dd519ad42dd94a2c61a57b97061 : SUCCESS in 1h 41m 04s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/84ccd23a881b48feaf5cd61e60d4ccbc : SUCCESS in 30m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/accdc5d149a349c6ad0b8d0035ce4fa8 : SUCCESS in 42m 54s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/19b6325906ec47238be8cf630d62de9a : SUCCESS in 1h 54m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0c19ba8c046547638d0ac319835cb5e1 : SUCCESS in 45m 20s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a8c083f58d364fbea04b015d1e9ed511 : SUCCESS in 1h 07m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8cfb3f3cb4694aeba7059b47ec57ace7 : SUCCESS in 57m 31s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/6db4e282146242e5a82a06aa028512cd : SUCCESS in 46m 44s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/5f495eb699364f11a411a75a3bdcc242 : SUCCESS in 5m 03s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/776f659cd3514259b21e08b3019c1d5a : FAILURE in 57m 40s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/886c6fc47c8a455485bdd15d5c6149f1 : FAILURE in 1h 02m 22s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f4454e3c7b3a4647a4d813bb49abcf7f : SUCCESS in 5m 23s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/2e4f03670d194ae9b88972dde48c5b20 : FAILURE in 1h 11m 46s","accounts_in_message":[],"_revision_number":4},{"id":"90d6c3cedf2868dd846109098b45e835c264c9a6","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-12-21 11:16:57.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e93e54957b3a8ac5dd27fa886e15b570b3a916ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-21 13:48:22.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/02916be32578438082da3a91be3c8aa6 : SUCCESS in 1h 02m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2a4c46735a5b4dd58d9798896a5d8c26 : SUCCESS in 1h 27m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35411683db8a42ca85552aa7ce45517f : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da9596874e5146e39a8dcabbb5247cb1 : SUCCESS in 14m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8d7ae3bfcbe4d60a5a50369b61f862f : SUCCESS in 7m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/964aac8c5c5c41d9a96cdebc89369565 : SUCCESS in 5m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67fd3b9a2ea2425cb2b72ccfd462cc1c : SUCCESS in 5m 01s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2f849ed43d1c4c1ab816bb7b92ba596b : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/338ef8d10a39431aa4c38e156cdf9038 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70582c58b0a6417588098c3f07b48959 : SUCCESS in 6m 47s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f46a3ae5ea76423fb6b37f254aff74f0 : SUCCESS in 1h 45m 20s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/48245861b6344544815c5ffa24b35711 : SUCCESS in 1h 13m 06s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/199d6d0fcccf4dafa946e46feec87a5c : SUCCESS in 1h 16m 49s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/fad03a850a4b461facd5dc5b3838e030 : SUCCESS in 1h 27m 26s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/7a49989edcc34500b6e2b3bee6871dcf : SUCCESS in 2h 06m 57s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/23e371e0a5264e0ca1b326e44b754d72 : SUCCESS in 1h 21m 15s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/8cd731cb861042baa3c8f6301b5fce01 : SUCCESS in 1h 16m 48s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/6815de003cce4c77b11db8b2e1670539 : SUCCESS in 1h 14m 24s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/463820eff41c41ef97a684ad3ff0805e : SUCCESS in 1h 26m 49s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/2d9a10609ff84107bc6243b4f04f1563 : SUCCESS in 29m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ec2414652a54cfda2ee98525807cc0b : SUCCESS in 46m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9a71d82dffbf450595a30669eec6df98 : SUCCESS in 2h 24m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/011033af4a674b8184f29de3c480ffb4 : SUCCESS in 54m 37s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bf780d4fe33946098a05eb3134371933 : SUCCESS in 1h 05m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/12310670b94f4a8997be06ac5286a389 : POST_FAILURE in 1h 13m 06s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/635596088d614917a7cabcf46f345ad7 : SUCCESS in 1h 05m 08s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b217f560a0d6493ca90e0280cb563246 : SUCCESS in 4m 28s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ff694b4712a84438906cf62d94e1de92 : FAILURE in 1h 08m 38s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/22e377f321994aadb47829049088e3bf : FAILURE in 1h 08m 10s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f3f84439ab57456aa0d76781f0dd845f : SUCCESS in 3m 43s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/4a12ad3cd01c47a785a903003fd9ba00 : SUCCESS in 1h 29m 59s","accounts_in_message":[],"_revision_number":4},{"id":"2fcb9bbf645b19cbad04c2e964cc497d7bb78f63","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2021-12-22 06:48:18.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e417963c82d94216fc1303969cd59fa4384d06ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-22 08:47:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/3467ed96bb154f4ca7ed14143754d4cd : SUCCESS in 49m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf4a4c05a3dd49e1b4fbc086c77de6dd : SUCCESS in 1h 47m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1561de3763aa4a84bd5db9eb22bb8830 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6579e81070ef4115902e3e0f10b84284 : SUCCESS in 13m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0108b17e7fea4edb9ed59435eef3cc46 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1c3c7415c4743b9863d9272cba19f66 : SUCCESS in 6m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e7c939bafed4c4daa288070d8826d15 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/194d6f303f5f41aca241c3e6f0931d69 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/313ca5dbb55946839255675e7606ac86 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6794c9c0d2d04e1dad6bbebc0ba92030 : SUCCESS in 4m 38s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f3e5ff54e9d544188029abd51da52556 : SUCCESS in 1h 29m 12s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/288512fd8aca41fcb21cddd1ab8f8e04 : SUCCESS in 1h 01m 21s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/649e49b43f6d48f0a43bb642e495fe35 : SUCCESS in 1h 16m 17s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/891c9b8d53d741d5891f4268eaa9cc30 : SUCCESS in 1h 34m 05s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/c5c3fc444ae544d1a3aec338e068e625 : SUCCESS in 1h 42m 20s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/dfa4a7ba40e641baaec64dbab66ed208 : SUCCESS in 1h 40m 02s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/92bb81cf00e64b67ae448882d900e9a5 : SUCCESS in 1h 09m 28s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/1cb0d92c8c784be7a17dba406458cc28 : SUCCESS in 1h 45m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/86e95725219b45d3b451b20a92f17de5 : SUCCESS in 1h 46m 11s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/872ac7ffd67345fc8938af8e173924b3 : SUCCESS in 30m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18047a027f51472e8ccdea31756fd3d0 : SUCCESS in 46m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a4730a1e83d94992a53bd5b1c091e67e : SUCCESS in 1h 49m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fa53b56a61e24354959d8783256424a3 : SUCCESS in 48m 11s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/95a97af1815c4963a091f6165852867e : SUCCESS in 1h 33m 31s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5c5f14be2ad2435cacd9a2b7ad1976e0 : SUCCESS in 47m 47s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/995cb0984e45406ca5a174676eef1981 : SUCCESS in 1h 00m 37s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/0af2d4269a23430dbd8994e71f90956e : SUCCESS in 4m 14s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7cfd0fae222d4dbda41a94cac63ad873 : FAILURE in 1h 35m 38s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/1c5299b5c8044082ab489bbb314155be : FAILURE in 1h 04m 46s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/17cf3f2df332497e8ed060fb6ec37962 : SUCCESS in 3m 27s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/530114c7bca545f7b00dfdbf924ad699 : SUCCESS in 1h 24m 07s","accounts_in_message":[],"_revision_number":4},{"id":"3f54629b222a3c495409497196e20f9d84966f05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-19 04:38:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"659629d433e68c805d11462c59083cdb3045f777","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-19 04:39:08.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"0d9b81619bc9421982f32536adad2e3e47c5585e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-19 07:02:32.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/ec571a96dbc04a2b9ef906fb90bdfdd3 : SUCCESS in 55m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34b1daf21a5e4da981cd053f92333b9e : SUCCESS in 1h 32m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eee05982c2ea427e8c90d496988312f6 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ef28e7bd3b741b09040a526bd5c3892 : SUCCESS in 13m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7cf1bc078e64f51b02c1edca0bfb01a : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea79501955ea46d5a07901774a457924 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/147abf342d5b4265842b8166f549c978 : SUCCESS in 5m 02s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ed39a98695a4a0aa9c52a389837621c : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c90bae139bc5437aa91fafd7bf112bc7 : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be3d9c0fa2484d47b619e717815bfc0b : SUCCESS in 5m 41s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b2d102d93af144a4a574d4083669a354 : SUCCESS in 1h 38m 53s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/bfb6289557e242b99fc337a40050886c : SUCCESS in 1h 23m 43s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/14d7739f48c7485aaea365fc75038268 : SUCCESS in 1h 24m 53s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/fccc634e762c451ab35f8c494b2b3274 : SUCCESS in 1h 20m 54s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/9a3f7b17660d4649b4f1f5a15385c086 : SUCCESS in 1h 24m 13s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/0f1ed046ce9642089f9007932c2b119f : SUCCESS in 1h 28m 25s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/fdc4ad92991847e3bd43f19ca0dd71a3 : SUCCESS in 1h 25m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cc35560453eb403a8057db452a523a38 : SUCCESS in 1h 32m 57s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c27ab956fc6d4ab7afbddea6d95fb134 : SUCCESS in 29m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b71886c1dc14149bd293b0b488d90de : SUCCESS in 51m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6cc2af33612a47ceb7fae7251c9c99ff : SUCCESS in 2h 18m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6a8fa26089ab4d2b9f883932ac7d8463 : SUCCESS in 48m 49s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/835ae63ac51048cb95cad3e0c236fa51 : SUCCESS in 1h 13m 58s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ef880bb7c20b42c68bbbfd66826e3091 : SUCCESS in 56m 08s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/872619b7019d4a4391ae0e2b21c13ee9 : SUCCESS in 58m 26s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/bf162586eb4d4b19886c000f9b475e52 : SUCCESS in 5m 31s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5716711c6a2f4fdfb6835eda7f2a0f80 : FAILURE in 1h 10m 54s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/fd1e9df0d28243dfaafc1213b3804695 : FAILURE in 59m 09s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c7f619dd8022468c85b3ba8f190ebe02 : SUCCESS in 4m 46s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/38c3c6ef5bc04d3ca3230cec614390de : FAILURE in 1h 12m 26s","accounts_in_message":[],"_revision_number":5},{"id":"ead19e45464b9bbca14e7a9539d02919bf789d5c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2022-01-27 16:24:36.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1ee11f00bf806168113c744047cea6beef0c9296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-27 19:21:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/93a80464f7cf4f8fa5fe98aec04feacd : SUCCESS in 59m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8d7e1b151e348c9ab6abbbadc0c729e : SUCCESS in 1h 28m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8a4347e5f3a4c5592a41461c1de276d : SUCCESS in 7m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b5c502559ee46a1983146ff8f37ee5d : SUCCESS in 13m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55a84a7996214e3aa5922bce10985614 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47ecb1103ce441d9b671f490ecd112f7 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/98f874f509ec42ca966e73c566e59437 : SUCCESS in 6m 11s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ef6a8b7d3d742bc825d2144a03db25c : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/afa0d5392ca441279936c318e2c0a5a6 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e7261732fce4481923cd98a07fcaf16 : SUCCESS in 5m 45s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/77885197577e4331bd98e0c5ed158b87 : SUCCESS in 1h 37m 27s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/30e8db138e2e42dda8df0b9d82963d1d : SUCCESS in 1h 21m 15s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/31bc13d2bf3e4286b418deb0735c2e2d : SUCCESS in 1h 29m 21s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/ccd5c6bd88ed42db92822f5b6b8fb2cc : SUCCESS in 1h 24m 53s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/475f61883bf54e3a8e07f8ea69cb63a6 : SUCCESS in 1h 29m 02s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/bcf24cbeaa7d432c886e9db14e9de3d5 : SUCCESS in 1h 23m 32s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/47bef798f5e24f19aaa119406b659d53 : SUCCESS in 1h 19m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d510ec0669ac42d98e0162014439fcd4 : SUCCESS in 1h 37m 30s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c02e55cf95474cb08cdaf5e0cbfc294a : SUCCESS in 29m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69344d20cb864b318ed9d17601240b2b : SUCCESS in 45m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a8825eaeb6c346a68e444f1edb108c7c : SUCCESS in 2h 04m 20s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/049dbcedcd8647698fcd6e53d69628c1 : SUCCESS in 52m 35s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/eb4bfd088d224abf8f4a49bb6c97e2e8 : SUCCESS in 1h 24m 55s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/432b979420364fae8ee028d5d802c8f6 : SUCCESS in 1h 07m 16s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/c6a3a5e0e6ad48da8fcb9ed14ba35db7 : SUCCESS in 1h 04m 37s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/42a0ab7b61254f6a9bfeeff1072ef690 : SUCCESS in 4m 45s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/4d745b3aef3c4335aaa62328b2a526d1 : FAILURE in 1h 03m 01s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ce96e0ce70ab47ebb44a4723cedc8302 : FAILURE in 1h 08m 41s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/da002579dcad493ba4015d04d4f7b0ac : SUCCESS in 3m 51s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/f0081e2aafbe4c14ac8a0b887755f165 : SUCCESS in 1h 31m 06s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dee9337a05a641fe477b30b1d43dd968726a9625","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-02-14 02:15:32.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3096fe7e5a1c9d7a54713f2292ae6841e981666c","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-03-02 10:51:24.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"21a6c55b8491eed3b856d3dfe3551b3efaed6035","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 10:52:25.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f4ef3596d1469505958babee90df070b0b2a5fde","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:15:05.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/42bf11a6dc9c4fd48a6a61f38476bdef : SUCCESS in 56m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e777b43409b54a79949f931a4d37c8e8 : SUCCESS in 1h 39m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc5f8982baa242599e75e54371e73aea : SUCCESS in 15m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0401535d5c94139bc6e0c60c3e79e5a : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd2ab866b6f441d8e0fab9b57ab229c : SUCCESS in 6m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33ca8fe2680541e1894a75235b3195a3 : SUCCESS in 7m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e884bc1204347b5b75a13c763f6df36 : SUCCESS in 6m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6d1a666eb84c4ef9906e05da4e107e7a : SUCCESS in 7m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37aec1cfbafd4fc2b394bf965b909e7e : SUCCESS in 7m 30s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4ef611eaf5ce4adf88dfabdfd4d27341 : SUCCESS in 2h 21m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/65f9bec5ec8947d19f8a67f326781468 : SUCCESS in 47m 09s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/7b0575c734dc4736b63dd9e65c4e6068 : FAILURE in 1h 06m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90034c46c95349fbb867612b4bef818f : SUCCESS in 37m 52s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/93968a04ab4a4b229634f3be80a690e0 : SUCCESS in 1h 14m 33s","accounts_in_message":[],"_revision_number":5},{"id":"4d677f6f2c767292170fe052979d73e81a9434ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:15:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"77019bbb8a593a1ac9505e8d79d1d807a2e2bc99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:16:55.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d15499ff31d499984b3d43403d91b92 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f84233a53dc14b2cab8a01bba48b8b83 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"2a5ef1b9339e5c739d9271b0f991e24f73495fc5","revisions":{"45b01d75b98384079228a48fb39dad09de1052f7":{"kind":"REWORK","_number":1,"created":"2021-11-29 06:48:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/819594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/819594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/819594/1"}}},"commit":{"parents":[{"commit":"daeae9c8a586ac60f7e9dfc486822fd80f1da136","subject":"Merge \"Update Interop doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/daeae9c8a586ac60f7e9dfc486822fd80f1da136"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 06:48:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 06:48:25.000000000","tz":540},"subject":"Do not use deprecated LOG.warn","message":"Do not use deprecated LOG.warn\n\n... and use LOG.warning instead.\n\nChange-Id: I43e2269d4b1c425119c86919eb869941d621c6de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/45b01d75b98384079228a48fb39dad09de1052f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/45b01d75b98384079228a48fb39dad09de1052f7"}]},"branch":"refs/heads/master"},"0718f33775659de4b3c62ce92f3757ac7d4f6ffd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-11-29 06:50:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/819594/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/819594/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/819594/2"}}},"commit":{"parents":[{"commit":"daeae9c8a586ac60f7e9dfc486822fd80f1da136","subject":"Merge \"Update Interop doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/daeae9c8a586ac60f7e9dfc486822fd80f1da136"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 06:48:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 06:50:14.000000000","tz":540},"subject":"Do not use deprecated LOG.warn","message":"Do not use deprecated LOG.warn\n\n... but use LOG.warning instead.\n\nChange-Id: I43e2269d4b1c425119c86919eb869941d621c6de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0718f33775659de4b3c62ce92f3757ac7d4f6ffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0718f33775659de4b3c62ce92f3757ac7d4f6ffd"}]},"branch":"refs/heads/master"},"ad8b15306f1e5534f1db01faf3ecbc56470c7830":{"kind":"REWORK","_number":3,"created":"2021-11-29 09:24:45.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/819594/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/819594/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/819594/3"}}},"commit":{"parents":[{"commit":"daeae9c8a586ac60f7e9dfc486822fd80f1da136","subject":"Merge \"Update Interop doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/daeae9c8a586ac60f7e9dfc486822fd80f1da136"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 06:48:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 09:24:36.000000000","tz":540},"subject":"Use LOG.warning instead of deprecated LOG.warn","message":"Use LOG.warning instead of deprecated LOG.warn\n\nThe LOG.warn method is deprecated[1] and the LOG.warning method should\nbe used instead.\n\n[1] https://docs.python.org/3/library/logging.html#logging.warning\n\nChange-Id: I43e2269d4b1c425119c86919eb869941d621c6de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad8b15306f1e5534f1db01faf3ecbc56470c7830"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ad8b15306f1e5534f1db01faf3ecbc56470c7830"}]},"branch":"refs/heads/master"},"e47e68ac211d6966c315ab105c0f366de815c187":{"kind":"REWORK","_number":4,"created":"2021-11-29 12:05:20.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/819594/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/819594/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/819594/4"}}},"commit":{"parents":[{"commit":"daeae9c8a586ac60f7e9dfc486822fd80f1da136","subject":"Merge \"Update Interop doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/daeae9c8a586ac60f7e9dfc486822fd80f1da136"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 06:48:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 12:05:11.000000000","tz":540},"subject":"Use LOG.warning instead of deprecated LOG.warn","message":"Use LOG.warning instead of deprecated LOG.warn\n\nThe LOG.warn method is deprecated[1] and the LOG.warning method should\nbe used instead.\n\n[1] https://docs.python.org/3/library/logging.html#logging.warning\n\nChange-Id: I43e2269d4b1c425119c86919eb869941d621c6de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e47e68ac211d6966c315ab105c0f366de815c187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e47e68ac211d6966c315ab105c0f366de815c187"}]},"branch":"refs/heads/master"},"2a5ef1b9339e5c739d9271b0f991e24f73495fc5":{"kind":"REWORK","_number":5,"created":"2022-01-19 04:38:31.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/819594/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/819594/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/819594/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/819594/5"}}},"commit":{"parents":[{"commit":"daeae9c8a586ac60f7e9dfc486822fd80f1da136","subject":"Merge \"Update Interop doc\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/daeae9c8a586ac60f7e9dfc486822fd80f1da136"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-29 06:48:25.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2022-01-19 04:38:21.000000000","tz":540},"subject":"Use LOG.warning instead of deprecated LOG.warn","message":"Use LOG.warning instead of deprecated LOG.warn\n\nThe LOG.warn method is deprecated[1] and the LOG.warning method should\nbe used instead.\n\n[1] https://docs.python.org/3/library/logging.html#logging.warning\n\nChange-Id: I43e2269d4b1c425119c86919eb869941d621c6de\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2a5ef1b9339e5c739d9271b0f991e24f73495fc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2a5ef1b9339e5c739d9271b0f991e24f73495fc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
