)]}'
{"id":"openstack%2Ftempest~819751","triplet_id":"openstack%2Ftempest~master~I2e868d13d52a2d3c6081a8b756fbe83f8b57cf66","project":"openstack/tempest","branch":"master","topic":"bug/1938103","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2021-12-01 15:32:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e868d13d52a2d3c6081a8b756fbe83f8b57cf66","subject":"Replace deprecated assertDictContainsSubset","status":"MERGED","created":"2021-11-30 04:29:30.000000000","updated":"2021-12-01 15:34:48.000000000","submitted":"2021-12-01 15:32:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"819751-bug/1938103","meta_rev_id":"63425bd2a6b03c30c03887bf3cd080ee88095474","_number":819751,"virtual_id_number":819751,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-01 15:32:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2021-12-01 13:24:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2021-12-01 13:24:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-30 06:32:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-01 13:24:37.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"a03e1762df61e91b173b002037008fee71a5ac72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-30 04:29:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"547ac9a879ed245ab58fadedbb664a4344e9054b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 06:32:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/0b91dc9f0f9f4db798bdd1c2d62cfcef : SUCCESS in 1h 03m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d82b402f5f844a88ace009731c6c1266 : SUCCESS in 1h 23m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/366c029f79a74e68b170ba4cb34dccb4 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7db43d7f37743d6a5bb2e5c8c9b8601 : SUCCESS in 17m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd4393c2bb22496e834eb1ce26ff3661 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1edb7896648d4e9296bb51b4341dda2f : FAILURE in 5m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fc592a2c1e9543c48a574369f74002d0 : SUCCESS in 6m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/04755a3608a7450886b10d44d0400de1 : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5011c620204b476cbe76e589391dd870 : SUCCESS in 4m 27s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/6786d41cb66c46878d4e0e16c5d827e0 : SUCCESS in 1h 28m 38s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/46f7c1090e8f41a993914b2e4d1d7031 : SUCCESS in 51m 18s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e3a631717d91463692af56e04074db0a : SUCCESS in 1h 12m 34s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/d1283d54a0e549dfbe94c40d653a08fd : SUCCESS in 1h 20m 02s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/8305b439199d4fd2b94f75d8bdb813fc : SUCCESS in 1h 22m 16s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/4692ba53b1d64e6d8d328086c7b60b35 : SUCCESS in 1h 25m 44s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/767cdbdc916a4099bfdc358bf455c903 : SUCCESS in 1h 00m 45s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/347bcde5bf2c4217a09b5194dd8a780f : SUCCESS in 1h 37m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8347f273ed424ebf829c5fe7d2fe229c : SUCCESS in 1h 29m 21s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/359c6f37f87a42a58c7f206bfbdfba97 : SUCCESS in 27m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8aa9dbab43e64da187b74fb911e01be0 : SUCCESS in 55m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/a2021e12c9e14ddd9b42407c8c67de6d : SUCCESS in 1h 58m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/be4196b17b6b4327963cc5ae41a42924 : SUCCESS in 46m 19s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b79f46ac3cd745529a480e4c1717f9a9 : SUCCESS in 1h 13m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b49dcdea2d6b4d2d873b090f2ccae0c3 : SUCCESS in 1h 02m 35s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/a3078f3e332e449ca869cfd4375bee5d : SUCCESS in 1h 03m 04s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/ec2dd04a285a4f54b95eb524e497951a : SUCCESS in 5m 15s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1198535d65bf4a1fb13ed0d4d8757911 : FAILURE in 1h 08m 20s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ff2ae527a2f14c9cbeb4b91986b6e055 : FAILURE in 1h 00m 16s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d64d275b981d4edb941418410594bb91 : SUCCESS in 4m 25s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/949c9114ddbf42cb925d993dd6927693 : SUCCESS in 1h 46m 13s","accounts_in_message":[],"_revision_number":1},{"id":"9c269b7cf11ec9de84b1b23e647cf8d626d52f8f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2021-11-30 12:28:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"64053304274e9afffa486c77c6235cb5a41bc6e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-30 15:07:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/294cb3caf5d445c4a608534ffc294cfd : SUCCESS in 1h 03m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/000ded08d54846aa92e6fc3800ea5d1a : SUCCESS in 1h 34m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66ca7620bb394f88adb116ceac58ff6b : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d5ecd3d1f3417ea67e9a43095e4336 : SUCCESS in 12m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae40c7f81a84f9b85a5c688fa327d99 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93d51707f8bf465f99cbaab1d45d006b : SUCCESS in 4m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4185393a1c904c0cb207bf8d05a805cc : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e94fa5767894b138e033583d9567313 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd19c98bc3a241b4a1fb6faf020f921e : SUCCESS in 5m 20s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/92b07671b2ec426ba7b256477ae6e64e : SUCCESS in 1h 50m 40s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/78f7eb9687fb4cc19491a59639e25ea0 : SUCCESS in 1h 18m 04s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dddbf50134194d94b11fa496ff131fdd : SUCCESS in 1h 19m 48s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/1fe9759a4d42440295c68e74eb3e19d8 : SUCCESS in 1h 27m 46s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/c76b23d6d1ef46939b366e9acd41b9b5 : SUCCESS in 1h 25m 14s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/3f83c2c174c54f278d491f49a158eb78 : SUCCESS in 1h 30m 14s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/d26c6d74cdeb4b42a80b1680eb3b4334 : SUCCESS in 1h 36m 42s\n- tempest-full-train-py3 https://zuul.opendev.org/t/openstack/build/b63236bcee5c48fb8d669e0c8b37bbba : SUCCESS in 1h 35m 25s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e9e36069be3249899ac752dff2614f18 : SUCCESS in 1h 43m 21s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/dfba1ed092de48ed821147eb6ab7bf6f : SUCCESS in 34m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49d6e7880f484d75a5c873835ce6dbb8 : SUCCESS in 54m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/dec74a2b5f8946739c9b95bf4bab3184 : SUCCESS in 2h 32m 01s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/89997b5243b74c6db7a8121f0964bb9d : SUCCESS in 56m 04s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/aa00f33eb22d48feb2d215c1641b707e : SUCCESS in 1h 24m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/cff7344cf31f41a6ab918accdd31ac72 : SUCCESS in 1h 02m 47s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/0eedc4b456cc49dab19311de43d53f83 : SUCCESS in 1h 04m 20s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/37ca10932c174b119dcb98b2b8907b83 : SUCCESS in 5m 38s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/71a977be971d400ca81d9aad5efc27fa : FAILURE in 1h 20m 32s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/cc8f946a0b88482c815d72d0146b929b : FAILURE in 1h 09m 09s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/58d5f93468734b49a53c40a70568b808 : SUCCESS in 4m 42s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/edd3c1e88e464592af89a84bb54810e9 : SUCCESS in 1h 28m 52s","accounts_in_message":[],"_revision_number":1},{"id":"d98ab8407740cf9534ea318bafbff38ccb4de36c","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2021-12-01 13:24:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e710c06504ecbfb7a8a8b0c11bbb619c66b57ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 13:25:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"461a794f7e15bcfb6cca1ec3fe7e362b4a05440e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 15:32:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/835d067363ce423285e2d457acada1a3 : SUCCESS in 46m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a05e385bcf654c4f97a767bc9fb8e57b : SUCCESS in 1h 17m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd6a91a9fbee4bfdb9c596a597ad6dd3 : SUCCESS in 12m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f818a0e4b0ab4ddcab5716de2b62ecae : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d29f04bc43749f397a01a0c1416cbe6 : SUCCESS in 4m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2335f09445bd45648142740806b1ba75 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9204d0a811441aea61182de67df2f0e : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/856c1c4137ac4d6dbc924d824d7e5621 : SUCCESS in 4m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7365f2fe227a4582ac8f030a9874bca2 : SUCCESS in 2h 06m 07s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b53b2f5612e64471b9b626732737c47f : SUCCESS in 46m 00s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/e2127593f0654c8fa2d9ec2ecaf49dfe : SUCCESS in 1h 38m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6ac04887c9d04f8497ca8c4ea5956823 : SUCCESS in 44m 24s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/57f190efc5f94e638cd9bb669487143d : SUCCESS in 1h 23m 49s","accounts_in_message":[],"_revision_number":1},{"id":"fcf7bdf607f3fdcbf50e5c11325bee12f24d0449","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 15:32:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"63425bd2a6b03c30c03887bf3cd080ee88095474","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-01 15:34:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/556cae4939134d1c92bb2119c10433ba : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bbe4f8c3054d1f71fd5057e6da096a99753e5754","revisions":{"bbe4f8c3054d1f71fd5057e6da096a99753e5754":{"kind":"REWORK","_number":1,"created":"2021-11-30 04:29:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/51/819751/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/51/819751/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/819751/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/819751/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/819751/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/51/819751/1"}}},"commit":{"parents":[{"commit":"c08dfd20142fc845a9a7fc4502eeecc33ddb680b","subject":"This patch creates the network v2.0 floating-ips-port-forwarding client.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c08dfd20142fc845a9a7fc4502eeecc33ddb680b"}]}],"author":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 04:29:18.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"tkajinam@redhat.com","date":"2021-11-30 04:29:18.000000000","tz":540},"subject":"Replace deprecated assertDictContainsSubset","message":"Replace deprecated assertDictContainsSubset\n\nThe method is deprecated since Python 3.2[1] and shows the following\nDeprecationWarning.\n\n/usr/lib/python3.9/unittest/case.py:1134: DeprecationWarning:\nassertDictContainsSubset is deprecated\n  warnings.warn(\u0027assertDictContainsSubset is deprecated\u0027,\n\n[1] https://docs.python.org/3/whatsnew/3.2.html#unittest\n\nCloses-Bug: #1938103\nChange-Id: I2e868d13d52a2d3c6081a8b756fbe83f8b57cf66\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bbe4f8c3054d1f71fd5057e6da096a99753e5754"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bbe4f8c3054d1f71fd5057e6da096a99753e5754"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
