)]}'
{"id":"openstack%2Ftempest~824594","triplet_id":"openstack%2Ftempest~master~I986f1b76274322b2a3e3a675190aa7acbe4ef952","project":"openstack/tempest","branch":"master","topic":"objects","attention_set":{},"removed_from_attention_set":{"29422":{"account":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"last_update":"2022-01-14 10:42:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I986f1b76274322b2a3e3a675190aa7acbe4ef952","subject":"do not check transfer-encoding nor content-lengh for DELETE RFC doesn\u0027t say it must send transfer-encoding for DELETE 200 uwsgi decided not to send it doesn\u0027t thus breaks the test object_storage.test_object_slo.ObjectSloTest.test_delete_large_object","status":"ABANDONED","created":"2022-01-13 15:22:26.000000000","updated":"2022-01-14 10:42:01.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0067d8f7591528df2b8917362fd2f8882b79c9cb","_number":824594,"virtual_id_number":824594,"owner":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-01-13 19:10:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-13 19:10:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1f907d2a28a34afa0695ae458638c4f393083fc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"date":"2022-01-13 15:22:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c565753a1c8b44353703b7a5a6d1925c5ea0657b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 19:10:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/a2fe13ea17c14a859b6637fde4bc0c97 : SUCCESS in 59m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9c370c2eda934477af28fa78220d0d11 : SUCCESS in 1h 42m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6d1772f64e1491fad5f18bb1a269d6a : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c55b7318b6fd4e95b27edd83ed4fdcf1 : SUCCESS in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/159153d5ac1645c78ef292f90249d68e : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fbc8a81eac3468c874fe2d5caff8113 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/09d5019fc07041a18086b8913b8bf709 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f067b0f709254a788c9fcfb211e2dc75 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c16a828aa2684e6a8ef9cee7d2bfb6d0 : SUCCESS in 7m 28s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/b66dbb8533d64eeaa955cbd8e47485f5 : SUCCESS in 1h 44m 40s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/ec7d425dcf804081b6b2eb486ebc8486 : SUCCESS in 1h 25m 50s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c1cb3187b0884e789e82d14aa19cce01 : SUCCESS in 1h 35m 36s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/23ae080a67134ce3beb6b72cf74b40d5 : SUCCESS in 1h 25m 35s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/3d16b4ca481d42cd99ceeaa905ad17bb : SUCCESS in 1h 20m 14s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/e11fd24617174750831617ac7b4f144d : SUCCESS in 1h 21m 31s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/c5db000d85bb45b1a45e4a143bd11fc9 : SUCCESS in 1h 25m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4f3a344f528b4777afb1f91e20d4bc67 : SUCCESS in 1h 43m 06s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0ccd6c18118c46eebb0fa3dfdc749808 : SUCCESS in 30m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b77c1e6e7f364431867d2614308e291c : SUCCESS in 1h 08m 15s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/723ec991acac4fdb8aac674a2068e08a : SUCCESS in 2h 29m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3b4c5e1e13c04fcf8a269b92c90810eb : POST_FAILURE in 30m 44s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6be13cbdd81b461cb326f0dca13ea705 : SUCCESS in 1h 12m 11s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4208f7657c645bb955eff2820bddc76 : SUCCESS in 1h 01m 05s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/959ba12fb9a341bd9314a3c386122a45 : SUCCESS in 59m 22s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e7db96fde78c41c68aee083e6f96e812 : SUCCESS in 5m 06s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/43d3871194fd41aeb54d34df2a894e09 : FAILURE in 1h 14m 33s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/fea638fb2ca343f19e0bd460cc2dbcb3 : FAILURE in 1h 06m 22s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7fbc1bdb454e460fa13323f5304a2419 : SUCCESS in 4m 13s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/e8a8a6ea329f41408f1dad71dc022c6e : FAILURE in 1h 35m 59s","accounts_in_message":[],"_revision_number":1},{"id":"8cf24483a94bd413b6068d648de108c6dfd3f874","author":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"date":"2022-01-14 04:35:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"45524472735a4bcfa2813f9733ef350ba62a2458","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-14 07:25:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/84cc36e2d1ef4f06adf4a499c262cd73 : SUCCESS in 59m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc28391500134e1dadf4cf43f79ddf2e : SUCCESS in 1h 22m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74565897cb5e45a5806e225d100bb488 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad51e2825b994118a1fadbe1485ab4be : SUCCESS in 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/955ec67f78ac484f888430f5a24e7b17 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/55bfb917d9684745b4595281c5f53afb : SUCCESS in 7m 09s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d295482ccca0452aa6166939e451c467 : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6d22ba39c9c4423b166c27b797f8f14 : SUCCESS in 6m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/201fe75d69dd4f3089fb08b958829d34 : SUCCESS in 7m 03s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/e7947363198847f28559ef0a47991bde : SUCCESS in 1h 48m 11s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/b2a3a5833e594a5aa589d54c1cc5a61f : SUCCESS in 1h 11m 45s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/43f948b63ef54e2aa1e0e00e28914c0e : SUCCESS in 1h 24m 57s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/c19fcc7dd03343bea987bda2fdc4c406 : SUCCESS in 1h 06m 56s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/1aaf8160dfa34df38587a22e4406c9ee : SUCCESS in 1h 31m 13s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/b77480d87f2d40a09ede47ba12aa97ea : SUCCESS in 1h 16m 52s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/b6c95d8e1d0f447583465958138096b9 : SUCCESS in 1h 31m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/64d8b039d3664614a8945887d2974b2d : SUCCESS in 1h 28m 41s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/d254d65abcbe4689a3b73cc567663141 : SUCCESS in 30m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e2a20716e374392b4a7ce04649e5a97 : SUCCESS in 52m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/aac792b989394f06bd1b0ea6aa1612a3 : SUCCESS in 2h 42m 33s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/900362427eb049698daf0f3f9f417554 : SUCCESS in 52m 34s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c2f61adc2ce54a9189fe8df565af74df : SUCCESS in 1h 10m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d7d880a6418d43c99934d039695b15f6 : SUCCESS in 56m 43s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/01623c3e4da3412a8446c1c407666efc : SUCCESS in 56m 49s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/23ef75b2d6994434893313906b1152c2 : SUCCESS in 3m 57s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1741560ada5d4aa7b1d58fe3b6185789 : FAILURE in 1h 12m 06s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/f7b7280f3eca4246939dcb45834bdd0b : FAILURE in 1h 05m 33s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b5e5052a11364d5383c27a682a551e1a : SUCCESS in 3m 15s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/83bca968de414537a4be611e7a41c8f2 : FAILURE in 1h 34m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0067d8f7591528df2b8917362fd2f8882b79c9cb","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"date":"2022-01-14 10:42:01.000000000","message":"Abandoned\n\nseems to be a uwsgi issue as transfer-encoding or content-length are also missing in response of a 200 PUT request","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fcef58b5957cf5fbc309baf1144e2718af50cfcd","revisions":{"fcef58b5957cf5fbc309baf1144e2718af50cfcd":{"kind":"REWORK","_number":1,"created":"2022-01-13 15:22:26.000000000","uploader":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"ref":"refs/changes/94/824594/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/94/824594/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/824594/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/824594/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/94/824594/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/94/824594/1"}}},"commit":{"parents":[{"commit":"fe0ac89a5a1c43fa908a76759cd99eea3b1f9853","subject":"Merge \"Test Network QoS minimum packet rate scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fe0ac89a5a1c43fa908a76759cd99eea3b1f9853"}]}],"author":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2022-01-13 15:19:10.000000000","tz":60},"committer":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2022-01-13 15:22:22.000000000","tz":60},"subject":"do not check transfer-encoding nor content-lengh for DELETE RFC doesn\u0027t say it must send transfer-encoding for DELETE 200 uwsgi decided not to send it doesn\u0027t thus breaks the test object_storage.test_object_slo.ObjectSloTest.test_delete_large_object","message":"do not check transfer-encoding nor content-lengh for DELETE\nRFC doesn\u0027t say it must send transfer-encoding for DELETE 200\nuwsgi decided not to send it doesn\u0027t thus breaks the test\nobject_storage.test_object_slo.ObjectSloTest.test_delete_large_object\n\nChange-Id: I986f1b76274322b2a3e3a675190aa7acbe4ef952\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fcef58b5957cf5fbc309baf1144e2718af50cfcd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fcef58b5957cf5fbc309baf1144e2718af50cfcd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
