)]}'
{"id":"openstack%2Ftempest~828109","triplet_id":"openstack%2Ftempest~master~If8fcfd97e7f49cb1248e997375a16d3917cc4605","project":"openstack/tempest","branch":"master","topic":"acpi-neg-vol","attention_set":{},"removed_from_attention_set":{"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2023-07-18 01:54:54.000000000","reason":"Change was abandoned"},"5803":{"account":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"last_update":"2022-02-08 08:13:36.000000000","reason":"Attila Fazekas replied on the change"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-07-18 01:54:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If8fcfd97e7f49cb1248e997375a16d3917cc4605","subject":"Add instance validation to volume negative","status":"ABANDONED","created":"2022-02-07 10:53:08.000000000","updated":"2023-07-18 01:54:54.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5451e17485463ba1f557cc310064b849ccca6a58","_number":828109,"virtual_id_number":828109,"owner":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-02-07 13:18:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-07 13:18:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-07 13:53:38.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2022-02-07 13:53:38.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"5bab9d04653288efcd855d07e1f8f5c34ded6f1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2022-02-07 10:53:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65c7dd03803318ca2560731b5b34f5ad9b4bf895","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-07 13:18:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/02ae5ef4f14144798334e79d214ba33d : SUCCESS in 1h 02m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1ce1a75f329b40058733f7dade9fc185 : SUCCESS in 1h 27m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/494a6eab20744ce5a58a2880af9cda99 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41e20898913f4b6da67a399ca766e1e9 : SUCCESS in 13m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f6f5a4b3b944ae78023f663e28b2d5f : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dab3de36c3114c03a1b44b78a00647ea : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d94d9aab56c143a799ca585ffd15fd25 : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7dbed9644f2543ceb67fab06ec2ae7d6 : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccb0f9bbb3a24ecdb3d5042fa49dadcb : SUCCESS in 5m 47s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/968fdcbab8ce422c8868b3292157b53d : SUCCESS in 1h 48m 47s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/85ef50ce590e4ee59187f98f45a14918 : SUCCESS in 1h 21m 30s (non-voting)\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/1e061650ce764808963b990e8ed268dd : SUCCESS in 1h 39m 38s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/8b539626379b4b10a920e0af54f71f6c : SUCCESS in 1h 23m 42s\n- tempest-full-victoria-py3 https://zuul.opendev.org/t/openstack/build/78d515e5871342d884a379c7e5cf3e27 : SUCCESS in 1h 43m 10s\n- tempest-full-ussuri-py3 https://zuul.opendev.org/t/openstack/build/401d4e6e53304f63b93919422be93646 : SUCCESS in 1h 30m 36s (non-voting)\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e77bd983729147eeaf979531a853aef3 : SUCCESS in 1h 45m 56s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f8b827d9c6cc456ca5e4c13cfc77f614 : SUCCESS in 28m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e9019a9bdc70483fba87e49d8dab2dc4 : SUCCESS in 33m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fbddcf1f19ef430f86975af3c4a3cab0 : SUCCESS in 2h 21m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/be52f9053725411ba4c0a9a94aee4275 : SUCCESS in 52m 07s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/079a6e95ea2d433081956c21d9ec3a23 : SUCCESS in 1h 05m 14s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e581bb550c884f5eaf14aea81f7fbe28 : SUCCESS in 54m 52s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/1f10dd59fbf44b3296a351e50c2ff7ff : SUCCESS in 53m 37s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/cea3d33eb3b0448d9138818f7cb8ad83 : SUCCESS in 5m 24s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/50cb99da70a44af78444b297237864f1 : FAILURE in 1h 02m 41s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/2172f5c8661146458bc84a63d85b362f : FAILURE in 1h 01m 30s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3cb362348b0a4009997aa103167dbee8 : SUCCESS in 3m 52s\n- tempest-full-py3-centos-8-stream https://zuul.opendev.org/t/openstack/build/c2905585641b4035a364d3bcebf09c5f : SUCCESS in 1h 28m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f37bff6036dab59eadf63558703a16c66560f303","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-02-07 13:53:38.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"942a8d5d07fad5e3bca13ceb04f6ae14f68be865","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2022-02-08 08:13:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29cc3ee64cac9608c903591c8a6e030d1f55c04b","author":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"date":"2022-02-08 12:10:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5451e17485463ba1f557cc310064b849ccca6a58","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-18 01:54:54.000000000","message":"Abandoned\n\nThanks. No activity on this since long, feel free to restore it if it is still needed.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a100fc00f986b98ca8ebc3b01586540b2b16de64","revisions":{"a100fc00f986b98ca8ebc3b01586540b2b16de64":{"kind":"REWORK","_number":1,"created":"2022-02-07 10:53:08.000000000","uploader":{"_account_id":5803,"name":"Attila Fazekas","email":"afazekas@redhat.com","username":"afazekas"},"ref":"refs/changes/09/828109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/09/828109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/828109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/828109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/09/828109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/09/828109/1"}}},"commit":{"parents":[{"commit":"115065a239a8e1ab50b6070a3c6ebfeabd0ad6fc","subject":"Fix TypeError in QoSBandwidthAndPacketRateTests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/115065a239a8e1ab50b6070a3c6ebfeabd0ad6fc"}]}],"author":{"name":"afazekas","email":"afazekas@redhat.com","date":"2022-02-07 10:49:53.000000000","tz":60},"committer":{"name":"afazekas","email":"afazekas@redhat.com","date":"2022-02-07 10:53:06.000000000","tz":60},"subject":"Add instance validation to volume negative","message":"Add instance validation to volume negative\n\nPCI device add/removal has to have live Operating system,\nadding instance validation to test_attach_volume_negative.py .\n\nhttps: //docs.google.com/presentation/d/1Im-iYVzroKwXKP23p12Q5vsUGdk2V26SPpLWF3I5dbA/edit#slide\u003did.g40c5dc74a9_0_30\nChange-Id: If8fcfd97e7f49cb1248e997375a16d3917cc4605\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a100fc00f986b98ca8ebc3b01586540b2b16de64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a100fc00f986b98ca8ebc3b01586540b2b16de64"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
