)]}'
{"id":"openstack%2Ftempest~83952","triplet_id":"openstack%2Ftempest~master~I150f4e9f2e79f6e20f18dee69cf061c68a5bed68","project":"openstack/tempest","branch":"master","topic":"bug/1299350","hashtags":[],"change_id":"I150f4e9f2e79f6e20f18dee69cf061c68a5bed68","subject":"Remove variable part of agent dict to fix AgentManagementTest","status":"MERGED","created":"2014-03-29 05:44:13.000000000","updated":"2014-04-15 12:07:49.000000000","submitted":"2014-04-15 12:07:48.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"bece11cd9613fedd6e74c62a12fe323d845f17e9","_number":83952,"virtual_id_number":83952,"owner":{"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":0,"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},{"value":2,"date":"2014-04-15 12:07:48.000000000","_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"recommended":{"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},"all":[{"value":1,"date":"2014-04-02 08:00:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2014-04-10 10:42:16.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"date":"2014-04-02 09:42:31.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},{"value":0,"date":"2014-04-14 20:09:32.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"all":[{"value":0,"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"date":"2014-04-10 10:42:16.000000000","_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},{"value":1,"date":"2014-04-14 20:09:32.000000000","_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"value":0,"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},{"value":0,"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},{"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},{"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2014-04-02 08:00:07.000000000","updated_by":{"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},"reviewer":{"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},"state":"REVIEWER"},{"updated":"2014-04-10 10:42:16.000000000","updated_by":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"reviewer":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"state":"REVIEWER"},{"updated":"2014-04-14 20:09:32.000000000","updated_by":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"reviewer":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"state":"REVIEWER"},{"updated":"2014-04-15 12:07:48.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"e89d9ac6f06f71471e7b9f77b58555b678f71cf6","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-29 05:44:14.000000000","message":"Patch Set 1:\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"0f757e117f22e8f2837587c67d7c6ccacdec06be","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-03-29 06:46:59.000000000","message":"Patch Set 1:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://70dcd673047f0ebed6a7-dda4e480e005aaa13ec303551d2d8155.r49.cf5.rackcdn.com/52/83952/1/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":1},{"id":"e2b316a73fcca5fb2d899928239cd07d96207211","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-03-29 06:48:50.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- check-grenade-dsvm http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm/119c945 : SUCCESS in 42m 43s\n- check-tempest-dsvm-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-full/a1dbe36 : SUCCESS in 1h 04m 31s\n- check-tempest-dsvm-neutron http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron/732fb51 : SUCCESS in 48m 22s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron-heat-slow/0de21c1 : SUCCESS in 22m 54s\n- gate-tempest-pep8 http://logs.openstack.org/52/83952/1/check/gate-tempest-pep8/f695ccc : SUCCESS in 3m 58s\n- gate-tempest-docs http://logs.openstack.org/52/83952/1/check/gate-tempest-docs/ce0fdc0 : SUCCESS in 2m 39s\n- gate-tempest-python27 http://logs.openstack.org/52/83952/1/check/gate-tempest-python27/b510593 : SUCCESS in 3m 05s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-postgres-full/ca3ce2c : SUCCESS in 54m 05s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron-full/6d3a1ca : SUCCESS in 1h 04m 04s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-cells-full/8211b79 : FAILURE in 32m 47s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/52/83952/1/check/gate-tempest-dsvm-large-ops/f209532 : SUCCESS in 15m 18s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/52/83952/1/check/gate-tempest-dsvm-neutron-large-ops/705438b : SUCCESS in 18m 13s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm-partial-ncpu/1a899ed : SUCCESS in 36m 02s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-ironic-nv/5cf3031 : SUCCESS in 32m 06s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-ironic-postgres-nv/896deb0 : SUCCESS in 36m 54s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-full-nv/5e48155 : SUCCESS in 1h 03m 53s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-postgres-nv/fa0d174 : SUCCESS in 27m 42s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-neutron-nv/ca7ec7a : SUCCESS in 52m 37s (non-voting)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"2575a7aa8156be2656891add0d3373a60f5c1682","author":{"_account_id":7139,"name":"Zhi Kun Liu","username":"zhikunliu","inactive":true},"date":"2014-04-02 08:00:07.000000000","message":"Patch Set 1: Looks good to me, but someone else must approve\n\n","accounts_in_message":[],"_revision_number":1},{"id":"d580352bbf0404cc03fc4a00f1e8177c5ffd493d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-02 08:00:08.000000000","message":"Patch Set 1: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"e7a2dc28c4f14a1fed11404b322af1bdd5d0cfd9","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-02 08:59:39.000000000","message":"Patch Set 1: Doesn\u0027t seem to work\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- check-grenade-dsvm http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm/d74a3e5 : SUCCESS in 39m 56s\n- check-tempest-dsvm-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-full/a5a37d8 : FAILURE in 37m 14s\n- check-tempest-dsvm-neutron http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron/e10eca6 : SUCCESS in 49m 13s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron-heat-slow/ddaba6c : SUCCESS in 21m 37s\n- gate-tempest-pep8 http://logs.openstack.org/52/83952/1/check/gate-tempest-pep8/a2aab53 : SUCCESS in 3m 35s\n- gate-tempest-docs http://docs-draft.openstack.org/52/83952/1/check/gate-tempest-docs/c521532/doc/build/html/ : SUCCESS in 4m 04s\n- gate-tempest-python27 http://logs.openstack.org/52/83952/1/check/gate-tempest-python27/7f778b2 : SUCCESS in 3m 24s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-postgres-full/7aec6fa : SUCCESS in 40m 07s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron-full/8670b1a : SUCCESS in 45m 13s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-cells-full/7229f90 : FAILURE in 32m 05s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/52/83952/1/check/gate-tempest-dsvm-large-ops/8f5626b : SUCCESS in 13m 58s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/52/83952/1/check/gate-tempest-dsvm-neutron-large-ops/1905e00 : SUCCESS in 16m 53s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm-partial-ncpu/947f369 : SUCCESS in 35m 21s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-ironic-nv/818afeb : SUCCESS in 34m 18s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-ironic-postgres-nv/3233196 : SUCCESS in 34m 54s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-full-nv/bce91e8 : SUCCESS in 53m 36s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-postgres-nv/6f6bdb1 : SUCCESS in 24m 52s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-neutron-nv/9e8adbb : SUCCESS in 57m 12s (non-voting)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"b9671dfaf3acdc684e8884485618f6bf7fee0b26","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-04-02 09:00:00.000000000","message":"Patch Set 1:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- check-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1295876\n\nWe don\u0027t automatically recheck or reverify, so please consider\ndoing that manually if someone hasn\u0027t already. For a code review\nwhich is not yet approved, you can recheck by leaving a code\nreview comment with just the text:\n\n    recheck bug 1295876\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":1},{"id":"02a4aac851ce91f78818843b8ac3c6d688a5b5cb","author":{"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},"date":"2014-04-02 09:42:31.000000000","message":"Patch Set 1:\n\nrecheck bug 1295876","accounts_in_message":[],"_revision_number":1},{"id":"ee910d89a9219a825d856ae7e27bb76c1cea5884","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-02 09:42:32.000000000","message":"Patch Set 1: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"43277541dceaacd2c904f6088ac31ee36611ae3a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-02 10:49:32.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- check-grenade-dsvm http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm/5a5adbf : SUCCESS in 42m 31s\n- check-tempest-dsvm-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-full/f889b4e : SUCCESS in 1h 00m 17s\n- check-tempest-dsvm-neutron http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron/4f00633 : SUCCESS in 54m 13s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron-heat-slow/ccfb355 : SUCCESS in 26m 46s\n- gate-tempest-pep8 http://logs.openstack.org/52/83952/1/check/gate-tempest-pep8/c729d16 : SUCCESS in 4m 13s\n- gate-tempest-docs http://docs-draft.openstack.org/52/83952/1/check/gate-tempest-docs/87b1893/doc/build/html/ : SUCCESS in 3m 35s\n- gate-tempest-python27 http://logs.openstack.org/52/83952/1/check/gate-tempest-python27/dcf26e6 : SUCCESS in 3m 31s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-postgres-full/0ad5c90 : SUCCESS in 50m 34s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron-full/d331944 : FAILURE in 56m 36s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-cells-full/c48e780 : FAILURE in 27m 39s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/52/83952/1/check/gate-tempest-dsvm-large-ops/b19fc4f : SUCCESS in 17m 03s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/52/83952/1/check/gate-tempest-dsvm-neutron-large-ops/b744721 : SUCCESS in 16m 17s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm-partial-ncpu/35c0b6c : SUCCESS in 35m 23s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-ironic-nv/0e5bfa0 : SUCCESS in 33m 12s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-ironic-postgres-nv/2c1bd36 : SUCCESS in 32m 48s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-full-nv/94a96e1 : SUCCESS in 1h 01m 56s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-postgres-nv/72c21d0 : SUCCESS in 21m 10s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-neutron-nv/1a346d7 : SUCCESS in 56m 46s (non-voting)\n\n","accounts_in_message":[],"_revision_number":1},{"id":"0e3cd6d63024cf0922f800b5adb92759341aacce","author":{"_account_id":10385,"name":"Citrix XenServer CI","username":"citrix_xenserver_ci","tags":["SERVICE_USER"]},"date":"2014-04-02 18:45:29.000000000","message":"Patch Set 1:\n\nPassed using XenAPI driver with XenServer 6.2: Logs at http://dd6b71949550285df7dc-dda4e480e005aaa13ec303551d2d8155.r49.cf1.rackcdn.com/52/83952/1/results.html\n\nStandard recheck supported; use \"recheck xenserver\" to trigger only xenserver re-check.  XenServer CI contact: openstack@citrix.com.","accounts_in_message":[],"_revision_number":1},{"id":"f16dddd30648afeca95ae27630066775f36f937d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-10 10:42:16.000000000","message":"Patch Set 1: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"abfbf13204d471ce8f758a790c42288c01fa20ae","author":{"_account_id":6167,"name":"Ken\u0027ichi Ohmichi","email":"ken1ohmichi@gmail.com","username":"oomichi"},"date":"2014-04-10 10:42:16.000000000","message":"Patch Set 1: Looks good to me (core reviewer)\n\nLGTM, thanks","accounts_in_message":[],"_revision_number":1},{"id":"bcde235f3f29d47449cbbfe1429db792c944e18d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-10 12:07:02.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- check-tempest-dsvm-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-full/3186f0f : SUCCESS in 56m 43s\n- check-tempest-dsvm-neutron http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron/c76fc35 : SUCCESS in 56m 30s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron-heat-slow/2be9b2e : SUCCESS in 20m 53s\n- check-grenade-dsvm http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm/72d4ac3 : SUCCESS in 37m 57s\n- check-grenade-dsvm-neutron http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm-neutron/f3ab680 : SUCCESS in 53m 26s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/52/83952/1/check/gate-tempest-pep8/8f438e5 : SUCCESS in 3m 38s\n- gate-tempest-docs http://docs-draft.openstack.org/52/83952/1/check/gate-tempest-docs/0a5ae9d/doc/build/html/ : SUCCESS in 3m 29s\n- gate-tempest-python27 http://logs.openstack.org/52/83952/1/check/gate-tempest-python27/eb8134a : SUCCESS in 2m 46s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-postgres-full/397b8f1 : SUCCESS in 57m 54s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron-full/4d6bba4 : SUCCESS in 59m 35s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-cells-full/3a4e45e : FAILURE in 31m 52s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/52/83952/1/check/gate-tempest-dsvm-large-ops/bf449f7 : SUCCESS in 14m 34s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/52/83952/1/check/gate-tempest-dsvm-neutron-large-ops/679a55a : SUCCESS in 15m 44s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm-partial-ncpu/b2e36c5 : SUCCESS in 35m 39s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-ironic-nv/80226bc : SUCCESS in 37m 18s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-ironic-postgres-nv/f2959bf : SUCCESS in 36m 09s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-full-nv/b3324ed : SUCCESS in 50m 40s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-postgres-nv/31c261b : SUCCESS in 25m 59s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-neutron-nv/15d4376 : SUCCESS in 55m 03s (non-voting)\n- check-tempest-master-dsvm-full-havana http://logs.openstack.org/52/83952/1/check/check-tempest-master-dsvm-full-havana/c62cc3f : FAILURE in 1h 17m 55s (non-voting)\n- check-tempest-master-dsvm-neutron-havana http://logs.openstack.org/52/83952/1/check/check-tempest-master-dsvm-neutron-havana/2fd5c6e : FAILURE in 1h 24m 40s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"030e5eb6fededbabf3f62eebb4342d9374d607ba","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-14 20:09:32.000000000","message":"Patch Set 1: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"a68eb3f673a2d3577b4f6bfa7542cd573191f962","author":{"_account_id":1192,"name":"David Kranz","email":"dkranz@redhat.com","username":"david-kranz"},"date":"2014-04-14 20:09:32.000000000","message":"Patch Set 1: Approved\n\n","accounts_in_message":[],"_revision_number":1},{"id":"6b2cf608a6ebafa993c5db1483ca471a3ee80a01","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-14 21:47:07.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- check-tempest-dsvm-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-full/7945fba : SUCCESS in 1h 08m 11s\n- check-tempest-dsvm-neutron http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron/ef21360 : SUCCESS in 49m 23s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron-heat-slow/88a158f : SUCCESS in 21m 33s\n- check-grenade-dsvm http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm/2db794e : SUCCESS in 35m 59s\n- check-grenade-dsvm-neutron http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm-neutron/02d6154 : FAILURE in 27m 11s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/52/83952/1/check/gate-tempest-pep8/1891e52 : SUCCESS in 3m 57s\n- gate-tempest-docs http://docs-draft.openstack.org/52/83952/1/check/gate-tempest-docs/903c497/doc/build/html/ : SUCCESS in 4m 06s\n- gate-tempest-python27 http://logs.openstack.org/52/83952/1/check/gate-tempest-python27/4cc8628 : SUCCESS in 4m 04s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-postgres-full/061ef19 : SUCCESS in 49m 40s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron-full/649a853 : SUCCESS in 1h 10m 43s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-cells-full/dd0b3fe : FAILURE in 27m 25s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/52/83952/1/check/gate-tempest-dsvm-large-ops/3683878 : SUCCESS in 15m 55s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/52/83952/1/check/gate-tempest-dsvm-neutron-large-ops/7226d12 : SUCCESS in 18m 26s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm-partial-ncpu/e4907d7 : SUCCESS in 40m 21s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-ironic-nv/69918b7 : FAILURE in 35m 20s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-ironic-postgres-nv/1a7b930 : SUCCESS in 36m 34s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-full-nv/caece58 : FAILURE in 1h 08m 53s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-postgres-nv/3741c90 : SUCCESS in 25m 06s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-neutron-nv/42b98cf : SUCCESS in 55m 37s (non-voting)\n- check-tempest-master-dsvm-full-havana http://logs.openstack.org/52/83952/1/check/check-tempest-master-dsvm-full-havana/f2bff84 : FAILURE in 1h 17m 14s (non-voting)\n- check-tempest-master-dsvm-neutron-havana http://logs.openstack.org/52/83952/1/check/check-tempest-master-dsvm-neutron-havana/90608bd : FAILURE in 1h 21m 58s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"1fa658c92c79700c476e9d5f14a0c501708efc01","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-14 21:47:10.000000000","message":"Patch Set 1: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"5ff79ad18ed40e25c2daa7efa5863f7c742bf86c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-14 23:05:58.000000000","message":"Patch Set 1:\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"9b0fcaf7986e13a155de7767136dfa06f560ab74","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-15 00:13:23.000000000","message":"Patch Set 1: Fails\n\nBuild failed.  For information on how to proceed, see https://wiki.openstack.org/wiki/GerritJenkinsGit#Test_Failures\n\n- gate-tempest-dsvm-full http://logs.openstack.org/52/83952/1/gate/gate-tempest-dsvm-full/a76c309 : FAILURE in 1h 01m 05s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/52/83952/1/gate/gate-tempest-dsvm-neutron/466aa54 : SUCCESS in 1h 05m 32s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/52/83952/1/gate/gate-tempest-dsvm-neutron-heat-slow/8c96313 : SUCCESS in 19m 03s\n- gate-grenade-dsvm http://logs.openstack.org/52/83952/1/gate/gate-grenade-dsvm/1acc981 : SUCCESS in 44m 41s\n- gate-grenade-dsvm-neutron http://logs.openstack.org/52/83952/1/gate/gate-grenade-dsvm-neutron/a1e4df9 : FAILURE in 28m 27s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/52/83952/1/gate/gate-tempest-pep8/224d1f8 : SUCCESS in 3m 29s\n- gate-tempest-docs http://docs-draft.openstack.org/52/83952/1/gate/gate-tempest-docs/d7adc34/doc/build/html/ : SUCCESS in 3m 23s\n- gate-tempest-python27 http://logs.openstack.org/52/83952/1/gate/gate-tempest-python27/300397a : SUCCESS in 4m 00s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/52/83952/1/gate/gate-tempest-dsvm-postgres-full/07230cd : SUCCESS in 1h 01m 00s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/52/83952/1/gate/gate-tempest-dsvm-large-ops/0c16456 : SUCCESS in 14m 53s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/52/83952/1/gate/gate-tempest-dsvm-neutron-large-ops/72e2b19 : SUCCESS in 18m 37s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/52/83952/1/gate/gate-grenade-dsvm-partial-ncpu/a9fe454 : SUCCESS in 41m 18s\n","accounts_in_message":[],"_revision_number":1},{"id":"f5f1250f49df55159050e6374e84582b4d21810a","author":{"_account_id":8871,"name":"Elastic Recheck","username":"elasticrecheck"},"date":"2014-04-15 00:24:28.000000000","message":"Patch Set 1:\n\nI noticed jenkins failed, I think you hit bug(s):\n\n- gate-grenade-dsvm-neutron: unrecognized error\n- gate-tempest-dsvm-full: https://bugs.launchpad.net/bugs/1302774\n\nYou have some unrecognized errors.\nFor bug details see: http://status.openstack.org/elastic-recheck/","accounts_in_message":[],"_revision_number":1},{"id":"f7b60cc72081b28fd9d86aa7a438f16bac50318a","author":{"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},"date":"2014-04-15 09:36:37.000000000","message":"Patch Set 1:\n\nreverify bug 1302774","accounts_in_message":[],"_revision_number":1},{"id":"a03823517f90f84ad7c701c4ea4feb3b232ed999","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-15 09:36:38.000000000","message":"Patch Set 1: No score\n\nStarting check jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"b63eec7c7ae2e1881c65926a2abd9a7fcbded89f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-15 11:04:41.000000000","message":"Patch Set 1: Works for me\n\nBuild succeeded.\n\n- check-tempest-dsvm-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-full/f4b22d3 : SUCCESS in 1h 02m 05s\n- check-tempest-dsvm-neutron http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron/7c6eac5 : SUCCESS in 51m 55s\n- check-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron-heat-slow/08d165c : SUCCESS in 23m 06s\n- check-grenade-dsvm http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm/7de53a5 : SUCCESS in 39m 04s\n- check-grenade-dsvm-neutron http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm-neutron/d923338 : FAILURE in 25m 50s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/52/83952/1/check/gate-tempest-pep8/32be45f : SUCCESS in 3m 47s\n- gate-tempest-docs http://docs-draft.openstack.org/52/83952/1/check/gate-tempest-docs/a1e51d2/doc/build/html/ : SUCCESS in 4m 28s\n- gate-tempest-python27 http://logs.openstack.org/52/83952/1/check/gate-tempest-python27/f28b4df : SUCCESS in 3m 27s\n- check-tempest-dsvm-postgres-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-postgres-full/fe5ad7d : SUCCESS in 1h 05m 30s\n- check-tempest-dsvm-neutron-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-neutron-full/81ffa97 : SUCCESS in 52m 06s (non-voting)\n- check-tempest-dsvm-cells-full http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-cells-full/ca5b64a : FAILURE in 43m 26s (non-voting)\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/52/83952/1/check/gate-tempest-dsvm-large-ops/f1bdc97 : SUCCESS in 15m 05s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/52/83952/1/check/gate-tempest-dsvm-neutron-large-ops/875c7bc : SUCCESS in 17m 14s\n- check-grenade-dsvm-partial-ncpu http://logs.openstack.org/52/83952/1/check/check-grenade-dsvm-partial-ncpu/b55b792 : SUCCESS in 45m 57s\n- check-tempest-dsvm-ironic-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-ironic-nv/b22bd6e : SUCCESS in 36m 59s (non-voting)\n- check-tempest-dsvm-ironic-postgres-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-ironic-postgres-nv/90cf96b : SUCCESS in 32m 12s (non-voting)\n- check-tempest-dsvm-sahara-full-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-full-nv/0f8aa25 : SUCCESS in 50m 31s (non-voting)\n- check-tempest-dsvm-sahara-postgres-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-postgres-nv/23bb235 : SUCCESS in 29m 00s (non-voting)\n- check-tempest-dsvm-sahara-neutron-nv http://logs.openstack.org/52/83952/1/check/check-tempest-dsvm-sahara-neutron-nv/a832904 : SUCCESS in 53m 57s (non-voting)\n- check-tempest-master-dsvm-full-havana http://logs.openstack.org/52/83952/1/check/check-tempest-master-dsvm-full-havana/c42d9bc : FAILURE in 1h 13m 01s (non-voting)\n- check-tempest-master-dsvm-neutron-havana http://logs.openstack.org/52/83952/1/check/check-tempest-master-dsvm-neutron-havana/9471c8e : FAILURE in 1h 15m 03s (non-voting)\n","accounts_in_message":[],"_revision_number":1},{"id":"272fed277e2cd41806ae4938b31a09f06eda008d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-15 11:04:44.000000000","message":"Patch Set 1: No score\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"81fea573297a04271bd13b95494cc6cdc7fa0568","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-15 12:07:48.000000000","message":"Patch Set 1: Verified\n\nBuild succeeded.\n\n- gate-tempest-dsvm-full http://logs.openstack.org/52/83952/1/gate/gate-tempest-dsvm-full/7924c38 : SUCCESS in 1h 02m 43s\n- gate-tempest-dsvm-neutron http://logs.openstack.org/52/83952/1/gate/gate-tempest-dsvm-neutron/c3ff24c : SUCCESS in 50m 57s\n- gate-tempest-dsvm-neutron-heat-slow http://logs.openstack.org/52/83952/1/gate/gate-tempest-dsvm-neutron-heat-slow/caef3f3 : SUCCESS in 21m 37s\n- gate-grenade-dsvm http://logs.openstack.org/52/83952/1/gate/gate-grenade-dsvm/996aca5 : SUCCESS in 43m 52s\n- gate-grenade-dsvm-neutron http://logs.openstack.org/52/83952/1/gate/gate-grenade-dsvm-neutron/3b9313e : FAILURE in 27m 44s (non-voting)\n- gate-tempest-pep8 http://logs.openstack.org/52/83952/1/gate/gate-tempest-pep8/ae08541 : SUCCESS in 3m 47s\n- gate-tempest-docs http://docs-draft.openstack.org/52/83952/1/gate/gate-tempest-docs/c34a37b/doc/build/html/ : SUCCESS in 4m 29s\n- gate-tempest-python27 http://logs.openstack.org/52/83952/1/gate/gate-tempest-python27/280c638 : SUCCESS in 2m 52s\n- gate-tempest-dsvm-postgres-full http://logs.openstack.org/52/83952/1/gate/gate-tempest-dsvm-postgres-full/7aedb4c : SUCCESS in 52m 51s\n- gate-tempest-dsvm-large-ops http://logs.openstack.org/52/83952/1/gate/gate-tempest-dsvm-large-ops/cae23d9 : SUCCESS in 19m 53s\n- gate-tempest-dsvm-neutron-large-ops http://logs.openstack.org/52/83952/1/gate/gate-tempest-dsvm-neutron-large-ops/02c204c : SUCCESS in 16m 24s\n- gate-grenade-dsvm-partial-ncpu http://logs.openstack.org/52/83952/1/gate/gate-grenade-dsvm-partial-ncpu/0dc13d7 : SUCCESS in 38m 27s\n","accounts_in_message":[],"_revision_number":1},{"id":"db07edb615fb1adc94cdc0c938b02739f6a98b94","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2014-04-15 12:07:49.000000000","message":"Change has been successfully merged into the git repository.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6c2080180c3544d42530e53ba22daac4fb7781e8","revisions":{"6c2080180c3544d42530e53ba22daac4fb7781e8":{"kind":"REWORK","_number":1,"created":"2014-03-29 05:44:13.000000000","uploader":{"_account_id":6072,"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","username":"enikanorov"},"ref":"refs/changes/52/83952/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/52/83952/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/83952/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/83952/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/52/83952/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/52/83952/1"}}},"commit":{"parents":[{"commit":"59dd4247ada671aef05adfc7ed32efe9a4b6ac77","subject":"Merge \"Typo in config.py\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/59dd4247ada671aef05adfc7ed32efe9a4b6ac77"}]}],"author":{"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","date":"2014-03-29 05:34:20.000000000","tz":240},"committer":{"name":"Eugene Nikanorov","email":"enikanorov@mirantis.com","date":"2014-03-29 05:34:20.000000000","tz":240},"subject":"Remove variable part of agent dict to fix AgentManagementTest","message":"Remove variable part of agent dict to fix AgentManagementTest\n\nAgents may vary their configuration while API tests blindly\ncompare agent dicts. Need to pop items that may change.\n\nChange-Id: I150f4e9f2e79f6e20f18dee69cf061c68a5bed68\nCloses-Bug: #1299350\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6c2080180c3544d42530e53ba22daac4fb7781e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6c2080180c3544d42530e53ba22daac4fb7781e8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"59dd4247ada671aef05adfc7ed32efe9a4b6ac77","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
