)]}'
{"id":"openstack%2Ftempest~846242","triplet_id":"openstack%2Ftempest~master~I0214df9b32e8773706d436aa252cf490aef5af44","project":"openstack/tempest","branch":"master","topic":"master2","attention_set":{},"removed_from_attention_set":{"11075":{"account":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"last_update":"2023-06-07 23:48:13.000000000","reason":"Change was abandoned"},"30409":{"account":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"last_update":"2023-06-07 23:48:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I0214df9b32e8773706d436aa252cf490aef5af44","subject":"Test update volume type details with wrong volume name","status":"ABANDONED","created":"2022-06-17 02:15:58.000000000","updated":"2023-06-07 23:48:13.000000000","total_comment_count":5,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"7e8f2fe22b93baac6cef3eafb040e03175939f3b","_number":846242,"virtual_id_number":846242,"owner":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-06-20 08:10:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2022-06-20 07:38:35.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},{"value":1,"date":"2022-07-14 10:05:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":-1,"date":"2022-08-06 15:38:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-07-12 10:40:11.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-17 04:52:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-20 07:38:31.000000000","updated_by":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"reviewer":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"state":"CC"},{"updated":"2022-06-20 07:38:35.000000000","updated_by":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"reviewer":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"state":"REVIEWER"},{"updated":"2022-07-12 10:40:11.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2022-07-14 10:05:03.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-08-06 15:38:06.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"8bd757f253381b56ddcca1ea6f284c246463165d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2022-06-17 02:15:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67f806b6d47daa57d76c9391882dc0b6dcf5c6ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 04:52:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4e42fe788fcd4ed08ae2f99cebb50135 : SUCCESS in 54m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a6850a0d1234bb3af0449699328dec4 : FAILURE in 1h 08m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64a877908f4f4e8aa7656d510a32f1a1 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e8e9ce0461f4e4cb6e5c4f2b075f722 : SUCCESS in 15m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c734b513bba478c8cdcae2e89e94450 : FAILURE in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c2db895ac4b404e8f69cfd17bae883c : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/41314bc728d8423990f46a22330b1b03 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3220b4c515548549c79131e9918ed1d : SUCCESS in 5m 11s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/0d9ddc1dcd144e48a43d9232d23274c4 : FAILURE in 1h 39m 41s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/a4f2a364f8534be98b78a3018e2566e6 : FAILURE in 1h 06m 41s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0064f0b973db401eabfa6fc42036a5ad : FAILURE in 1h 10m 59s (non-voting)\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/2d6b121c4a1f4ef089aabf16f1781a46 : FAILURE in 1h 21m 14s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/1139baecdbb34da68da2b61bacfdbb4e : FAILURE in 1h 06m 23s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/6aa86c1316954d2884ddd0f36446b635 : FAILURE in 1h 15m 51s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/bfa2a17e8a524b44870c0cec33d29a23 : SUCCESS in 2h 22m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6641e7e3037b4b5d8e8cd95e5b1d6358 : FAILURE in 1h 18m 24s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e7267c07e18e45ab8f211f384968e6e8 : SUCCESS in 32m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/299f21f9519749a5a2486714e33edf8c : SUCCESS in 45m 59s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b0bf9e58e204472a8201f9634e91b444 : SUCCESS in 2h 07m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/48511ca584d74a3085ee9e69fd28be29 : FAILURE in 53m 16s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d25589ddc4ab4a3284d7ded4a862c3eb : FAILURE in 1h 18m 39s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/19f1eeb6756f49589191d5380f8019ea : SUCCESS in 57m 35s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/873c6b27c5dd4e9391ac063d8f390938 : SUCCESS in 59m 34s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/267f89332cac450988dbd68c9832249c : SUCCESS in 4m 17s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2fb7eb9365214859b025cf872a962f52 : FAILURE in 1h 07m 01s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/bf23784b776b44388b2439489758d684 : FAILURE in 1h 02m 32s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/feee7734a3164825aa425952e57aac4f : SUCCESS in 4m 39s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/615b120d30494254986df2ffb81e7be2 : FAILURE in 1h 18m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4a533f3062bf01ca526e32abc38d3d2aeebfdc18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2022-06-17 08:36:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ca99f91a2ac48b05eb43ea994f93d5eac128bc51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2022-06-17 10:03:32.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8b407be4906d9c117dca241d9c8425dc5e655c08","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 12:32:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/22158b006f1c482da5971365ec14fa51 : SUCCESS in 46m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9f05c7a61469405c96950718be0a05ac : SUCCESS in 1h 32m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae5e2f8932584164a3aabc820853f54f : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026530b44f8c4775bab41eef57075a7f : SUCCESS in 12m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f73e0cc49c95485d9762ea3c50f2253b : FAILURE in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e22c378984fa4091acf376de915c27ab : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be89e70051444b78a8eef6cd18189b4a : SUCCESS in 4m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5da57795faf44a4a9343cc4b7e0a1b6f : SUCCESS in 4m 24s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/a5d0e15b46164c9e9cc7b938ee6a732c : SUCCESS in 1h 49m 18s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/f610062bb7a94058a228800a4286d623 : SUCCESS in 1h 22m 36s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3320b93adfa8491c917c843bbf5e9817 : SUCCESS in 1h 17m 48s (non-voting)\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/b97a74488e3b4102b8238d99b3414a87 : SUCCESS in 1h 39m 13s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/ee6ffc5f141b449a96e673a4cca406d2 : SUCCESS in 1h 19m 56s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/d241653403254bfeab3a8555d50c04d1 : SUCCESS in 1h 21m 59s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/28eab84c397a4e55b857396688b3f7f1 : SUCCESS in 2h 26m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5ef91dfae46d42d2807902b1756f4207 : SUCCESS in 1h 35m 35s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/78b15e53fa1d4a48b12908d4f73b7165 : SUCCESS in 26m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eb24b3a4c9594c09911eda2ed4c4d75b : SUCCESS in 43m 23s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bbac95797884443b92747b0e4675b665 : SUCCESS in 2h 24m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bc2f1343b71648bda86bb041b984dbfd : FAILURE in 57m 10s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/967591f0eb7b4f6f9da270d6dd0667ef : SUCCESS in 1h 13m 50s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fdeec394803e43f2b31fa883d6521912 : SUCCESS in 45m 20s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2f33f06a81b7400583894f32ce804186 : SUCCESS in 54m 42s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/5ca3c7736b7942cd913b9c5d292ea2ca : SUCCESS in 4m 54s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/689f7938464849e081d42a6ec55dfcc8 : FAILURE in 1h 09m 42s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ae90b8e82919493eb67e41115cbbf299 : FAILURE in 1h 06m 51s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/0990a6d087444c43b1299fbbd71f76a9 : SUCCESS in 3m 55s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/a1af5dd8ce104efea0116aa15f5e5b76 : FAILURE in 1h 33m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a9c4ee298a040be05a5bad522b0ef8be0fa3caa3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"date":"2022-06-20 06:00:17.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5259817c1c661fda42eb16ff25411f1228b6702d","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-06-20 07:38:31.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"712ca1b28e2dd97148bb999ddbcaff87c5164116","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-06-20 07:38:35.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"5d38a56d065c36e09d9cfd5b3dbc441ac49e49dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-20 08:10:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/95dfa0b4914a4152a363ba7a9f556d75 : SUCCESS in 48m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7466053bd4af4ea8b8c8c0db385e164d : SUCCESS in 1h 15m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef6cc99d0145485bb02ce6144ea67193 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d20d427afb614c479a15797dfa7b51f1 : SUCCESS in 12m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2114385e61441b19bc2bb3223b7d9b3 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1191468dda1e485e982e3440f8411e93 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f07be3d69ef04d799303556c4c837374 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21c69c505b7d4bf1aefc7be8eddfef38 : SUCCESS in 4m 33s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f90cb1f8e0d3432bae808978b45f0255 : SUCCESS in 1h 51m 41s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/e7e8bd8a1c774708a72764247a567450 : SUCCESS in 1h 04m 04s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ea893b6e575048d0a23b832c7ec2aeed : SUCCESS in 1h 06m 45s (non-voting)\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/dcdf526abd794f0a8dae6415dec3327d : SUCCESS in 1h 16m 09s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/ae53471d42ed491d8e7b1061c023f0bd : SUCCESS in 1h 07m 45s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/28b81c6265004fa5b3a0fe4b435b0a89 : SUCCESS in 1h 18m 22s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/0782ecb3d9b74e518dfed48ab7b88995 : SUCCESS in 2h 03m 28s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1120e5d87d7f48dc8f48fe6c0e27d73b : SUCCESS in 1h 20m 05s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/02a4855f7e5a440bb0625458132d1c93 : SUCCESS in 25m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ad8b0175976e44e1b51ce2838a527ce2 : SUCCESS in 49m 56s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/63c90a8cb6ef47f29ef02b7146db4dbb : SUCCESS in 1h 58m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3ce03291899e4f14b7c76a3311469b36 : SUCCESS in 48m 15s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d35c9e28501a4ddabe61dc84b2bd0741 : SUCCESS in 1h 15m 00s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/48a633c5586345eb80bbdbd69f86d2b3 : SUCCESS in 49m 15s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/b01af2c0d2fd4e42b8cc3c490e7173c9 : SUCCESS in 47m 44s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/c919b7e77ef743e18e687a877e0af74a : SUCCESS in 3m 39s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/446a689b74f64127abe40ce7b766790d : FAILURE in 1h 23m 28s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/7f7cb646727c49ae928d7b5ff788bcf5 : FAILURE in 1h 04m 51s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b342a62916f7452991e8754540413398 : SUCCESS in 2m 50s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/626690b4553e4514aa0a21cec9952a17 : FAILURE in 1h 28m 44s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"616ee516042b9d63436a7908c4d16f1db9a1c502","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2022-07-12 10:40:11.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b413b0b3cb926cd3965279040fdf600abd524716","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-07-14 10:05:03.000000000","message":"Patch Set 4: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"03cb6220bf56d5a651195152bc905eaf86833703","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-08-06 15:38:06.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7e8f2fe22b93baac6cef3eafb040e03175939f3b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-06-07 23:48:13.000000000","message":"Abandoned\n\nfeel free to restore it if you add other scenario positive case in this tests","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b07c328599d398c5fda012cfef357db6c13f8ac7","revisions":{"91bf273cc0c123a1c140fc50a381d0438a0fcfcc":{"kind":"REWORK","_number":1,"created":"2022-06-17 02:15:58.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/42/846242/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/42/846242/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/846242/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/846242/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/846242/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/42/846242/1"}}},"commit":{"parents":[{"commit":"9e5595f8baa4fd12617b4b5dd5f9dab98b6f1a20","subject":"Add release notes page for version 31.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9e5595f8baa4fd12617b4b5dd5f9dab98b6f1a20"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2022-06-17 02:11:12.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2022-06-17 02:13:43.000000000","tz":480},"subject":"Test update volume type details with wrong volume name","message":"Test update volume type details with wrong volume name\n\nChange-Id: I0214df9b32e8773706d436aa252cf490aef5af44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/91bf273cc0c123a1c140fc50a381d0438a0fcfcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/91bf273cc0c123a1c140fc50a381d0438a0fcfcc"}]},"branch":"refs/heads/master"},"5467ca9a4016e67b1b01a94e4c7951f6870ec4ac":{"kind":"REWORK","_number":2,"created":"2022-06-17 08:36:43.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/42/846242/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/42/846242/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/846242/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/846242/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/846242/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/42/846242/2"}}},"commit":{"parents":[{"commit":"9e5595f8baa4fd12617b4b5dd5f9dab98b6f1a20","subject":"Add release notes page for version 31.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9e5595f8baa4fd12617b4b5dd5f9dab98b6f1a20"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2022-06-17 02:11:12.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2022-06-17 08:26:17.000000000","tz":480},"subject":"Test update volume type details with wrong volume name","message":"Test update volume type details with wrong volume name\n\nUpdate volume type details with test with empty name\n\nChange-Id: I0214df9b32e8773706d436aa252cf490aef5af44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5467ca9a4016e67b1b01a94e4c7951f6870ec4ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5467ca9a4016e67b1b01a94e4c7951f6870ec4ac"}]},"branch":"refs/heads/master"},"cf0482005e2cf426648664383bf1c9d1dd396c8f":{"kind":"REWORK","_number":3,"created":"2022-06-17 10:03:32.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/42/846242/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/42/846242/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/846242/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/846242/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/846242/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/42/846242/3"}}},"commit":{"parents":[{"commit":"9e5595f8baa4fd12617b4b5dd5f9dab98b6f1a20","subject":"Add release notes page for version 31.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9e5595f8baa4fd12617b4b5dd5f9dab98b6f1a20"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2022-06-17 02:11:12.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2022-06-17 10:02:15.000000000","tz":480},"subject":"Test update volume type details with wrong volume name","message":"Test update volume type details with wrong volume name\n\nUpdate volume type details with test with empty name\n\nChange-Id: I0214df9b32e8773706d436aa252cf490aef5af44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cf0482005e2cf426648664383bf1c9d1dd396c8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cf0482005e2cf426648664383bf1c9d1dd396c8f"}]},"branch":"refs/heads/master"},"b07c328599d398c5fda012cfef357db6c13f8ac7":{"kind":"REWORK","_number":4,"created":"2022-06-20 06:00:17.000000000","uploader":{"_account_id":30409,"name":"wangzhiguang","email":"wangzhiguang@inspur.com","username":"wangzhiguang"},"ref":"refs/changes/42/846242/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/42/846242/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/846242/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/846242/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/42/846242/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/42/846242/4"}}},"commit":{"parents":[{"commit":"9e5595f8baa4fd12617b4b5dd5f9dab98b6f1a20","subject":"Add release notes page for version 31.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9e5595f8baa4fd12617b4b5dd5f9dab98b6f1a20"}]}],"author":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2022-06-17 02:11:12.000000000","tz":480},"committer":{"name":"wangzhiguang","email":"wangzhiguang@inspur.com","date":"2022-06-20 05:59:11.000000000","tz":480},"subject":"Test update volume type details with wrong volume name","message":"Test update volume type details with wrong volume name\n\nUpdate volume type details with test with empty name\n\nChange-Id: I0214df9b32e8773706d436aa252cf490aef5af44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b07c328599d398c5fda012cfef357db6c13f8ac7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b07c328599d398c5fda012cfef357db6c13f8ac7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
