)]}'
{"id":"openstack%2Ftempest~846251","triplet_id":"openstack%2Ftempest~master~I005755b0e527255362923a6b4bbbba2f7c1dfbd1","project":"openstack/tempest","branch":"master","topic":"wait_until_sshable_pingable","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2022-06-20 15:18:06.000000000","reason":"Change was abandoned"},"10135":{"account":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"last_update":"2022-06-20 15:18:06.000000000","reason":"Change was abandoned"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-06-17 12:06:43.000000000","reason":"Martin Kopec replied on the change"},"9708":{"account":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"last_update":"2022-06-20 15:18:06.000000000","reason":"Change was abandoned"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-06-17 14:54:30.000000000","reason":"Ghanshyam Mann replied on the change"},"8367":{"account":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"last_update":"2022-06-20 15:18:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I005755b0e527255362923a6b4bbbba2f7c1dfbd1","subject":"Revert \"compute: Move volume attached live migration tests to use SSHABLE\"","status":"ABANDONED","created":"2022-06-17 11:03:16.000000000","updated":"2022-06-20 15:18:06.000000000","total_comment_count":10,"unresolved_comment_count":2,"has_review_started":true,"revert_of":817772,"meta_rev_id":"ccc889f298629bc0c57fa1431b25d64095029ea1","_number":846251,"virtual_id_number":846251,"owner":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2022-06-17 17:25:50.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"date":"2022-06-20 15:17:19.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-06-17 14:47:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2022-06-17 17:52:34.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"date":"2022-06-17 14:50:55.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-06-17 11:03:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":12393,"name":"chandan kumar","display_name":"Chandan Kumar","email":"chkumar@redhat.com","username":"chkumar246"},"state":"CC"},{"updated":"2022-06-17 11:03:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2022-06-17 11:03:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":10135,"name":"Lee Yarwood","display_name":"Lee Yarwood","email":"lyarwood@redhat.com","username":"lyarwood"},"state":"REVIEWER"},{"updated":"2022-06-17 11:03:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-06-17 11:03:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2022-06-17 11:03:16.000000000","updated_by":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-06-17 12:24:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2022-06-17 14:39:31.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"1a1a6b1ffb44cae2c4995eaac3a585f12687c4dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-06-17 11:03:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7fc551e6a2e8672218d66521c9a8edc3d5cb18a4","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-06-17 12:06:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27a2526ecc581afb7fa9717ed9c0e7020922cd80","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-06-17 12:24:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"066bcdc36229eee4c4fcdb7bce5f1d3343921ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 13:11:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- grenade https://zuul.opendev.org/t/openstack/build/853ebf6fb6c142ae9a9b4560c5f22f3b : SUCCESS in 1h 00m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8a111252c7434e1491b95ade12658345 : SUCCESS in 1h 34m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0904044b8c7f43e1891752746f541309 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19ec0c0e619846e89122a4612c04c3ca : SUCCESS in 17m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/600dd5430ef14bb88ad6ba39d59e9f8e : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab9c16c9527d43abb59f82dc4929099f : SUCCESS in 5m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f434e2e699543d79c902eed67b0c0c3 : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa89a14a43cd49f4abca282906dc1590 : SUCCESS in 5m 26s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/c0f6f19be496427d808543c8704d0761 : SUCCESS in 1h 52m 47s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/67b7159be2b244b8a735e1ea4bc6a1e2 : SUCCESS in 1h 03m 04s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6cbeada9e24143d3baf318fc38b413e8 : SUCCESS in 1h 15m 56s (non-voting)\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/473f0db0a5664aad99a05c70772aa11a : SUCCESS in 1h 44m 00s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/edf9a203307f42a9ba41f4d811b3ea5a : SUCCESS in 1h 10m 33s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/c92d3a02a3764f208ba497eb7bf3df73 : SUCCESS in 1h 21m 19s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/c5a95724a63c4cb2a942e4b67f6c8315 : SUCCESS in 1h 58m 42s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/51cac9ab0199477fa4608ee8a50c954e : SUCCESS in 1h 31m 40s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/2b4ac8c583ae4249a8808ade37dc5185 : SUCCESS in 25m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14af2cd805a94a8592736637121230dc : SUCCESS in 48m 11s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/7f0635b920a04c04897897ed897bde15 : SUCCESS in 2h 03m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/43c7c48feba84b2f8ef3d29c594ab990 : SUCCESS in 49m 06s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/693f607021b940e99e4c7e866c78d734 : TIMED_OUT in 2h 04m 37s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4b4eff261d15415cb1b6d697e82c2697 : SUCCESS in 48m 36s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2df86a32146c41f6ac67e06ce69a3a02 : SUCCESS in 54m 56s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/7466e416629d423a8af63c32c827c81b : SUCCESS in 5m 12s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/e1796927e04c4c4bb237c3a9c91682cb : FAILURE in 1h 04m 10s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/7c31b782abd149c988c838e62d94d168 : FAILURE in 57m 50s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/37c038779dde402e87680d1366af7a26 : SUCCESS in 4m 24s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2586753c21a4be5a4480ee90c47cfd2 : FAILURE in 1h 20m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cc6429f076609a92408f51fd8cba16ddcf267607","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 13:12:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6d3127cc227b2b21776a89e2751c476e62ac451c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-06-17 14:39:31.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ab71c25705357bbfff4057c965e9a6401f406bfa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-06-17 14:43:06.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"c4e21950baa3993a8406c3ee329f69de7418ecde","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-06-17 14:43:22.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"ea47e5f90e4098fefa645132e1062ec3846aa8ef","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-06-17 14:47:08.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4ac6c51c5cacbbcc116130814f5b26f6736cafd0","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-06-17 14:47:37.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"3b150728d64a0c807aa9cb7120559e6b20c5120e","author":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"date":"2022-06-17 14:50:55.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"553c6c6d294e8e60fc473579ff8d20a328e633fd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-06-17 14:54:30.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"cc8f3b500505cfa5a8d299e8868971a33f17df3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 17:25:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/123655d98dcb4345b9bd771063d04f2d : SUCCESS in 1h 00m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b985fc0edbe14b19afd5d3c7c2bcb1a6 : SUCCESS in 1h 32m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/11667713c9294c128f83c6cbcef72d1f : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5247b36ff19413c8ce7aad8e7caa128 : SUCCESS in 13m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94e4a074b95c4b09b71e72349888fa65 : SUCCESS in 5m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3d364f411ae9436bb5ffd507f3146d80 : SUCCESS in 5m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdbeb74cb87240f8bd4429c0dd87ff2e : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7e9da997ff64d9face3db76bcaed254 : SUCCESS in 4m 59s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/2c62a6fcc5924562bdb8bfdd0108beab : SUCCESS in 1h 43m 41s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/f471251639cb436990b2987709e14bac : SUCCESS in 1h 19m 38s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1d8e03b9d6204aebb71e12536f3a0cd2 : SUCCESS in 1h 28m 39s (non-voting)\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/9ec66ccccc304c92820cf94a8aee7252 : SUCCESS in 1h 30m 18s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/45fc7a90424e46e6b0ab6eda5d208409 : FAILURE in 1h 20m 24s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/b008992c6b1e4cb180870daada65641c : SUCCESS in 1h 31m 01s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/a4a5ae101c3646b8adfab2587a8e387a : SUCCESS in 2h 11m 45s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cddf17baa6af4aa18de4b534bbf0edeb : SUCCESS in 1h 37m 54s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/8ff3480a2d684c758a42954c9bc76a54 : SUCCESS in 30m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/86107517b2014a308ae6e36c893e9f0c : SUCCESS in 52m 09s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/250402ed2d0648bf9b05169f94b9f1aa : SUCCESS in 2h 33m 56s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/48f691a62f2343d4bff3e3066c1ce6b7 : SUCCESS in 49m 24s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6107dd5c71c549e8b92fdc2bfff135d8 : SUCCESS in 1h 22m 51s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3aad03f3e06b4d88a09ff21df7f59dfc : SUCCESS in 55m 41s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/da02f0c3a6914cf2a3267a5b460f6eb1 : SUCCESS in 1h 02m 23s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/221517e4c8714991b639cd1e53223dce : SUCCESS in 4m 44s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ca434fb9627249af8320c4c4515565ec : FAILURE in 1h 13m 09s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/1aff85892fe8453791aca14034c4b0c6 : FAILURE in 1h 02m 08s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b5bdd1e59a864c00a39e247e50563163 : SUCCESS in 4m 37s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/9db74ffdb98340808fe2a712448a6e6f : FAILURE in 1h 23m 55s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"346a9d8cbd83bc3f6af0178f77e134b7c08047fe","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-06-17 17:39:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4581430f4d94937e2dfa7a72f39488c8a3b22da9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-06-17 17:52:34.000000000","message":"Patch Set 2: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"29ba5157c1da70c7f0836dde14e989414aa6cf5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 20:17:55.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- grenade https://zuul.opendev.org/t/openstack/build/4b8406520cd24a4e9c9c871d00b71694 : SUCCESS in 56m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb2c8539926b4155bb3c666e0f254068 : SUCCESS in 1h 24m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71e5dfea45fe49569dfb6aea29c974f8 : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95de4abcbaee4ab38580ad7ca6a00b53 : SUCCESS in 17m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b1c8789dd484ee48e5e0145d800f6e9 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5201dfeba59447c94f88085514f0bb5 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da5c7988f95f469f9d5fd321b23b0072 : FAILURE in 6m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02560d3462374725b8737683b938ae1c : SUCCESS in 7m 15s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f99eca3b41c2457d877d30c4e22137b2 : SUCCESS in 1h 54m 44s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/1c8ef217b40d4281888b560889d8a823 : SUCCESS in 1h 21m 11s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7969d288a58f4d9790804239840f42a6 : SUCCESS in 1h 39m 23s (non-voting)\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/81042de7f8b34734a3a2195975d3f4a7 : SUCCESS in 1h 32m 27s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/5b17c91c927c459ca93d1055faa8a8fb : SUCCESS in 1h 21m 54s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/cf6cf74c475c490487c9703a054e78e8 : SUCCESS in 2h 06m 55s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/5e872fe9b0f34398a9c2ff9d7acf525f : SUCCESS in 2h 31m 08s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dd8c6140012f43869bf8518cf09add93 : SUCCESS in 1h 37m 52s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c3b536a1e1c24a6e839f3feafb89a20f : SUCCESS in 31m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/699b0ae7b03544f188e1353a812aee2a : SUCCESS in 1h 01m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/614be92d6d3140bc875a4e19eae3f337 : SUCCESS in 2h 31m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/be85f2848aeb48ec9399b6e6af9b38a9 : FAILURE in 1h 04m 47s (non-voting)\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/cfcaea4a40724ad8af1c6a276a33edd3 : SUCCESS in 1h 10m 10s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1752b15ff74f405aafe986f6429761da : SUCCESS in 1h 08m 32s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2600dddf59cc4fc6b70eaa0ce957bab8 : SUCCESS in 1h 07m 35s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d92ca104725749b3a4203f164c566a27 : SUCCESS in 4m 19s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/6854fef2732e4ac9904918120a8b3249 : FAILURE in 1h 24m 54s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/e90e0dfce86440bc8f219ceed6266bca : FAILURE in 1h 06m 21s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ba197fbe85914e20b6bd7725d3ecc641 : SUCCESS in 4m 34s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/668130318fa7455b8af404f3ffa2dda0 : FAILURE in 1h 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6daced9a493bad77cf96c99628d93ccab9bb5796","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-06-20 15:17:19.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ccc889f298629bc0c57fa1431b25d64095029ea1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-06-20 15:18:06.000000000","message":"Abandoned\n\nabandoning as it\u0027s not needed anymore","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"56a0c5c7c814449dee6ab365f8208dfdb4e75ea3","revisions":{"d5d26ec07957be8efdc7d9d6390a33c85ec9e944":{"kind":"REWORK","_number":1,"created":"2022-06-17 11:03:16.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/51/846251/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/51/846251/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/846251/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/846251/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/846251/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/51/846251/1"}}},"commit":{"parents":[{"commit":"00bdb22cc31db962986ca910a909c0a25b3f590f","subject":"compute: Move volume attached live migration tests to use SSHABLE","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/00bdb22cc31db962986ca910a909c0a25b3f590f"}]}],"author":{"name":"Balazs Gibizer","email":"gibizer@gmail.com","date":"2022-06-17 11:03:16.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"gibizer@gmail.com","date":"2022-06-17 11:03:16.000000000","tz":0},"subject":"Revert \"compute: Move volume attached live migration tests to use SSHABLE\"","message":"Revert \"compute: Move volume attached live migration tests to use SSHABLE\"\n\nThis reverts commit 00bdb22cc31db962986ca910a909c0a25b3f590f.\n\nReason for revert: this test now fails while trying to clean up the validation resource.\n\nChange-Id: I005755b0e527255362923a6b4bbbba2f7c1dfbd1\nCloses-Bug: #1979052\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d5d26ec07957be8efdc7d9d6390a33c85ec9e944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d5d26ec07957be8efdc7d9d6390a33c85ec9e944"}]},"branch":"refs/heads/master"},"56a0c5c7c814449dee6ab365f8208dfdb4e75ea3":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-06-17 14:43:06.000000000","uploader":{"_account_id":9708,"name":"Balazs Gibizer","display_name":"gibi","email":"gibizer@gmail.com","username":"gibi"},"ref":"refs/changes/51/846251/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/51/846251/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/846251/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/846251/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/51/846251/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/51/846251/2"}}},"commit":{"parents":[{"commit":"d720340c44e67856274235a47ebae38efd0b7625","subject":"Merge \"Make test_server_actions.resource_setup() wait for SSHABLE\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d720340c44e67856274235a47ebae38efd0b7625"}]}],"author":{"name":"Balazs Gibizer","email":"gibizer@gmail.com","date":"2022-06-17 11:03:16.000000000","tz":0},"committer":{"name":"Balazs Gibizer","email":"gibizer@gmail.com","date":"2022-06-17 14:43:06.000000000","tz":0},"subject":"Revert \"compute: Move volume attached live migration tests to use SSHABLE\"","message":"Revert \"compute: Move volume attached live migration tests to use SSHABLE\"\n\nThis reverts commit 00bdb22cc31db962986ca910a909c0a25b3f590f.\n\nReason for revert: this test now fails while trying to clean up the validation resource.\n\nChange-Id: I005755b0e527255362923a6b4bbbba2f7c1dfbd1\nCloses-Bug: #1979052\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/56a0c5c7c814449dee6ab365f8208dfdb4e75ea3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/56a0c5c7c814449dee6ab365f8208dfdb4e75ea3"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
