)]}'
{"id":"openstack%2Ftempest~856275","triplet_id":"openstack%2Ftempest~master~Ia96e2f3b52ad72d79a97ae7583feee08da3c0f1c","project":"openstack/tempest","branch":"master","attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2022-09-08 03:27:20.000000000","reason":"\u003cGERRIT_ACCOUNT_11075\u003e replied on the change","reason_account":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"}},"11075":{"account":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"last_update":"2022-09-19 07:58:08.000000000","reason":"Someone else replied on the change"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-09-07 17:19:54.000000000","reason":"\u003cGERRIT_ACCOUNT_11075\u003e replied on the change","reason_account":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"}},"10459":{"account":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"last_update":"2022-09-07 17:19:54.000000000","reason":"\u003cGERRIT_ACCOUNT_11075\u003e replied on the change","reason_account":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"}},"20813":{"account":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"last_update":"2022-09-07 14:24:24.000000000","reason":"\u003cGERRIT_ACCOUNT_11075\u003e replied on the change","reason_account":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"}},"9535":{"account":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"last_update":"2022-09-07 17:19:54.000000000","reason":"\u003cGERRIT_ACCOUNT_11075\u003e replied on the change","reason_account":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"}}},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-07-20 21:58:16.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"Ia96e2f3b52ad72d79a97ae7583feee08da3c0f1c","subject":"Add backend name for volume type","status":"NEW","created":"2022-09-07 12:27:03.000000000","updated":"2023-07-20 21:58:16.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":25,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"c5f8a7edf12263100c22b5886646789cbb5c30cd","_number":856275,"virtual_id_number":856275,"owner":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-08 06:08:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"disliked":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2022-10-07 16:00:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":-1,"date":"2023-07-20 21:58:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-09-19 07:58:08.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},{"_account_id":21187,"name":"Tzach Shefi","email":"tshefi@redhat.com","username":"tshefi"},{"_account_id":23003,"name":"Gal Amado","email":"gamado@redhat.com","username":"gamado"}],"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-07 12:34:07.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"CC"},{"updated":"2022-09-07 14:10:34.000000000","updated_by":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"reviewer":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"state":"CC"},{"updated":"2022-09-07 16:53:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-07 17:19:54.000000000","updated_by":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"reviewer":{"_account_id":21187,"name":"Tzach Shefi","email":"tshefi@redhat.com","username":"tshefi"},"state":"CC"},{"updated":"2022-09-07 17:19:54.000000000","updated_by":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"reviewer":{"_account_id":9535,"name":"Gorka Eguileor","email":"geguileo@redhat.com","username":"Gorka"},"state":"REVIEWER"},{"updated":"2022-09-07 17:19:54.000000000","updated_by":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-09-07 17:19:54.000000000","updated_by":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-09-07 22:46:54.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2022-09-07 22:47:23.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2022-09-12 12:09:51.000000000","updated_by":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"CC"},{"updated":"2022-09-12 12:10:06.000000000","updated_by":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"reviewer":{"_account_id":23003,"name":"Gal Amado","email":"gamado@redhat.com","username":"gamado"},"state":"CC"},{"updated":"2022-09-19 07:58:08.000000000","updated_by":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"reviewer":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"state":"REVIEWER"},{"updated":"2022-10-07 16:00:14.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"}],"messages":[{"id":"d3dc975f70846c0ab4ac7ec85d19140dd27cf718","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-09-07 12:27:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"919fd0270360717efa04d1e1c1a18dc2b21619b7","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-09-07 12:34:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bcaf8ee13f711c9a9d413e932e33d7355a35ace","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-09-07 12:40:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ffb6b0aa517e248f4163fdf94a31ca5b2600ae92","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-09-07 12:45:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f00014593f48cbd78a3bdbb20494edda74723897","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-09-07 13:42:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"12032168b2fcf0fd66c77ecf394e8d05a0954695","author":{"_account_id":20813,"name":"Sofia Enriquez","email":"lsofia.enriquez@gmail.com","username":"enriquetaso"},"date":"2022-09-07 14:10:34.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"74cb19b330d9f29eee95c105b587c643a61763b3","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-09-07 14:24:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84353997c45879fd68a1a6f9eff72fc44192ce7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-09-07 14:27:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"38be80fd9722e2bb209f5b054803c09eb816f73d","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-09-07 14:36:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"54c0ee72764bba4f0556d23a4a71bec189afa0c3","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-09-07 14:40:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"168a29e2d0f5fbd677bdac4d14a17339266462e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-09-07 14:49:38.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f18d7b1dc4afeeee80945968bcf13b32847927ff","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2022-09-07 15:02:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4ad88fc9134ad93c1eeefac912284dc7e0dd548c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 16:53:46.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dda76105cb2f4a109722f534ae90f94c\n\n- grenade https://zuul.opendev.org/t/openstack/build/6daef0ba9d474c1fbcd34c34a9e71328 : SUCCESS in 48m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b3f102f8b5974217a85edbabd5666137 : SUCCESS in 1h 37m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d4659ba350d460b875a0a777712b736 : SUCCESS in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eadebd2c20f6442e96c081c0db8957e8 : SUCCESS in 14m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79ee959ccec6457b80878a95d715b25f : FAILURE in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5be1e260ebd496f98fee1311641b9b3 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4bf59daf2c8441e9b0899a506d85cbe : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9879d14ee3be445b9ab9468ab358c5ee : SUCCESS in 6m 17s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/7f0ac8a54dae46959e1222cf0f408ca2 : SUCCESS in 1h 49m 38s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/0484636f4b504d6f99f36a5e8282755c : SUCCESS in 1h 26m 34s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d390c47552e7454b9fa6d8d17cf29df2 : SUCCESS in 1h 27m 30s (non-voting)\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/9eeb64c3c3944940bde326ce50619f5f : SUCCESS in 1h 32m 49s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/ec3c908d45f54211b1d3799d1829941c : SUCCESS in 1h 25m 57s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/6f14cdad5239463c82046d7ba5060eb5 : SUCCESS in 1h 27m 51s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/b27ab78d72324859b564a214e2a6b4ea : SUCCESS in 1h 56m 17s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/eadc3907138e4594993e0c5b6850f1d1 : SUCCESS in 1h 58m 00s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/54e3d7cd2606485e9e4bb1840fe2a908 : SUCCESS in 33m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6562ae4caeb4f2cbc16ae5eff22d48d : SUCCESS in 1h 04m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1f641da3c7b742a1b83665acdb96acfb : SUCCESS in 1h 59m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/40033ea645d64082a3477c11631efd97 : SUCCESS in 1h 14m 40s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/78ea9ca4a4cf41a3b254f1d0b89a7099 : SUCCESS in 1h 30m 53s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e31a1ad1d28f43edbc666fe9d193cc74 : SUCCESS in 1h 04m 12s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2696b3901ade4b28a04664b0ca4f38eb : SUCCESS in 1h 04m 27s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/34a10b6ee9e341aca12a58d4d78aa9f1 : SUCCESS in 4m 20s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/f08df1dc77794f80b1b6f6ef7704f359 : FAILURE in 1h 48m 02s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/39fef6802f7f4135941b05b92d2945da : FAILURE in 1h 13m 23s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e9567b0dc23f4ea98b8e08f174de4c9a : SUCCESS in 4m 26s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/b7792d61de4b46778396da16a7034999 : SUCCESS in 1h 56m 40s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"926583ede193769178edc772a5b03aa46d89e15f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-09-07 17:18:23.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e95cd50a7b0165bb357e1b957530f9940ea648a3","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-09-07 17:19:54.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ac9a09035a4c63ca07f9a54050ef9659c5d2f289","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 19:43:21.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/899d79aeac3d454498c39bd3d56af6db\n\n- grenade https://zuul.opendev.org/t/openstack/build/1a191e0f214a45a3b39bc563ae710cb2 : SUCCESS in 1h 02m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a315acedd01147ccb58f3303117cd21c : FAILURE in 1h 40m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28e6a6cf3ada41c184a104c5ab9419e4 : SUCCESS in 6m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5a671170fa944309568bd5a07d110c8 : SUCCESS in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d65efbc69b184b3d9cf7900440b0c28d : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e0880108c174ad8bc30635c655bdc5c : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7abae47bb50f4c9db23bc40d56018fe3 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86e8f3727f874088a0fb6778c4d86b04 : SUCCESS in 5m 13s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/837ae8965905416294169b80b3887555 : FAILURE in 2h 23m 38s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3038d067bd524317bb3d8c5523ab4111 : FAILURE in 1h 35m 58s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4940569dde594c0281877790ea4e7416 : FAILURE in 2h 09m 37s (non-voting)\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/77d52a9e514748c6a361fda7ead120ad : FAILURE in 1h 53m 17s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/0f46ddae901541119ff782389706ee0a : FAILURE in 1h 33m 34s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/a71e7d78c7e64a2b901aa568a3e3513f : FAILURE in 1h 37m 33s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/211373e3ffd74794be79397dc7fd95bf : FAILURE in 2h 09m 12s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d081a01eb24d48268c2f7a88598883f7 : FAILURE in 1h 46m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a2d76ecd561a4bc19950af6d78e8bf63 : SUCCESS in 25m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e498a0751e04655b0e8830927667585 : SUCCESS in 56m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9780dbddfc7943039489c1038fa418aa : FAILURE in 2h 13m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/67f9d2ab608c4c17826cfe71e66b693c : SUCCESS in 1h 14m 43s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/fd3b58820dd14535b63bd494ae3001fb : FAILURE in 1h 28m 45s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/315cd6dfd3044e6bb7a76a75ab7f2149 : SUCCESS in 1h 10m 42s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/135e3e5ded674d84b3afd433bc4a6354 : SUCCESS in 1h 11m 06s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e0ca91bd2d384e9b98c612b8fbb2787a : SUCCESS in 4m 51s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/4928ec46d2594a40a001bd38f8191ae3 : FAILURE in 1h 38m 13s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/4638ed20bede47ca893f5ce77958c9f6 : FAILURE in 1h 14m 10s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5f29ad36f0f542a7b8a660b2a8ec6724 : SUCCESS in 3m 45s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/057d334c96ed4de6b856f8f4b7718d81 : TIMED_OUT in 2h 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"454f083bb78ec2c2483fd3185479ac9db9d02d2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-09-07 20:07:09.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b43d39c98a8bae54d7e2807f343487e9c12f9614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-07 22:39:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75b37f55aaa945769ba50110d77c5202\n\n- grenade https://zuul.opendev.org/t/openstack/build/e1ac4754d6a843f19bd3415586bfe614 : SUCCESS in 1h 00m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6460f3dea67f4cda903ab2f417107e1d : SUCCESS in 1h 21m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0600852257041ae987a64607d624387 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92105447d1140bc99e6436b95351149 : SUCCESS in 13m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28c569bbf4a54e78b38e4a342f529122 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d07a0114434c4e3f8da79d29b502831d : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8fc41d7642c545ffbec0cb9dd1ca51dc : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75b2b5041ae64379bc751b18c69dd77d : SUCCESS in 6m 03s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/3f03dac8477a4a73a840dd5713225b16 : SUCCESS in 1h 44m 43s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/04534196252d4064af11ad2d7d5336bb : SUCCESS in 1h 32m 44s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0bff2a27c8744178ba27056b3a8171f5 : SUCCESS in 1h 35m 44s (non-voting)\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/c88cb670388747f5aac5e1e7811aa06d : SUCCESS in 1h 41m 01s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/081d7478adcd47fd967c3874a416a793 : SUCCESS in 1h 30m 12s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/1ee9882a8fcc4ba2b3896b96e1ce0cdc : SUCCESS in 1h 45m 34s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/b48b16143b1549fa8fbb3aee71964580 : SUCCESS in 2h 26m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f47899fde87248ceb2c23ea5fbeb64c2 : SUCCESS in 1h 48m 10s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/8d29549bb1c5465c89e6baf1b0cea318 : SUCCESS in 29m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4691cc1774e4080bf02b06e52043bd7 : SUCCESS in 59m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/72fd6d0af60d42b186301278db160eac : SUCCESS in 1h 58m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c6063ea904a24f14be4e492681942257 : SUCCESS in 1h 02m 05s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c95b457102c4454c8f0533332af0220f : FAILURE in 1h 27m 50s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/294cf9f384544f07aec64486b46006eb : SUCCESS in 1h 08m 58s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/63c95f6008f34b6fa934b618a5001ccb : SUCCESS in 1h 02m 45s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/be98a340625d45fd82f9182cfff65d55 : SUCCESS in 4m 55s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/e831f645c78f48ab8ac7902af5a1d94c : FAILURE in 1h 17m 57s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/54fdd0ed079d4a79807f4e876a7736eb : FAILURE in 1h 12m 39s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9c123af0b89143809bfc97a57ca695d3 : SUCCESS in 4m 05s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c82c0aaeb654cae91335e8bf9c7fe47 : SUCCESS in 1h 44m 35s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"15783c2046c26e61cd3fe46738ab8bb8bf80ad3a","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-09-07 22:46:54.000000000","message":"Patch Set 6:\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4a051cd8fc1fa6f83bb8da14580fb41ebc072825","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2022-09-07 22:47:23.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2c85766f3963e7c68ebb09367ce120a18f071f5d","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-09-08 03:27:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"812a3ef54229b4f1160e8d399643cb6a89b2094a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-09-08 03:34:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"231aa8b90463637335466629967bfce0804d4b6e","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-09-08 04:07:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0c0c9f6afaa4a55c96470be3cd24c07ef91022bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-08 06:08:34.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35f5cdfc771843f2899555614f077171\n\n- grenade https://zuul.opendev.org/t/openstack/build/7d996354224d48dd909bf2327cbcf461 : SUCCESS in 1h 28m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cb73c66a20c342c9a13340fcb2fc7540 : SUCCESS in 1h 42m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f74b5ed3b30e48e4a3c25993754726d4 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1d0ca8a3e5b4e1ea9544e804ec371ae : SUCCESS in 12m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b458bf0cdd1e4a6291ed779acc613b1b : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/53d3b1674719497ab2b19fa83f63e129 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad489830d818412eb99d265e1ce3f569 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8fb79f85d1f74272999ccb2527d53422 : SUCCESS in 5m 29s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/5d3cb0d7fe834fe295a2ec3be78cbab9 : SUCCESS in 1h 37m 07s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/13bdabb5ba5d455d94499bc32b14d146 : SUCCESS in 1h 45m 50s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/72c9d7de690c411fb6b44ac76208ffc2 : SUCCESS in 1h 31m 12s (non-voting)\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/1c489ab7f44149a3a7e0f7498aed63d5 : SUCCESS in 1h 33m 49s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/ce7480acfac64a12a7d65dc04ae201b3 : SUCCESS in 1h 44m 51s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/e07e23552de145c58588026c00b77b39 : SUCCESS in 1h 28m 57s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/b811843ca72c4b7eabcdae908f86195c : SUCCESS in 2h 28m 18s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9b15dc6db99045dcaee848259c34936e : SUCCESS in 1h 27m 36s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/14cd98a97b23429d8bbaab96f2972530 : SUCCESS in 29m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/64cab7b6c4904024abc455a194bb67c7 : SUCCESS in 46m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/75decfd70a8d40d6bca34367a1c1303f : SUCCESS in 1h 58m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/762b2f2aa0d04627abb31176cda5ff79 : SUCCESS in 48m 48s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/7f18d62657f54b2dba5c57e6d753f047 : SUCCESS in 1h 06m 17s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9eb2975aedd14cb29cd932800cb4bdc3 : SUCCESS in 51m 30s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/8ae71a7cccbf44ba990827e11411ec62 : SUCCESS in 1h 08m 45s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/33296d5b35be46a4831b0c8f8f70887c : SUCCESS in 5m 30s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/63ff76e362c948f68641f54d6707a86e : FAILURE in 1h 26m 30s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/c284bc6a08644739ae82c55cb4bed274 : FAILURE in 1h 16m 49s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ed61d89874a243e7b71744edc5a76c65 : SUCCESS in 4m 00s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/ad919e05df8f426a9e01e6a47f76e4cd : SUCCESS in 1h 47m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6df4df9bf3d1cdcdc8cec8edaf1eaea80fdf7bb6","author":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"date":"2022-09-12 12:09:51.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a5c22f866d115d2d7581af9af7ed4815473c85b6","author":{"_account_id":8367,"name":"Arx Cruz","email":"arxcruz@redhat.com","username":"arxcruz"},"date":"2022-09-19 07:58:08.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"e40d2998a792eae3f7a17795354de65d58237e66","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-10-07 16:00:14.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c5f8a7edf12263100c22b5886646789cbb5c30cd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-20 21:58:16.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"9154265c1f5d2533aa719306b0f7130e56f8e75c","revisions":{"7c8d4d2bd60502d2f58e3316fa2933b8167f72c8":{"kind":"REWORK","_number":1,"created":"2022-09-07 12:27:03.000000000","uploader":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"ref":"refs/changes/75/856275/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/75/856275/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/75/856275/1"}}},"commit":{"parents":[{"commit":"6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345","subject":"Merge \"Allow multiple instances on create_test_server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345"}]}],"author":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-07 12:19:00.000000000","tz":180},"committer":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-07 12:27:01.000000000","tz":180},"subject":"Add backend name for volume type","message":"Add backend name for volume type\n\nAllow to volume type creation use a default backend name\nCurrent code does not set volume backend name for volume-types.\nIn case we set a default volume backend, volume type will not\nuse it.\n\nIn cinder multi-backend setup we will be able to pick a backend\nname to create a type and volumes.\n\nChange-Id: Ia96e2f3b52ad72d79a97ae7583feee08da3c0f1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7c8d4d2bd60502d2f58e3316fa2933b8167f72c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/7c8d4d2bd60502d2f58e3316fa2933b8167f72c8"}]},"branch":"refs/heads/master"},"a8566aa2595cb117b4c2247a89a57316543d3607":{"kind":"REWORK","_number":2,"created":"2022-09-07 13:42:58.000000000","uploader":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"ref":"refs/changes/75/856275/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/75/856275/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/75/856275/2"}}},"commit":{"parents":[{"commit":"6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345","subject":"Merge \"Allow multiple instances on create_test_server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345"}]}],"author":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-07 12:19:00.000000000","tz":180},"committer":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-07 13:41:49.000000000","tz":180},"subject":"Add backend name for volume type","message":"Add backend name for volume type\n\nAllow to volume type creation use a default backend name\nCurrent code does not set volume backend name for volume-types.\nIn case we set a default volume backend, volume type will not\nuse it.\n\nIn cinder multi-backend setup we will be able to pick a backend\nname to create a type and volumes.\n\nChange-Id: Ia96e2f3b52ad72d79a97ae7583feee08da3c0f1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a8566aa2595cb117b4c2247a89a57316543d3607"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a8566aa2595cb117b4c2247a89a57316543d3607"}]},"branch":"refs/heads/master"},"122cd84fd954bc5f1deb0b9e0c7883fbe35c7cda":{"kind":"REWORK","_number":3,"created":"2022-09-07 14:27:11.000000000","uploader":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"ref":"refs/changes/75/856275/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/75/856275/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/75/856275/3"}}},"commit":{"parents":[{"commit":"6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345","subject":"Merge \"Allow multiple instances on create_test_server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345"}]}],"author":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-07 12:19:00.000000000","tz":180},"committer":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-07 14:26:53.000000000","tz":180},"subject":"Add backend name for volume type","message":"Add backend name for volume type\n\nAllow to volume type creation use a default backend name\nCurrent code does not set volume backend name for volume-types.\nIn case we set a default volume backend, volume type will not\nuse it.\n\nIn cinder multi-backend setup we will be able to pick a backend\nname to create a type and volumes.\n\nChange-Id: Ia96e2f3b52ad72d79a97ae7583feee08da3c0f1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/122cd84fd954bc5f1deb0b9e0c7883fbe35c7cda"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/122cd84fd954bc5f1deb0b9e0c7883fbe35c7cda"}]},"branch":"refs/heads/master"},"33584806b8b0074f82b6f86c09c17640abc06b2a":{"kind":"REWORK","_number":4,"created":"2022-09-07 14:49:38.000000000","uploader":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"ref":"refs/changes/75/856275/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/75/856275/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/75/856275/4"}}},"commit":{"parents":[{"commit":"6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345","subject":"Merge \"Allow multiple instances on create_test_server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345"}]}],"author":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-07 12:19:00.000000000","tz":180},"committer":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-07 14:49:22.000000000","tz":180},"subject":"Add backend name for volume type","message":"Add backend name for volume type\n\nAllow to volume type creation use a default backend name\nCurrent code does not set volume backend name for volume-types.\nIn case we set a default volume backend, volume type will not\nuse it.\n\nIn cinder multi-backend setup we will be able to pick a backend\nname to create a type and volumes.\n\nChange-Id: Ia96e2f3b52ad72d79a97ae7583feee08da3c0f1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/33584806b8b0074f82b6f86c09c17640abc06b2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/33584806b8b0074f82b6f86c09c17640abc06b2a"}]},"branch":"refs/heads/master"},"ba8c734e089e867c8a627617fbad4b49172e0394":{"kind":"REWORK","_number":5,"created":"2022-09-07 17:18:23.000000000","uploader":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"ref":"refs/changes/75/856275/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/75/856275/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/75/856275/5"}}},"commit":{"parents":[{"commit":"6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345","subject":"Merge \"Allow multiple instances on create_test_server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345"}]}],"author":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-07 12:19:00.000000000","tz":180},"committer":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-07 17:18:11.000000000","tz":180},"subject":"Add backend name for volume type","message":"Add backend name for volume type\n\nAllow to volume type creation use a default backend name\nCurrent code does not set volume backend name for volume-types.\nIn case we set a default volume backend, volume type will not\nuse it.\n\nIn cinder multi-backend setup we will be able to pick a backend\nname to create a type and volumes.\n\nChange-Id: Ia96e2f3b52ad72d79a97ae7583feee08da3c0f1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ba8c734e089e867c8a627617fbad4b49172e0394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/ba8c734e089e867c8a627617fbad4b49172e0394"}]},"branch":"refs/heads/master"},"bec3e8d550bbfd61401bd3541abb093eddfd4188":{"kind":"REWORK","_number":6,"created":"2022-09-07 20:07:09.000000000","uploader":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"ref":"refs/changes/75/856275/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/75/856275/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/75/856275/6"}}},"commit":{"parents":[{"commit":"6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345","subject":"Merge \"Allow multiple instances on create_test_server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345"}]}],"author":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-07 12:19:00.000000000","tz":180},"committer":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-07 20:06:57.000000000","tz":180},"subject":"Add backend name for volume type","message":"Add backend name for volume type\n\nAllow to volume type creation use a default backend name\nCurrent code does not set volume backend name for volume-types.\nIn case we set a default volume backend, volume type will not\nuse it.\n\nIn cinder multi-backend setup we will be able to pick a backend\nname to create a type and volumes.\n\nChange-Id: Ia96e2f3b52ad72d79a97ae7583feee08da3c0f1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bec3e8d550bbfd61401bd3541abb093eddfd4188"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bec3e8d550bbfd61401bd3541abb093eddfd4188"}]},"branch":"refs/heads/master"},"9154265c1f5d2533aa719306b0f7130e56f8e75c":{"kind":"REWORK","_number":7,"created":"2022-09-08 03:34:36.000000000","uploader":{"_account_id":11075,"name":"Benny Kopilov","email":"bkopilov@redhat.com","username":"bkopilov"},"ref":"refs/changes/75/856275/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/75/856275/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/75/856275/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/75/856275/7"}}},"commit":{"parents":[{"commit":"6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345","subject":"Merge \"Allow multiple instances on create_test_server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345"}]}],"author":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-07 12:19:00.000000000","tz":180},"committer":{"name":"Benny Kopilov","email":"bkopilov@redhat.com","date":"2022-09-08 03:33:54.000000000","tz":180},"subject":"Add backend name for volume type","message":"Add backend name for volume type\n\nAllow to volume type creation use a default backend name\nCurrent code does not set volume backend name for volume-types.\nIn case we set a default volume backend, volume type will not\nuse it.\n\nIn cinder multi-backend setup we will be able to pick a backend\nname to create a type and volumes.\n\nChange-Id: Ia96e2f3b52ad72d79a97ae7583feee08da3c0f1c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9154265c1f5d2533aa719306b0f7130e56f8e75c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9154265c1f5d2533aa719306b0f7130e56f8e75c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
