)]}'
{"id":"openstack%2Ftempest~857602","triplet_id":"openstack%2Ftempest~master~I188a23bf77855358535986256b17fcc4cc6405f9","project":"openstack/tempest","branch":"master","topic":"qa-zed-release","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-09-14 15:45:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I188a23bf77855358535986256b17fcc4cc6405f9","subject":"Use stable constraint in tox to release new tag for Zed","status":"MERGED","created":"2022-09-14 09:41:19.000000000","updated":"2022-09-14 15:47:35.000000000","submitted":"2022-09-14 15:45:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"857602-qa-zed-release","meta_rev_id":"7098ed19cd8f7d8c02d55db3811189fe5bac819b","_number":857602,"virtual_id_number":857602,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-09-14 15:45:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2022-09-14 13:29:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2022-09-14 13:29:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-14 11:39:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-14 13:29:19.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"ffce7f7c91e40467a2883290b71576734990dd8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-09-14 09:41:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34bbd2190801a389171a37e2d6b8b799f57d68a9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-09-14 09:42:21.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"5cc82840e73b213ff549186e557aabdf23186576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 11:39:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1729bb6b1c3b416ab524e9c772b1ff23\n\n- grenade https://zuul.opendev.org/t/openstack/build/478b439f481743e590bf17e5ab2d1f04 : SUCCESS in 1h 05m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/41f206d7ae38450b949c81357581de35 : SUCCESS in 1h 39m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e2c71a0179b42cb8cf92103737b5132 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05161d7ea0084ca5a7f5ef3a36e692da : SUCCESS in 14m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/069ba60f4cbd4ef8980d121cf3c9533e : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/10e6a54f662449fbb75f191c3359c10c : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46082545b7f449bcaba6f218b663d4b8 : SUCCESS in 7m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4083307a4e874f978d1dde88348ed6b2 : SUCCESS in 5m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba14971666bd455b9e242309b5c206ce : SUCCESS in 4m 49s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/97d01a7c437d45ab8efc9066e161daba : SUCCESS in 1h 45m 24s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/10b4e70207054ca79b49c38fcbb74f76 : SUCCESS in 1h 27m 26s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0647a1dba2cf4babb23e5c137d40fcda : SUCCESS in 1h 17m 12s (non-voting)\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/3716b93051ee429da6bb07e73fae7e3b : SUCCESS in 1h 44m 27s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/fe2105038f6441fab19c1ba5c48c9cbc : SUCCESS in 1h 27m 24s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/57b377714e7544abbfe265869c04b71b : SUCCESS in 1h 38m 57s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/e0751fe6a4d3447b9b38776029012799 : SUCCESS in 1h 50m 51s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c1277835189946649c4b2e1af3c886fb : SUCCESS in 1h 47m 02s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/461ff45fba0e4cea9315d37055a4abb0 : SUCCESS in 24m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4b95c1f80c0e4bc38376a9489d76821b : SUCCESS in 48m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/bf46c074e3284d0d8946c740707d200f : SUCCESS in 1h 49m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ee5ce0bc0aab46aca9ff3e765d79fa7b : SUCCESS in 55m 55s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6eb22e1709544b5d8e6367f60f5442c4 : SUCCESS in 1h 08m 40s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/92098afe883d41b58f9ceaf408a37248 : SUCCESS in 58m 09s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/5ec6f10f459c4c1eb0b098681b82fdef : SUCCESS in 55m 57s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/096f91b625fb4e9c9e76ea866a02f403 : SUCCESS in 6m 08s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1207171729a241459325557696f8f622 : FAILURE in 1h 28m 52s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/766cb779e2204cac95f81f7f8d29a10a : FAILURE in 1h 07m 27s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/378e6d5acfec4de7a0abba575f04389f : SUCCESS in 4m 36s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/604ed007eb3e4dac8f70b76bf5e1ecf5 : SUCCESS in 1h 54m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"cc354d39d8b7f47548231341eca0444ea2cfdd8d","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-09-14 13:29:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f446e8784818e641b881153eadbb4b63dd9d39b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 13:30:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"bd5843d2d4857d51fa6a8e76df362fc8d83b2bb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 15:45:55.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8088ec6f1f0844e58fcdd3b07eaae698\n\n- grenade https://zuul.opendev.org/t/openstack/build/33455b129a394ee48e6af813b4d98661 : SUCCESS in 53m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c7fcd58f6874427b8a6f1677033558c : SUCCESS in 1h 22m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b3df241c777f4a1cbc3870e19ad15d06 : SUCCESS in 13m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/270caab39d9340d2b8997d251bccaa89 : SUCCESS in 11m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93e1f27cdcd442ba85b44aa56b12aa53 : SUCCESS in 8m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a19af9b6eccc455f8214d4a68f039277 : SUCCESS in 9m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/052fb508a8b14c36bd8d3337b4dc8d1e : SUCCESS in 8m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cde9a690535c471ea3a3122e9927c857 : SUCCESS in 5m 04s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/13b6cdd0d0a742a1b1748eff3bf7cacd : SUCCESS in 2h 11m 37s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/93f6349af72f47e39ac7f3ebe0b4faf0 : SUCCESS in 1h 07m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ce17bdcf0387485c97c398485fcf2941 : SUCCESS in 58m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7c82a4ddb5854fb0909f5f4043c5d30a : SUCCESS in 58m 21s","accounts_in_message":[],"_revision_number":2},{"id":"06145dd828ba5613f83becfa15922293da0c87b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 15:45:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7098ed19cd8f7d8c02d55db3811189fe5bac819b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 15:47:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/655f1fb0d5884d4ca68997748e68a40f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95509214979b44748dcdc86c8d28df39 : SUCCESS in 1m 07s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/00fae86964f84976a220d7ba004e219a : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"979f05a03b3f601b319fada515a523f0fab8c1a0","revisions":{"84783d155543d3f4d1ab8a8766194d8bb2cb4677":{"kind":"REWORK","_number":1,"created":"2022-09-14 09:41:19.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/857602/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/02/857602/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/857602/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/857602/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/857602/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/02/857602/1"}}},"commit":{"parents":[{"commit":"6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345","subject":"Merge \"Allow multiple instances on create_test_server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345"}]}],"author":{"name":"ghanshyam mann","email":"gmann@ghanshyammann.com","date":"2022-09-14 09:39:58.000000000","tz":330},"committer":{"name":"ghanshyam mann","email":"gmann@ghanshyammann.com","date":"2022-09-14 09:39:58.000000000","tz":330},"subject":"Use stable constraint in tox to release new tag for Yoga","message":"Use stable constraint in tox to release new tag for Yoga\n\nWe are going to release Tempest new tag to declare the start of\nsupport for stable/zed. So that new tag use stable/zed\nconstraint in the tox env.\n\nFor example: if anyone use Tempest new tag (32.0.0) in future say\n1 year later then tox env also should use the zed contstraint for\ncompatibility instead master.\n\nThese need to be move back to master constraint once release is done.\n\nChange-Id: I188a23bf77855358535986256b17fcc4cc6405f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/84783d155543d3f4d1ab8a8766194d8bb2cb4677"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/84783d155543d3f4d1ab8a8766194d8bb2cb4677"}]},"branch":"refs/heads/master"},"979f05a03b3f601b319fada515a523f0fab8c1a0":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-09-14 09:42:21.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/02/857602/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/02/857602/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/857602/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/857602/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/857602/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/02/857602/2"}}},"commit":{"parents":[{"commit":"6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345","subject":"Merge \"Allow multiple instances on create_test_server\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6ec23b64dc844e0f7c2f01dc7e1f6d87fdb50345"}]}],"author":{"name":"ghanshyam mann","email":"gmann@ghanshyammann.com","date":"2022-09-14 09:39:58.000000000","tz":330},"committer":{"name":"ghanshyam mann","email":"gmann@ghanshyammann.com","date":"2022-09-14 09:42:03.000000000","tz":330},"subject":"Use stable constraint in tox to release new tag for Zed","message":"Use stable constraint in tox to release new tag for Zed\n\nWe are going to release Tempest new tag to declare the start of\nsupport for stable/zed. So that new tag use stable/zed\nconstraint in the tox env.\n\nFor example: if anyone use Tempest new tag (32.0.0) in future say\n1 year later then tox env also should use the zed contstraint for\ncompatibility instead master.\n\nThese need to be move back to master constraint once release is done.\n\nChange-Id: I188a23bf77855358535986256b17fcc4cc6405f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/979f05a03b3f601b319fada515a523f0fab8c1a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/979f05a03b3f601b319fada515a523f0fab8c1a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
