)]}'
{"id":"openstack%2Ftempest~860502","triplet_id":"openstack%2Ftempest~master~I613357e8c9cc6d382caaf7291e150448148f32dd","project":"openstack/tempest","branch":"master","topic":"image-behavior","attention_set":{},"removed_from_attention_set":{"30742":{"account":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"last_update":"2022-10-12 17:51:30.000000000","reason":"\u003cGERRIT_ACCOUNT_30742\u003e replied on the change","reason_account":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-10-12 19:44:54.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-10-12 17:53:05.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I613357e8c9cc6d382caaf7291e150448148f32dd","subject":"Add a boot-snapshot-boot test","status":"MERGED","created":"2022-10-05 22:06:29.000000000","updated":"2022-10-12 19:45:59.000000000","submitted":"2022-10-12 19:44:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860502-image-behavior","meta_rev_id":"f7fb770318beffd3548785d46d4fc3650ccf07de","_number":860502,"virtual_id_number":860502,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-12 19:44:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"recommended":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"all":[{"value":1,"date":"2022-10-12 17:51:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":2,"date":"2022-10-12 17:41:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-12 17:53:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-10-12 17:53:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-06 01:27:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-07 16:36:52.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"},{"updated":"2022-10-07 20:32:20.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2022-10-12 17:41:16.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-10-12 17:53:05.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"d9fbea508f0a5a9af892ac54ef64dfd4d5fb7390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-05 22:06:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"420d258004a3ff8424c99d05438b059bcee5c71f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-05 23:12:40.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"506628621110eda168b84f4a4bd05dd5d7540c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 01:27:23.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1db5a2ec59db4ce7a005b3053c63c386\n\n- grenade https://zuul.opendev.org/t/openstack/build/529ee478f27f45369e0abce94a096c7b : SUCCESS in 51m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e65b6afb3f9c4fd2a35f0f377683e517 : SUCCESS in 1h 26m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a326f39c6744d1cb827a9013c44de1d : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03c8bb18e4b2484b9a08428d6c579740 : SUCCESS in 14m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9781ba0d12424fbbafc034b0f1085292 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6314bdaf4a584c05bb1674f87539969e : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2a0bd7a62aa4602a34fb62a63d3f63e : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/636b3c86be9442d1b90b757201ef6253 : SUCCESS in 6m 19s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/2ccb837c58f8441c9f94814e78b51257 : FAILURE in 1h 32m 28s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/01f62e27e817404098bb5f5bf9375182 : SUCCESS in 1h 13m 13s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e8b77be3e4f347bd93b39c587bf087d9 : SUCCESS in 1h 30m 40s (non-voting)\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/6b23273bd571496288e4937847393c45 : TIMED_OUT in 2h 06m 06s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/9c52b553bf13462396e6737a803f00a5 : SUCCESS in 1h 42m 55s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/2c74c303081a48ccbf3f215be72b2986 : SUCCESS in 1h 13m 30s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/0b7bb5291388419b8e859a6cd09dc080 : SUCCESS in 1h 38m 32s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/52d2ed8ca71141f4a001af58e0f25033 : SUCCESS in 2h 13m 13s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8fe5b991d1c24b02b57c2be28ef9526c : SUCCESS in 1h 45m 25s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/dbdc1c31926941cca9595350998b8387 : SUCCESS in 22m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7938f3442e434d90a96a3fe45a60266c : SUCCESS in 52m 59s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c53ca1c9ad924fb090698a98279e18d3 : SUCCESS in 1h 51m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/6836165d2d09449e8602ea2c881c7e77 : SUCCESS in 52m 55s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/fe22e94a40504fe3851e0fa0af056d96 : FAILURE in 1h 21m 15s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/aeda0d1f5f6b40f697aa65d4d2cb3598 : SUCCESS in 55m 27s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/290d4392026b4a7d8bebc9b45a292f0e : SUCCESS in 51m 47s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/2e6d5aebe2274097b500cb0c52aab962 : SUCCESS in 5m 06s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/949c437a0f4b4c5b92cc17a5bb7dedb3 : FAILURE in 1h 34m 27s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/aa6d57be47cf4471a02bbb77ca0f8350 : FAILURE in 57m 33s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ad7c8956b3fc46c28382161ee4d22141 : SUCCESS in 4m 24s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/e43c291758d94b57b059b114cc2c49c6 : SUCCESS in 1h 38m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"99620bb16ac3877fcbb1d6929075bb287d9afe15","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-06 13:21:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b1d2da3d3472d5039945563696736b0e54cc5170","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-06 13:33:11.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e3d7f73692b9b486dbc3195d13b668025c5d37f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-06 16:06:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cf98e9519054c68b76ec09accba59f0\n\n- grenade https://zuul.opendev.org/t/openstack/build/f66a352ec75c49db9f680b20bb533ca7 : SUCCESS in 41m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6a7714821126469abb03750db99465b1 : FAILURE in 31m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0d9ed0c14e5142988512e44ad9bae528 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a488559f40240ce9637c9855fd6a156 : SUCCESS in 13m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ace72d3369ff47618850c40bcadb9f99 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/64023c8cac3a4a0db306073c6d70de25 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9cc95bbe60a478f80846b481253d88f : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0ca1701849043788a649ed04d6b9943 : SUCCESS in 4m 43s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/a39343643ff6428e8a082a6ad930a559 : FAILURE in 20m 43s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/e540466f34bb46ea9cfd7572628f0cf2 : FAILURE in 20m 33s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6d4c0999fde746c487ffd34085ce94c9 : POST_FAILURE in 19m 49s (non-voting)\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/5c9dc126353c49d68b2f00d3a947688d : SUCCESS in 1h 30m 28s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/5b5df978ffe343eba243b95457cac463 : SUCCESS in 1h 28m 51s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/e669cbab23d74b1e8d01ecc81c269ba6 : SUCCESS in 1h 38m 54s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/f658b60f978a4fa09938b92cbd3191da : SUCCESS in 1h 18m 10s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/9d944740738a4d469cd2abe8da71e345 : SUCCESS in 2h 27m 23s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/bbc3c8644e464000b52e9c335c85f74e : FAILURE in 29m 37s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e5cda6a78edb49ad8cd0f804d5fe9dc6 : SUCCESS in 34m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/656561bb4bdc457298bea28f429d81ae : FAILURE in 34m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5cc38258693b41dbb6e8c8a63355342b : FAILURE in 38m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/466802e640d540f986655633f8aadcad : POST_FAILURE in 26m 47s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6c00f999cece45289b5b006cc4e0e059 : FAILURE in 22m 08s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a51af14f50f445f58895c49cb08d0733 : SUCCESS in 1h 02m 24s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/7159751f35d44d439f8d8ff735bfcc37 : SUCCESS in 48m 42s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/cdb208765f1e4a82a5684a78d1eac06e : SUCCESS in 4m 27s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/03704cff1db342fb87a93f09765a7ac9 : FAILURE in 21m 58s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/d3f61ac7ad7a44d89ffd987a474a96a0 : FAILURE in 30m 14s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/418b6d481fc846348cc58bb76d65f1aa : SUCCESS in 4m 05s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c42e9f16dbd4ffc86851e47a847399c : SUCCESS in 1h 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2666a72f10cee90563de628c4ced5ce60733ec30","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-07 13:34:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2b2edfe0030580191860ce5c61c871a6d433f21f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 15:29:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff49df9a59f143d489c2d7b41c5e9058\n\n- grenade https://zuul.opendev.org/t/openstack/build/2fb3e246fabd4766b74def7f2c2a0ab5 : SUCCESS in 46m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9da4f5364f494d3188b16b5a476571cf : SUCCESS in 1h 27m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28b1d8771793495da930c5b1dab45717 : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ebf6296cf124c8384c7c48d341cf0d6 : SUCCESS in 12m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/373af83f1ded417bbd1775c16a0bbdba : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb62ac97a1db482f9d15b12290b3dabf : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6139a88b879949b8a4417f28102eb0a5 : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94cc7d90a98d4edd8b9381e770ae6336 : SUCCESS in 4m 32s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/20a0f43ebc4c4473a2e173db438522be : FAILURE in 1h 26m 09s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/06adf130536d4534b9088b05c9f0e6c8 : SUCCESS in 1h 09m 55s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/802b2437e9cd4d249b549d79f1326ce1 : SUCCESS in 1h 34m 47s (non-voting)\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/b346dafce3e443acb7dae33b1153c987 : SUCCESS in 1h 30m 11s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/5befd7402ad54620908ec6037d3699df : SUCCESS in 1h 25m 35s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/41fbfcef10ec4c0f9a38eb28e922cb4e : SUCCESS in 1h 20m 48s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/631e3c44b68941afb4a0c1399f26461a : SUCCESS in 1h 22m 31s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/0830445773064188bb80ef4cdfab94c3 : SUCCESS in 1h 51m 59s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/679472e06166416482f76208c100b48a : SUCCESS in 1h 36m 24s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/d330eb96f7084a48b596c140c3535efc : SUCCESS in 28m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6cb9ce7fcfe841268da971909567fa40 : SUCCESS in 45m 18s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8220af80c44d489481a496b5e8a67093 : SUCCESS in 1h 45m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ad91b33e9ed448228532bf8a570077c3 : SUCCESS in 44m 23s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/64b3f6ee91ea4834a8f428b4741a48f9 : FAILURE in 1h 15m 38s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dd5fd15397594306a40c297bf7423690 : SUCCESS in 41m 15s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/ea87b2d161d2471091b77c9f6088ceea : SUCCESS in 56m 53s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/a78e1598e1d4421ba72f2791d8475aec : SUCCESS in 4m 00s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ad2ad8315de54eb4bd92065ba143050e : FAILURE in 1h 20m 24s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/eb90271447ce427d85116a91cd88e7d4 : FAILURE in 58m 17s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f70aeea15029416fbd23954ec06b7ed0 : SUCCESS in 3m 23s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/875b8d4429fc4fcb9ac70e6e4ba18087 : SUCCESS in 1h 35m 03s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a631041d30783536367c378537ed7630faf706a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-07 16:29:28.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"8365354fd80f8bd94c89b7fcb1ee9a6d03adb44b","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-10-07 16:36:52.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7aa831025063fda9d36ae01ca2ee4d7faec02096","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-07 18:08:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"3b27d4951787e16d7137ea39bee16ab35b5ceb54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 20:11:27.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc7801b20da3416689e3f0616955deaa\n\n- grenade https://zuul.opendev.org/t/openstack/build/02579f366a874ae0a12ac16d9e8e3cc7 : SUCCESS in 50m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86c31c16c362492b82eceab5f629678a : SUCCESS in 1h 34m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ca3e0972ca94bc38ada7a9a961b0260 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/367bf9f7b7ab44df883eb2032fd01b24 : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59d8dec80ba0477a891b004b0c71fd07 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/024eec41461d41caac92b51655e886f8 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74fc613d91e044e3be292fe67baf86b9 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bbe9ef102a844d67b3dab0b95c259e76 : SUCCESS in 4m 01s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/f509faf62f1b4cfba99a2b48e7546c06 : SUCCESS in 1h 30m 16s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/4c6f0fdc0c7d43ad84c993b7d326bb3c : SUCCESS in 1h 18m 16s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b03977a4fc504751ab6e3bc69d792b38 : SUCCESS in 1h 26m 59s (non-voting)\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/822d1da39514430ca9c2afdf722b10c1 : SUCCESS in 1h 14m 55s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/9d5e0c9026eb4fa0b7e7f01e4006e12d : SUCCESS in 1h 28m 14s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/40238abc6ccf43f882690ec970650c93 : SUCCESS in 1h 16m 02s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/86a72a5a1eb649dba33c74068933034c : SUCCESS in 1h 15m 47s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/dcdab36cabce47edb3ebd90fcec7c7a1 : SUCCESS in 2h 01m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/691593e17f154facbfa31afb7edf39f6 : SUCCESS in 1h 48m 12s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3c2e40414d424ce79627ab5be5eed92a : SUCCESS in 27m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f11cbf198e804d2ca408224dc8fc9689 : SUCCESS in 55m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8d2c56e9ee4242ed9870cfa8c838e3bf : SUCCESS in 1h 55m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b3fcf5fceed04bdab4d69845dceb0475 : SUCCESS in 53m 26s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/7b3a287aa5a44c439b926aaf1fb06448 : FAILURE in 1h 06m 54s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9222757074ae403d9d2dff41335dc757 : SUCCESS in 40m 28s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/6c9c774e34404f04bab88ccb6bfbccf9 : SUCCESS in 57m 20s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/baff8684c17e4b06b7d50c22ed174d30 : SUCCESS in 3m 20s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a67de5cff43f4de09e28b8833d14a9dd : FAILURE in 1h 40m 07s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/bc7d1f0ba2eb416a84cbe30a352166d2 : FAILURE in 1h 01m 03s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/89bd2bc56e2d4dc99bf5404f2a6ca6dd : SUCCESS in 2m 43s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/bb852f92abcc4de6aa6d7bdf6803aa72 : SUCCESS in 1h 38m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8fdeb37b72274bb833b81c07fdcd9c97c9966b5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-07 20:30:21.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2186772edfa034e4979bfa4ae0ed36a30d2fa787","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-07 20:32:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"323d5105a2ba386f110885f847a158cc141ae962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-07 20:33:29.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"0c5bc774a8552dc221bb2a511278163f965d3f93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-07 23:03:35.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a8a60e79d73474586a1566a4b640494\n\n- grenade https://zuul.opendev.org/t/openstack/build/1fbbd3f0366d4995aaa958bc8128875c : SUCCESS in 51m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a499d0a2ff1d4aa1b6522558bd48739d : SUCCESS in 1h 28m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/833323bbadfa412c91917a1fb5d8fcd9 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e1f3ea12d924f4ab8a6747c1fefda58 : SUCCESS in 17m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52b28614c19b4021830512ce9d201eaa : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4245ef1cd1ae4c3cb363d821f586f383 : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/869c354aa70548c884bda9321dce1949 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2b543e74dbbf495ebb8962ee80bb458b : SUCCESS in 6m 51s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/9d19b77390ef40ecb1c12e0427475d6b : SUCCESS in 1h 36m 18s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3782747c165c41f18496d0e7bb96dda3 : SUCCESS in 1h 20m 52s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/60b7c6c2305f442cba7f97d61bb35406 : SUCCESS in 1h 26m 38s (non-voting)\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/11a4191ac848478da1b4772543c3d11a : SUCCESS in 1h 17m 23s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/16d61bfae959412f805927c936cb9bc6 : SUCCESS in 1h 27m 36s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/6b366b35433844a39174b0949c99edc3 : SUCCESS in 1h 10m 15s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/ba142aadb46e468c9449049808409f51 : SUCCESS in 1h 29m 54s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/d347995685ac42918687e1f17176e217 : SUCCESS in 2h 24m 29s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/262fa5a00d9c4fb2a7545aad4810389b : SUCCESS in 1h 37m 59s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/7e32db1101ef4af5b45f7f6bba839e5b : SUCCESS in 22m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3219dd9a557c4703a51a7bdb2aa758de : SUCCESS in 45m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/148e90e6619f41c6985c2231fc0729bb : SUCCESS in 2h 17m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b620865350734fd1b75cf2fcc39e5ae4 : SUCCESS in 47m 15s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/74ce94ea48304f698448b1a75cf2b4a1 : SUCCESS in 1h 24m 15s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/63227c915b5340018c4ff3d58fa6574a : SUCCESS in 53m 48s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/46f32f39c8e3463c91c45e2c2e2ffa1c : SUCCESS in 59m 09s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/c270e74c86a94e328e6ab6d20db567ce : SUCCESS in 3m 26s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/65775fe6e8e940acae8ab94665277b35 : FAILURE in 1h 24m 56s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/82d2ad2113774c47b43a66890d8b43fb : FAILURE in 1h 12m 02s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7339f0101cf4424da61ce9b6481ffa05 : SUCCESS in 3m 15s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/17bc3b444c894ebc83e09afc9db91bdd : SUCCESS in 2h 00m 40s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"73fe523f208375667260acf97100c99f9243ea90","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-10 13:54:05.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"e2f846cd766f9514ff673310b981e00aab211018","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-10-11 08:39:41.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b51198c0d2ea9c4285c0822d15dcdeddbbf594b5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-11 14:44:28.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"5434f83b26357f087ace3135edc8a37a19d7e837","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-11 14:48:02.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"fc0636a8ce1cd5faeee120d0621f0ce8f00cba5b","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-10-12 10:34:30.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"759dad334203b294ae3aae8e7640d7d2b5876340","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-10-12 17:41:16.000000000","message":"Patch Set 7: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"2e9bc74aebfbfc99922af599212596226f1f663f","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-12 17:50:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"07f004a25741fe4a92a8c14dd71a278fe985c1c9","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-10-12 17:51:30.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"9afd46948fbab1f1250203c1ef53c64eca472205","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-12 17:53:05.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0e60d51da84f4a1d6a846df3ded6d3165a393506","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 17:53:35.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"dce2bae382e1e6dbc240fb89d321433a4ee6d35f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 19:44:54.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/589448fc527347ce8dc58c7e51b42ac4\n\n- grenade https://zuul.opendev.org/t/openstack/build/50e4c38e629c4d5ea197c8dda22ef047 : SUCCESS in 1h 02m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9fdb30acd0794ac495b63020f9431cf4 : SUCCESS in 1h 14m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21044eda036049d1a59c650b42aac7c1 : SUCCESS in 12m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94d869e717d8428f9b08d6f051e01e97 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d60e7adc97194826a59aa3f6b7143737 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fcef82bc0f3445fb9342a780b57a3d64 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/20eff42aa5244784aa5f3ae5c62a6fdc : SUCCESS in 4m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6fd408d6003d4fa8a7d6e6b92d4971ee : SUCCESS in 1h 48m 51s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ed7d30884184700847672fde12a07f8 : SUCCESS in 46m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ffdceacff0f4b06a00e7eadb0e69741 : SUCCESS in 37m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/425b7dd4a6144acbba5442709aea9b5a : SUCCESS in 1h 15m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/80926f17028542cc9702611e1cb86dbf : SUCCESS in 54m 57s","accounts_in_message":[],"_revision_number":7},{"id":"be903ff2876819827623c8f4098ae0fc759b1a92","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 19:44:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"f7fb770318beffd3548785d46d4fc3650ccf07de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-12 19:45:59.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd45bb5dfe2b471d89040bc8ffec7a9a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34ebc9ae081942d0aa365cf3957b2526 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"6714b65a210c5851591dfd5dfe9ad1cbf1fa49e9","revisions":{"fe5d150df6de982e2adc9746ded0e1719ef74e27":{"kind":"REWORK","_number":1,"created":"2022-10-05 22:06:29.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/860502/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/02/860502/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/02/860502/1"}}},"commit":{"parents":[{"commit":"f1d0e395e95d3502a9c3904ce56520d116e6bf48","subject":"Add Tempest gate job for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f1d0e395e95d3502a9c3904ce56520d116e6bf48"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-05 22:05:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-05 22:05:31.000000000","tz":-420},"subject":"Add a boot-snapshot-boot test","message":"Add a boot-snapshot-boot test\n\nIt occurred to me that we do not (AFAIK) have a test that shows that\nwe can boot servers from snapshots we create right after we create\nthem. This adds one.\n\nChange-Id: I613357e8c9cc6d382caaf7291e150448148f32dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fe5d150df6de982e2adc9746ded0e1719ef74e27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fe5d150df6de982e2adc9746ded0e1719ef74e27"}]},"branch":"refs/heads/master"},"6d12766dbc08b55861ab6a05c123e6969a53703b":{"kind":"REWORK","_number":2,"created":"2022-10-05 23:12:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/860502/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/02/860502/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/02/860502/2"}}},"commit":{"parents":[{"commit":"f1d0e395e95d3502a9c3904ce56520d116e6bf48","subject":"Add Tempest gate job for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f1d0e395e95d3502a9c3904ce56520d116e6bf48"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-05 22:05:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-05 23:12:34.000000000","tz":-420},"subject":"Add a boot-snapshot-boot test","message":"Add a boot-snapshot-boot test\n\nIt occurred to me that we do not (AFAIK) have a test that shows that\nwe can boot servers from snapshots we create right after we create\nthem. This adds one.\n\nChange-Id: I613357e8c9cc6d382caaf7291e150448148f32dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6d12766dbc08b55861ab6a05c123e6969a53703b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6d12766dbc08b55861ab6a05c123e6969a53703b"}]},"branch":"refs/heads/master"},"265b78b51b15eaf4cc76075613bed1330e4fa0ae":{"kind":"REWORK","_number":3,"created":"2022-10-06 13:33:11.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/860502/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/02/860502/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/02/860502/3"}}},"commit":{"parents":[{"commit":"f1d0e395e95d3502a9c3904ce56520d116e6bf48","subject":"Add Tempest gate job for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f1d0e395e95d3502a9c3904ce56520d116e6bf48"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-05 22:05:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-06 13:33:05.000000000","tz":-420},"subject":"Add a boot-snapshot-boot test","message":"Add a boot-snapshot-boot test\n\nIt occurred to me that we do not (AFAIK) have a test that shows that\nwe can boot servers from snapshots we create right after we create\nthem. This adds one.\n\nChange-Id: I613357e8c9cc6d382caaf7291e150448148f32dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/265b78b51b15eaf4cc76075613bed1330e4fa0ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/265b78b51b15eaf4cc76075613bed1330e4fa0ae"}]},"branch":"refs/heads/master"},"fe00e58d13e27d31de0ea2ebfa06b8c216475fae":{"kind":"REWORK","_number":4,"created":"2022-10-07 16:29:28.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/860502/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/02/860502/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/02/860502/4"}}},"commit":{"parents":[{"commit":"f1d0e395e95d3502a9c3904ce56520d116e6bf48","subject":"Add Tempest gate job for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f1d0e395e95d3502a9c3904ce56520d116e6bf48"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-05 22:05:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-07 16:28:19.000000000","tz":-420},"subject":"Add a boot-snapshot-boot test","message":"Add a boot-snapshot-boot test\n\nIt occurred to me that we do not (AFAIK) have a test that shows that\nwe can boot servers from snapshots we create right after we create\nthem. This adds one.\n\nChange-Id: I613357e8c9cc6d382caaf7291e150448148f32dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fe00e58d13e27d31de0ea2ebfa06b8c216475fae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/fe00e58d13e27d31de0ea2ebfa06b8c216475fae"}]},"branch":"refs/heads/master"},"574e0206113ba50300463b97d8d06a889e03aed0":{"kind":"REWORK","_number":5,"created":"2022-10-07 18:08:31.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/860502/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/02/860502/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/02/860502/5"}}},"commit":{"parents":[{"commit":"f1d0e395e95d3502a9c3904ce56520d116e6bf48","subject":"Add Tempest gate job for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f1d0e395e95d3502a9c3904ce56520d116e6bf48"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-05 22:05:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-07 18:08:23.000000000","tz":-420},"subject":"Add a boot-snapshot-boot test","message":"Add a boot-snapshot-boot test\n\nIt occurred to me that we do not (AFAIK) have a test that shows that\nwe can boot servers from snapshots we create right after we create\nthem. This adds one.\n\nChange-Id: I613357e8c9cc6d382caaf7291e150448148f32dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/574e0206113ba50300463b97d8d06a889e03aed0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/574e0206113ba50300463b97d8d06a889e03aed0"}]},"branch":"refs/heads/master"},"eda85800eae0c97a45bc4ad3c068d167b1f07ced":{"kind":"REWORK","_number":6,"created":"2022-10-07 20:30:21.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/860502/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/02/860502/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/02/860502/6"}}},"commit":{"parents":[{"commit":"f1d0e395e95d3502a9c3904ce56520d116e6bf48","subject":"Add Tempest gate job for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f1d0e395e95d3502a9c3904ce56520d116e6bf48"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-05 22:05:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-07 20:30:14.000000000","tz":-420},"subject":"Add a boot-snapshot-boot test","message":"Add a boot-snapshot-boot test\n\nIt occurred to me that we do not (AFAIK) have a test that shows that\nwe can boot servers from snapshots we create right after we create\nthem. This adds one.\n\nChange-Id: I613357e8c9cc6d382caaf7291e150448148f32dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eda85800eae0c97a45bc4ad3c068d167b1f07ced"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/eda85800eae0c97a45bc4ad3c068d167b1f07ced"}]},"branch":"refs/heads/master"},"6714b65a210c5851591dfd5dfe9ad1cbf1fa49e9":{"kind":"REWORK","_number":7,"created":"2022-10-07 20:33:29.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/02/860502/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/02/860502/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/02/860502/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/02/860502/7"}}},"commit":{"parents":[{"commit":"f1d0e395e95d3502a9c3904ce56520d116e6bf48","subject":"Add Tempest gate job for stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f1d0e395e95d3502a9c3904ce56520d116e6bf48"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-05 22:05:31.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-07 20:33:22.000000000","tz":-420},"subject":"Add a boot-snapshot-boot test","message":"Add a boot-snapshot-boot test\n\nIt occurred to me that we do not (AFAIK) have a test that shows that\nwe can boot servers from snapshots we create right after we create\nthem. This adds one.\n\nChange-Id: I613357e8c9cc6d382caaf7291e150448148f32dd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6714b65a210c5851591dfd5dfe9ad1cbf1fa49e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6714b65a210c5851591dfd5dfe9ad1cbf1fa49e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
