)]}'
{"id":"openstack%2Ftempest~860881","triplet_id":"openstack%2Ftempest~master~I8dd5e3256339ab1483e909fb5207d0da856e467e","project":"openstack/tempest","branch":"master","topic":"image-behavior","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2022-10-14 04:54:31.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2022-10-14 20:34:43.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2022-10-14 09:29:58.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-10-14 18:07:24.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I8dd5e3256339ab1483e909fb5207d0da856e467e","subject":"Fix image import tests for read-only stores","status":"MERGED","created":"2022-10-10 22:27:04.000000000","updated":"2022-10-14 20:35:50.000000000","submitted":"2022-10-14 20:34:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"860881-image-behavior","meta_rev_id":"04c0a3d920efb506e77993dcfd81c29b30804b18","_number":860881,"virtual_id_number":860881,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-14 20:34:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2022-10-12 09:56:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":1,"date":"2022-10-14 04:54:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2022-10-14 09:29:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2022-10-14 18:07:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2022-10-14 18:07:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-11 00:52:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-12 09:56:41.000000000","updated_by":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"reviewer":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"state":"REVIEWER"},{"updated":"2022-10-13 20:37:52.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2022-10-13 20:37:52.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2022-10-13 20:37:52.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"df2371c1be7d574baed71c254353bd9e76e7fe6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-10 22:27:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"918e38fd47bdbbd1bb56a83e7d828e32da8dbf15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-10 22:53:43.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"878aa824ebd80e5786804b5bad77e2747eff1ae8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 00:52:19.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eaef9c116e884d1d8a4416947392da29\n\n- grenade https://zuul.opendev.org/t/openstack/build/a7601768035d4dcca1bd23524a8b8710 : SUCCESS in 39m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a0ea36b20430459a913562161250c710 : FAILURE in 1h 07m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fddfb26669444834ad84cdf4545f17e0 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17853bf38ad840ba81c8482e2093213a : SUCCESS in 12m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c74d19fb8f954752a904cd6b30c8db0a : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/df756fbe7b17489dadefe3ccf34dce33 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e98faf206f946dbae95e4d8daac6bc2 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5471771f684246a8bdd31791ead77a9d : SUCCESS in 4m 46s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/4f985821eb0142d09737e2f3d28f3414 : FAILURE in 1h 23m 36s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/aa198d8d1ea840d5bfafd539beadb77a : SUCCESS in 1h 10m 31s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7b843150720543099dcda140efa73fda : SUCCESS in 1h 07m 36s (non-voting)\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/ca3f0e533d5a4b79aaf01c0242276262 : FAILURE in 1h 01m 59s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/290e2db6706f432f80b679f4b2063732 : FAILURE in 1h 05m 03s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/ace152474bd94c9a908d5362e19ee05a : FAILURE in 1h 01m 50s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/888e94f836b14edc9544743a8332a355 : FAILURE in 1h 06m 44s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/55431f49e1924f73af485ab8268c121a : SUCCESS in 1h 50m 06s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a105667a96c7478d83aef1890da45cf1 : SUCCESS in 1h 18m 44s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/229e2341fb3e4586b2f40363b1a63f76 : SUCCESS in 27m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a62ed5d6c00e4e7eb045c281667d8d9c : SUCCESS in 42m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b3231e6e917a43918ede4dd46b293cf3 : SUCCESS in 1h 56m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/cd4a6c07383c4f1cadbffca506587765 : SUCCESS in 40m 38s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/54634e9d922e455b877ca54dd30b6497 : FAILURE in 1h 13m 33s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a589ba5497274fad827e333694197798 : SUCCESS in 44m 57s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/80b2491b377944c586add35e534d9ea8 : SUCCESS in 50m 01s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d99c2217aad54a59ae86480f31e0ad21 : SUCCESS in 4m 14s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3a6e8f4c9e35437799e3172f2370754a : FAILURE in 1h 14m 52s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/4a8a56d6bd8e4c949829a7bf5c1b6951 : FAILURE in 55m 07s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/bf33a1c53a544e7ba1dc0f512012801d : SUCCESS in 3m 19s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/abe80c873e7c48c6b5b2b0ea38cb3a6a : FAILURE in 1h 05m 04s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ac3b030d15efeca84227b42aa1a249f630152ce4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2022-10-11 15:15:13.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"838d944b9dcb0ca48b4ca754aa6b04a91fc7a297","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-11 17:39:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3ceda369bb849fb9f0d6e13417c6502\n\n- grenade https://zuul.opendev.org/t/openstack/build/d8c00d2ca7c943e79a34b2dbfa907b48 : SUCCESS in 47m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/476a6d7bab014091ae6bf57abb857712 : SUCCESS in 1h 48m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/710e1835b5d04bb58e28373146f371f8 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12e8275139bc4544a5737b68b5161d69 : SUCCESS in 13m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a201f0bc02e4b5191f684231da35b41 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/14c01b8026cd4854868df49fd762d3df : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4e77b4e00714afd9c61613c9ad2465b : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/77eec9fdd40d40ce82cbfacf9e6cce1f : SUCCESS in 4m 26s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/862669ba7a854d50a07a7372bb200d36 : SUCCESS in 2h 12m 25s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/dd2f255511b8433f8a00e0c7699b486a : SUCCESS in 1h 22m 42s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b11217b20972427398efcdf7ac0de6bd : SUCCESS in 1h 13m 17s (non-voting)\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/46eb0f7d63144df188d963f036e7c6bc : SUCCESS in 1h 27m 54s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/2a1a05c0e2c945ba80098324165e17be : SUCCESS in 1h 46m 06s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/293555a4eeac45cab18ea6d4e7e8d9d6 : SUCCESS in 1h 34m 35s\n- tempest-full-wallaby-py3 https://zuul.opendev.org/t/openstack/build/8e3a837b5b054a36acf7d203bca8cc5e : SUCCESS in 1h 45m 01s\n- tempest-slow-wallaby https://zuul.opendev.org/t/openstack/build/c6c0463e2e214c4b971818021c8b4ce3 : SUCCESS in 2h 16m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/93e73881619b4822bef5d315a419039c : SUCCESS in 1h 41m 27s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/fe36cb5fa8e9483a82f84ea3e3967d25 : SUCCESS in 33m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9d09e280712848b3bf67379dac78e82a : SUCCESS in 59m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/46b780fa1d874ddfa933d4fb669bb5c8 : SUCCESS in 1h 47m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f4913c868a1f4334afd1a74ca8f76b22 : SUCCESS in 1h 02m 14s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/bf0bb72c73114f1e851b7ce8c6434196 : SUCCESS in 1h 17m 35s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ec34ed5f230b405895c1d46578cdb20c : SUCCESS in 1h 09m 19s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/840e6ada97cc46d7b072d2a0d7096c57 : SUCCESS in 1h 14m 35s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/96153c256f4e46bebdd0f36e07d5a848 : SUCCESS in 4m 20s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1e74eac00eb64d0a958f9f7d277ed141 : FAILURE in 1h 04m 16s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/63b1df3d17784acba4ac455f12d4d35f : FAILURE in 1h 11m 40s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/69aa608719cf4544a53bf6fe5b39bb1d : SUCCESS in 3m 08s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/d85f23f3c698402d98aa2ca0e108ef28 : FAILURE in 1h 40m 21s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"02eabc5882dc84c15e2daddd8a96acd39bb7cc5a","author":{"_account_id":30742,"name":"Soniya Murlidhar Vyas","email":"svyas@redhat.com","username":"svyas"},"date":"2022-10-12 09:56:41.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"49fc89c4f73019e3801dd70cab6296e8bc2cab2e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2022-10-14 04:54:31.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5d28216dc0523cddea961b4c189a3f96570174d6","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-10-14 09:29:58.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"544b81203ae73021b30da1b3874f81345e5de9bc","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-10-14 18:07:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"31d9fba9498158dd01fb918f10b49633797580e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 18:07:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"23addcb25e3fe2bf0571ff152f29887c52495afc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 20:34:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bde3de19dd84b5f8c423e8d6a86f8da\n\n- grenade https://zuul.opendev.org/t/openstack/build/00890d7694f04072b791c9f1a4c581bf : SUCCESS in 59m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/05e20676d2ea4b3890b10f614ee448cf : SUCCESS in 1h 28m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c8d03828d3c43f6af8bef05a6b171b9 : SUCCESS in 13m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71b74a450b4f441a88bddeb58f0385d1 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ba9881060d948cf8939a1036d44035b : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be0545c7d1904c9880943dd1ef38130c : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63cf11900345435db45b5e921f5b207f : SUCCESS in 4m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e3741c770acc4658a0a9918cdf7ae376 : SUCCESS in 2h 20m 16s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bd2a6c21bb4a45179c67741a9691e955 : SUCCESS in 54m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1ba26de9ebd44b2aee71c8da474bf0d : SUCCESS in 46m 38s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ae4abe918459425f92ea953424308192 : SUCCESS in 1h 39m 17s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d1297d74b318487b9e7a256040c95bc9 : SUCCESS in 44m 35s","accounts_in_message":[],"_revision_number":3},{"id":"970e6c04666dfe26430e84d10b00cbf868f54d1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 20:34:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"04c0a3d920efb506e77993dcfd81c29b30804b18","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 20:35:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f34d2ee972e4574b96ee73889b2df7c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32942be2b8bb4b7d887bbdb1e6a99aed : SUCCESS in 47s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"466f7060828b67c7a2482dd5f04ae2c59c5437db","revisions":{"26d44d6c0a7a87dec4f857b74c54e0298f1a77ff":{"kind":"REWORK","_number":1,"created":"2022-10-10 22:27:04.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/81/860881/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/81/860881/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/81/860881/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/81/860881/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/81/860881/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/81/860881/1"}}},"commit":{"parents":[{"commit":"6714b65a210c5851591dfd5dfe9ad1cbf1fa49e9","subject":"Add a boot-snapshot-boot test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6714b65a210c5851591dfd5dfe9ad1cbf1fa49e9"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-10 22:25:39.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-10 22:26:51.000000000","tz":-420},"subject":"Fix image import tests for read-only stores","message":"Fix image import tests for read-only stores\n\nIn order to test with one or more read-only store (such as an http\ntype store), we must exclude those from our list of stores we expect\nto see when doing an import. This fixes both the waiter for the\n\"all stores\" case, as well as our specific-store test from choosing\nthose stores for the list.\n\nChange-Id: I8dd5e3256339ab1483e909fb5207d0da856e467e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/26d44d6c0a7a87dec4f857b74c54e0298f1a77ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/26d44d6c0a7a87dec4f857b74c54e0298f1a77ff"}]},"branch":"refs/heads/master"},"5af75980ce197fa750a895d3b99c353ed01cb857":{"kind":"REWORK","_number":2,"created":"2022-10-10 22:53:43.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/81/860881/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/81/860881/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/81/860881/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/81/860881/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/81/860881/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/81/860881/2"}}},"commit":{"parents":[{"commit":"6714b65a210c5851591dfd5dfe9ad1cbf1fa49e9","subject":"Add a boot-snapshot-boot test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6714b65a210c5851591dfd5dfe9ad1cbf1fa49e9"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-10 22:25:39.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-10 22:53:36.000000000","tz":-420},"subject":"Fix image import tests for read-only stores","message":"Fix image import tests for read-only stores\n\nIn order to test with one or more read-only store (such as an http\ntype store), we must exclude those from our list of stores we expect\nto see when doing an import. This fixes both the waiter for the\n\"all stores\" case, as well as our specific-store test from choosing\nthose stores for the list.\n\nChange-Id: I8dd5e3256339ab1483e909fb5207d0da856e467e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5af75980ce197fa750a895d3b99c353ed01cb857"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5af75980ce197fa750a895d3b99c353ed01cb857"}]},"branch":"refs/heads/master"},"466f7060828b67c7a2482dd5f04ae2c59c5437db":{"kind":"REWORK","_number":3,"created":"2022-10-11 15:15:13.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/81/860881/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/81/860881/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/81/860881/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/81/860881/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/81/860881/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/81/860881/3"}}},"commit":{"parents":[{"commit":"6714b65a210c5851591dfd5dfe9ad1cbf1fa49e9","subject":"Add a boot-snapshot-boot test","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6714b65a210c5851591dfd5dfe9ad1cbf1fa49e9"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-10 22:25:39.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2022-10-11 15:15:01.000000000","tz":-420},"subject":"Fix image import tests for read-only stores","message":"Fix image import tests for read-only stores\n\nIn order to test with one or more read-only store (such as an http\ntype store), we must exclude those from our list of stores we expect\nto see when doing an import. This fixes both the waiter for the\n\"all stores\" case, as well as our specific-store test from choosing\nthose stores for the list.\n\nChange-Id: I8dd5e3256339ab1483e909fb5207d0da856e467e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/466f7060828b67c7a2482dd5f04ae2c59c5437db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/466f7060828b67c7a2482dd5f04ae2c59c5437db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
