)]}'
{"id":"openstack%2Ftempest~864371","triplet_id":"openstack%2Ftempest~master~Ie3bbbc80e4b2e7bca8e8b70fce1d31166220c6d3","project":"openstack/tempest","branch":"master","topic":"wallaby-last","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-12-07 10:08:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie3bbbc80e4b2e7bca8e8b70fce1d31166220c6d3","subject":"End support for stable/wallaby","status":"MERGED","created":"2022-11-13 01:29:30.000000000","updated":"2022-12-07 10:10:31.000000000","submitted":"2022-12-07 10:08:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864371-wallaby-last","meta_rev_id":"49eb317eff11b5b9543e6ce669e63cc1e470f0de","_number":864371,"virtual_id_number":864371,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-07 10:08:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2022-12-07 07:59:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2022-12-07 07:59:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-13 04:01:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-15 08:29:32.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"53364874595e31e971938968790f7201f92d04cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-13 01:29:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f887bb5aa6b668dfb0abd81a0af235088c043464","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-13 04:01:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88e6b7f11703450e880b373105a62450\n\n- grenade https://zuul.opendev.org/t/openstack/build/cfc2f037a70a4f9390614df744a077d3 : SUCCESS in 1h 00m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bb40e606d21c49f193cdcbb67c22af41 : SUCCESS in 1h 38m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f40f1f127b5749c1b77d4c063bb3dca4 : SUCCESS in 6m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaea12bcae9e45ec9fa81c37526a68fb : SUCCESS in 10m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/04b3d1fe344143d6807e307b7a1539f0 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f340ba234b924215998db7d3e0bb5be2 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b60d78f64f1649fcb640d796335df02c : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/485fb89ac0ce445ca55ed0c3cf8c624e : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45e0c75edec84e51af9eb3adf20d1bec : SUCCESS in 4m 51s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/7da94c734bcc4bad9d30e0ab159209ec : SUCCESS in 2h 17m 30s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/0e593eafbbe34f708de6e3f2bc4d5095 : SUCCESS in 1h 32m 25s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2db49617445e4621892e751fc823b53b : SUCCESS in 1h 31m 29s (non-voting)\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/648cf56db59049ac91552293ee9633f7 : SUCCESS in 1h 38m 53s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/58af5e37d57845dbada5a683486d8407 : SUCCESS in 1h 50m 25s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/213938629a4940d794c955642d242870 : SUCCESS in 1h 31m 30s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/9395b50d50e24748994a3288ae2917ac : SUCCESS in 1h 48m 50s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/dc8ebc8195354884a8ef4744eea40acb : SUCCESS in 33m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/321807b154904ccc980f0d4095a8f70c : SUCCESS in 1h 02m 06s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e57c612b82a04d3ab9784a3918e0a1a4 : SUCCESS in 2h 24m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c98b50d8c6a34380b5f528e5a0cb65d3 : SUCCESS in 56m 24s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9359a2122566473b87dd334e27bcd7e4 : SUCCESS in 1h 22m 01s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0f88e13d0064a71a40c9847d60be6c2 : SUCCESS in 1h 01m 28s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/4f988985a85540ad8baad8fa12b4c24c : SUCCESS in 1h 04m 28s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/16431c4e0ccb41529bfce62dc25e23ee : SUCCESS in 4m 41s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/d871ba531af4462c969822b2b21b099d : FAILURE in 1h 23m 54s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/376ae2392d80470f9cf04d05171193aa : FAILURE in 1h 19m 18s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2318fcc94fca4a18ae735f8bdd08590e : SUCCESS in 2m 38s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/903382560a024b7387e0a4bc50cb34db : SUCCESS in 1h 56m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9509cf29a5a14719b9a9a5176514613449807bf9","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-11-15 08:29:32.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b95539c1239b73c60b336165b8f11e6af620a903","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-15 08:29:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9a4f8d6d8a84e02a8127d6444c6909d\n\n\nWarning:\n  Change 862393 in project openstack/releases does not share a change queue with 864371 in project openstack/tempest","accounts_in_message":[],"_revision_number":1},{"id":"5ce4149e773e20928418da5387d6e68fe9f660c3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-15 18:42:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1211ea5ecb84c52f1f92f10d742ba124352d8489","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-07 00:37:31.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b65e250c9b5f1cf335a016451c751a611b0435d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 03:08:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6d8675d353140a89b466b39e4bfb3c7\n\n- grenade https://zuul.opendev.org/t/openstack/build/d98fdb7a732d4e7cab99601c04c62d3d : SUCCESS in 50m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8ef0c29a0c04fa78b3db7b037c063a5 : SUCCESS in 1h 27m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/de9d60e34658468d9f6097bcd144884d : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014dc51f2b924aa3bcea6e6ae2635ae2 : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/398f818419874e66a9c743702e700e70 : SUCCESS in 4m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979ee4f3957544fb92b0c382e0e8da18 : SUCCESS in 3m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/916a75b108dc4956b391392ed61e1eeb : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bc6ec100e1a74062a16129ca18ceb37e : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f046a612f1840fbac9c349e736bb867 : SUCCESS in 3m 52s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/92e1f1b09e1841ee8e1db7a565013e9a : SUCCESS in 2h 02m 02s (non-voting)\n- tempest-full-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5fb7e02ac6654d50a4a2bbba1db4d904 : SUCCESS in 1h 25m 49s\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/7df273954b5c41229c3c99e3d22b33dc : SUCCESS in 1h 19m 44s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/25c0007d9391417ab27e5e9ccf5e5b07 : SUCCESS in 1h 41m 44s (non-voting)\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/ac57677ab56e46b5bb16c84788386678 : SUCCESS in 1h 44m 18s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/1d745f7ce87e47f0bc508525bef2a88c : SUCCESS in 1h 38m 03s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/d44d3ceb19a84f519c50319762a05cd9 : SUCCESS in 1h 37m 23s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ae5e283762fb4d56ba5a12d43b330cfb : SUCCESS in 1h 42m 34s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b09941abbe994c86b14d832cea37b916 : SUCCESS in 32m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d2c9b54189a4c108b0f38efe4969aa2 : SUCCESS in 54m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ae2c5291e6f74f648da1e97844849200 : SUCCESS in 2h 25m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/73fcba783e18465dbc92ac5031052afd : SUCCESS in 40m 51s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a34f4416a2034512b94ba90f4626ab98 : SUCCESS in 1h 07m 12s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/14f1e0c9cf1b4e90b1b35deb14bf6333 : SUCCESS in 44m 24s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/ab38b05ce8cf418bb70f18262e6dc34d : SUCCESS in 1h 06m 57s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/34714ca04abd4fe3bb2e78e82e2003c6 : SUCCESS in 3m 39s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/41542786f9704bb9bf49d62f29ac1175 : FAILURE in 1h 36m 36s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/6ee99323953b4ace9f62d4cf9b3f3bb0 : FAILURE in 1h 19m 54s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f3a79a05eae343b1beb292e3b66f74ee : SUCCESS in 2m 50s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/4cebc4daadea4a25a5d92a5a6810cf8d : SUCCESS in 1h 55m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b73683d4856b162da32c24d9c213f7e5f98b6f94","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-12-07 07:59:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8fb9faf1a4839ac135f5d407a7a31e22ac2fe7a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 08:00:19.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ee377744bdf7817347a093dd87357519d0338002","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 10:08:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7c9513184a442b79ab65b0ee44d0769\n\n- grenade https://zuul.opendev.org/t/openstack/build/801fbe894ead43399d939d86e9cf78e3 : SUCCESS in 40m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/22fb9a51785349018ca612d86899f2a1 : SUCCESS in 1h 07m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54e31ac9d4ad485ebdb876adf02cab8f : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80066d31c6c54a5690bdb55ca7200db3 : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba4a190e239b4aa18f05dcfab03fe340 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7feddf681dd64dd89e27d0428053649f : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/131edf92f2d049f2987a6c8f13dbe9c1 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7487c7e716a4644979309095509db42 : SUCCESS in 4m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9af034390864750b0cf704ff5d6e891 : SUCCESS in 2h 01m 34s\n- tempest-full-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3cce5a353efb4f5e94b30e4760adabce : SUCCESS in 1h 45m 48s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/85aec018d8b64d8895ef069168a5c230 : SUCCESS in 52m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ddbd8233bd944961b9d6e0317777d5b7 : SUCCESS in 53m 41s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b8a21183599b4ca2a6a872b4f33f1169 : SUCCESS in 1h 42m 10s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/151b1bbe733847fd9b66f1c557d3bdd2 : SUCCESS in 1h 28m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5b4fc1e5e5b34098be575b5000ade9cd : SUCCESS in 41m 14s","accounts_in_message":[],"_revision_number":2},{"id":"d5e219e287de7a8c7258157d9cf19c91127adc8b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 10:08:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"49eb317eff11b5b9543e6ce669e63cc1e470f0de","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 10:10:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ed1c7ae7c70428cb5baad5378b80149\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0528d27df78c418590c16b12a3a5bbd8 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4ff7fb91b1b497eae2781081a857665 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5844fd9aeb307d31f6128f73b3e6163e22f8df88","revisions":{"a83e911dbf38d14c9a2efbc4cf8556608392969b":{"kind":"REWORK","_number":1,"created":"2022-11-13 01:29:30.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/864371/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/864371/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/864371/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/864371/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/864371/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/864371/1"}}},"commit":{"parents":[{"commit":"edd5a4cba37e94579ac5583b5bce38109c14bf75","subject":"Merge \"Fix server id reference in _rebuild_server_and_check()\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/edd5a4cba37e94579ac5583b5bce38109c14bf75"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-13 01:25:20.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-13 01:25:20.000000000","tz":-360},"subject":"End support for stable/wallaby","message":"End support for stable/wallaby\n\nThis marks the end of support for stable/wallaby\nwhich is EM now [1]. This remove the stable/wallaby\njobs and also adds relesenote to mark the end of\nsupport of Wallaby.\n\nDepends-On: https://review.opendev.org/c/openstack/releases/+/862393\n\n[1] https://releases.openstack.org/\n\nChange-Id: Ie3bbbc80e4b2e7bca8e8b70fce1d31166220c6d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a83e911dbf38d14c9a2efbc4cf8556608392969b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a83e911dbf38d14c9a2efbc4cf8556608392969b"}]},"branch":"refs/heads/master"},"5844fd9aeb307d31f6128f73b3e6163e22f8df88":{"kind":"REWORK","_number":2,"created":"2022-12-07 00:37:31.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/71/864371/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/71/864371/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/864371/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/864371/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/71/864371/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/71/864371/2"}}},"commit":{"parents":[{"commit":"96cd444cac4a0d2d1db619365f645a60c3de73a5","subject":"Merge \"Set back to use Bionic for Ussuri, Train and Stein\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/96cd444cac4a0d2d1db619365f645a60c3de73a5"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-13 01:25:20.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-12-07 00:37:17.000000000","tz":-360},"subject":"End support for stable/wallaby","message":"End support for stable/wallaby\n\nThis marks the end of support for stable/wallaby\nwhich is EM now [1]. This remove the stable/wallaby\njobs and also adds relesenote to mark the end of\nsupport of Wallaby.\n\nDepends-On: https://review.opendev.org/c/openstack/releases/+/862393\n\n[1] https://releases.openstack.org/\n\nChange-Id: Ie3bbbc80e4b2e7bca8e8b70fce1d31166220c6d3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5844fd9aeb307d31f6128f73b3e6163e22f8df88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5844fd9aeb307d31f6128f73b3e6163e22f8df88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
