)]}'
{"id":"openstack%2Ftempest~864373","triplet_id":"openstack%2Ftempest~master~I1a3cea4cc543aabe61483a32f71b2b9db178cecf","project":"openstack/tempest","branch":"master","topic":"wallaby-last","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-12-07 13:20:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1a3cea4cc543aabe61483a32f71b2b9db178cecf","subject":"Switch back the tox constraint to master","status":"MERGED","created":"2022-11-13 01:41:28.000000000","updated":"2022-12-07 13:21:56.000000000","submitted":"2022-12-07 13:20:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"864373-wallaby-last","meta_rev_id":"b056b03c89efb7d2f41800b81f32e227b0c47825","_number":864373,"virtual_id_number":864373,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-07 13:20:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2022-12-07 08:13:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2022-12-07 08:13:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-11-13 04:13:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-11-15 08:30:11.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"027150b5ba59aea339867b796f2acf8e3c7ed30f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-11-13 01:41:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c98a4cc4c01933d1f7abd7068aeb6c3957649b5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-13 04:13:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d53a96625ecc4851beb6347e81110c5d\n\n- grenade https://zuul.opendev.org/t/openstack/build/a041200b8a9a452b9c32cfa704d705f3 : SUCCESS in 1h 03m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb68b28f603647b78877e7a98e448ee1 : SUCCESS in 1h 45m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9a7272577954e548db82a576ab5b623 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c57e616abe2945df9b6605a4b56ec098 : SUCCESS in 11m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a39506fa22f4a1086efbdb3387d541c : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/304b0b3f9d6e4a30858ff5d901a15586 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b6188becd584ea09ef7403fd5357e2d : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ec36cec5d5144f6b871a16a2c3b1d66 : SUCCESS in 4m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4279c787f184d2e8318aa89cb137a0c : SUCCESS in 4m 48s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/135f044f21f74a7ebb7d4000994d3f41 : SUCCESS in 2h 06m 07s (non-voting)\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/6a1d555e569843f0ad2842113037e6a8 : SUCCESS in 1h 30m 37s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c429c3e5811d41798230260b36ae9a29 : SUCCESS in 1h 32m 59s (non-voting)\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/764692b4f32f4e779ed5c23f81dd460f : SUCCESS in 1h 48m 41s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/8f16d37208124ca2bbda37debe1a1f54 : SUCCESS in 1h 39m 26s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/d4197cdf927c4e908549683aacc971be : SUCCESS in 1h 35m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a89170d68d7b4a509b631ebdbd46bd18 : SUCCESS in 1h 53m 18s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/192252c44ff348079f6eeb331a40e66a : SUCCESS in 35m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/455e9df3177948d7bb018c5fa3236b5a : SUCCESS in 1h 02m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9de58934607440fb9164a0eee28e1599 : SUCCESS in 2h 23m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/56ca26da69e24470837f04e79a23d56c : SUCCESS in 57m 29s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/88d51480673648d2b2abec8403856000 : SUCCESS in 1h 31m 20s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0bbdb4b31ad44933b59161c3ea557195 : SUCCESS in 1h 10m 32s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/35eb29faa7ba49299dd3984c2c1b9ede : SUCCESS in 1h 15m 03s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/ee40b9fd738e4067acca29a48adcba55 : SUCCESS in 4m 01s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c53cc2717ea3489bbb22031864a760f1 : FAILURE in 1h 15m 31s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/de364912d9924429af3576c166095095 : FAILURE in 1h 28m 27s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c3a4ad3334dd492da739dfc77061c24f : SUCCESS in 2m 48s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5962292737a4493b7a2292362a3ec32 : SUCCESS in 1h 39m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14ca019b8611691e0bde1f5fa59fd9906d4faa52","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-11-15 08:30:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f2f3670cb2c4ac27ccaf9709fd5a08b6c2de0e34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-12-07 00:37:54.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cae0d42ba7d339bc6f5cb4790accbaf69b11a0ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 03:15:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7aa352670ffa461c800b45b51984902c\n\n- grenade https://zuul.opendev.org/t/openstack/build/eafcf65a6d9d402d8c5bc5a00c2509c2 : FAILURE in 41m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4e7f57f5c9234ee9a89317c38d9f5b89 : SUCCESS in 1h 52m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a56d43ead357465f83673f678f4b50a2 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec6610e3bf74167bd6a0c9a7441f861 : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e9b53bf75024cbfb1747ff9f6d7ac31 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e98ee69f8299419d8f7b74ebfe86cde6 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b80b2f5617024f88a6e11fe277e7e5bd : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40a46ff331fc48339eff07b6407bf7a6 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c941bcb68ea04f2e81706d3b9addc663 : SUCCESS in 4m 01s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/5b9ef1771f87490cafe8b722ce03763a : SUCCESS in 1h 48m 05s (non-voting)\n- tempest-full-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6afa1d809f494104a60259f1fc193802 : SUCCESS in 1h 37m 30s\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3c9ad9a0d179483b9c3e20afd822c7e3 : SUCCESS in 1h 16m 08s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/edc3315023db47e1a77188785dfe7f23 : SUCCESS in 1h 21m 45s (non-voting)\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/82d5d9013f6441128b7b90a6ad33bfcf : SUCCESS in 1h 54m 07s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/36aef12b547f4e4e801fbb2c0bda442a : TIMED_OUT in 2h 04m 40s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/e23c1904eed44b60bad5e5e377801f9c : SUCCESS in 1h 43m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e528901e76584e789ec3e4d9c1550595 : FAILURE in 1h 30m 12s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/586a27a58924481ea5fcb8267302c88f : SUCCESS in 26m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bece1da86dc456196867a8f248817b6 : SUCCESS in 1h 01m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/38cec5e1ba05499ba799188d7f1f6297 : SUCCESS in 2h 17m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5f1abdb1223540fab170383b69dc953c : SUCCESS in 50m 00s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/860d360dc7ff48b3824b1caa35ace92d : SUCCESS in 53m 48s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9cfff85b15174bde860108073b8663a0 : SUCCESS in 56m 34s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/8b3dad37cdd74cf9a6ababfe1dbce029 : SUCCESS in 53m 28s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/ed3b7f1f0c114fc2838748bed19e83a9 : SUCCESS in 3m 30s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0b2f1fcddaa4499398460e40f54805fe : FAILURE in 1h 19m 14s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/b65b70f461994f488934b66ed09c98bb : FAILURE in 1h 12m 12s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3c8aad600277440c88491d7dab11b242 : SUCCESS in 2m 52s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/72e61c80e6b04aa58d24ff5ca0b1cf3d : SUCCESS in 1h 51m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b6ea4045166ee65f24593b9b9ac22bec95f3e40","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2022-12-07 08:13:34.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5c315873150536b849915e0983110d1fbf8e502b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 10:58:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4c6c2b59552403a9bb5bb2ab921cad6\n\n- grenade https://zuul.opendev.org/t/openstack/build/f4742ed1182d413697b8d465ed548e90 : SUCCESS in 58m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c13d608b5cf3487b832a8d9e89cd97df : SUCCESS in 1h 42m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5ec9155c63c4b7ea811c6460e837b99 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49a0ad44e8f94b4ba367304728dfff54 : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12d20a9d1f9a4ced8d767144f110c61e : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d7f1a24ee2f4c81a49f1b2ac187de9e : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22d3d1f1848f4c1c9b16e87a7526803c : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5f96aab3dc5e4dc6b387399801849a3f : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb8586ddd1124304bce61dbfd6aeced4 : SUCCESS in 4m 05s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/6dfc696fd6274b018f853f9aed7efafc : TIMED_OUT in 2h 35m 32s (non-voting)\n- tempest-full-ubuntu-focal https://zuul.opendev.org/t/openstack/build/19254b42e7f34832ab4d5323b5c3773f : SUCCESS in 1h 20m 48s\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/3947eea457b848a3a9ad6818fb4de495 : SUCCESS in 1h 30m 24s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1520117aa07545f4898ae45eee95a1bf : TIMED_OUT in 2h 05m 57s (non-voting)\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/b776f9c3ce7e45388a4616663594d952 : SUCCESS in 1h 25m 12s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/f7eab5c0f8184092bd24892ee9da470f : SUCCESS in 1h 31m 11s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/af21bfe0a0af4c7da018db968f227a1a : SUCCESS in 1h 18m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a0f6f854c5b44716af2561b116e4f1cc : SUCCESS in 1h 29m 36s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/97a9a3a7d6c645a0a66557685178b94d : SUCCESS in 35m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5646f67505941b8a49f685270f1be75 : SUCCESS in 43m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4b2e90a163fd420f853fbc407dc81a34 : SUCCESS in 1h 52m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8589e2dd626442e3aec73e34ab347838 : SUCCESS in 52m 42s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9b7a090447c041679f9fc31b20e9dd34 : SUCCESS in 1h 23m 55s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/fda1cfe1960e4158a9e749e9fe54d07d : SUCCESS in 1h 20m 49s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6619b9844a904e31a5d733fe47b860b4 : SUCCESS in 55m 08s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/b92b5ad43e8841dc974487b3043313c1 : SUCCESS in 1h 13m 12s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8be9fbdb892344b6a3399c6b48f533ac : SUCCESS in 3m 13s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/e99ff068ad6e43c183613d025a0a112f : FAILURE in 1h 20m 46s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/2ec2762cc045428d8b52bba3492831ea : FAILURE in 1h 11m 49s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/6d64ae7d2d9a413781bb2a6eadc6c7d7 : SUCCESS in 3m 14s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/0ebe337497a94772836512b741a87946 : SUCCESS in 1h 23m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5d2a53d4cbf78ee86509dd961d26120465ed1398","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 10:59:11.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e4bd739255f29bf4492cf8b87b1bc53a54967278","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:20:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f7cbe88f9704150832d478b676896bb\n\n- grenade https://zuul.opendev.org/t/openstack/build/4ed3dfcc0fdb4a87bb55a2b94d3cd714 : SUCCESS in 56m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/60b14b44ce524a148c66b7a5c6ac97c1 : SUCCESS in 1h 36m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af62e3a6c22e4478a0b33d35a21a6895 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/700ff149613e4fd8adcf6ab3a21cb435 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90ecaf474e2043a990467b357dd6dd58 : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad0ddf5ce374413881e3bd43e0bf465a : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62e90c81e0304d08b08d183434755491 : SUCCESS in 4m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/caee0024b5bf4fe29693e9b8d47a483b : SUCCESS in 5m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0b7772e3990548d5acabfdd3c280c9a5 : SUCCESS in 2h 15m 10s\n- tempest-full-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f81515e2395c48ad828070bd86bad0ae : SUCCESS in 1h 29m 43s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b46ee53114fb4fb0ab5871139939ed30 : SUCCESS in 51m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/126e026b1f2e4b56acd30df8e93789e1 : SUCCESS in 54m 56s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a8520f9441c24565a6cef21d5c06d0ff : SUCCESS in 1h 40m 49s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/292b79a381d043c69a8ad9ac5ca030ec : SUCCESS in 1h 34m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a2bfb8e9b694490481f731c7a6d7c70b : SUCCESS in 45m 46s","accounts_in_message":[],"_revision_number":2},{"id":"65c34bc9f510659869e026d8b06d2db82edf9253","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:20:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b056b03c89efb7d2f41800b81f32e227b0c47825","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-07 13:21:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d10f6c92975145978b96b8f1a0e63600\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/938865a859804f59948b69701e93b497 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4eb6a1d43b38432f9803d5bc9df72205 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5c075ccef2a3a6dd151756d1244fa5541f26a406","revisions":{"c5a007ea70557e12616b8be61c2ca84019185ea1":{"kind":"REWORK","_number":1,"created":"2022-11-13 01:41:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/864373/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/864373/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/864373/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/864373/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/864373/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/864373/1"}}},"commit":{"parents":[{"commit":"c7ae676c03e8f10051b4a43d299c65bd8e9b0a24","subject":"Pin stable/zed constraint in tox to release 33.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c7ae676c03e8f10051b4a43d299c65bd8e9b0a24"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-13 01:40:11.000000000","tz":-360},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-13 01:40:11.000000000","tz":-360},"subject":"Switch back the tox constraint to master","message":"Switch back the tox constraint to master\n\nTempest 33.0.0 release hash is prepared[1] so we can\nrevert back to use the master constraints by default.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/864372/\n\nChange-Id: I1a3cea4cc543aabe61483a32f71b2b9db178cecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c5a007ea70557e12616b8be61c2ca84019185ea1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c5a007ea70557e12616b8be61c2ca84019185ea1"}]},"branch":"refs/heads/master"},"5c075ccef2a3a6dd151756d1244fa5541f26a406":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2022-12-07 00:37:54.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/73/864373/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/73/864373/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/864373/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/864373/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/73/864373/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/73/864373/2"}}},"commit":{"parents":[{"commit":"633bd0d13a3c0b15b69c8e0e4bf063f0836af14d","subject":"Pin stable/zed constraint in tox to release 33.0.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/633bd0d13a3c0b15b69c8e0e4bf063f0836af14d"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-11-13 01:40:11.000000000","tz":-360},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2022-12-07 00:37:54.000000000","tz":0},"subject":"Switch back the tox constraint to master","message":"Switch back the tox constraint to master\n\nTempest 33.0.0 release hash is prepared[1] so we can\nrevert back to use the master constraints by default.\n\n[1] https://review.opendev.org/c/openstack/tempest/+/864372/\n\nChange-Id: I1a3cea4cc543aabe61483a32f71b2b9db178cecf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5c075ccef2a3a6dd151756d1244fa5541f26a406"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5c075ccef2a3a6dd151756d1244fa5541f26a406"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[]}
