)]}'
{"id":"openstack%2Ftempest~877714","triplet_id":"openstack%2Ftempest~master~Ic91e443ae1948d71fa276be03dcef4576c64df33","project":"openstack/tempest","branch":"master","topic":"qa-2023-1-release","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2023-03-17 16:37:28.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"Ic91e443ae1948d71fa276be03dcef4576c64df33","subject":"Switch to the master constraint in tox.ini","status":"MERGED","created":"2023-03-16 19:53:50.000000000","updated":"2023-03-17 21:35:24.000000000","submitted":"2023-03-17 21:33:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877714-qa-2023-1-release","meta_rev_id":"471a97390efd4e9be847646f210ea847504f28f5","_number":877714,"virtual_id_number":877714,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-17 21:33:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2023-03-17 00:11:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2023-03-17 17:21:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2023-03-17 17:21:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-16 22:30:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-17 00:11:44.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"72d7fd968942c9aad5fcf481dd04247d4fe50c3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-16 19:53:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7ba0ab2f037f0cf3f8f0f901aa5b338dd4aad4f7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-16 19:54:38.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3fcaa3e6ba055c82c6cd141fe29f94e1473f0494","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 22:30:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd08079f3db94ac4bc501d0ebecc80e2\n\n- grenade https://zuul.opendev.org/t/openstack/build/1287d09e141a4caeb41d8d6480e85d40 : SUCCESS in 1h 06m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a593b280248453bb5a7260d577f70ad : SUCCESS in 1h 55m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3cf2b30152094b468a61adff1224eadd : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/115387b01a124ff3b2b213f55aaa490b : SUCCESS in 12m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7ba8b7cb62b4213a381603ef0a08b82 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcf3b65689534779a84b17750098576b : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/596d93e9bb5b42f8bff91c516d0a4e31 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eed4e38e21ce4c0da3279227d07fe114 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99b21bc0b9694eeeb35f617cb2cf2fa5 : SUCCESS in 4m 49s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/4b780cd0ffd849348e15f97f037f26e0 : SUCCESS in 52m 53s\n- tempest-full-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c3112f5815984f4d8830ede7e55448a5 : SUCCESS in 1h 38m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5794a927b414484a9bd6dc04114464de : SUCCESS in 1h 34m 18s (non-voting)\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/1adba35dca0041a3b3fbcaeec096abe2 : SUCCESS in 1h 21m 21s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/54024accb5df4cee98a12198d57b22bd : SUCCESS in 1h 14m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/28a8b390bd954c7fb9d77d8b0062827e : SUCCESS in 1h 04m 13s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/4f3a99316d1d458dab3720fb01c01f6e : SUCCESS in 32m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7aeafeb8a817495194b9e78887e02959 : SUCCESS in 48m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2e0c2f67dc7b4b4e9824e0564d4d96d8 : SUCCESS in 2h 28m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/542170169eb54546836c6405e61c0db3 : SUCCESS in 1h 00m 37s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6d5a5aae729f4d5da67dc40217f8b71b : SUCCESS in 1h 53m 33s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5fbdfe234cd9425998a0a614f5cbe9f1 : SUCCESS in 1h 14m 21s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e80f21ad479849e384daf6d83dbbd053 : SUCCESS in 1h 07m 50s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2ed4c02dd4e14b88a5b08cd00691312a : SUCCESS in 1h 29m 24s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/44610060f42c466dbecf519c2f8e30d5 : SUCCESS in 3m 48s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/cb5530dddcd34fa29efdabdecada6c49 : FAILURE in 1h 27m 32s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/838eea610b034be7aabd5b6764b7bb1a : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"1debc1e25fe230cc26f0ba03a2f3073cdebb1f07","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-03-17 00:11:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"545c86718446510bdd521c85990c96a2b5ad6129","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-17 16:37:28.000000000","message":"Patch Set 1: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06de3b03087548eea959c35e966ddd85eb3e76ee","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-03-17 17:21:09.000000000","message":"Patch Set 1: Workflow+1 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6db96589b227d8f788e660aa6ab0add329a029a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 18:34:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e05a27df561331993645f9360bc7009b45d41a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 21:33:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0667564c7823450a93deea921e421360\n\n- grenade https://zuul.opendev.org/t/openstack/build/207a78261bdb4043bbd543149571c7a3 : SUCCESS in 1h 11m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/83c53be0cc0044dfa3fc74aa979f8683 : SUCCESS in 1h 26m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a75ae605c28a4732800b0719264effc0 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57cd3ecfed1047aaa107d1e95cbc3d08 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94792864ec634d8e8f2b991ab1bc1e05 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7f5b7c122c94300aba8a83f51fa7896 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/547da93379b94722935007ef1b67dc83 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87d84b29702b451eae0240850cfa9e17 : SUCCESS in 4m 13s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/4fa014474e0044e6aa2c0132910b513f : SUCCESS in 2h 16m 23s\n- tempest-full-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c69b5b29fe95436db91150c7d3eafb72 : SUCCESS in 57m 47s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e857ad6446d44adb9940de88c1b53acc : SUCCESS in 1h 05m 17s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/cba85c7aebac4afe95ad2444bf1f81dc : SUCCESS in 51m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f8c0aaf6cec4a8fb0a1fee9ebc6ad2b : SUCCESS in 37m 53s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/125d96d5b09a47b6975cbe325953a3c4 : SUCCESS in 57m 51s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/45059d9cae624504b5e665c93e7ab903 : SUCCESS in 1h 14m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c4d6941a511e43f28914fa3100652b54 : SUCCESS in 57m 01s","accounts_in_message":[],"_revision_number":1},{"id":"5a9eab2473b4796c96d7b263b3964903f67dedf5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 21:33:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"471a97390efd4e9be847646f210ea847504f28f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 21:35:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87ba4c22ae564a1aa05723cda0ab6464\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c35d7dfe156148c99c952791ef8de9a1 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/451cdf34b49341be9bcc4ffa42e80f7a : SUCCESS in 35s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"91e2be9ebfebfa9b1695b2216b570991adc0f08c","revisions":{"91e2be9ebfebfa9b1695b2216b570991adc0f08c":{"kind":"REWORK","_number":1,"created":"2023-03-16 19:53:50.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/14/877714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/14/877714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/877714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/877714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/14/877714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/14/877714/1"}}},"commit":{"parents":[{"commit":"00a315dc9636fb33332ef96ec504660f471e0ec1","subject":"Use stable constraint in tox to release new tag for 2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/00a315dc9636fb33332ef96ec504660f471e0ec1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-03-16 19:52:29.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-03-16 19:52:29.000000000","tz":-300},"subject":"Switch to the master constraint in tox.ini","message":"Switch to the master constraint in tox.ini\n\nTo releas a new tag for Tempest 2023.1, we modified the\ncompatible constraint in tox.ini\n- https://review.opendev.org/c/openstack/tempest/+/877713\n\nAs new tag is released now so we can switch back to the master\nconstraint in tox.ini.\n\nChange-Id: Ic91e443ae1948d71fa276be03dcef4576c64df33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/91e2be9ebfebfa9b1695b2216b570991adc0f08c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/91e2be9ebfebfa9b1695b2216b570991adc0f08c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
