)]}'
{"id":"openstack%2Ftempest~880630","triplet_id":"openstack%2Ftempest~master~Idfdbe3dc276fad53c50421f5e4ec32972e052ebc","project":"openstack/tempest","branch":"master","topic":"fix/object_storage","attention_set":{},"removed_from_attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2023-05-04 14:47:43.000000000","reason":"\u003cGERRIT_ACCOUNT_30674\u003e replied on the change","reason_account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}},"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-05-13 08:39:14.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_6968\u003e by clicking the attention icon","reason_account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}},"hashtags":[],"change_id":"Idfdbe3dc276fad53c50421f5e4ec32972e052ebc","subject":"[WIP] Catching broken tests in tempest-full-test-account-* jobs","status":"NEW","created":"2023-04-17 13:30:51.000000000","updated":"2026-05-13 08:39:14.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"03397fa93d42d5dc24b0bbe9b6165e373d30315d","_number":880630,"virtual_id_number":880630,"owner":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-06-05 09:54:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"CC":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2023-04-17 13:31:08.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-17 13:31:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-05-04 14:32:08.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"CC"}],"messages":[{"id":"7038a589bf220d75f9f556761b2d327524b986c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-04-17 13:30:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"efdaa31cb58a4046c9f3f9ad5f51b23257a9c03f","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-04-17 13:31:08.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/tempest on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/tempest/zuul.d/project.yaml@master\", line 12, column 31","accounts_in_message":[],"_revision_number":1},{"id":"5806c8c2585fe899da0ac0534b05da7998964c76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:31:22.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/tempest on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/tempest/zuul.d/project.yaml@master\", line 12, column 31","accounts_in_message":[],"_revision_number":1},{"id":"ba55d1d3b85c2be74359ea51d633f89b64b9e279","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-04-17 13:35:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f9e7e101964228a5bf809ff5437f2c8d77abdfc3","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-04-17 13:35:59.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"411ca6647611b35dd0439792716e5512d35350fd","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2023-04-17 13:36:03.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/tempest on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/tempest/zuul.d/project.yaml@master\", line 46, column 31","accounts_in_message":[],"_revision_number":2},{"id":"31cd32931d745165757384fe0edf1b6d090f3f0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 13:36:16.000000000","message":"Patch Set 2: Verified-1\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/tempest on branch master.  The error was:\n\n  found undefined alias\n  in \"openstack/tempest/zuul.d/project.yaml@master\", line 46, column 31","accounts_in_message":[],"_revision_number":2},{"id":"f3f418adaa4e4c2f78c5e0484e9769f3dcd9685f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-04-17 13:37:54.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3eb92bbd7057b44e7599b0c305de6b343ec4df41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 14:57:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcba093639f94333830ddcf445cb9f05\n\n- grenade https://zuul.opendev.org/t/openstack/build/7c90075e428f4d9a88d55620d231a08f : FAILURE in 26m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e574a8a509e543bb8d780217ff9628a7 : FAILURE in 46m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c237c8ef98374125929c3a68eb06dc03 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f09ff16a824479fb087a3671ac81d18 : SUCCESS in 13m 50s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1d4aa897bdf24fffbcd53b6cafe5f232 : FAILURE in 1h 10m 08s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/42980b1e858346aea464c1365fc3e1d7 : FAILURE in 1h 17m 17s","accounts_in_message":[],"_revision_number":3},{"id":"a36e9d8d73aa8236b13e34e5473e3f808df61781","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-04-17 15:51:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"0132c39be1416f0eceb9d928d39a6bce95314737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-17 17:13:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/45020513e9594ca7bb1e184e12ad2db2\n\n- grenade https://zuul.opendev.org/t/openstack/build/13676c97ff30432dab656abce772fa99 : FAILURE in 39m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c119f36273c14ccb9a8d244973156bfa : FAILURE in 1h 09m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d9f52ad11814a498105c129035a33c6 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c5ba050c358043d9ab709d062880bee0 : SUCCESS in 13m 22s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0600a4139c5f4a07befee60f0a08ed88 : FAILURE in 1h 12m 59s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/a9f45455d68c440e85a4fe6fbf6a6cfd : FAILURE in 1h 20m 11s","accounts_in_message":[],"_revision_number":4},{"id":"779062c1967f02544a2b8527a8b6580df50fcbe3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-04-18 08:13:24.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"8983bd2f7b2fbcddb1d881e556f3d144ccbcaf11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 09:41:46.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0cbe750964cb4b869021b880fc596b8b\n\n- grenade https://zuul.opendev.org/t/openstack/build/97c9588d00a64ac0803bf4fa58749616 : SUCCESS in 1h 01m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3b514574c3f47a3993835ad0278b336 : FAILURE in 1h 22m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97071860c7054bba81997817fcfdc1e1 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ef11d26a3494de0955d78a64628f88d : SUCCESS in 11m 23s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/57edb50f6a1645f5acf650f0d826b8f9 : FAILURE in 1h 14m 45s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ac8e81db36c0496ca850acfa704da002 : FAILURE in 1h 06m 52s","accounts_in_message":[],"_revision_number":5},{"id":"7b1499b876884df16df14089c0be8ee8f6a68653","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-04-18 13:13:05.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"8b40f0ad43b8f9794d00e166f7e35d984d127b9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-18 14:38:24.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8a8268819194300928f20cfa96c3b0e\n\n- grenade https://zuul.opendev.org/t/openstack/build/89e3042a5b4f4437b9414989b400ea47 : SUCCESS in 47m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/516c87bf01ea4aa7a83259b576dcd7e6 : SUCCESS in 1h 22m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/84397d2a9bf14df2a99b09c3c6014712 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ec47555951640b3aaaa383bc8094de2 : SUCCESS in 16m 04s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/db7c7de267334065925dece1d8b996e4 : FAILURE in 1h 17m 49s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/a6dcb58528f24ec99399169a9f8ae16a : FAILURE in 1h 03m 05s","accounts_in_message":[],"_revision_number":6},{"id":"6fb252b2e63640b091243babf1594e9c8571b58b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-04-26 15:43:04.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"feec7a62ca1cba73dee9d321a069eb04cd94c14e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 16:57:33.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df457c8480b94473b388601d9c966572\n\n- grenade https://zuul.opendev.org/t/openstack/build/837c7d83eb914f84a1d9750fefbd1b06 : SUCCESS in 1h 03m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4096c68db60e467dbd19dd0ddf2e85ae : SUCCESS in 1h 08m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7330d50459b94d7ea472d63987297355 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22124fd08e774700a7bc5e01dcb233cd : SUCCESS in 12m 20s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/291d992887744e13ba7bdafec450ae95 : FAILURE in 1h 09m 20s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/c54a4bb5f9134d888d970cecf8a3a954 : FAILURE in 1h 06m 08s","accounts_in_message":[],"_revision_number":7},{"id":"814ecbc1ea18b5a928c89e9b434d1804020f2679","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-04-27 07:09:36.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"5c7bb25b4eaf6b19fa753a65f594cfde34e7424d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-27 08:32:47.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d15b04d33d240219b6a9e9b38b244c2\n\n- grenade https://zuul.opendev.org/t/openstack/build/c3825e96ee5c4242a7d5ccd74666d6fd : SUCCESS in 1h 17m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8574e5d23b4e4412adf48fca5b46e970 : SUCCESS in 1h 11m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2380d3977594dc58e559861caf84bd6 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57ae975c047d4fd2826a8dab9ca5f910 : SUCCESS in 11m 28s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/3f06b6b040bb44af9ce3d5b382cefafc : FAILURE in 1h 04m 55s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/89852605af5747609f9b654b521c5ae2 : FAILURE in 1h 03m 26s","accounts_in_message":[],"_revision_number":8},{"id":"7273a31e795d0f409c37c0f6318ae7603d730451","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-04-28 15:21:35.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"18fcee48c7d0b2d5da65fef85bd6ac2ec74cabed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-28 17:39:37.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a953f7d683e4fdb9f85468c86f5d5db\n\n- grenade https://zuul.opendev.org/t/openstack/build/446b0e2fe1764cf7a3e65e0165a3a6fc : SUCCESS in 57m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e3d3c52474de4120b5413ad493aff90c : FAILURE in 1h 06m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6b7daec6ee546fc87a860adfad9e281 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f95f8eda1cc54c4fb51c3170b2bb7282 : SUCCESS in 12m 45s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/e388e2f625644bc8b4ced0452a89f9c1 : FAILURE in 1h 18m 32s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/60f237c066f6463d99f214036272e6f8 : FAILURE in 1h 14m 42s","accounts_in_message":[],"_revision_number":9},{"id":"db56617cebe88ec17a3de23bf90e8cec3bed3bb8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-05-02 10:03:04.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"8dcae9ee7010b5a2c86c639f364ca6dedd2f0b6b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-02 11:16:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/851d2fc664304cb89abfb25308c52d39\n\n- grenade https://zuul.opendev.org/t/openstack/build/fe75a2cb962e4020a534688e033a8fd2 : SUCCESS in 1h 00m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/00fc3d15a3f64e30bede43a59f7a5519 : FAILURE in 1h 11m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08b4e8c4477c4b56959c63875af87d22 : SUCCESS in 4m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f0a09719d544b7bf8859bedab7a04f : SUCCESS in 12m 45s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/52972884d42a4b14be5e976378ffb442 : FAILURE in 1h 07m 53s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/74d8fb297ba446a19ac4a6a96ea86437 : FAILURE in 47m 20s","accounts_in_message":[],"_revision_number":10},{"id":"b0c2f86813a2ce4b651a0e08f34cd61d178da383","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2023-05-04 14:32:08.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e2fce72d81d6b298652a55bb53df789422b5ee5a","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-05-04 14:47:43.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"d8b4847e5739493fabaf41bd00004b4390b3c7ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-04 17:48:40.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e8a2ea7b7f7d42c0aefec1938a380d53\n\n- grenade https://zuul.opendev.org/t/openstack/build/c33f18d591744b3cb610202259d91d97 : SUCCESS in 48m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4d7192ddf2f41198bcc312c9a950f1e : FAILURE in 1h 41m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/415c9d42f0874d9e8368f3cd3dadd0a7 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0e8b00f21ed4a3790d31c98bb79959d : SUCCESS in 16m 11s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/0df90d851e87481b8f1ac48e6aa89bf0 : FAILURE in 1h 05m 17s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/246ba5a9412047368b4d57eff3590961 : FAILURE in 1h 14m 39s","accounts_in_message":[],"_revision_number":10},{"id":"71633fa882be8b2387788076a0630c35255090a1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-05-23 11:33:37.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"bb25bc5ac4c059618f863c196e9391deaa0f687d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-23 13:14:24.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1871e84915f94e039697585cad54163b\n\n- grenade https://zuul.opendev.org/t/openstack/build/fca55e2bed2e4fa4817881091c3edfed : SUCCESS in 1h 13m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50e9e91abec240a2a7cd7b1d1310adf4 : SUCCESS in 1h 35m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10d3f76bd1844f52bfdda6d54dda10a7 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b413c50c22e4a0ca5680569f9895975 : SUCCESS in 9m 52s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/da39a588af5c4a9c86c2783ad295c31e : FAILURE in 1h 08m 09s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/ef4c1d0780144a38993a5b6feef072f7 : FAILURE in 1h 04m 46s","accounts_in_message":[],"_revision_number":11},{"id":"eea1b2c46b849618688108241c67788ef5462859","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-05-25 15:17:23.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"d25e641dd5c96c1d565e6cfc08820906a8e4a26f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 17:33:11.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/595d7a746a144e7987a8264f1c713e08\n\n- grenade https://zuul.opendev.org/t/openstack/build/5bd479cd271345cda4ff821ee8a39a35 : FAILURE in 32m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d44e25a0b09e4e039e67fdd16d89e38d : FAILURE in 1h 33m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2b1d80f218743439fc9319bdb473e66 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3bfc6265006145dfb23992cfd7c5cdf4 : SUCCESS in 12m 57s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/27ea8fa3563d49eebbc19360e48e1cfe : FAILURE in 1h 12m 52s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/47837271d9ad4933b5e1e40550d2a423 : FAILURE in 1h 19m 59s","accounts_in_message":[],"_revision_number":12},{"id":"377b2398c3d1332102a05a2326b7b3f6cc6c8b73","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-05-25 19:52:56.000000000","message":"Patch Set 13: Patch Set 12 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"8b12daaf27633ec836d3ba08d3b5ff937480608d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-25 21:49:26.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c4898a16abf492fb50c203274a5fba3\n\n- grenade https://zuul.opendev.org/t/openstack/build/a8123dadb71d4fbaa9c75b5ef5aad00c : SUCCESS in 57m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e2c0fd22535f4a7cb99fcec2b452e6cc : SUCCESS in 1h 50m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/460c450e92174a99a1264677af40ce3d : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa6b771967254b69a6d4958a09bae08f : SUCCESS in 15m 00s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/982332db593f4f88a9636b470ea5967d : FAILURE in 1h 04m 47s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/70fe77112be5462dabc6773cc1842fdd : FAILURE in 1h 21m 51s","accounts_in_message":[],"_revision_number":13},{"id":"6b89c2cc373bbe354da862cea8a663f4ef99caec","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-05-26 07:08:23.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"ab2b7a3c42a5258b656be8bd9d8b2406cfc88ebc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-26 08:28:00.000000000","message":"Patch Set 13:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f1367eb58a549c2bd548f1d7c1722bb\n\n- grenade https://zuul.opendev.org/t/openstack/build/80561ccc2d194ed18c0969b6c28157b5 : SUCCESS in 49m 46s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d3f41d79d9e418fb4165aac21985eb8 : SUCCESS in 1h 17m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40ca2f3dfe00400eb9c207ab384b79a0 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a317614259804bdca71cd90c56ccd658 : SUCCESS in 12m 13s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1149db7a6c55429781288790f80df101 : FAILURE in 1h 09m 37s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/5d2abdc9f47e42fca5132892633516e3 : FAILURE in 48m 35s","accounts_in_message":[],"_revision_number":13},{"id":"e2596fc494f141b7a79374534fd8e5ec52ea41e8","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"date":"2023-06-05 08:07:25.000000000","message":"Patch Set 14: Patch Set 13 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"86c23da8d0f50a9c0fe40265a01e0bc0a6f6889d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-05 09:54:03.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b91957209bc45769bb34190973eddb7\n\n- grenade https://zuul.opendev.org/t/openstack/build/ca05e9fa16a34f98bcde2dd798057ac2 : FAILURE in 57m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3137672cc1a34bc79ab2dbf87ab355f5 : SUCCESS in 1h 41m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/abb17b37e14c4bbcacdbfcf21a476def : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bae42af547984befac7f1d2e8987cdce : TIMED_OUT in 30m 59s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1bc5911e5de6485bb7a1cf9fc50eb978 : FAILURE in 1h 12m 20s (non-voting)\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/f826b8ae8f994f918664e9cec3941fbf : FAILURE in 1h 01m 16s","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"36dee00d4f32a0f93d11b61b54762df2939149d4","revisions":{"2517095a8a41b5ed54cc7d0120b27444e99cf844":{"kind":"REWORK","_number":1,"created":"2023-04-17 13:30:51.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/1"}}},"commit":{"parents":[{"commit":"3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab","subject":"Merge \"Add project reader to account-generator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:30:06.000000000","tz":0},"subject":"[WIP] Add cleanup for create_containers()","message":"[WIP] Add cleanup for create_containers()\n\nRelated-Bug: #1996624\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2517095a8a41b5ed54cc7d0120b27444e99cf844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2517095a8a41b5ed54cc7d0120b27444e99cf844"}]},"branch":"refs/heads/master"},"319a3b39452bcec85a67e3b58b0fd70a35e4b8af":{"kind":"REWORK","_number":2,"created":"2023-04-17 13:35:45.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/2"}}},"commit":{"parents":[{"commit":"3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab","subject":"Merge \"Add project reader to account-generator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:34:46.000000000","tz":0},"subject":"[WIP] Add cleanup for create_containers()","message":"[WIP] Add cleanup for create_containers()\n\nRelated-Bug: #1996624\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/319a3b39452bcec85a67e3b58b0fd70a35e4b8af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/319a3b39452bcec85a67e3b58b0fd70a35e4b8af"}]},"branch":"refs/heads/master"},"69c5e3854171d377c7cbea978094e3473d157b09":{"kind":"REWORK","_number":3,"created":"2023-04-17 13:37:54.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/3"}}},"commit":{"parents":[{"commit":"3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab","subject":"Merge \"Add project reader to account-generator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:37:02.000000000","tz":0},"subject":"[WIP] Add cleanup for create_containers()","message":"[WIP] Add cleanup for create_containers()\n\nRelated-Bug: #1996624\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/69c5e3854171d377c7cbea978094e3473d157b09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/69c5e3854171d377c7cbea978094e3473d157b09"}]},"branch":"refs/heads/master"},"bf4b8591ea8e342c363854b9365120f0ecdf0da3":{"kind":"REWORK","_number":4,"created":"2023-04-17 15:51:07.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/4"}}},"commit":{"parents":[{"commit":"3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab","subject":"Merge \"Add project reader to account-generator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 15:49:51.000000000","tz":0},"subject":"[WIP] Add cleanup for create_containers()","message":"[WIP] Add cleanup for create_containers()\n\nRelated-Bug: #1996624\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bf4b8591ea8e342c363854b9365120f0ecdf0da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bf4b8591ea8e342c363854b9365120f0ecdf0da3"}]},"branch":"refs/heads/master"},"1485a09d31bae4384549c8a860e801604f5ab072":{"kind":"REWORK","_number":5,"created":"2023-04-18 08:13:24.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/5"}}},"commit":{"parents":[{"commit":"3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab","subject":"Merge \"Add project reader to account-generator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-18 08:12:06.000000000","tz":120},"subject":"[WIP] Add cleanup for create_containers()","message":"[WIP] Add cleanup for create_containers()\n\nRelated-Bug: #1996624\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1485a09d31bae4384549c8a860e801604f5ab072"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1485a09d31bae4384549c8a860e801604f5ab072"}]},"branch":"refs/heads/master"},"8cef57f10541b207a75eabf270a658a58a3663fd":{"kind":"REWORK","_number":6,"created":"2023-04-18 13:13:05.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/6"}}},"commit":{"parents":[{"commit":"3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab","subject":"Merge \"Add project reader to account-generator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-18 13:12:37.000000000","tz":120},"subject":"[WIP] Add cleanup for create_containers()","message":"[WIP] Add cleanup for create_containers()\n\nRelated-Bug: #1996624\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8cef57f10541b207a75eabf270a658a58a3663fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8cef57f10541b207a75eabf270a658a58a3663fd"}]},"branch":"refs/heads/master"},"84fd692fa0a4e09fc96e2acdd58618168ace551b":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2023-04-26 15:43:04.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/7"}}},"commit":{"parents":[{"commit":"3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab","subject":"Merge \"Add project reader to account-generator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-26 14:24:59.000000000","tz":0},"subject":"[WIP] Catching broken tests in tempest-full-test-account-* jobs","message":"[WIP] Catching broken tests in tempest-full-test-account-* jobs\n\nRelated-Bug: #1996624\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/84fd692fa0a4e09fc96e2acdd58618168ace551b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/84fd692fa0a4e09fc96e2acdd58618168ace551b"}]},"branch":"refs/heads/master","description":"Edit commit message"},"af15f16bf1d508418e5641c28a7feae4b5f7fde4":{"kind":"REWORK","_number":8,"created":"2023-04-27 07:09:36.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/8"}}},"commit":{"parents":[{"commit":"3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab","subject":"Merge \"Add project reader to account-generator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-27 07:09:01.000000000","tz":120},"subject":"[WIP] Catching broken tests in tempest-full-test-account-* jobs","message":"[WIP] Catching broken tests in tempest-full-test-account-* jobs\n\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/af15f16bf1d508418e5641c28a7feae4b5f7fde4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/af15f16bf1d508418e5641c28a7feae4b5f7fde4"}]},"branch":"refs/heads/master"},"da025baed26ace3fba145f491fbc3f92fc0cb6d1":{"kind":"REWORK","_number":9,"created":"2023-04-28 15:21:35.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/9"}}},"commit":{"parents":[{"commit":"3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab","subject":"Merge \"Add project reader to account-generator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-28 15:21:26.000000000","tz":120},"subject":"[WIP] Catching broken tests in tempest-full-test-account-* jobs","message":"[WIP] Catching broken tests in tempest-full-test-account-* jobs\n\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/da025baed26ace3fba145f491fbc3f92fc0cb6d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/da025baed26ace3fba145f491fbc3f92fc0cb6d1"}]},"branch":"refs/heads/master"},"c06ffdf6b6ddf6b875548169e51afc101fd6be61":{"kind":"REWORK","_number":10,"created":"2023-05-02 10:03:04.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/10"}}},"commit":{"parents":[{"commit":"3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab","subject":"Merge \"Add project reader to account-generator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-05-02 10:02:53.000000000","tz":120},"subject":"[WIP] Catching broken tests in tempest-full-test-account-* jobs","message":"[WIP] Catching broken tests in tempest-full-test-account-* jobs\n\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c06ffdf6b6ddf6b875548169e51afc101fd6be61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c06ffdf6b6ddf6b875548169e51afc101fd6be61"}]},"branch":"refs/heads/master"},"4a3cb978bc63a5e18a359940107cef27bda53366":{"kind":"REWORK","_number":11,"created":"2023-05-23 11:33:37.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/11"}}},"commit":{"parents":[{"commit":"3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab","subject":"Merge \"Add project reader to account-generator\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3c7eebaaf35c9e8a3f00c76cd1741457bdec9fab"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukáš Piwowarski","email":"lpiwowar@redhat.com","date":"2023-05-23 11:33:28.000000000","tz":120},"subject":"[WIP] Catching broken tests in tempest-full-test-account-* jobs","message":"[WIP] Catching broken tests in tempest-full-test-account-* jobs\n\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4a3cb978bc63a5e18a359940107cef27bda53366"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4a3cb978bc63a5e18a359940107cef27bda53366"}]},"branch":"refs/heads/master"},"1727efd581f28578c808f175e05570fed691bde1":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2023-05-25 15:17:23.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/12"}}},"commit":{"parents":[{"commit":"347376aa98b2d542ccfbce6f9352c6799055e93a","subject":"Fix cleanup of resources for object_storage tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/347376aa98b2d542ccfbce6f9352c6799055e93a"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-05-25 15:17:23.000000000","tz":0},"subject":"[WIP] Catching broken tests in tempest-full-test-account-* jobs","message":"[WIP] Catching broken tests in tempest-full-test-account-* jobs\n\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1727efd581f28578c808f175e05570fed691bde1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1727efd581f28578c808f175e05570fed691bde1"}]},"branch":"refs/heads/master","description":"Rebase"},"0c2f07d1aa73e6af1bded4322e45a323e09b29ee":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2023-05-25 19:52:56.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/13"}}},"commit":{"parents":[{"commit":"8597ed2de39cc4b9dd5aea6e6af4b94c8283c3a0","subject":"Fix cleanup of resources for object_storage tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8597ed2de39cc4b9dd5aea6e6af4b94c8283c3a0"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-05-25 19:52:56.000000000","tz":0},"subject":"[WIP] Catching broken tests in tempest-full-test-account-* jobs","message":"[WIP] Catching broken tests in tempest-full-test-account-* jobs\n\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0c2f07d1aa73e6af1bded4322e45a323e09b29ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0c2f07d1aa73e6af1bded4322e45a323e09b29ee"}]},"branch":"refs/heads/master","description":"Rebase"},"36dee00d4f32a0f93d11b61b54762df2939149d4":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2023-06-05 08:07:25.000000000","uploader":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"ref":"refs/changes/30/880630/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/880630/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/880630/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/880630/14"}}},"commit":{"parents":[{"commit":"037ea4efe07cbc78aa81388dccc06eed21874d4b","subject":"Enable Nova, Glance RBAC old defaults in pre provisioned account testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/037ea4efe07cbc78aa81388dccc06eed21874d4b"}]}],"author":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-04-17 13:16:36.000000000","tz":0},"committer":{"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","date":"2023-06-05 08:07:25.000000000","tz":0},"subject":"[WIP] Catching broken tests in tempest-full-test-account-* jobs","message":"[WIP] Catching broken tests in tempest-full-test-account-* jobs\n\nChange-Id: Idfdbe3dc276fad53c50421f5e4ec32972e052ebc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/36dee00d4f32a0f93d11b61b54762df2939149d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/36dee00d4f32a0f93d11b61b54762df2939149d4"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
