)]}'
{"id":"openstack%2Ftempest~881536","triplet_id":"openstack%2Ftempest~master~Ieca98a0df1179db4d9c95435178bacfe41cc3cdb","project":"openstack/tempest","branch":"master","hashtags":[],"change_id":"Ieca98a0df1179db4d9c95435178bacfe41cc3cdb","subject":"DNM testing out","status":"ABANDONED","created":"2023-04-26 07:25:55.000000000","updated":"2023-07-20 22:39:22.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"efee8a49f424531b66a4b74361634088810c952a","_number":881536,"virtual_id_number":881536,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-04-26 10:00:32.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-04-26 09:22:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2023-04-26 10:00:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"60a76d99e1af8a2e1196153a29c12533e98445dd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-04-26 07:25:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b1c7e9589515aa3c3068b0355efb1835e712c147","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2023-04-26 07:26:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ad155bb43e1d2412957dd6399f5f4d6c11b15584","tag":"autogenerated:zuul:experimental","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 09:22:43.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4aff4d4fcf684c36a016482086326b1b\n\n- nova-multi-cell https://zuul.opendev.org/t/openstack/build/8e23c04a2f2e4289bc50bf47d6a47ffd : SUCCESS in 1h 35m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c9df0d40a3cd46ef813855e953d19213 : SUCCESS in 1h 06m 29s\n- tempest-with-latest-microversion https://zuul.opendev.org/t/openstack/build/db3efa84e0024f11b6a7439d4c9c0479 : FAILURE in 1h 39m 50s\n- tempest-stestr-master https://zuul.opendev.org/t/openstack/build/0a9f75389dd44296b524711918de9e0d : SUCCESS in 25m 10s\n- tempest-cinder-v2-api https://zuul.opendev.org/t/openstack/build/e6d5bbd70bd049218146a92a947ca389 : SUCCESS in 41m 58s\n- tempest-all https://zuul.opendev.org/t/openstack/build/2b1c9896adc541ab8c355ec94502b103 : SUCCESS in 1h 43m 25s\n- tempest-slow-parallel https://zuul.opendev.org/t/openstack/build/f36ed52ee3554487abf061d6d16eabc9 : SUCCESS in 1h 41m 07s\n- tempest-full-parallel https://zuul.opendev.org/t/openstack/build/bbfa070b144247ef8fbfd771fc09f83b : SUCCESS in 1h 47m 01s\n- tempest-full-zed-extra-tests https://zuul.opendev.org/t/openstack/build/3d532747d60746888b08ea853eaae4ea : SUCCESS in 51m 58s\n- tempest-full-yoga-extra-tests https://zuul.opendev.org/t/openstack/build/0a97b578e87d4985a4bf610b9f30d4e6 : SUCCESS in 42m 50s\n- tempest-full-xena-extra-tests https://zuul.opendev.org/t/openstack/build/b7accd3e43ab4b379f849235849d2111 : SUCCESS in 39m 39s\n- tempest-full-enforce-scope-new-defaults-zed https://zuul.opendev.org/t/openstack/build/cbda458ac72049c89e2cbd14cc2b0f1d : SUCCESS in 1h 27m 56s\n- neutron-ovs-tempest-dvr-ha-multinode-full https://zuul.opendev.org/t/openstack/build/289ec8e9f598465e97fa4fef179c2b04 : SUCCESS in 1h 05m 46s\n- nova-tempest-v2-api https://zuul.opendev.org/t/openstack/build/68796d24038645dbb55d5a9ef5f18485 : FAILURE in 1h 39m 07s\n- cinder-tempest-lvm-multibackend https://zuul.opendev.org/t/openstack/build/4b1dbca6b73846c5828c50826ffd5538 : SUCCESS in 1h 50m 24s\n- tempest-pg-full https://zuul.opendev.org/t/openstack/build/f8d1580edaa0487b8776894ad5d8ec25 : SUCCESS in 1h 43m 57s\n- tempest-full-py3-ipv6 https://zuul.opendev.org/t/openstack/build/45c93b756d9641a2b7bad660f7175b12 : SUCCESS in 1h 45m 14s\n- tempest-full-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e138e70dbb44ea6a009fb11d3303a23 : SUCCESS in 1h 49m 46s\n- tempest-centos9-stream-fips https://zuul.opendev.org/t/openstack/build/cf8490601a924de88952c9726c04d0d3 : SUCCESS in 1h 34m 33s\n- tempest-full-test-account-no-admin-py3 https://zuul.opendev.org/t/openstack/build/f5e0f230c6a34967a269bd4656d464b8 : FAILURE in 1h 28m 23s","accounts_in_message":[],"_revision_number":1},{"id":"627f73bff2c5de0272e5302188c0bdc08775899b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-04-26 10:00:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90532a34b080430cbdcc20472304d397\n\n- grenade https://zuul.opendev.org/t/openstack/build/8fd5a3d025f14360a1236a02907ca483 : SUCCESS in 1h 08m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4aa9d8171c7f4feda894abd17b25ebfe : SUCCESS in 1h 29m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e45b0e03e3894c8d9d1e8c81fae96229 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9777f36622843a7b1a7a2d80572b752 : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83ec61b3cb7e43ec81fc5d624991fdfe : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a185073d92f9411aaedb597c68ec8b47 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aed96b7cbe8945b68527f8fb8bcbdf9b : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ff55e6df0654ddeb40f1f44deae69d2 : SUCCESS in 4m 06s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/97db06e26c8f40a3b0c2dcec4500e47a : SUCCESS in 1h 00m 24s\n- tempest-full-ubuntu-focal https://zuul.opendev.org/t/openstack/build/17590867733d479eb1e540e930a3292c : SUCCESS in 1h 36m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f56a08f207644366b8b7a6ab94d55029 : FAILURE in 1h 15m 00s (non-voting)\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/772ee139fecd4b3aa6fca9db3e850c69 : SUCCESS in 1h 14m 13s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/3d22f4b08e674b25adefc5d6efab8ca9 : SUCCESS in 1h 13m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e3054481d87742b6916980c2e6058721 : SUCCESS in 1h 03m 46s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/13804611fa694ede87340481ebac3348 : SUCCESS in 27m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/192995dceb2f463ab95194a1c0d17de8 : SUCCESS in 51m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/b86dff6b843d4b05ba388e4e5cbd1719 : SUCCESS in 2h 32m 46s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ca9baaef6a1447aca2fb5724af4bac1c : SUCCESS in 54m 26s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f0bbeea5d0a1457080c2ae2e0ded3b16 : SUCCESS in 1h 29m 25s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5d88125f203341aa92ba1db24806efa8 : SUCCESS in 1h 09m 22s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c4b784c467a54ede9c1dbba228053113 : SUCCESS in 54m 02s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/1534c8973c664af39a249f9488f7d552 : SUCCESS in 1h 19m 23s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/395951e96c0445e6b5d3ce13b8dc62c2 : SUCCESS in 3m 45s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ad002f55591a49f1b1a37d2bff734fc6 : FAILURE in 1h 32m 09s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/c1f3956e34624e45a10742489777577d : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":1},{"id":"efee8a49f424531b66a4b74361634088810c952a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-07-20 22:39:22.000000000","message":"Abandoned\n\ntesting done, depends-on merged.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5e926bc3bbb70a0bdcea7416ced38be527aaf9dc","revisions":{"5e926bc3bbb70a0bdcea7416ced38be527aaf9dc":{"kind":"REWORK","_number":1,"created":"2023-04-26 07:25:55.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/36/881536/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/36/881536/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/36/881536/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/36/881536/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/36/881536/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/36/881536/1"}}},"commit":{"parents":[{"commit":"e64f4c2b07e048ea77d08a29b7afbd17448abc32","subject":"Simplify definition of fips jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e64f4c2b07e048ea77d08a29b7afbd17448abc32"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-04-26 07:24:50.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2023-04-26 07:24:50.000000000","tz":120},"subject":"DNM testing out","message":"DNM testing out\n\nTesting out the cirros version bump to 0.6.1\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/881437/\nChange-Id: Ieca98a0df1179db4d9c95435178bacfe41cc3cdb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5e926bc3bbb70a0bdcea7416ced38be527aaf9dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5e926bc3bbb70a0bdcea7416ced38be527aaf9dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
