)]}'
{"id":"openstack%2Ftempest~883558","triplet_id":"openstack%2Ftempest~master~I7fd475c0cb92968f8dcac69e79c01aba155fa412","project":"openstack/tempest","branch":"master","topic":"secure-rbac","hashtags":[],"change_id":"I7fd475c0cb92968f8dcac69e79c01aba155fa412","subject":"DNM: testing nova new defaults setting change in devstack","status":"ABANDONED","created":"2023-05-18 18:05:43.000000000","updated":"2023-05-19 00:31:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d530f73b5d9662dea68901a73996d048f3280eaa","_number":883558,"virtual_id_number":883558,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-05-18 21:00:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-18 21:00:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d9e5c3c0dca4c864534ced2b40395b1f39e64ed8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-18 18:05:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6a1fab5c95508827dd38ba6388560f09e27eb3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-18 21:00:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/178ecffa388243e895d071e4c06897da\n\n- grenade https://zuul.opendev.org/t/openstack/build/3eadf2b427a04d818d646a432899f4e6 : SUCCESS in 1h 02m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dd504e27184a4a44b69a099f1939fd41 : SUCCESS in 1h 35m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc0ba5ced0ae42ce9e03c9cb506df526 : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b08d38116f874f648d45cd428b01f9f6 : SUCCESS in 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/597510c25a1e4224be928d96a1bdcc49 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf3c86d912904cbd89ba1d2be1320862 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40e9541713d746ffa559152aeb110a83 : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b643cdf731c54dc096a328bcaf958dbe : SUCCESS in 4m 13s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/8a298a6dc86845cba58220cee1ad9c32 : SUCCESS in 1h 23m 28s\n- tempest-full-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4b37abbfa6be4c32a2e387ec81105fff : SUCCESS in 1h 19m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/056b4168a10f43a4ab91c781f10cfc7b : TIMED_OUT in 2h 05m 49s (non-voting)\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/c652da0377fb44fb90af17fdba4026e1 : SUCCESS in 1h 33m 18s\n- tempest-full-xena https://zuul.opendev.org/t/openstack/build/5e7e75c6efce4ec9a75c8520f3950446 : SUCCESS in 1h 11m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/4091d6c68e44406e81e8c704d9430354 : SUCCESS in 1h 03m 57s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a17d42deb5ac48db8c350f0e3f416e9a : SUCCESS in 34m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c9838c5675d54dbd8564f2a62eb7a407 : SUCCESS in 55m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/74e68e7e02c5422096f12b33471a1833 : SUCCESS in 2h 43m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ec45c7c155ae435cac66d4116c8c1703 : SUCCESS in 38m 41s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ea126fa83f464eeba4192c0bf6e0d210 : SUCCESS in 1h 33m 01s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/cc87bb16bbbe4fe89871d5c968634fb0 : SUCCESS in 1h 34m 02s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c230370506584ef192be441f66d523c7 : SUCCESS in 50m 02s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2d90d563f4724a56bd9ef83bc0d9a414 : SUCCESS in 1h 00m 51s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/7cefe26ec18c4145aa4ffe7c083be9b8 : SUCCESS in 3m 49s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ec025f3b2f6b413cac76465d81a74959 : FAILURE in 1h 04m 01s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5fa2f37271c341cc9e970ab993f99f3d : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":1},{"id":"d530f73b5d9662dea68901a73996d048f3280eaa","tag":"autogenerated:gerrit:abandon","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-05-19 00:31:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"73863499490934cb99d9bba89b41e343b7bd223a","revisions":{"73863499490934cb99d9bba89b41e343b7bd223a":{"kind":"REWORK","_number":1,"created":"2023-05-18 18:05:43.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/58/883558/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/58/883558/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/883558/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/883558/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/58/883558/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/58/883558/1"}}},"commit":{"parents":[{"commit":"4054e13e65f7d11c9097946a36976bf9c6edea45","subject":"Create releasenotes page for tempest 34.2.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4054e13e65f7d11c9097946a36976bf9c6edea45"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-05-18 18:05:05.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2023-05-18 18:05:05.000000000","tz":-300},"subject":"DNM: testing nova new defaults setting change in devstack","message":"DNM: testing nova new defaults setting change in devstack\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/883556\nChange-Id: I7fd475c0cb92968f8dcac69e79c01aba155fa412\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/73863499490934cb99d9bba89b41e343b7bd223a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/73863499490934cb99d9bba89b41e343b7bd223a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
