)]}'
{"id":"openstack%2Ftempest~886991","triplet_id":"openstack%2Ftempest~master~I6693421ef569f8a977d8a3fdb52dbfdf33ebfeeb","project":"openstack/tempest","branch":"master","topic":"bug/2024859","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2023-08-02 16:49:08.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I6693421ef569f8a977d8a3fdb52dbfdf33ebfeeb","subject":"Disable immediate discard of unused blocks in make_fs","status":"ABANDONED","created":"2023-06-26 16:45:55.000000000","updated":"2023-08-02 16:49:08.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"14992d12be0cddfcf97c7a996623ceeaf09ec07f","_number":886991,"virtual_id_number":886991,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-07-25 11:35:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-06-26 19:58:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c08333cc481b28f255e9c2f73272506cc9fa2ee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2023-06-26 16:45:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a2d1cfb7760bb1fe98a2b79c33f7b1b2a5fc125","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-06-26 19:58:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6392f691350c4874b00279af9ee84043\n\n- grenade https://zuul.opendev.org/t/openstack/build/6f8a9738ade144ef9e9fb8a3c0f09c8f : SUCCESS in 1h 05m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4964f918c4284b388555b98ee50ebea6 : TIMED_OUT in 2h 04m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/149c4555e97d4001a7de83f583f7c7c5 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f9592e259844c0f83052070c22382b8 : SUCCESS in 19m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d2be53df4254e4aa30160af0b01fe79 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/291781f46b6541a5b76518298fc091a8 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/028dfa6f533d4b89984e90d61824167f : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c723a405bc1e402e9de8d4cb7f156678 : SUCCESS in 4m 47s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/fed7f6e9d49649bdb38de958b8d1273e : SUCCESS in 59m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/769d90a532d44d8aaf31a151a14e605a : FAILURE in 1h 53m 14s (non-voting)\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/0c217d47d2fb4790a090e79067f748bc : SUCCESS in 1h 39m 45s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/48cd79cab831449faaf91fe3b996d9f8 : SUCCESS in 1h 53m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/242e67a5cbf8427cb186d5e1fc5d807e : SUCCESS in 1h 19m 24s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/70f34afbac4544ae8f69fb3e060e73c1 : SUCCESS in 29m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a3aef66c95e7428eabca0110e9541159 : SUCCESS in 1h 03m 49s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/be61b5fcff3042b196bbffd2e8dac980 : TIMED_OUT in 3h 03m 19s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/1e4994774a924ff59fb163dcca4b61d0 : SUCCESS in 57m 17s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/db713503eb594ccca86d803f10ff3899 : SUCCESS in 1h 44m 51s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/7247589f79de4bc499595fc21f2059f0 : TIMED_OUT in 2h 03m 01s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/22726240b1174f329140229cadc5839f : SUCCESS in 1h 08m 46s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/0b5f61fd897c4ab2af2c7111fee634f4 : SUCCESS in 1h 28m 19s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/03eec74dae184e9cbb4ba6e013800790 : SUCCESS in 4m 31s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5fde11437fdf40b08b7ee07fbf86161f : FAILURE in 1h 29m 52s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5de431d85f3c45aab95a6909455019f8 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6e3651020ca58a3648f4842da4ded2f09d992204","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2023-06-26 20:09:22.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b1ca2362421989af808b2d10383fdf1b89f4a69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2023-07-25 09:08:01.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"997fe706fc4c414c34c2de6757d63c7e1400699f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-07-25 11:35:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0a812bdcb3448b8912e59da18e6a17a\n\n- grenade https://zuul.opendev.org/t/openstack/build/427ee6a7fdef44898f7abcd6f9108dd4 : SUCCESS in 59m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5fa2faeb074d4974afa1d43ba706617f : SUCCESS in 1h 34m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29ecb0f11f6b4059ad996c2e18b953ee : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0dd94dc2b54ff2a0efb913f8811123 : SUCCESS in 13m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7e897eea97c4304915c38954d900c84 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/915dc8602ac3485c8041e082ab4441d4 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76d347582f83477ab0c3b177cb06b6bb : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bc4497a817347f2bbeb807b8dbdbca4 : SUCCESS in 4m 01s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/8ab1991d62d74a6e9fe23c049e3f6b91 : SUCCESS in 51m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/948b5ccb46da4ac7a74c38e37c5c58a7 : TIMED_OUT in 2h 06m 21s (non-voting)\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/c9b0cd6d8bc743c9b08fd931263510d5 : SUCCESS in 1h 22m 50s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/a2e52d5ac5e0450887461c09f5ed315b : SUCCESS in 1h 22m 28s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1c392654382d414cb3a7c42adde894f5 : SUCCESS in 1h 20m 59s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ff5d3da29aac479795f50f7f360fc60c : SUCCESS in 42m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dae315f5e2e4fe9aa8514a5968aaa11 : SUCCESS in 1h 03m 01s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2d38d3fd65e0444b90817e1c55b06077 : SUCCESS in 1h 36m 40s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c0cf2280515341b9b4ab32255b0de566 : SUCCESS in 1h 04m 38s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1cbed41aaaae49a2bf7263ca057aa93c : SUCCESS in 2h 02m 18s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/35d2e9f3d5824653b71aa58c0aa8571b : TIMED_OUT in 2h 03m 12s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c0ae2d7356724c53ac630aa585cd1b54 : SUCCESS in 54m 49s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/bf2ebd7b78884076b4849f0ce2361a2f : SUCCESS in 1h 15m 42s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/a88ec8bfa98d43a59912ace66421ed40 : SUCCESS in 4m 04s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7d64a58ff0774f25b9460120bd846c59 : FAILURE in 1h 27m 38s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/96d236bcccdb4ac386f460bab7e1b1cb : SUCCESS in 3m 05s","accounts_in_message":[],"_revision_number":2},{"id":"14992d12be0cddfcf97c7a996623ceeaf09ec07f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2023-08-02 16:49:08.000000000","message":"Abandoned\n\nPS1 itself failed CI with the mke2fs bug https://launchpad.net/bugs/2024859, so this doesn\u0027t seem to help anything","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f7e92290e652f4078b0d91ed0e2996f1636b5a47","revisions":{"38ca8781ed57c2715164a98bafb02ddd5e47c772":{"kind":"REWORK","_number":1,"created":"2023-06-26 16:45:55.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/91/886991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/91/886991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/91/886991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/91/886991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/91/886991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/91/886991/1"}}},"commit":{"parents":[{"commit":"2a68b13277e15efd4b4a8e8b399b7af6826f82a8","subject":"Merge \"Add test to rebuild volume backed instance\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/2a68b13277e15efd4b4a8e8b399b7af6826f82a8"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2023-06-26 16:30:01.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2023-06-26 16:42:32.000000000","tz":0},"subject":"Disable immediate discard of unused blocks in make_fs","message":"Disable immediate discard of unused blocks in make_fs\n\nLately in the gate we are seeing job failures while trying to\ncreate_timestamp() on volumes during the make_fs step.\n\nThe mke2fs command is timing out, not returning within the default\ntimeout of 300 seconds and causing the test to fail as a result. By\ndefault, mke2fs discards unused blocks upfront which can potentially\ntake a long time. For our purposes, we don\u0027t need to try to optimize\nwith discard for our short-lived CI test volumes.\n\nThis adds the option \u0027-E nodiscard\u0027 to the command in an effort to\nspeed up creation of the filesystem and avoid timeouts.\n\nCloses-Bug: #2024859\n\nChange-Id: I6693421ef569f8a977d8a3fdb52dbfdf33ebfeeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/38ca8781ed57c2715164a98bafb02ddd5e47c772"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/38ca8781ed57c2715164a98bafb02ddd5e47c772"}]},"branch":"refs/heads/master"},"f7e92290e652f4078b0d91ed0e2996f1636b5a47":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-07-25 09:08:01.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/91/886991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/91/886991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/91/886991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/91/886991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/91/886991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/91/886991/2"}}},"commit":{"parents":[{"commit":"b7c55ad0d8c726df88de947aab417bd9b095240e","subject":"Merge \"Optimize the resource creation and status waiting in scenario tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b7c55ad0d8c726df88de947aab417bd9b095240e"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2023-06-26 16:30:01.000000000","tz":0},"committer":{"name":"sean mooney","email":"smooney@redhat.com","date":"2023-07-25 09:08:01.000000000","tz":0},"subject":"Disable immediate discard of unused blocks in make_fs","message":"Disable immediate discard of unused blocks in make_fs\n\nLately in the gate we are seeing job failures while trying to\ncreate_timestamp() on volumes during the make_fs step.\n\nThe mke2fs command is timing out, not returning within the default\ntimeout of 300 seconds and causing the test to fail as a result. By\ndefault, mke2fs discards unused blocks upfront which can potentially\ntake a long time. For our purposes, we don\u0027t need to try to optimize\nwith discard for our short-lived CI test volumes.\n\nThis adds the option \u0027-E nodiscard\u0027 to the command in an effort to\nspeed up creation of the filesystem and avoid timeouts.\n\nCloses-Bug: #2024859\n\nChange-Id: I6693421ef569f8a977d8a3fdb52dbfdf33ebfeeb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f7e92290e652f4078b0d91ed0e2996f1636b5a47"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f7e92290e652f4078b0d91ed0e2996f1636b5a47"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
