)]}'
{"id":"openstack%2Ftempest~890687","triplet_id":"openstack%2Ftempest~master~I1f13bd11fc7cb521d1b995dbd4a4d9c662221428","project":"openstack/tempest","branch":"master","topic":"skip-tasks","attention_set":{},"removed_from_attention_set":{"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2023-08-08 00:29:16.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2023-08-07 20:50:05.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}}},"hashtags":[],"change_id":"I1f13bd11fc7cb521d1b995dbd4a4d9c662221428","subject":"Skip test_image_tasks_create() for bug 2030527","status":"MERGED","created":"2023-08-07 16:31:40.000000000","updated":"2023-08-08 00:30:31.000000000","submitted":"2023-08-08 00:29:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"890687-skip-tasks","meta_rev_id":"0137c017b7571901fe88cd5d2fc1c7951595d1f5","_number":890687,"virtual_id_number":890687,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-08 00:29:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2023-08-07 16:51:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":1,"date":"2023-08-07 16:51:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":2,"date":"2023-08-07 16:51:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-07 16:46:32.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2023-08-07 17:23:10.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2023-08-07 18:35:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"be0df005ae37de080158678f86cc981278a3258f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-08-07 16:31:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"446e9cd55acc948e396992bbb02eeed7ed1f6ff4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-08-07 16:46:32.000000000","message":"Patch Set 1: Review-Priority+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"87708066fcbbb5618f540e80fc691362b25c5814","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-08-07 16:50:45.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"66d1e4b690c943f72b2c469fa140ea0c6c5e60c7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2023-08-07 16:51:45.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1 Review-Priority+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"219463bfd405a416afcbfb8268e9e2e4ec66b5ea","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-08-07 17:23:39.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c6ffd7d8b757fb2b92c7111e60fe1a3a77e4281e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 18:35:33.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2433ee1b5cd04e9e86dc80944ec5cc90\n\n- grenade https://zuul.opendev.org/t/openstack/build/0ab92ab934e640d19793fff1aaa707e2 : FAILURE in 47m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/91d753f804674f0e8206b6a4fe8bd0ca : SUCCESS in 1h 09m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/981b19d8f24b4545ae4e001d3a5402d6 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d221285b2e14f85bcab9952c1498c6e : SUCCESS in 16m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bda376d89be048dc8ce03a82a8d9b109 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/def9f0f9d20241bbb7dd17b54ad70e95 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0d63786c99e84e5a847cdaf5ef63777a : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3ba3ffe06e644f1ebb6643bd3af13656 : SUCCESS in 4m 43s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/6be5ef6dada949238502532858b1e006 : SUCCESS in 1h 03m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cf9d63bd7e8442a6b767258b7b9c55f4 : FAILURE in 1h 34m 30s (non-voting)\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/fadddac599444e1ab3f5db86fa9ba766 : SUCCESS in 1h 33m 13s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/7e256b82d10644ed9c00cbf9c7f6baa8 : SUCCESS in 1h 37m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/414db7e872d14dea84256a9057c5f546 : SUCCESS in 53m 28s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e41a80f960dc49069afc95e2079a87e2 : SUCCESS in 33m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b93a68574ae44622bc416c9de3284174 : SUCCESS in 1h 02m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1ca9a1d696024b14923cf0011ffe6329 : SUCCESS in 1h 13m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d2f75794af1f494baa6d721ba9968e47 : SUCCESS in 58m 55s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2c90769eb676442e9b2525dea5698337 : SUCCESS in 1h 11m 43s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c7b6dce836e94125908778bf4db12930 : SUCCESS in 1h 36m 10s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8fee0c816d724f2bb920f88bf0761db8 : SUCCESS in 48m 02s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/1f64c3e28c944fe4921591b24bbcf7e7 : SUCCESS in 58m 50s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/2e62f906d102435fbcd86c1dbebc7f42 : SUCCESS in 3m 28s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ed85cb940ac843aa9e3550c64713eacc : FAILURE in 55m 45s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/e82a98dd91f248a99375cc7ff4b378f2 : SUCCESS in 48m 14s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/fae3cfcd0a0b4862b19273a91ce1b697 : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":2},{"id":"f9ecf512326918778420d7e8afc9e3ef9be220fe","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-08-07 18:51:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"deabba75125eb9efebfa67b8d67411825b1e459c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 20:47:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/df961fd93f77476f92b9ff7047ebd30f\n\n- grenade https://zuul.opendev.org/t/openstack/build/8c94a9362d634e6ebc65e10333542fda : SUCCESS in 44m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/27e8f74fff3c43968dc9d7ee20049d5d : SUCCESS in 1h 34m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38df04b942954b23946be9e6e988cce5 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0bf327d7b3a4df395d927aae26a1810 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b293a9de8f7e4b3e8dbf7896782b0da5 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bcec3153910477581433558b17514a7 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9a9afd463eb64bc38496b6cfea2dee9b : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90ff1cf240e543459e9bf4ac53c013c9 : SUCCESS in 3m 56s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/5556fbd7ca0842f9a0b0cff3b9b1150b : SUCCESS in 45m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4a62900ceaee41c2a34b8ed88520fa37 : SUCCESS in 1h 21m 31s (non-voting)\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/bd11035e2a0346d5bcd2c19145ade314 : SUCCESS in 1h 26m 45s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/253e3ca88f0d4d1b972d8d1d5dd91025 : SUCCESS in 49m 02s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5f45c23b4d3840838c8b27608bcb444b : SUCCESS in 1h 15m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3d19a6fd10ac41af8584df6b4a2031b3 : SUCCESS in 24m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b52d8ffc7754485bb569011c7aba66e4 : FAILURE in 1h 21m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/531f49f4579b4b7781623dc85ca1fb24 : FAILURE in 1h 15m 15s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/5c57831696064486951c09dece65e08d : SUCCESS in 38m 23s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0aefa6bae1294134a2dda772076a98c7 : SUCCESS in 1h 04m 44s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d92af3e310eb4382bf37fdf9d5163d36 : SUCCESS in 1h 53m 33s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/374377350257434190cc303229ea7e95 : SUCCESS in 59m 15s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/bcaad40c002344a9bbc73c844c7eeb8d : SUCCESS in 1h 21m 39s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/bddaf4b0ec654000871440c1a3a44a0a : SUCCESS in 3m 16s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/2efb423a9a894f6fa0c2be9974e71ca4 : FAILURE in 1h 48m 24s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/9d73098098644d7e86eb5ad98c6236b1 : SUCCESS in 48m 50s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/97b56908e80e448d82067bde293476f8 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"bab6f8c36710df53b7e7b9aaf40918761016c120","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2023-08-07 20:50:05.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"759eeb3c2cb0ad33cac0c1d2492594448dfb0af7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 22:43:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/368dad2371fa4091bb24d4d5b605721d\n\n- grenade https://zuul.opendev.org/t/openstack/build/a0a073cfe0bd49a0bd3116fb5257c3fc : SUCCESS in 35m 54s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e8cbf2e986154d2fb7e6a78dd1f6e056 : SUCCESS in 53m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cfa87a3fdf543b191e8484040175881 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85760206a8d1430d9192fa250032a8f8 : SUCCESS in 13m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3012e2fe6b44c9da18c5dd51e80503e : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4513d9e1f5d142738e60e5fce765d9d8 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7844d0639e849c9a837bf7b8fc8f44b : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7921fce55ed846c79c2eb2bbc4bf217d : SUCCESS in 4m 01s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/6f178d582b57412db44fcb1fc44aaaae : SUCCESS in 48m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/793aa316a6d84e75973a0cc8b5cfb176 : FAILURE in 1h 29m 56s (non-voting)\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/150db18b24064a1aaeeb94c6f503b926 : SUCCESS in 1h 28m 22s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/f8de43aeddb544e69a601c969f087647 : SUCCESS in 1h 12m 32s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/247f121a3e964aaf9cad23813a7d1243 : SUCCESS in 56m 58s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/853dc5c2fd954b59ab83e3e8d89cc71f : SUCCESS in 29m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/261df233f52f4ab99be7bbc316d7e888 : SUCCESS in 1h 04m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/84dcc82f4dd14796a584193f86d86ff3 : SUCCESS in 1h 10m 02s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d27606f880e94a8eaada45b21878188a : SUCCESS in 50m 00s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0ecef1bc5c174166987b65c197666b38 : SUCCESS in 1h 27m 18s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a5c4c826a5df4edbb128336bb3f0b430 : SUCCESS in 1h 51m 39s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/3ea4f60cd1524674ad2f13007af8c7a2 : SUCCESS in 51m 18s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/d390cfc3c84b4e2586ae8bd5aa03cd8b : SUCCESS in 1h 19m 32s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/34df10326727404a9bff8586e2c69cbb : SUCCESS in 4m 18s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/acf3e139a4084ca5b72bd49ab656ceaa : FAILURE in 1h 10m 57s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/27b03935d47e44239be14a005108d01b : SUCCESS in 1h 04m 43s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/77e9a60fb6a1478ca9b17320409bbef0 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":2},{"id":"2bdfa46e5102a32fbafc8075f3fef3df938c687d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-07 22:43:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"992cbf89cd1b9c734152280c807d30bac5d98fc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 00:29:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/955f9074915a472298d71d0a7805d072\n\n- grenade https://zuul.opendev.org/t/openstack/build/9ff781fe300842af9c6809cf33379341 : SUCCESS in 1h 00m 53s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0180ab37365445a0bed31d2f4d48adf8 : SUCCESS in 1h 25m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b41d901cc43446f9d834c107cfcc8e4 : SUCCESS in 10m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0cc854f8c6d44ccba2195fcdca768c36 : SUCCESS in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/628fae4d5d254107b4d428b434e0ca7e : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de80285f3ed64005829d74f508543e7d : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/264c422db1824b0a9dd8ef95af15d3b0 : SUCCESS in 4m 10s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6d2d21b1a3674ccc8f6227db2cae587c : SUCCESS in 1h 45m 19s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c6e3cf6a21f149bbb03fdf2be5e0ae56 : SUCCESS in 52m 29s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/6740e549e47f435aa4ccf1e189dd8aad : SUCCESS in 1h 04m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d14f71ab0a714c4491d703e535f0adc6 : SUCCESS in 1h 04m 45s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/96486aa56edd4bb7b4c1aa9540f21fda : SUCCESS in 44m 55s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/12362a391071499e842aed81cedb33d3 : SUCCESS in 1h 35m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/82ad10ca78bd44d2ae33f2ea284618ad : SUCCESS in 54m 07s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/043db7a7b78641a3b86ad6894bac17bf : SUCCESS in 53m 20s","accounts_in_message":[],"_revision_number":2},{"id":"7003098b3c9f00f382d76723e02d4499ab1411fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 00:29:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0137c017b7571901fe88cd5d2fc1c7951595d1f5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-08 00:30:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3ab7272cbda4f28b46decfbd77dd398\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9eb7e62b39e4ba58166c432302d5b0f : SUCCESS in 58s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"58bb991974bddec0df5a94535e371df971c47840","revisions":{"0dc55274f9f425822e65beff155fb7043f065205":{"kind":"REWORK","_number":1,"created":"2023-08-07 16:31:40.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/87/890687/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/87/890687/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/87/890687/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/87/890687/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/87/890687/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/87/890687/1"}}},"commit":{"parents":[{"commit":"46b78980eec98ceaade313c559887480a9af9759","subject":"Merge \"Skip scenario tests early to avoid unnecessary setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/46b78980eec98ceaade313c559887480a9af9759"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-08-07 16:31:12.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-08-07 16:31:12.000000000","tz":-420},"subject":"Skip test_image_tasks_create() for bug 2030527","message":"Skip test_image_tasks_create() for bug 2030527\n\nChange-Id: I1f13bd11fc7cb521d1b995dbd4a4d9c662221428\nRelated-Bug: #2030527\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0dc55274f9f425822e65beff155fb7043f065205"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0dc55274f9f425822e65beff155fb7043f065205"}]},"branch":"refs/heads/master"},"58bb991974bddec0df5a94535e371df971c47840":{"kind":"REWORK","_number":2,"created":"2023-08-07 16:50:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/87/890687/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/87/890687/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/87/890687/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/87/890687/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/87/890687/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/87/890687/2"}}},"commit":{"parents":[{"commit":"46b78980eec98ceaade313c559887480a9af9759","subject":"Merge \"Skip scenario tests early to avoid unnecessary setup\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/46b78980eec98ceaade313c559887480a9af9759"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-08-07 16:31:12.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2023-08-07 16:50:38.000000000","tz":-420},"subject":"Skip test_image_tasks_create() for bug 2030527","message":"Skip test_image_tasks_create() for bug 2030527\n\nChange-Id: I1f13bd11fc7cb521d1b995dbd4a4d9c662221428\nRelated-Bug: #2030527\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/58bb991974bddec0df5a94535e371df971c47840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/58bb991974bddec0df5a94535e371df971c47840"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
