)]}'
{"id":"openstack%2Ftempest~900099","triplet_id":"openstack%2Ftempest~master~Ib8072397b873c7856523228efaa4ccd7911deda5","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-08 13:22:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib8072397b873c7856523228efaa4ccd7911deda5","subject":"Extend retry sleep after server fault","status":"MERGED","created":"2023-11-04 16:47:15.000000000","updated":"2024-02-08 13:24:04.000000000","submitted":"2024-02-08 13:22:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900099","meta_rev_id":"42926c6b827b4f11904dc7aa66121d41f87a67ec","_number":900099,"virtual_id_number":900099,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-08 13:22:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"recommended":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"all":[{"value":1,"date":"2024-01-11 06:49:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":2,"date":"2024-02-07 08:43:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2024-01-11 00:56:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":1,"date":"2024-02-07 08:43:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-04 18:21:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-11 00:56:02.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-01-11 06:49:34.000000000","updated_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"reviewer":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"state":"REVIEWER"},{"updated":"2024-02-07 08:43:04.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"507fac6d1db0058bf90a287a589ca2eb418eff5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-04 16:47:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50ef27965e7267b4f18a2c01774155eb842c4988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-04 18:21:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2daf0a07f19842dea2215216b1e43b47\n\n- grenade https://zuul.opendev.org/t/openstack/build/d2f87e60d668403e9f770a245d8e35c0 : SUCCESS in 47m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bf251b754b4d4b2984966121fb0d517a : SUCCESS in 1h 12m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73704aa132d1404d9b5d85f23fa98513 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/417adbb151b045caa7366626d95a8a67 : SUCCESS in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2ed35578c0d44dca306d3a84aaedff3 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8cab24d7aae4f16995da9b9e2d2c806 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cf7e51388af4960b64d56441fe1dfc9 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3c798e1aa3a34858bcfe0d7a2968a240 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/05819cb2952a461ab15fad11d405c579 : SUCCESS in 4m 21s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/3aad5af8f99d4aa2a60e797828191033 : SUCCESS in 59m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c3c26e08b0ca4b23851c773163566e61 : SUCCESS in 1h 24m 54s (non-voting)\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/4f54e3acaf0247529559c50ecf474fe9 : SUCCESS in 56m 19s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/c4f607c1b069469896bb674db60c7108 : SUCCESS in 1h 22m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dd6b8333caba45c986841fe8288ecf94 : SUCCESS in 38m 43s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0ae1b223d3814be89ecb6684c657ef13 : SUCCESS in 22m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d9305db6aafb4cd584b09a50ffd98a08 : SUCCESS in 1h 00m 19s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0e30032015e640548cbd04be83b10152 : SUCCESS in 1h 06m 31s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f1901bd24e924ee8beab274921da3f31 : SUCCESS in 32m 30s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0fe26796687a435cb3c681f101eca7a9 : SUCCESS in 1h 06m 24s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/ae2768b92d02416aa631f9304453cf9b : SUCCESS in 1h 32m 48s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/da843cc72eae419c8d4e69fdd395ce13 : SUCCESS in 38m 22s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9597cbc51b8945efb30b4f895cf7eef6 : SUCCESS in 41m 49s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/23d124392803462285e87acbb96c6a8b : SUCCESS in 49m 49s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f601d9f817a44899a03bbc7c278de0f2 : SUCCESS in 3m 33s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/71fc5cf61a5a4645a040574b2c38f8fc : FAILURE in 1h 06m 56s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/cf18ea52f48c4175a9939b8a837cd266 : SUCCESS in 44m 29s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/d16a0e16ebea4105813b2d637b969e29 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ff921a46ea0bce35619a8183ae122544ee4dd0b8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-01-11 00:56:02.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41961028641006f36a352891cdaaa57a2cd27f92","author":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"},"date":"2024-01-11 06:49:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c1193506219cc9dafa101bf76860254c0eaf8d38","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-02-07 08:43:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4188a192741d55b4dfb13bd00819770a7d1c0365","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 08:43:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"99a5065c72878842991029f9e637e6d1ef8a5c49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-07 12:24:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fb70491f589744eb94d4d7fb6c990db5\n\n- grenade https://zuul.opendev.org/t/openstack/build/ef1be5731c07422082e8b083d50ccbbe : SUCCESS in 1h 07m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/efeaddd588b64eb98e81849f90db756c : SUCCESS in 1h 32m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06eb745654714cc380b041879c96ba0e : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69a81509ec694bd5b9ddf5e568d46f61 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45fce155a2fc494da78de93ffcc5cd29 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c36758bf42ba4af288470e7ab9c2a9bf : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed5d29a9201241c89d8d7fb419013386 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c608efe006a2434b83bda256ff278953 : SUCCESS in 3m 55s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/19e5dbcc853a4e15a24ad65bea131dd7 : SUCCESS in 1h 36m 53s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d9b98991df5945fabd7a000dcb75361a : SUCCESS in 54m 35s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/8157e229d30b478d93d30e147a49b33b : SUCCESS in 1h 04m 53s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/26d1a092a28645939b5fc80cea62c2f8 : SUCCESS in 1h 12m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7ee7e4f74894235b611b68124a9de2b : SUCCESS in 44m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/eac01baa94164c31bb287cb445d57ffb : SUCCESS in 1h 10m 06s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bc4d057b517b47ada4d121e5af20cacb : FAILURE in 1h 19m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a35c9dd05aee4ad68b9c99dca328d859 : SUCCESS in 58m 24s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/dbb4961151164aab8746980cab14cab0 : SUCCESS in 54m 19s","accounts_in_message":[],"_revision_number":1},{"id":"d7aea353ee52235474a9ee8d6afb475cf1cdb978","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-02-08 09:39:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"223b6f1bdb93c13a67a501bd43f16c6c6ce40744","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 11:47:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a984a8d3b7e84291816d2ce2bb7f1b6f\n\n- grenade https://zuul.opendev.org/t/openstack/build/b378946fb76f494db9b4e790cc955b00 : SUCCESS in 1h 03m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/598b355c4eb14191ad1a75606fb65f96 : SUCCESS in 1h 10m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d780d241724f46c9bcec756ba0d2ab16 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea1bb219abc43e3bd489a1820e0aa23 : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6949cbcb25d04390bf31f9a24d188f1a : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b0a2c47396ad40199f2ed30d522134da : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1b2b8802eb32473499b441c4d2d4c610 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50ac5584cd7a422baddb561fe0c80bba : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc8dfead7339412d995563534c319fd8 : SUCCESS in 4m 36s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/a8399702623d40ff8e2ab423dd291544 : SUCCESS in 53m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/efb9b8a4ff7b4dc9a0e434bed0c25853 : SUCCESS in 1h 44m 44s (non-voting)\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/8af3876662864650b578f72a82782727 : SUCCESS in 1h 21m 57s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/8dcbd5e1a78146c3acb67bed5d03792c : SUCCESS in 1h 18m 16s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7e7c4e68114b4b278edb9b61a2ade693 : SUCCESS in 1h 08m 11s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0e26fa629c4a411c948b383921227935 : SUCCESS in 26m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ae1c4c3c73a442cdba56a11814754433 : SUCCESS in 49m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/87439d73772d4416b491445eebb5675e : SUCCESS in 1h 31m 37s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/898913e314dd45e7afd793c0ea94bd05 : SUCCESS in 38m 55s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5a7365bd329c40498a93155325945b79 : SUCCESS in 1h 22m 37s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/7caa8d3dead04fb8997a157635a6a6d6 : TIMED_OUT in 2h 04m 18s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bc313b4fe3294a8ebeb723fcf9f592cf : SUCCESS in 1h 02m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6c3468d48a854232acc0c2d907dd7109 : SUCCESS in 1h 01m 21s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/972553a672e64585a86810d41f4699ed : SUCCESS in 1h 13m 05s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/eb5d92fdfa7b4950a762f7501757fb9b : SUCCESS in 5m 52s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/390a805607414481aa46f2e3a916d0f9 : FAILURE in 58m 10s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/18e5623f321e422681f8dbe4a9608a2a : SUCCESS in 50m 46s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f8317e5922444245a6633371b8f6eba6 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"3d14c48b2d9fa34f63dfc9a49e5c2eab234d5143","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 11:48:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8ff220995c5fc86a7bc06d611cc1555c27fdc68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 13:22:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2a984d4991644029231e92795679b0f\n\n- grenade https://zuul.opendev.org/t/openstack/build/2ebca193d8f04375ab8090aaba3a2585 : SUCCESS in 51m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b38ed6ec49e48efbd85e791d3bcf042 : SUCCESS in 1h 10m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25d933fc8dfd48b0bff604c95387fbda : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1fc72663fed43cc94ad85dcfdceb5de : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/456b4887c5cb4d52b16871cd5bbd786c : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3136ce3c2c04cafa48d8fd7a87a34d2 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fe06ead188924b08a67e2ae2dfb87184 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f24ad2d021844226a156f8964fac717c : SUCCESS in 4m 28s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3591df42eb35499483acb681318bca03 : SUCCESS in 1h 21m 44s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2f3c3200bc894138b0c43e591a1310c2 : SUCCESS in 1h 06m 14s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/33d0487361184f1e87f0fc14855ff28e : SUCCESS in 52m 35s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/28b1bfb860784188a4bcfb8fc1fcb4d6 : SUCCESS in 56m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/58d8129a5e4b4d34847ed25e732d7384 : SUCCESS in 1h 09m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d7e8acc3b29b40f795205cbbc662fc48 : SUCCESS in 1h 08m 21s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/eb6396b3cb1740a79f69686859fd8496 : SUCCESS in 1h 28m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b3bc5cdbf92142b182a406e5fc19bd23 : SUCCESS in 38m 35s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/f5cf74bfffda421aab070c298265ac00 : SUCCESS in 52m 45s","accounts_in_message":[],"_revision_number":1},{"id":"1a26739efb6c15c050183906d8b0686ca70a9896","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 13:22:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"42926c6b827b4f11904dc7aa66121d41f87a67ec","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-08 13:24:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb9a3b299958401e9b5c862d3071d58f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/732997d95bc743feb59c328a1aacbe20 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c75642699174b2ae9a150d8732386d1352e48113","revisions":{"c75642699174b2ae9a150d8732386d1352e48113":{"kind":"REWORK","_number":1,"created":"2023-11-04 16:47:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/900099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/99/900099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/99/900099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/99/900099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/99/900099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/99/900099/1"}}},"commit":{"parents":[{"commit":"14f893e822e342ca64d50f4d86d0097225aa77c6","subject":"Skip test_dhcp_agent_scheduler tests if there is no DHCP agent available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/14f893e822e342ca64d50f4d86d0097225aa77c6"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-04 16:43:10.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-04 16:45:21.000000000","tz":540},"subject":"Extend retry sleep after server fault","message":"Extend retry sleep after server fault\n\nChange Ic9eed8352c64f183ca4077e141c6b94f81a9b91a introduced retry sleep\nfor server errors, but the 1 second sleep is not enough for some cases.\n\nRelated-Bug: #2030855\nChange-Id: Ib8072397b873c7856523228efaa4ccd7911deda5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c75642699174b2ae9a150d8732386d1352e48113"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c75642699174b2ae9a150d8732386d1352e48113"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36201,"name":"Shanjiayi","email":"shanjiayi@yovole.com","username":"Shanjiayi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
