)]}'
{"id":"openstack%2Ftempest~906919","triplet_id":"openstack%2Ftempest~master~Iab973e018c88bfeadf1b942767291fef04ac528c","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2024-01-27 01:09:58.000000000","reason":"Change was marked work in progress"}},"hashtags":[],"change_id":"Iab973e018c88bfeadf1b942767291fef04ac528c","subject":"Use /var/tmp instead of /tmp for (create|get)_timestamp()","status":"NEW","created":"2024-01-26 20:37:42.000000000","updated":"2024-03-21 01:27:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"048ebe0a74793b58baef3ca282dd461a9ee2e92f","_number":906919,"virtual_id_number":906919,"owner":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-03-21 01:27:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-01-26 22:31:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fc504adf768803bc0825f082b7de2309307f8649","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-01-26 20:37:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6010a2db787806cb091a76ab372aeea197592ad4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 22:31:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/290f86274c5941f0818bc22570a6c71d\n\n- grenade https://zuul.opendev.org/t/openstack/build/d8d12dc0e07f497b9967cf1d12f0772d : SUCCESS in 44m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6c4b053e9b3404a99af7e05da234b6f : SUCCESS in 1h 08m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5f32df3dcbee42cdba33ebbb4639411f : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68028a37201c481eb774e50b088b2889 : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37d2a0c559924795a8eb48159b30fa14 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8dd883fe92ad4623b7a69ac9d161c5ad : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5215a94f30864052a87cc60a243b8f00 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e3633f9af46844c1975f27fb2eaa93bd : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2057653b5ba94aeaa7109a81380ca3bf : SUCCESS in 4m 29s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/446ef453e7d74fd397d7e191bf3230a0 : SUCCESS in 58m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f5b0d7b7fe5d4373a4b0008f9a5bcecf : FAILURE in 1h 44m 53s (non-voting)\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/5a8638685cb943149d208b58224e1165 : SUCCESS in 1h 36m 38s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/e7dea2a3e62c4d539dd74e2ae5f9ac80 : SUCCESS in 1h 28m 22s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fe59d2c40137487b9ff427f9159fcd46 : FAILURE in 1h 11m 29s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/5a8735b496a54cfbb0303b66232b2af7 : SUCCESS in 29m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14472504d5604781b3a440c8bc7e0d73 : SUCCESS in 1h 02m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/736f21d1902c469080ba34cc2b4ad20e : FAILURE in 1h 28m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/3446e569b93e4907b81a0c1fac738373 : SUCCESS in 45m 43s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ccebc28b6b5d4fa595d714e6d573e39f : SUCCESS in 1h 14m 14s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b52e32fc836c44218b0e028fb05a03ca : FAILURE in 1h 34m 52s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/4217642503e84c388acf0d39e212a4d9 : SUCCESS in 46m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c2fb1d74fb1441bd9827e26cbee27659 : SUCCESS in 52m 28s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/0f13247944a54eda8c4f62d411878769 : SUCCESS in 55m 55s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b75717b9cd6b45128d184efeb7be4cd1 : SUCCESS in 3m 49s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/abfd0bc5ef2a44aca1c8f2ee7dff565a : FAILURE in 1h 00m 57s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/5a6d4e978d5344bbb057821c514560c2 : SUCCESS in 45m 08s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/f0a4ba5ec2a0459da9c46a965faa99ec : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"aee8f25bd332ee4b849b7d539ce6a81e0ffb8aa4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-01-26 22:41:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e5ca39b456c11714d509899dade2af53060db041","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-27 00:37:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e3015f489634eceabffee0485eeb273\n\n- grenade https://zuul.opendev.org/t/openstack/build/b830398c0f28408bb4e5daa6d6ced9a7 : SUCCESS in 38m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a22095c3f12e4c8092c7f888fde3e222 : SUCCESS in 58m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a2326c9fe06d4bff99b6c9b5087abfdb : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a26b0b91136a46e886e3c84f2728a6b2 : SUCCESS in 11m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8abf293e1c74ae2b4626ef3302e86c7 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7ebcd561bdf4cbcaae3db50f735271c : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/edd51de67d7f4b8988421a5c60cc0b60 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aae1de333b9246ad9679daf719a3316d : SUCCESS in 6m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a39c9302469a45ad9fc22b34bc2594db : SUCCESS in 4m 09s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/269cbb2f75444e42bbc9ac185fa4a638 : SUCCESS in 51m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/441cfa3161854602826258f77cf78949 : SUCCESS in 1h 50m 15s (non-voting)\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/a6c37bf6375b4abbbace813ccfde4b14 : SUCCESS in 1h 10m 14s\n- tempest-full-yoga https://zuul.opendev.org/t/openstack/build/0795d07f71da480fbdf65e6595f87a3d : SUCCESS in 57m 05s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3984edecb27b4ebaa89329de5445ae54 : FAILURE in 49m 15s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ae5326a1f03b4ea882a17b03e3eed861 : SUCCESS in 26m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f38ee45f6ed4dfa9c7b6f46c849108a : SUCCESS in 49m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8f239e870488493ca6007ac38b57b194 : FAILURE in 1h 32m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ad92b2208b874ad8b36ceda90e7d1502 : SUCCESS in 55m 52s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/377c58387f0841b89d8ff1c17a44aad5 : SUCCESS in 1h 41m 31s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/5e57b7b02abb47c7bfa4da3ec7ae8d37 : FAILURE in 1h 27m 32s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b2cd8499254f416a91455da26c9fb274 : SUCCESS in 1h 08m 18s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/df4c8f4ad8444e968a5c4387e9ef1c9e : SUCCESS in 1h 13m 57s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/a173eb9e7d78446a8595c4a067762940 : SUCCESS in 56m 20s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/866621fc17b949ec9112178e3779e60d : SUCCESS in 3m 25s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/d228fdad2ad646bda2216b643196b9dc : FAILURE in 1h 05m 54s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/020bf70de8ae47ebba333ed6f179e1c4 : SUCCESS in 1h 05m 15s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/304f82684bbf4349af9b6a0f06b914c4 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":2},{"id":"1291e9df292d9dcf96d20620beee3f25fc342220","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-01-27 01:09:58.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"c957ca7dcbd04aad9d6ed8ca50124422ab4d4035","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-01-27 01:10:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6511b8cb31a89a42429375a5d041bd5df303602f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2024-03-20 23:29:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"048ebe0a74793b58baef3ca282dd461a9ee2e92f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-21 01:27:33.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a0bac53788344c2bae2160efd02ed0a1\n\n- grenade https://zuul.opendev.org/t/openstack/build/ae3a1331fd3348c3b01bfc52a663c2c4 : SUCCESS in 40m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5306ffb272b64187ae0026d215c755fb : SUCCESS in 1h 36m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2e497c4d1524cd88cb3d16f3554336b : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3efb4afe5e14ef492865115a9c51827 : SUCCESS in 12m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc1afe0537743c4addd1777768c75d1 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1eff1489c40c4e21912831295e7c7008 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8cb4fe45ab384b6d95eec56673f876c4 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/597c6c845132456c84fc15b0fb44fa98 : SUCCESS in 3m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec3ce2e8a3c344e3814d5fe959293644 : SUCCESS in 6m 03s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/25f335b397e64544b1498f1661df85ab : SUCCESS in 35m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8d26b30464d64ae5ba5807ad650f6a34 : SUCCESS in 1h 24m 01s (non-voting)\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/8bdf9a4cb6614847971583ad5cefe234 : SUCCESS in 1h 28m 09s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/a29a0151935344a7a1e3f5ceb4074902 : SUCCESS in 1h 20m 11s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/90d144a449b747f3a331ae18499a41ad : FAILURE in 1h 06m 04s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/59d099c72c5c4cd59400634b3119a0e7 : SUCCESS in 20m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3fc6c2930644a26938948ddae639f90 : SUCCESS in 37m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1145d9bcfb0a42a09557973c7a27fdd2 : FAILURE in 1h 30m 16s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d2bb9f254bf14f1682af36e61694fefc : SUCCESS in 49m 13s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1f5797a6390140f4aae30e202b0d7deb : SUCCESS in 1h 39m 47s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/8878cb3437794791b55e42c3d1b841b1 : FAILURE in 1h 55m 48s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e3d13627fc43423496bd923dcdff8273 : SUCCESS in 59m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/1a9bab2ab6b94f4d9c4b0ebc2ee014d6 : SUCCESS in 52m 24s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/a600623736f84fec8fbcce9ed78cdafb : SUCCESS in 58m 37s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e4b843570be642f5a042af0ad6be0050 : SUCCESS in 3m 34s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/112d29514f634bc5a9d5b4c4e2a678cc : FAILURE in 42m 40s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/24a861faee654a4286d67c6ec9c46021 : SUCCESS in 48m 22s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/97f745f2f1ea464c91d26d940b8faa8e : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e04e8a10c5ce1328d1dd5c24969c76a650157f6a","revisions":{"e0e6b80e0659f71f0f7aff269eb791c0e54308cc":{"kind":"REWORK","_number":1,"created":"2024-01-26 20:37:42.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/19/906919/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/19/906919/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/906919/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/906919/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/906919/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/19/906919/1"}}},"commit":{"parents":[{"commit":"a605aff46ddc9ebb7e467c079ada6b3ed49a6c2d","subject":"Merge \"Clean up [identity-feature-enable] options for old releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a605aff46ddc9ebb7e467c079ada6b3ed49a6c2d"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-01-26 20:25:20.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-01-26 20:37:11.000000000","tz":0},"subject":"Use /var/tmp instead of /tmp for (create|get)_timestamp()","message":"Use /var/tmp instead of /tmp for (create|get)_timestamp()\n\nAwhile back, change Iec41ff7cba616fd469de0019f76ab54072f6fc28 made /tmp\nthe default location for writing and reading the timestamp file.\n\nHowever occasionally, gate failures are observed where the timestamp\nfile can\u0027t be opened, for example:\n\n  ERROR    [tempest.lib.common.utils.linux.remote_client]\n    (TestVolumeBootPattern:test_volume_boot_pattern) Executing command\n    on 172.24.5.161 failed. Error: Command \u0027set -eu -o pipefail;\n    PATH\u003d$PATH:/sbin:/usr/sbin; sudo cat /tmp/timestamp\u0027,\n    exit status: 1, stderr: cat: can\u0027t open \u0027/tmp/timestamp\u0027:\n    No such file or directory\n\nThis changes the default directory from /tmp to /var/tmp in an effort\nto improve stability [1]:\n\n  \"/tmp/ and /var/tmp/ are two world-writable directories Linux systems\n  provide for temporary files. The former is typically on tmpfs and\n  thus backed by RAM/swap, and flushed out on each reboot. The latter\n  is typically a proper, persistent file system, and thus backed by\n  physical storage.\"\n\n[1] https://systemd.io/TEMPORARY_DIRECTORIES\n\nChange-Id: Iab973e018c88bfeadf1b942767291fef04ac528c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e0e6b80e0659f71f0f7aff269eb791c0e54308cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e0e6b80e0659f71f0f7aff269eb791c0e54308cc"}]},"branch":"refs/heads/master"},"5ce056da1e061851cae03997b0680010cc4f9d24":{"kind":"REWORK","_number":2,"created":"2024-01-26 22:41:47.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/19/906919/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/19/906919/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/906919/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/906919/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/906919/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/19/906919/2"}}},"commit":{"parents":[{"commit":"a605aff46ddc9ebb7e467c079ada6b3ed49a6c2d","subject":"Merge \"Clean up [identity-feature-enable] options for old releases\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a605aff46ddc9ebb7e467c079ada6b3ed49a6c2d"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-01-26 20:25:20.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-01-26 22:41:39.000000000","tz":0},"subject":"Use /var/tmp instead of /tmp for (create|get)_timestamp()","message":"Use /var/tmp instead of /tmp for (create|get)_timestamp()\n\nAwhile back, change Iec41ff7cba616fd469de0019f76ab54072f6fc28 made /tmp\nthe default location for writing and reading the timestamp file.\n\nHowever occasionally, gate failures are observed where the timestamp\nfile can\u0027t be opened, for example:\n\n  ERROR    [tempest.lib.common.utils.linux.remote_client]\n    (TestVolumeBootPattern:test_volume_boot_pattern) Executing command\n    on 172.24.5.161 failed. Error: Command \u0027set -eu -o pipefail;\n    PATH\u003d$PATH:/sbin:/usr/sbin; sudo cat /tmp/timestamp\u0027,\n    exit status: 1, stderr: cat: can\u0027t open \u0027/tmp/timestamp\u0027:\n    No such file or directory\n\nThis changes the default directory from /tmp to /var/tmp in an effort\nto improve stability [1]:\n\n  \"/tmp/ and /var/tmp/ are two world-writable directories Linux systems\n  provide for temporary files. The former is typically on tmpfs and\n  thus backed by RAM/swap, and flushed out on each reboot. The latter\n  is typically a proper, persistent file system, and thus backed by\n  physical storage.\"\n\n[1] https://systemd.io/TEMPORARY_DIRECTORIES\n\nChange-Id: Iab973e018c88bfeadf1b942767291fef04ac528c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5ce056da1e061851cae03997b0680010cc4f9d24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/5ce056da1e061851cae03997b0680010cc4f9d24"}]},"branch":"refs/heads/master"},"e04e8a10c5ce1328d1dd5c24969c76a650157f6a":{"kind":"REWORK","_number":3,"created":"2024-03-20 23:29:35.000000000","uploader":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"ref":"refs/changes/19/906919/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/19/906919/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/906919/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/906919/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/19/906919/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/19/906919/3"}}},"commit":{"parents":[{"commit":"905da972c9f5c8c78b656b50dfeaad6237c93551","subject":"Merge \"Fix branch regex for unmaintained branch model\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/905da972c9f5c8c78b656b50dfeaad6237c93551"}]}],"author":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-01-26 20:25:20.000000000","tz":0},"committer":{"name":"melanie witt","email":"melwittt@gmail.com","date":"2024-03-20 23:28:48.000000000","tz":0},"subject":"Use /var/tmp instead of /tmp for (create|get)_timestamp()","message":"Use /var/tmp instead of /tmp for (create|get)_timestamp()\n\nAwhile back, change Iec41ff7cba616fd469de0019f76ab54072f6fc28 made /tmp\nthe default location for writing and reading the timestamp file.\n\nHowever occasionally, gate failures are observed where the timestamp\nfile can\u0027t be opened, for example:\n\n  ERROR    [tempest.lib.common.utils.linux.remote_client]\n    (TestVolumeBootPattern:test_volume_boot_pattern) Executing command\n    on 172.24.5.161 failed. Error: Command \u0027set -eu -o pipefail;\n    PATH\u003d$PATH:/sbin:/usr/sbin; sudo cat /tmp/timestamp\u0027,\n    exit status: 1, stderr: cat: can\u0027t open \u0027/tmp/timestamp\u0027:\n    No such file or directory\n\nThis changes the default directory from /tmp to /var/tmp in an effort\nto improve stability [1]:\n\n  \"/tmp/ and /var/tmp/ are two world-writable directories Linux systems\n  provide for temporary files. The former is typically on tmpfs and\n  thus backed by RAM/swap, and flushed out on each reboot. The latter\n  is typically a proper, persistent file system, and thus backed by\n  physical storage.\"\n\n[1] https://systemd.io/TEMPORARY_DIRECTORIES\n\nChange-Id: Iab973e018c88bfeadf1b942767291fef04ac528c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e04e8a10c5ce1328d1dd5c24969c76a650157f6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e04e8a10c5ce1328d1dd5c24969c76a650157f6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
