)]}'
{"id":"openstack%2Ftempest~913950","triplet_id":"openstack%2Ftempest~master~I5ee1ce6b3ffbbb1f47e448fc15d540f74b21684f","project":"openstack/tempest","branch":"master","topic":"qa-2024-1-release","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-03-23 00:56:06.000000000","reason":"Change was submitted"},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2024-03-22 22:52:04.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-03-23 00:56:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5ee1ce6b3ffbbb1f47e448fc15d540f74b21684f","subject":"Use stable constraint in tox to release new tag for 2024.1","status":"MERGED","created":"2024-03-22 06:44:01.000000000","updated":"2024-03-23 00:58:17.000000000","submitted":"2024-03-23 00:56:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"913950-qa-2024-1-release","meta_rev_id":"54880012c90b8fcd8577ec7d625518d0972f9876","_number":913950,"virtual_id_number":913950,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-23 00:56:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2024-03-22 16:15:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2024-03-22 22:52:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-22 06:44:30.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-03-22 07:39:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-22 09:21:07.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"CC"},{"updated":"2024-03-22 12:59:16.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2024-03-22 16:15:30.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"ab88db76901d7a27d5f7bb3cb44684a84abc7b4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-22 06:44:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2d67af109f0f2008f497964a6cd079f623f26438","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-22 06:44:30.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/tempest on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027branches\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: tempest-full-py3\n      parent: devstack-tempest\n      # This job version is with swift enabled on py3\n      # as swift is ready on py3 from stable/ussuri onwards.\n      # As this use \u0027integrated-full\u0027 tox env which is not\n      # available in old tempest used till stable/wallaby,\n      # this job definition is only for stable/xena onwards\n      # and separate job definition until stable/wallaby\n      branches:\n        regex: ^.*/(victoria|wallaby)$\n        negate: true\n      description: |\n        Base integration test with Neutron networking, horizon, swift enable,\n        and py3.\n        Former names for this job where:\n          * legacy-tempest-dsvm-py35\n          * gate-tempest-dsvm-py35\n      required-projects:\n        - openstack/horizon\n      vars:\n        # NOTE(gmann): Default concurrency is higher (number of cpu -2) which\n        # end up 6 in upstream CI. Higher concurrency means high parallel\n        # requests to services and can cause more oom issues. To avoid the\n        # oom issue, setting the concurrency to 4 in this job.\n        tempest_concurrency: 4\n        tox_envlist: integrated-full\n        devstack_localrc:\n          USE_PYTHON3: true\n          FORCE_CONFIG_DRIVE: true\n          ENABLE_VOLUME_MULTIATTACH: true\n          GLANCE_USE_IMPORT_WORKFLOW: True\n        devstack_plugins:\n          neutron: https://opendev.org/openstack/neutron\n        devstack_services:\n          # Enbale horizon so that we can run horizon test.\n          horizon: true\n\n  in \"openstack/tempest/zuul.d/integrated-gate.yaml@master\", line 41, column 3","accounts_in_message":[],"_revision_number":1},{"id":"9f871dc3c8727f50ae1be4185c4775f3f973db5c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-22 07:00:26.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"e471703de717675fdc9ef93889222178afd92639","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 07:39:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8eaf5f85711844d5bc7225a1ab60c208\n\n- grenade https://zuul.opendev.org/t/openstack/build/7f0725bf0eb6415ba35928bdaadeab76 : FAILURE in 19m 03s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2639f1b365e40e293995160a4d3a998 : FAILURE in 30m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1860861306c348a5a45ed447459c95f4 : FAILURE in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f056adb7b89f4fbcbb061a6ed318daea : FAILURE in 3m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b1488fee0214a4d98acba5370d04acb : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83df5d55be4e4a4f8115ad1d1c304c43 : FAILURE in 2m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a85b9b5c4ed7411aba1a00b6f4936630 : FAILURE in 1m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82786b2f1da147ef917fe8d0c470df89 : FAILURE in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a1e428bb4aaf4c7ba138cbe06efcaad8 : FAILURE in 2m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/332e6725e6b8435aaf0d1712cd1d50b9 : FAILURE in 2m 00s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/3a209fa66cad4ec99c47865b268d6743 : FAILURE in 29m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3e2f66b996d3414b8bf0e503ac4bc790 : FAILURE in 31m 21s (non-voting)\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/9adf266ed95747d58a908bee379e65d7 : FAILURE in 33m 45s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/b98becb465b7414bbadf4e910aae4121 : FAILURE in 33m 55s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ecddc66c860240c4bf949d61d4c83bcd : FAILURE in 34m 01s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/feaf1c310f704ad0b29fbe4a81571d61 : FAILURE in 2m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6e80f83254b4359b20296a84f5231e8 : FAILURE in 23m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/eeb387a682dd4e4ea17ad15ff0cbd8bc : FAILURE in 18m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bbf5c34de793476c9166f69a36cc2a76 : POST_FAILURE in 25m 05s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/110b85c996fa48bd998569521e6b1af4 : FAILURE in 18m 44s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9571c0a941924aa0aa806a9e6c50c879 : FAILURE in 34m 40s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6b775e8a7f8940fd9fe1294bc34c9b16 : FAILURE in 34m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9ac3a5b145464189b6caca6811b54fe9 : FAILURE in 16m 41s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/6f426fd9a12c4ae4adc27cf56c649266 : FAILURE in 21m 40s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/bc0638f98cf94657bf97430273fd86d3 : SUCCESS in 4m 56s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/fb9b8bca8fb44c5fbee1ba15072c25ec : FAILURE in 23m 05s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/7ecb3f73af144e7f9e8e31c88d54b49f : FAILURE in 26m 36s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2ef6923354f94a329f224cf3b975da92 : FAILURE in 2m 45s","accounts_in_message":[],"_revision_number":1},{"id":"ca287ea0c4cbd79e765a8526354b87c521a1a84e","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-03-22 09:21:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"376e460f852cd9181f96be4c525ef508b3445e85","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-22 09:21:36.000000000","message":"Patch Set 1:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/tempest on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027branches\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: tempest-full-py3\n      parent: devstack-tempest\n      # This job version is with swift enabled on py3\n      # as swift is ready on py3 from stable/ussuri onwards.\n      # As this use \u0027integrated-full\u0027 tox env which is not\n      # available in old tempest used till stable/wallaby,\n      # this job definition is only for stable/xena onwards\n      # and separate job definition until stable/wallaby\n      branches:\n        regex: ^.*/(victoria|wallaby)$\n        negate: true\n      description: |\n        Base integration test with Neutron networking, horizon, swift enable,\n        and py3.\n        Former names for this job where:\n          * legacy-tempest-dsvm-py35\n          * gate-tempest-dsvm-py35\n      required-projects:\n        - openstack/horizon\n      vars:\n        # NOTE(gmann): Default concurrency is higher (number of cpu -2) which\n        # end up 6 in upstream CI. Higher concurrency means high parallel\n        # requests to services and can cause more oom issues. To avoid the\n        # oom issue, setting the concurrency to 4 in this job.\n        tempest_concurrency: 4\n        tox_envlist: integrated-full\n        devstack_localrc:\n          USE_PYTHON3: true\n          FORCE_CONFIG_DRIVE: true\n          ENABLE_VOLUME_MULTIATTACH: true\n          GLANCE_USE_IMPORT_WORKFLOW: True\n        devstack_plugins:\n          neutron: https://opendev.org/openstack/neutron\n        devstack_services:\n          # Enbale horizon so that we can run horizon test.\n          horizon: true\n\n  in \"openstack/tempest/zuul.d/integrated-gate.yaml@master\", line 41, column 3","accounts_in_message":[],"_revision_number":1},{"id":"e8ba02a0369f4038381f593b2a92ac3f031f003d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 10:19:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9cdcf19d690b470cab7b15843f9ee413\n\n- grenade https://zuul.opendev.org/t/openstack/build/399490843b4f4088b7a28958eb8d97ef : FAILURE in 32m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5059e533c2c54620beeb3bf3a2b83a15 : FAILURE in 32m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f89feb47a4bb410d993059683892e4d0 : FAILURE in 1m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52f14013202d484eb05ecaf62a4da382 : FAILURE in 3m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6aace4e141fb431e8a24ffaa466aa5ef : SUCCESS in 5m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4bb2ada5620429b9fde753a2b373d4d : FAILURE in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/67ff0a4aaa5f428d83dc1c39e3c8ee34 : FAILURE in 2m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ec50a4bac7a48fc9ae47540f01f630d : FAILURE in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a2ef138a3eb481a88da85495fe0fc02 : FAILURE in 2m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/edfe87e92a2a4fec91097e17349ea532 : FAILURE in 1m 57s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/3f58b153cf81424bb1696d9e8e22c286 : FAILURE in 21m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/523931b695d04b07a7195768c991906f : FAILURE in 25m 19s (non-voting)\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/7279b785449d485fb48689eafa75d649 : FAILURE in 33m 06s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/b822c66338fa4c44a3703cbbb0220307 : FAILURE in 35m 32s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/280016b2ac8a4ddfaaa0a5f7ed5cf481 : FAILURE in 19m 42s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/86b54a5b6aa5491185e267362cf9c12c : FAILURE in 2m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/155eec3876d64b6a8f96d8c46f64bb83 : FAILURE in 22m 12s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1b1c416ddac1423dbd00294f79830d68 : FAILURE in 33m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8ceb6a1b78e248caa222fb7fa75789d5 : POST_FAILURE in 35m 14s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/109c38c3d4704913bfa062d4d85375ff : FAILURE in 21m 48s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c258baf969a44ba0a867330ab8a045d0 : FAILURE in 22m 41s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7df3fa1b6a524641b9987776f3b50a0e : FAILURE in 35m 32s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/fc49620985b94e2fa00bc528175bd817 : FAILURE in 31m 13s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/6bcce7d2a7194a9fb6940ed96cb7d550 : FAILURE in 28m 26s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/77a8eca8d55c4995a0d9f9030c938188 : SUCCESS in 3m 41s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/6abd4b643ffa463e87ce77be9a0a569b : FAILURE in 32m 00s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/8f856182ce9c47aa929cb00eb38d95e9 : FAILURE in 47m 36s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/cf7fbfc918f6460bbaff5dc599a8e71b : FAILURE in 2m 09s","accounts_in_message":[],"_revision_number":1},{"id":"399d9af047b91beee3d42d17c319b1f31358b6b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-22 12:58:58.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"19f644e8eaf9a45bd302f517f6b9e657419a5e1a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-03-22 12:59:16.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"38eda9271e83e37224f631e40c4a94c25c3482e4","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-22 12:59:17.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/tempest on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027branches\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: tempest-full-py3\n      parent: devstack-tempest\n      # This job version is with swift enabled on py3\n      # as swift is ready on py3 from stable/ussuri onwards.\n      # As this use \u0027integrated-full\u0027 tox env which is not\n      # available in old tempest used till stable/wallaby,\n      # this job definition is only for stable/xena onwards\n      # and separate job definition until stable/wallaby\n      branches:\n        regex: ^.*/(victoria|wallaby)$\n        negate: true\n      description: |\n        Base integration test with Neutron networking, horizon, swift enable,\n        and py3.\n        Former names for this job where:\n          * legacy-tempest-dsvm-py35\n          * gate-tempest-dsvm-py35\n      required-projects:\n        - openstack/horizon\n      vars:\n        # NOTE(gmann): Default concurrency is higher (number of cpu -2) which\n        # end up 6 in upstream CI. Higher concurrency means high parallel\n        # requests to services and can cause more oom issues. To avoid the\n        # oom issue, setting the concurrency to 4 in this job.\n        tempest_concurrency: 4\n        tox_envlist: integrated-full\n        devstack_localrc:\n          USE_PYTHON3: true\n          FORCE_CONFIG_DRIVE: true\n          ENABLE_VOLUME_MULTIATTACH: true\n          GLANCE_USE_IMPORT_WORKFLOW: True\n        devstack_plugins:\n          neutron: https://opendev.org/openstack/neutron\n        devstack_services:\n          # Enbale horizon so that we can run horizon test.\n          horizon: true\n\n  in \"openstack/tempest/zuul.d/integrated-gate.yaml@master\", line 41, column 3","accounts_in_message":[],"_revision_number":2},{"id":"a9b2da6c43e07124088e7fd3911c8cd7250681d4","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-03-22 14:30:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8da9152f7c02547d1c41f8599c07bdd675c617ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 14:58:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f22c56da35e444d597d8977a7f1d11b8\n\n- grenade https://zuul.opendev.org/t/openstack/build/bdc772113b1d46c185319c48c8a47ffb : SUCCESS in 59m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4d3bbb04b808460b8d4c20e755e4b7e3 : SUCCESS in 1h 32m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1139d72985a6452186ad90c157db29f4 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0fbe4c602a3414891ca7c7344de6ed3 : SUCCESS in 10m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ed0ed2ed8950418ba33b102252952661 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02a4f3008d58421a83bb79c0069deba3 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/44417060ef284e70829df61fac145989 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c8d2a5777bfe4d2eb1f6c11cd26dbdc6 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29990a73cf534206b45b8543c84c7de0 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db5cfd7463c24e40ae1162e16f93969e : SUCCESS in 5m 18s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/819b87adb566499ab4ab365336683c35 : SUCCESS in 47m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a6c18c12503b40ab9714b9c045124206 : SUCCESS in 1h 54m 13s (non-voting)\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/7e497f7ad23447f1a3d2432c19387f50 : SUCCESS in 1h 33m 18s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/c56f9a272ad6409da1bc0a584c29819e : SUCCESS in 1h 18m 09s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/266f24b4a1a844eebe6735a51bd6c4c2 : SUCCESS in 1h 04m 53s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/20c0824e27634fff9deef94c08fed91f : SUCCESS in 33m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/84cb99b2176c4f4382c11ed4621dfbc1 : SUCCESS in 47m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0d05f45bf25947df940efe115b562fce : FAILURE in 1h 45m 05s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/873bad2872da4ff0924a05be733dbfa1 : SUCCESS in 53m 31s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2515a02e956042f1b6c08f99c196e473 : SUCCESS in 1h 31m 31s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9160608966f84a21b53ed41c06b31c0d : FAILURE in 1h 55m 14s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c3203c0ed5254214a3b2f28ac0f9e99f : SUCCESS in 58m 52s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e70f2a0259d34892b2487c4a084e6f1f : SUCCESS in 1h 05m 11s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/6849dc33cece4bfeb30c994e5b3f0dc4 : SUCCESS in 1h 12m 45s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b043708047d0472598f9d147913b5015 : SUCCESS in 5m 08s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1e88e90171504f55ac9cc7d11eccf1ce : FAILURE in 1h 01m 39s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/5314206e4e6a41f2b243e0a5c37d705e : FAILURE in 40m 29s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/bd521ed98937409e94e82b400e2bae74 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"65984ae4c883e99e14139310138bdf613cc385bf","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-03-22 16:15:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a0f2f42e1d484bbb2c1d3283a66608301abed7b6","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-03-22 16:15:30.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dd65839ff2bf42999b220d8c169029e3ae921d7e","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-03-22 16:15:46.000000000","message":"Patch Set 2:\n\nZuul encountered a syntax error while parsing its configuration in the\nrepo openstack/tempest on branch master.  The error was:\n\n  expected a list for dictionary value @ data[\u0027branches\u0027]\n\nThe error appears in the following job stanza:\n\n  job:\n      name: tempest-full-py3\n      parent: devstack-tempest\n      # This job version is with swift enabled on py3\n      # as swift is ready on py3 from stable/ussuri onwards.\n      # As this use \u0027integrated-full\u0027 tox env which is not\n      # available in old tempest used till stable/wallaby,\n      # this job definition is only for stable/xena onwards\n      # and separate job definition until stable/wallaby\n      branches:\n        regex: ^.*/(victoria|wallaby)$\n        negate: true\n      description: |\n        Base integration test with Neutron networking, horizon, swift enable,\n        and py3.\n        Former names for this job where:\n          * legacy-tempest-dsvm-py35\n          * gate-tempest-dsvm-py35\n      required-projects:\n        - openstack/horizon\n      vars:\n        # NOTE(gmann): Default concurrency is higher (number of cpu -2) which\n        # end up 6 in upstream CI. Higher concurrency means high parallel\n        # requests to services and can cause more oom issues. To avoid the\n        # oom issue, setting the concurrency to 4 in this job.\n        tempest_concurrency: 4\n        tox_envlist: integrated-full\n        devstack_localrc:\n          USE_PYTHON3: true\n          FORCE_CONFIG_DRIVE: true\n          ENABLE_VOLUME_MULTIATTACH: true\n          GLANCE_USE_IMPORT_WORKFLOW: True\n        devstack_plugins:\n          neutron: https://opendev.org/openstack/neutron\n        devstack_services:\n          # Enbale horizon so that we can run horizon test.\n          horizon: true\n\n  in \"openstack/tempest/zuul.d/integrated-gate.yaml@master\", line 41, column 3","accounts_in_message":[],"_revision_number":2},{"id":"08698d0532a8e66a5f851c6b26c31a545736a171","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-22 17:00:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f0dad04949cee5f7fcc896b71bb70365b00fcd8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 18:35:34.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc39a38ce97e4a45b6d9b45b4d1e5f2b\n\n- grenade https://zuul.opendev.org/t/openstack/build/32e575f5289145d18b393378cb65d283 : SUCCESS in 1h 11m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/112e9c776ef44df8af44a6aa5c8598cc : SUCCESS in 1h 09m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a997480a87e34ff59dae873977f8c06a : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f82ea21a92354cb69a47976ec1d10c4a : SUCCESS in 10m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a27df49b721f4cc9922d91b73cf82243 : SUCCESS in 6m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f8133ec92914b909692af73def3519c : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65e07c3720df4c2296eba3991ad859f4 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c815b0f546b44b279bd10a8f72583841 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a5077a0f3ac46c6ad480fd04e2ba1ab : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9d9078cddf6a4462a842cfde70e4dd5c : SUCCESS in 4m 25s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/c2ac22b87167463a9bbd3c3f48e685d1 : SUCCESS in 52m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0fa48fc52b434c859c91c59fbfda1c84 : SUCCESS in 2h 13m 47s (non-voting)\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/5375cdc8aa1e4bfba7ba5cfeaf20af69 : SUCCESS in 1h 42m 01s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/28dc57b814134bad8d91fdda493f24b5 : SUCCESS in 1h 19m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/14a9087b491d4424801d4453ca4c9c3f : SUCCESS in 1h 09m 45s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/368981a6ebb34a8fbcc50fb7a6ae1b56 : SUCCESS in 30m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05b7c62bc78c4afba33c8d5ca05ee7fe : SUCCESS in 53m 53s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3e58f9e4c3f84b708f4acee80b28d821 : SUCCESS in 1h 24m 28s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/452698a500bb4a778d64bb4c745f22dd : SUCCESS in 51m 06s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8e085709d3414b419529c98cd6e2a6c0 : SUCCESS in 1h 20m 28s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/10524bf306b94b1e9125d1ac4367ac6a : SUCCESS in 1h 35m 01s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d234dfa6366f4e20adee44b18f78d52e : SUCCESS in 47m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9b60f57a3ca64c81aa5d452573d6b877 : SUCCESS in 51m 32s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/f2c98e7b1bdc4977bf2b43d0a29ccbcb : SUCCESS in 59m 34s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/2be069a483ce4ab89e9742c071c16bd0 : SUCCESS in 3m 29s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/def9348337e14a63943426ce8111c155 : FAILURE in 1h 01m 35s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/22b2cadd5d10411ea7e134523d3434b9 : FAILURE in 33m 36s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1214169a4fe6496bbcf36367c7ef253f : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":2},{"id":"1a8fd2362dbb6b3ba1b4796f3b4d90fd5c4435ca","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-03-22 22:52:04.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"9e10145baf68823b1928fd0923e2398f3273c00d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-22 22:52:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fb535916fc6bf3c1f30b3f2b74cefe66b92d9a37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-23 00:56:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b3274f6fc524544a304f070c5ac95fa\n\n- grenade https://zuul.opendev.org/t/openstack/build/423f09c6fbc34a2896f464143c82ff8e : SUCCESS in 1h 05m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3c812e8b52b44379901b264c0b41e2a8 : SUCCESS in 1h 57m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4556f532882d41f78ca0b2f495b93bab : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e5ce25a2c724ae0a6137454e0a13f0e : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7607d77d95b4b4e976d193cc865162c : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89550c1420144e9bbafd9155c76ae6bb : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ab7d2d0804441a7b9b1e44600165466 : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8b0eda2e11546c38cc22dbce6c449a9 : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bfdf9dd62275466f995525febb58b69d : SUCCESS in 4m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6c3a45e5088b42bc832201dd5fa67b94 : SUCCESS in 1h 01m 01s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/64f797455e3845b8affe4b4334659370 : SUCCESS in 59m 03s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/adb624a8dd4e4ff68d1cce50560b2026 : SUCCESS in 57m 11s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ba03672154b64df8904b2c786c0b14cd : SUCCESS in 46m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7da0deb9aea941d2aab34344d8b2909b : SUCCESS in 41m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/92581774ec294eef8884c0999aa4ae87 : SUCCESS in 1h 01m 23s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/649104974d6f4971a8fc1dd8eab01226 : SUCCESS in 1h 28m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bd035d1b68ab4a8988b8405f7c2745de : SUCCESS in 43m 23s","accounts_in_message":[],"_revision_number":2},{"id":"3d4184868d0cf568fcc4488ab9e938f833c4094e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-23 00:56:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"54880012c90b8fcd8577ec7d625518d0972f9876","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-23 00:58:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db97642b6cd2439daab91bd7cdd4d580\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28529ae6d80044a5a3ae56b5c4b05c23 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d533250c1afe41d4bf2f4609ef9a1401 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d5f63dbe3332076ba32ab43740b49b647e182bc5","revisions":{"064a08467799421470a8e5371d60166a64bed95f":{"kind":"REWORK","_number":1,"created":"2024-03-22 06:44:01.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/50/913950/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/50/913950/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/913950/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/913950/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/913950/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/50/913950/1"}}},"commit":{"parents":[{"commit":"cfdaf9a4b9c8592c08e64206bef5e98828538662","subject":"Add releasenote to tag the Tempest for 2024.1 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cfdaf9a4b9c8592c08e64206bef5e98828538662"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-03-22 06:42:41.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-03-22 06:42:41.000000000","tz":-420},"subject":"Use stable constraint in tox to release new tag for 2024.1","message":"Use stable constraint in tox to release new tag for 2024.1\n\nWe are going to release Tempest new tag to declare the start of\nsupport for stable/2024.1. So that new tag use stable/2024.1\nconstraint in the tox env.\n\nFor example: if anyone use Tempest new tag in future say\n1 year later then tox env also should use the 2024.1 contstraint for\ncompatibility instead master.\n\nAlso, making ironic-tempest-bios-ipmi-direct-tinyipa as\nnon voting to merge his change.\n\nThese need to be reverted once release is done.\n\nChange-Id: I5ee1ce6b3ffbbb1f47e448fc15d540f74b21684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/064a08467799421470a8e5371d60166a64bed95f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/064a08467799421470a8e5371d60166a64bed95f"}]},"branch":"refs/heads/master"},"d5f63dbe3332076ba32ab43740b49b647e182bc5":{"kind":"REWORK","_number":2,"created":"2024-03-22 12:58:58.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/50/913950/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/50/913950/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/913950/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/913950/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/50/913950/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/50/913950/2"}}},"commit":{"parents":[{"commit":"cfdaf9a4b9c8592c08e64206bef5e98828538662","subject":"Add releasenote to tag the Tempest for 2024.1 release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cfdaf9a4b9c8592c08e64206bef5e98828538662"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-03-22 06:42:41.000000000","tz":-420},"committer":{"name":"Jens Harbott","email":"frickler@offenerstapel.de","date":"2024-03-22 12:58:58.000000000","tz":0},"subject":"Use stable constraint in tox to release new tag for 2024.1","message":"Use stable constraint in tox to release new tag for 2024.1\n\nWe are going to release Tempest new tag to declare the start of\nsupport for stable/2024.1. So that new tag use stable/2024.1\nconstraint in the tox env.\n\nFor example: if anyone use Tempest new tag in future say\n1 year later then tox env also should use the 2024.1 contstraint for\ncompatibility instead master.\n\nAlso, making ironic-tempest-bios-ipmi-direct-tinyipa as\nnon voting to merge his change.\n\nThese need to be reverted once release is done.\n\nChange-Id: I5ee1ce6b3ffbbb1f47e448fc15d540f74b21684f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d5f63dbe3332076ba32ab43740b49b647e182bc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d5f63dbe3332076ba32ab43740b49b647e182bc5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
