)]}'
{"id":"openstack%2Ftempest~914695","triplet_id":"openstack%2Ftempest~master~I68521a4ffcac52144308f17cd1062b99630e4dd2","project":"openstack/tempest","branch":"master","topic":"qa-2024-1-release","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-03-28 23:06:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I68521a4ffcac52144308f17cd1062b99630e4dd2","subject":"Update integrated template for grenade-skip-level job","status":"MERGED","created":"2024-03-28 18:19:00.000000000","updated":"2024-03-28 23:10:15.000000000","submitted":"2024-03-28 23:06:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"914695-qa-2024-1-release","meta_rev_id":"d788f77640e4ace7da76ee5d98ab85ebd5e9b5f6","_number":914695,"virtual_id_number":914695,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-03-28 23:06:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2024-03-28 21:16:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2024-03-28 21:16:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2024-03-28 18:19:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-03-28 20:34:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-28 21:16:06.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"af108ff0582ce966e846436b543d39204bc87619","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-28 18:19:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"219e60a90ba762dc11adfafd28a0b124ea727a47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-28 18:19:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"0fbb0692da79154a3d58687e88f0ed66fe3ecca5","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-03-28 18:19:48.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"5bc53a20a5288762f22843941b15309c3c879aa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 20:34:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80e9de86aed94fff92d6cd230ebe5f54\n\n- grenade https://zuul.opendev.org/t/openstack/build/73a18a65e9164691b966bb2a3e066a19 : SUCCESS in 44m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/430066615241410d861b4d2f57187e82 : SUCCESS in 1h 08m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/895f3c10bc4944ea8b09c5f1fa757371 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1887b5350ac641618dfe5d829a24ecee : SUCCESS in 11m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c89d4cdf0df0410cb297f41970119bef : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9927e527796645fc964616f5e4aa0700 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bffa7665eb8147ae9e21826964b20fa7 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/658dc357e38a424e91f6ad4480809ea3 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f8993dede66e424ba532b10a2e5ce53f : SUCCESS in 4m 17s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/dc87c7a99e164dcf8a7bbd7007f8c8ea : SUCCESS in 53m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a4f0556f55f246f1ad8ec03f3cc7c682 : FAILURE in 2h 01m 12s (non-voting)\n- tempest-full-2023-2 https://zuul.opendev.org/t/openstack/build/a203d1acdb0b4d53afb4c88ebb8fafbb : SUCCESS in 1h 28m 34s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/f3945756a79543f8a0065222ed4358c5 : SUCCESS in 1h 13m 58s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5d3412c178d94c6c8878a02d458ce97a : SUCCESS in 1h 05m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/19725e8447da436181bbf26a81120408 : SUCCESS in 28m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed30c790268d41f18f72613d0820676f : SUCCESS in 1h 01m 54s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2a3f9c1aa17941d5a37b63242badb3d9 : SUCCESS in 1h 22m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4060e02f88724e89a75f219820e437fc : SUCCESS in 50m 18s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/29b91e3326fe4565becf293762162d71 : SUCCESS in 1h 36m 43s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a7e8a4a8152745c29db29cc49ec85772 : TIMED_OUT in 2h 03m 18s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/13565cf88fec476ead6016446ec1965f : SUCCESS in 53m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/545202bfba7248ff8561a58274beb5ce : SUCCESS in 54m 52s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/6760c5cb1c304a0690a739e33b578041 : SUCCESS in 1h 19m 02s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b4680a8a154b4070aa48c723a840f1ef : SUCCESS in 4m 04s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5829e0839484487a87dddd8a116288dc : FAILURE in 50m 27s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/ec665b5697074790af963d6d4bdf4ab0 : SUCCESS in 1h 04m 44s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/246c71df49cb4691a3bab1cd84a4e9df : SUCCESS in 3m 59s","accounts_in_message":[],"_revision_number":2},{"id":"78ed182da99a53d93e592f1f419d34b252462bd2","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-03-28 21:16:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff9b321080c73234455468adefd0c36b24205ee9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 21:16:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9199033e44f92ac8f1d7fd51a13cc6c0059d2c6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 23:06:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35efa632c90f4cd4956523deff0eea15\n\n- grenade https://zuul.opendev.org/t/openstack/build/15bc17fa77464d2b89d7e26e16e57b6d : SUCCESS in 1h 07m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2b0ee9c6894c4e87970f9b72c1986ac4 : SUCCESS in 1h 38m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a77c88acb20641a1bc3e37d61fecd330 : SUCCESS in 10m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86c87df3009344c0903a535def6c6864 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ecfb692c687c4c52a0b65196da522c6d : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a567739bc91a4f51bf6fd6041d59412d : SUCCESS in 3m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/36a47ea05b574b02bd907520f2516b0e : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bb01320233ab4c68989497e0f760cbc6 : SUCCESS in 3m 24s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fc5c7dee02084e9eb93e72682011629b : SUCCESS in 1h 12m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/a0c1053affd745f89c886f3df75aa1e0 : SUCCESS in 59m 16s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/e6cd9dd97d7a45da8e095862b3b6228a : SUCCESS in 40m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5b079a61254d4b2cace7f1d515645851 : SUCCESS in 1h 00m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eaccfbaa124747c790024f9111162e06 : SUCCESS in 40m 30s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/325929f92f2445ffa2b053106728f323 : SUCCESS in 1h 08m 07s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cfc33355e00747d499c733864ebc0447 : SUCCESS in 1h 32m 49s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/93e02576b7ce424ab3595d36f83c0157 : SUCCESS in 34m 37s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/193c09b168754e9db70c9fc35d585041 : SUCCESS in 1h 02m 33s","accounts_in_message":[],"_revision_number":2},{"id":"6d996a29a5b9fa7d5eb43aefed434d38a9a8e4d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 23:06:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d788f77640e4ace7da76ee5d98ab85ebd5e9b5f6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-28 23:10:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da757ff5ff30465b93cbce8da982a726\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e40d0e0ee9f14e94a4e11bae3d9b51d0 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0765568e792fa21a7faa796e4e15111c627f2849","revisions":{"f62410a31850fb0a15f1e50942827c3489f86811":{"kind":"REWORK","_number":1,"created":"2024-03-28 18:19:00.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/95/914695/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/95/914695/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/95/914695/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/95/914695/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/95/914695/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/95/914695/1"}}},"commit":{"parents":[{"commit":"3585604be877c3a0ea9481a93617cfa728702eaa","subject":"Switch to the master constraint in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3585604be877c3a0ea9481a93617cfa728702eaa"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-03-28 18:14:07.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-03-28 18:14:07.000000000","tz":-420},"subject":"Update integrated template for grenade-skip-level job","message":"Update integrated template for grenade-skip-level job\n\nCurrent development cycle (2024.2) is non SLURP release and\nwe do not need run grenade-skip-level on non SLURP. If any\nproject want to run skip level upgrade testing in non SLURP\nrelease they can use grenade-skip-level-always job (for example\nNova does).\n\nChange-Id: I68521a4ffcac52144308f17cd1062b99630e4dd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f62410a31850fb0a15f1e50942827c3489f86811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/f62410a31850fb0a15f1e50942827c3489f86811"}]},"branch":"refs/heads/master"},"0765568e792fa21a7faa796e4e15111c627f2849":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-03-28 18:19:23.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/95/914695/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/95/914695/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/95/914695/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/95/914695/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/95/914695/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/95/914695/2"}}},"commit":{"parents":[{"commit":"16f6a0d55f880b4b9b4d7bb198f266eefe56daf1","subject":"Switch to the master constraint in tox.ini","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/16f6a0d55f880b4b9b4d7bb198f266eefe56daf1"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-03-28 18:14:07.000000000","tz":-420},"committer":{"name":"Ghanshyam","email":"gmann@ghanshyammann.com","date":"2024-03-28 18:19:23.000000000","tz":0},"subject":"Update integrated template for grenade-skip-level job","message":"Update integrated template for grenade-skip-level job\n\nCurrent development cycle (2024.2) is non SLURP release and\nwe do not need run grenade-skip-level on non SLURP. If any\nproject want to run skip level upgrade testing in non SLURP\nrelease they can use grenade-skip-level-always job (for example\nNova does).\n\nChange-Id: I68521a4ffcac52144308f17cd1062b99630e4dd2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0765568e792fa21a7faa796e4e15111c627f2849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0765568e792fa21a7faa796e4e15111c627f2849"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
