)]}'
{"id":"openstack%2Ftempest~916330","triplet_id":"openstack%2Ftempest~master~I595c3530461173fe3c9ba258aeb1ae90314874bf","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-04-30 11:42:30.000000000","reason":"Change was submitted"},"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2024-04-19 04:17:32.000000000","reason":"\u003cGERRIT_ACCOUNT_13861\u003e replied on the change","reason_account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-04-19 08:28:40.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2024-04-18 20:46:37.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-04-30 11:42:30.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I595c3530461173fe3c9ba258aeb1ae90314874bf","subject":"Fix constraint usage for unmaintained/victoria|wallaby","status":"MERGED","created":"2024-04-18 19:00:56.000000000","updated":"2024-04-30 11:43:43.000000000","submitted":"2024-04-30 11:42:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"916330","meta_rev_id":"3e5be37dbcf14e6a7070d05803bf3429478d7a99","_number":916330,"virtual_id_number":916330,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-30 11:42:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":1,"date":"2024-04-19 04:17:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":2,"date":"2024-04-18 20:46:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":1,"date":"2024-04-19 08:28:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-30 07:59:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":1,"date":"2024-04-19 06:32:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":2,"date":"2024-04-18 19:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"reviewers":{"REVIEWER":[{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-04-18 19:02:07.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2024-04-18 19:02:07.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"},{"updated":"2024-04-18 19:02:07.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-04-18 21:14:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-19 08:47:08.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"},{"updated":"2024-04-30 07:59:18.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"}],"messages":[{"id":"71d535c33137bf40c89e1d6e2c2afeb68a2a8bd6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-18 19:00:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a98b11ed893e2acf2f0099890d6e38b8239c0b7","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-18 19:02:07.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"d04d618294ba633ca0a1a34658de6a6a5d29f46e","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-04-18 20:46:37.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"81d30d2ca72052c15845ba4d08847fc474acda94","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-18 21:14:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a234d237630249aaa745fead044d9f25\n\n- grenade https://zuul.opendev.org/t/openstack/build/2508159004864c8d93f1ddc29e1af6a3 : SUCCESS in 47m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/34b17d14d891471db5dc2d4c68bcac59 : SUCCESS in 1h 40m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a706c15f78d462a81a8c18964b55890 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a96501c29cbd4a14ad18a3f1b0ec0d71 : SUCCESS in 19m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/613eb998ac824c44ae5df118135a4fef : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d4987fb405374a84be6c98e831e3f6de : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0f562bf222d647e5adc4f2e3d140b53d : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f63f8e5b11c146ce9bd9d3d22fc89a76 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c4f9ac9a2414374aeae57e908ba13ab : SUCCESS in 4m 03s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/fc4ded2c1f154c98b75e03d1fd4a400b : SUCCESS in 41m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5e7e612ecfce42d19f82099e7125eb6c : FAILURE in 2h 04m 19s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/2d9ea93514a54ee4a3ac6bbad89c0bf7 : SUCCESS in 1h 33m 05s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/438095af72014ff79840b2b3200050bd : SUCCESS in 1h 06m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/8d5e881ab91643ccb18882a29214bfb1 : FAILURE in 52m 55s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a86ad2785a0948c1888c3fb5adb4b1c2 : SUCCESS in 22m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5df082bf2f604301b584059999586b5f : SUCCESS in 1h 02m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ba0aef16fed24183b9d1d74f519e677d : SUCCESS in 1h 13m 22s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/599c57b4228b470786c5eaa78c0c3f00 : SUCCESS in 49m 34s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f226668a742d40659473e8e87ea11cff : SUCCESS in 1h 31m 46s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/591ed8d269734fd6bbc252c4ab23bd85 : SUCCESS in 1h 33m 08s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/96c11f8f227342608209a0cc22edcbf8 : SUCCESS in 1h 01m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e82c431a483f4cf3896a53acfb723a75 : SUCCESS in 1h 03m 35s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/24b13a359cb74a29a6a4ca6c51866f1e : SUCCESS in 1h 10m 01s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/ac5ae3c9fedc49fbb1af510d46f5f107 : SUCCESS in 3m 53s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/47656a80f38b423ba1db4cf866420473 : FAILURE in 1h 15m 50s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c3d3e035313844bba71b2404808d11e7 : SUCCESS in 1h 04m 39s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/238373edd6244d7093005280a750e807 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":1},{"id":"7f60bb25f65b59642455f8bdb7c502b4dddb4abe","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-18 23:08:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"343565d1f78412e3f68071c7a4ce45048f169d01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 01:01:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/777520b8d47646beb22330e08cc6f202\n\n- grenade https://zuul.opendev.org/t/openstack/build/9410e4f375364431832f5670bf11879f : SUCCESS in 1h 00m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68a43655ae894feda44348fe2542a5bd : SUCCESS in 1h 34m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24bde9fc281842df827072b53d9983d5 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/090a6acc58f94424bca37e6f17509418 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cf3dbd424ae4f46a62f4a55afc53df7 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/73dfe9ddf3fd43a29426d86db92b6ec5 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83dbe1cd13964d1d9fb37134402ee8b1 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38dad256b23a4153847640860b64bd10 : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25f5b3a6c5ab47eabab5c0826647cb54 : SUCCESS in 3m 58s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/cbc0cb6ce3d04880a254f9af4e9133c1 : SUCCESS in 47m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fdc439ed5d334dd0bee3cf0dabac1864 : SUCCESS in 1h 30m 32s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/504740b08e3c45eaa4a7ed8f3e7fb04f : SUCCESS in 1h 20m 49s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/8168c12a549348a1890ee1654646cf9f : SUCCESS in 1h 00m 59s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a75a7388276645f09febeafcdfcb31ba : SUCCESS in 1h 02m 29s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/38fd6701fdd64f2a908b3f3aa8872d16 : SUCCESS in 27m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f095a3d69a1a4d15a69166f17049b746 : SUCCESS in 45m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/3c7401dfb97c4d9c96b1990495f58fdd : SUCCESS in 1h 30m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0e9547fc2f80461da64c62fa2d58d27b : SUCCESS in 45m 15s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/707be42685ea404db7a990356f629667 : SUCCESS in 1h 17m 56s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9ace0802a4594037aafbb339b1a00ff2 : SUCCESS in 1h 41m 53s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7ca57ec0b3bc4b3384aea6386966d0a8 : SUCCESS in 39m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ed4701f9a37b4b5c87f91732e4dc7c6a : SUCCESS in 1h 01m 48s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/4751b69f33334a4981ade98f6a8501e1 : SUCCESS in 1h 06m 00s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/9de603cb40094bc49260c075aa873f11 : SUCCESS in 3m 54s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/fcb56195727e4c36b437eb22e536fc75 : FAILURE in 46m 35s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/2494bf5624a5461a86945c330da3868c : SUCCESS in 1h 07m 02s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/71376c2b308948d68a3dffbbdbb0d584 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"3a8e28b68b58ab1267411ec3f881bcb550e9a937","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2024-04-19 04:17:32.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e1eb085119926efbb69b3235d1c3f094271a8608","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-04-19 06:32:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"986c5d0a5934589bdb14e4c67995a8466f377d40","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 06:32:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"259af5989e2cdd257a5a7cf84b217c9c8f0f16d0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 08:17:23.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/538c6493f253437094afb505deb5ad51\n\n- grenade https://zuul.opendev.org/t/openstack/build/40f71998129946c38f86bc4583dc03bc : SUCCESS in 50m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/131cbc06277b4cc6b03dc40783aafa19 : SUCCESS in 1h 34m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d243fb84e544ecb8b67258b2a95f7d1 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/718a6b89bc634a1ea4b2f88cb80fc5d5 : SUCCESS in 4m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6961e7b4628044b6b0e69802b274a00d : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/823c7301110a452bb40ba9be972d1f5b : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a8f8b9c008fd48858dd897b0b163b727 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84267bd7434a4084a6463175f2ac9245 : SUCCESS in 4m 32s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/ff4f4afec0cf4748aeafedad8b2e5d84 : FAILURE in 1h 35m 54s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/5d114e9186ec480596516633f2d8d7ba : SUCCESS in 56m 30s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/bc5b62f3c2044ae49e72e65d3085905e : SUCCESS in 1h 18m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15f60cfb8e63408ebb46e2edf45bc305 : SUCCESS in 1h 02m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed436979597349039894d7d8334abb54 : SUCCESS in 57m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b7f36d2274dc43ce868905871904115b : FAILURE in 1h 20m 17s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/007fe74c52104f4cbba0be5d1186d6ec : SUCCESS in 1h 30m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/98740be5b7b84c818f0cd93690c8edc5 : SUCCESS in 49m 37s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/f4a0b8215eab419c9f9d66747a26e3bb : SUCCESS in 1h 05m 23s","accounts_in_message":[],"_revision_number":1},{"id":"4a6db9d140959f0660e155f08dfb36558fae002a","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-19 08:28:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09ba7d4d135cc2c83c993aaef60d9e0bf34e5017","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 10:34:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60b5b22c598f48fa88ca71c662866321\n\n- grenade https://zuul.opendev.org/t/openstack/build/c3ef93967d7e431b9746654b082705a6 : SUCCESS in 1h 00m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3684bef6285c4303bf1e99233e329cc9 : SUCCESS in 1h 11m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4421ecfeaaf49848fc2bcf91153e669 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f5445740466468e9f76010f52f3a308 : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/874b3ddae31b441b892f9293c7ca581d : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f64b0981c5f4b1dbffd4501a5c24339 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74d4548429424dd1b9819be8089a0c4c : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07b4ba03e261417a8bb9e97cda25cf75 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8797421e3dab4c37ae4295ca703786bd : SUCCESS in 4m 53s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/9880be2116ee4c69810837303c45a3c2 : SUCCESS in 55m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f4b60d084a3d4f8f92b0a7206aa150e4 : SUCCESS in 1h 48m 19s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/7297aec81c3b4026867212359791d863 : SUCCESS in 1h 21m 25s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/2334e803d39b4b578a7a5ff9367217ac : SUCCESS in 1h 17m 36s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/5f030ed09d9346b3b7f74f945ddf2e0f : SUCCESS in 1h 12m 57s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3f1f67c7e484425693bd2a54adf61356 : SUCCESS in 29m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03f0120b90bd4a19b6c3734b0e410832 : SUCCESS in 46m 57s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/d89ce35580fb475eac846b35f4bbe562 : SUCCESS in 1h 22m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/457827f1cbe34e419b57784f241055fc : SUCCESS in 50m 46s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5b6e0ca4e1d64f5883314274aec135e9 : SUCCESS in 1h 31m 51s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/473d7f9da8dd4c4fa15d7561aea787b0 : SUCCESS in 1h 51m 20s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b817ad6820ef4782a6e3949f10886aed : SUCCESS in 44m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/757dc656ee7c4a46b9a84a62e57e680f : SUCCESS in 1h 02m 43s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/5ba5f775986d4e3e9f62c0d2719aea74 : SUCCESS in 59m 38s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/3bf9d2e66c1b406988ac862beefa3763 : SUCCESS in 3m 47s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8ed3f42c6d534fe78ac7b885a2a539f6 : FAILURE in 1h 19m 52s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/43d78bab61444e42a2230e1c1ebe48c7 : SUCCESS in 1h 05m 03s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/564aefa6291f46919a81b52dbe7f678b : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"578160c1d4503afe0f8312eaccae14a164bb459a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 10:35:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a932fe8390d999149b787dbcff4fe91746893be1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 12:20:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1119d25664fd4d33948e317ecf6c7993\n\n- grenade https://zuul.opendev.org/t/openstack/build/79b7c87f6c264d8491c141f48d34a1c4 : SUCCESS in 1h 03m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc98e5bb03ff489681fa5e4bdfc601e8 : SUCCESS in 1h 12m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b4e45eb0f9841138eefed00f6358aba : SUCCESS in 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d9d6acb12434618aebd5266c00c8ab8 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9be80084eec4d5f8c2541616db16a1f : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c79524cf2d6a4204ab123caba7f0f8bd : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bad5cabafa0c49eea3db94bcff1954be : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d2ce615056a42ae840d273d0b8a16a1 : SUCCESS in 4m 27s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/76450f838769469fb9144a97d26cf1bb : FAILURE in 1h 36m 13s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/24a58603cf1549a5bda70859be595aae : SUCCESS in 58m 39s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/f3070a26e9e04e4a8d5d36413c04e4fa : SUCCESS in 56m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1c5b448bad4d4a8cb05b6453827e7f7c : SUCCESS in 1h 01m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1e2a0b663b44edab7e51b8260261df7 : SUCCESS in 37m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/314a90879c354681859213e8a99b5c3d : SUCCESS in 58m 39s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/83ef728efd48413097b82113d1f743ae : SUCCESS in 1h 23m 04s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b9d068a8f7d946419e64b24862944411 : SUCCESS in 36m 30s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/f1befffe3fa044b7b72fa5d3dd18b977 : SUCCESS in 1h 07m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0abc9ceb41198f392209608beff480a7f0d75f71","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-04-19 12:47:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7644e21e5d0f4bb045eec673c122ee5539d47e7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 14:52:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b71fb0eebd04803b285e07064019771\n\n- grenade https://zuul.opendev.org/t/openstack/build/29db78566a604e41be234d9033bf94d5 : SUCCESS in 52m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d6787a5d596649afb9972cbd5e8f3040 : SUCCESS in 1h 28m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1035e12ae8584c9ba7e456153ce900fa : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7694f65254f45ebb467b87a68ad250c : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b570e0d153d4a568d5a1db566398b44 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0bda304a279b4d3c9a101e6c9d9011a1 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe0d54fbe2984ca381ca21aa1075d1f1 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/782a4c6f05d94936a68f57f9e95154d7 : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d583301a1ffb471fad8ea90ca4aa9db6 : SUCCESS in 4m 22s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/7fc1e38a1d694e92b57479efd8adda62 : SUCCESS in 43m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f7a7109c8f374964ad5e3f49bb0a1cd0 : SUCCESS in 1h 58m 14s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/e61bd92e332b49c1943cddfa76bf012e : SUCCESS in 1h 27m 54s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/1125e57c8da84b9a8d4761832d0f25d5 : SUCCESS in 1h 10m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/589485953cff496aa35c9337196eade0 : SUCCESS in 1h 07m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/b463bd8506a846749c3edf22ae314e76 : SUCCESS in 29m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aac43da06ff042b980166434eb2133e6 : SUCCESS in 49m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/8d0c45340bda4880bc7badf900798e6c : SUCCESS in 1h 39m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/69ee981498a541d890cb65a6beef377e : SUCCESS in 45m 01s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/85922f45e4a44fe39a2d2189f6f1cf11 : SUCCESS in 1h 33m 26s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/38e9c2b102644f2db8d625919b32d5d3 : SUCCESS in 1h 54m 02s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c981d57bede24bb599a1a20e90fb2c02 : SUCCESS in 59m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c6180e78ecd9436d94519f3b6e48242f : SUCCESS in 1h 00m 12s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/ced15f40800b41318c671f9ca27e679a : SUCCESS in 1h 08m 14s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/b2187a1105224bc1a9fdf9f57c7dc174 : SUCCESS in 4m 34s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/9a92531a86df40cb9fad2f7e0bf06f9b : FAILURE in 1h 03m 42s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/48d607fd52ca4ca68d6af914987ada9e : SUCCESS in 1h 07m 08s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2430ed482c31426fabb9b228081eb912 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":1},{"id":"a0035802ac3183cc1ec44c50e571441a9ef18249","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 14:53:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3df4d4c3e05c11ce58e6407c96c7ca327a733681","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-19 16:43:43.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/81e7ca26bed9404b89fd63d2aaaa3f89\n\n- grenade https://zuul.opendev.org/t/openstack/build/8b19120bd05a4eab94961962711d655c : FAILURE in 37m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2359c69adfe40d0b25e18c7903c8234 : FAILURE in 1h 28m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c73442f13d2d4e4d9b550baa34852ca4 : SUCCESS in 13m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7b1fa52c86440e09e1581f305f3cdad : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8cb56ef9c5f425e916d4d87582883c5 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/137a3a29dad64fbb851d236619427c24 : SUCCESS in 4m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a18d65936b2b4240914b12a6348e152e : SUCCESS in 6m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/328f815f436047a685670697af7c2a8f : SUCCESS in 5m 20s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9b73df61b9dc49fc83e9d24bf242c63d : SUCCESS in 1h 41m 56s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bcfb525a513143629c2f2081b845e28f : SUCCESS in 54m 54s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/886d332df73942778211f846ee9c3dc2 : SUCCESS in 1h 05m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dbbeecd3cb094153807a9f4b3ca5f8ec : SUCCESS in 1h 01m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e68863c93b294489a25477a221feba6c : SUCCESS in 1h 08m 47s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/10c272dd968b440aa20029a4adbae683 : SUCCESS in 1h 08m 36s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/02837f16e6824855932770ee2f3b9c9c : SUCCESS in 1h 38m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/076f569904ef437096fae1017dbf170c : SUCCESS in 1h 01m 42s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/b2da5f741d3f4101902bc9b32a312709 : SUCCESS in 1h 01m 42s","accounts_in_message":[],"_revision_number":1},{"id":"f2c8e0d498b9b5c4aa4cacf538dc6f42151d04e1","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-22 11:47:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0ab97fee88de0f0211338ba7fd8cc1b38a288b7e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 14:05:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/17af62d0cfb64829b227915d13b2ae9f\n\n- grenade https://zuul.opendev.org/t/openstack/build/45fd8cc890734671abfa681be60ad6d0 : SUCCESS in 1h 01m 05s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0f8b3f5e83f241b4a1e627a43b3291e8 : SUCCESS in 1h 32m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b4b539f558774daa8bc6a0807338e7b0 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a657bf229344aebb6f251d52d7dab92 : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35eb51fb02e34b75ae78522ece9acef5 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/836cef9f5ef54bb49db0948dd3ccae4d : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adc5a120639541ea9c44abede3052f0e : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92dab4e7e320489f968b666a94a67346 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6471d45997224be0814893158da327bf : SUCCESS in 3m 46s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/4106eae70ca842c4afc710eae4fc6f33 : SUCCESS in 56m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9b1271406b5e469d8f6e8b0fa153e2cb : FAILURE in 2h 07m 26s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/a19eae67ccc2417692d15c9b86003366 : SUCCESS in 1h 19m 26s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/56a0abbb0b8c4725be803f9b668d184d : SUCCESS in 1h 02m 54s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6229a29c951c46ee8a07f54be16e32f8 : SUCCESS in 45m 57s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/6f47c17ade604b88a7fdbca243d6b88c : SUCCESS in 31m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/54398da5423449ffa61dc54162a5c101 : SUCCESS in 51m 07s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e68cf8690ccf4ef4a4e45505e2793a0f : SUCCESS in 1h 38m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/8539b31a45414653a66e3f686d7bd779 : POST_FAILURE in 23m 53s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/020b286404ba4e9c822e632fce3882b2 : SUCCESS in 1h 38m 33s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0615466e762f4763b47d4f388a5f88fe : SUCCESS in 1h 51m 42s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1f984437456b4178ad0e0a5641f81786 : SUCCESS in 56m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/368c9638f2b74abbafc57d03d4694ac1 : SUCCESS in 1h 04m 35s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/e3087594f10541f986e9eb7eaa60c88c : SUCCESS in 1h 00m 08s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/3b7f2053e1e04d82a5838d18930bc4be : SUCCESS in 3m 38s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/bbc2e0c17faa4318be5c18f6c36d6595 : FAILURE in 1h 21m 47s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/83c5e8f7cdb0485abe5328f2875f84a7 : SUCCESS in 49m 21s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/bb20c4c7199e4e61be4738c3cf94daab : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"37d83efe2ed5313eeed9bdee240d96e23f903bca","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-22 14:32:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bea91466bcb457a0863496671b3e60d61108cf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 16:52:56.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5db21054689b414898889a0b6481d621\n\n- grenade https://zuul.opendev.org/t/openstack/build/48e6f6141d5446f398d086b0b0bf0401 : SUCCESS in 49m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/53aa8fc3be334b51bfacb631f4671378 : SUCCESS in 1h 12m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6879c67a9304a1686b2c387fb5a4f81 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9153d2171ac84914b38e150e00b7e92f : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ce301b9e24b448e8efa09b2160882a7 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d92a52f80aad4c3c987e8689315a976d : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fef2c3c135314faf9dbca62c3df9c466 : SUCCESS in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6013440e142d47bca441bc0d75a7dc93 : SUCCESS in 4m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1b44f568cb4a46548b204bccc49c7607 : SUCCESS in 3m 56s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/9824570c331a476485d17432738f6ae0 : SUCCESS in 45m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f3bb3333ec7b452289548c5fc3e96ab3 : SUCCESS in 1h 58m 52s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/55330448b896482e8683109b5d8d5ad0 : FAILURE in 1h 21m 57s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/804318d3462846aab31ba1187621247a : SUCCESS in 1h 15m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/cb10a512759b4513a6500f4de0632293 : SUCCESS in 1h 11m 12s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/2253d7d110d14c6680482f9a23c84361 : SUCCESS in 32m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94478e4b6c4b4de9b6f01e96568f6bab : SUCCESS in 50m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/67f22545e1934c3da9921014e4ff1b35 : FAILURE in 1h 35m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/a22a8058e9884bc4b918beee71c758e6 : SUCCESS in 46m 27s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/68539f9f7bfb4c74bf90176396e52c8c : SUCCESS in 1h 19m 21s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1bd24eb1266443a8b97022ae84125c0a : SUCCESS in 1h 56m 14s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/572edc306922405f8bfe479a70525166 : SUCCESS in 53m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f48611819e8c44729872b4900c3445ff : SUCCESS in 1h 07m 14s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/1e78249ab3ae409fb1126859359e96f5 : SUCCESS in 1h 02m 21s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f90e4fbadd8e4d6a886fe671b17ac855 : SUCCESS in 5m 13s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5e8d358dd9d1417d942d58201dc9f3c3 : FAILURE in 1h 14m 52s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/bc80551fe45d4367ae3413e5c09bf6d1 : SUCCESS in 1h 08m 16s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/37fed2bad60746218f691c24a57da2e4 : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"a3bca93e355eefb745ba29b5f6ff99d42e3a5ee3","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-04-22 17:10:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d8bd36b205ca2c4b110c9ecb1f6af760da4329f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 19:10:01.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d28979be677d458687ce41bd8b5d8771\n\n- grenade https://zuul.opendev.org/t/openstack/build/d2132e1b19b04497910714685cb7ee0b : SUCCESS in 59m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/15e88bd1bf3c4783a56c99cb6a88a981 : SUCCESS in 1h 38m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d061a00c1ee148cfb554e62773dd94d1 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2bbec8c082d452fa1457eb3984180cf : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30eeb0a491424bf9a3985e915db5aded : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0adcd72e4ef841dd848954bc1370bc7f : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65d73c913acc4102bb8b78320d06b491 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6e090dc30f44f2582bd7e4f174b894a : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d359b57b67fb4ebeac0e454226f93d6e : SUCCESS in 4m 54s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/389beb23521842f585207253e4fcf91f : SUCCESS in 55m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ca0e8bc1cbf0493b9ab73ce02184fda3 : SUCCESS in 1h 58m 24s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/b34c778bba2744cfa381778dca2f922d : FAILURE in 1h 18m 26s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/31693f138d09444e8b4f5f3a1d38e9e7 : SUCCESS in 1h 20m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ccac2d9ac0604aeab7e9b2a3dd4ddaf8 : SUCCESS in 1h 04m 34s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/759f3532ea67401f9b0496fbf6e4c5eb : SUCCESS in 30m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/24a201b629464c329fc716978240fba0 : SUCCESS in 56m 41s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/db4e69b82ad247a9bedea853bea66462 : SUCCESS in 1h 51m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/544be59c953b4bdbbf05dcb2178573ea : SUCCESS in 55m 43s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/af677c02be3d44b49d2f0b30727edec1 : SUCCESS in 1h 14m 07s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/daf128ee620740b397afb78bfa5b5f9c : SUCCESS in 1h 35m 05s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f034c5e9ac604df8965bc3e4fc4e3f0d : SUCCESS in 58m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b835a8fb9fbc44479278e99ca3b1be36 : SUCCESS in 1h 00m 11s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/66c9fd85324e4084ab742e11ec4d9578 : SUCCESS in 51m 51s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f7afcf2f47ed4fba81b50c3fb6201f50 : SUCCESS in 4m 11s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/4910d1c9034c42a598a320c771806879 : FAILURE in 1h 22m 22s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/4f7d01c744b940acb475bef33eacd59d : SUCCESS in 1h 06m 15s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/4f8af45659904458a13d88d10dc6aace : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"bd97fba5b1e84cf0429bbecd235c1e91383eb9b0","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-23 15:11:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ae66ad09266aee638b4ddfc6808b1c4908abde0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-23 17:21:12.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4fa94713689749449a99aa681ab11d0e\n\n- grenade https://zuul.opendev.org/t/openstack/build/efd76505bb7e45db9c23bba57f0b3827 : SUCCESS in 50m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/851e74770d0b45a2ae7f43958c09be3b : SUCCESS in 1h 20m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5d8f61b5dee5404099f7ebe8f1bb26ce : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487e7199cf904f77a5a5610824d5f9c0 : SUCCESS in 11m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f78b3ae1edd48d8a485bc7cadca7766 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6cf7eafcb7584091907a3b148811d335 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e29f854a4d81484da8fa7aad6f01d6f4 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2969dd72ad34f0fb6287c44c07b7be2 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/340ae43685564d89abf17e1a813e5e30 : SUCCESS in 4m 24s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/c3a046a5a1d642f2a8a30f07d837c4ef : SUCCESS in 53m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/740ff03872944a68baba718bbc1bd047 : FAILURE in 1h 34m 43s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/fa0dddf3a6754cc297e5885c5ce9d832 : POST_FAILURE in 30m 16s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/88a682f2ae164dcd9e8ca7dc12993ba5 : SUCCESS in 1h 18m 53s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/dd66846eda934da8be697ece3cd12ac0 : SUCCESS in 1h 21m 24s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f6ed7405e07043b8887a2061289839cd : SUCCESS in 34m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a858fb9ccc994852a25b181dda599065 : SUCCESS in 54m 38s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6dc9143e200421f94846f1f2f895a16 : SUCCESS in 1h 21m 11s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/17fadf228f8347f1b1230540816ce2c6 : SUCCESS in 34m 19s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e238c0ae26034dc3acb7b67d303e3d51 : SUCCESS in 1h 19m 27s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/b1aea5c2507746c0a999ec3f1151b059 : TIMED_OUT in 2h 03m 41s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/dba763470a43402eaa55b41154efecfb : SUCCESS in 59m 34s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/adb51b6a08134756a2168d16c69e64cf : SUCCESS in 1h 04m 06s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/6bbdd83673d84ccaa9a9f8b35757171e : SUCCESS in 1h 08m 55s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/92027090457b4e109ef8ac63aaf84ac9 : SUCCESS in 3m 44s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/d607b1cc097b4a6caf2ee60cfce86412 : FAILURE in 1h 17m 26s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/aee5fdb0787b43c6acf033a4135ee856 : SUCCESS in 1h 09m 00s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3f4ae5e08c124065bdd612ccf7090957 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"9a36d365266c380f90501cb0ecd6528176ee377f","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-04-24 07:12:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"60d2bf18e92571bd5628951ebd9b4e3cf2c8a6c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 09:14:18.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b6b8c10edad42639a9ad2452b4860ed\n\n- grenade https://zuul.opendev.org/t/openstack/build/43281335554f4cf4b62ba64185a5bb48 : SUCCESS in 59m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/dec66162eaae4a03a7999624f08a6e54 : FAILURE in 1h 19m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/934dcf91f3874e418bd3854b8a3052df : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8edab294d17840e3baa5c6e44415b911 : SUCCESS in 11m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80aaa74ef6784ca0bc962f4ffd69c9d5 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/62604aea786940eb9c1347eedacc292b : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aca8bd2b6ed64af7b34f0eab36738816 : SUCCESS in 6m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fc26df80b1348c394841e7647844fe6 : SUCCESS in 6m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/35be1923e5e8443b9d4e5384353593fd : SUCCESS in 6m 22s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/92a764cc0681491eac99144e23a13810 : SUCCESS in 53m 36s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c884b430c61a4792a18634e2c23ba049 : SUCCESS in 1h 54m 16s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/a701887482704e9ab8b0e09c065c781b : SUCCESS in 1h 28m 38s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/44b4ab40841542c1825dd4a0d2eb326d : SUCCESS in 1h 00m 01s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/f8e7e2153105408d832514daa259c0f5 : SUCCESS in 52m 30s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/696ee398e41d494abf1a3d49b8735398 : SUCCESS in 21m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef3e3e9c474a486da8756d59800596d5 : SUCCESS in 1h 03m 42s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/f6c6afb9d91146f1875512837c9f8a33 : FAILURE in 1h 38m 29s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/dd4b124fd50740789cf35f1ce0e5afdb : SUCCESS in 38m 09s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/569d589a1d0e42c4bc439f443f4df5c5 : SUCCESS in 1h 29m 55s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/2d1c4ce21865471db6476764cc15960f : SUCCESS in 1h 33m 00s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/78b200c50b9e4f54a38fc20b4502b0e7 : SUCCESS in 54m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2df7f44f95b04520b2db32cd86acaa13 : SUCCESS in 58m 00s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/d2d246b9471249dbbe95514816a6554a : SUCCESS in 1h 08m 58s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/018e4323ab864e188a047aea473d5bf9 : SUCCESS in 3m 45s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/24743123d21843969301d8e17d913408 : FAILURE in 1h 14m 44s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/0bad3cad6e1f45c48df9857d08c20934 : SUCCESS in 48m 44s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9f129de3ddf04cc68201165e570bc5f2 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"d412c414ce1ff317d5c8effd6c70db2bc18efb95","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-04-24 12:57:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f44a2567b4a3d55a1ff4708b1879b8b090ed73b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 15:03:24.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/27264c2fb45545bbb7f0db04b110edac\n\n- grenade https://zuul.opendev.org/t/openstack/build/7e70105959854227a17ccc635b0d26b7 : SUCCESS in 1h 01m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2981a6727b0348a280d871d07b11da40 : SUCCESS in 1h 28m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/939f1cacc8d44cb1b85d99e5333563ee : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28f534b4b5e241a996144234e3afe6a9 : SUCCESS in 12m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6278de6f8dda4b2db39185fadbe52099 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d162838899174271827c33857e43f3ce : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a692ed5423934632bb09c24ecf296831 : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72b3cee896b44455adb757f1d34cd33c : SUCCESS in 4m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7930f55198604566820946bd118f20a5 : SUCCESS in 3m 57s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/16d377e7daaf450885d25f8e7c69119d : SUCCESS in 1h 10m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2661362c04164662b3ce98f55721be5b : FAILURE in 1h 58m 00s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/6b80e59a69a2433dbd7801fca7f2842b : SUCCESS in 1h 12m 43s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/c24901e68bc242fc8c9fbe02962e3339 : SUCCESS in 1h 18m 34s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ed976d70f68f4ed7b483de5d1bcbbb6e : FAILURE in 1h 07m 23s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/eb9eeb27ff7a4cb5b0ef82e92693277a : SUCCESS in 31m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a0689869dc04ef4be4c392f7d7ef322 : SUCCESS in 47m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/6b1ee68037a3480681d6154cefc580d9 : SUCCESS in 1h 31m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/7f45b7bdba824bef9efbe6e9ec2c587c : SUCCESS in 41m 07s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bdad74e868bc4e018379c168b6ece56c : SUCCESS in 1h 40m 13s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0e88445f21cf4f6cbf47e0defee9d481 : SUCCESS in 1h 52m 06s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/e40fadd71fd04283a85eccc207880357 : SUCCESS in 48m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b83b387d25c4420eb5a20e3eb0ec39df : SUCCESS in 1h 15m 10s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/48cabac2bd9d4b9da3ad1a9cc07c73f6 : SUCCESS in 1h 21m 27s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/231127028f144d4eb98fa10d9a8495b7 : SUCCESS in 4m 06s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/00e404ab381c40d0986c47373842e723 : FAILURE in 1h 16m 00s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/1f11fd3c858940d2bf3068ab73fafb67 : SUCCESS in 1h 03m 47s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/2cf09601acd748d9a3c748828f7fe774 : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2ae843cc1f81a20ce68f97f0ea68698b4fd3be1e","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-04-30 07:39:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c3973728a728396309eb20dcb54227bdfc487d8","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-04-30 07:59:18.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"28def5e5c16a373441a6fbd9e8f1b26fdbc9babc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 09:49:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/884e8e9623c44e2bb0bd2423be50cc26\n\n- grenade https://zuul.opendev.org/t/openstack/build/0daaff39e405457f9a7ff5b54066aff1 : SUCCESS in 59m 32s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac9ade2698ed438abe500d727ecd72f2 : SUCCESS in 1h 32m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/01dc31bfe98d4796a22091fc1ce5004e : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f4c7058661149ed99e56b1af88f2940 : SUCCESS in 10m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bf5904623e14e8facdc59dda156c816 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6fa5bc278bd4448b0808a2db374a26f : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39bd0b2a5ed14a83b14862b48f8606df : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f29e055269734e2ebbbdcf0f091d5d83 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a787b1f12f0b40aa8452004111800e7a : SUCCESS in 4m 20s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/da4b2eb8b3ef4db6b563da005ad193de : SUCCESS in 45m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3d25d5e6a4e24850bf091d3ec19acbc5 : SUCCESS in 1h 52m 42s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/a78a9a4137b84c19adda996999da4a4a : SUCCESS in 1h 32m 08s\n- tempest-full-zed https://zuul.opendev.org/t/openstack/build/319befa0364c4867b8648c359ac1a8f1 : SUCCESS in 1h 24m 50s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/43f987acb4bb42508c6f6afb68890131 : SUCCESS in 53m 26s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/42d51188f52543bf8a93d125dd2a1787 : SUCCESS in 31m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/371883e513be40a3917a485bf83cae6e : SUCCESS in 58m 37s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/780418cefb6a4f3b83edc7980b5de39e : SUCCESS in 1h 14m 10s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/f86b8dd458d7454ab82ce96117afb9d2 : SUCCESS in 50m 46s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0035278e716f4d51b4010004e860c599 : SUCCESS in 1h 37m 24s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a0b5a1b9e07c4f989787fa0b01acbab7 : SUCCESS in 2h 00m 55s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7651f540c54a4cd6b4b918dfc6e863bd : SUCCESS in 46m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8bc355f482074771845d3849f8525ecf : SUCCESS in 52m 53s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/726c40bde8d54173925844e9fe9acc98 : SUCCESS in 46m 41s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/945cf516d6a84bba8bb0f04ef670024f : SUCCESS in 3m 46s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/993142ca89124db0b95db4d3ba6a77b9 : FAILURE in 1h 07m 21s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/064e74b986e643bf8fb94496850e2e40 : SUCCESS in 52m 50s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b3caa0570e6a44ecbc6f41cffa107e82 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7401d33d558716754340dc48429365fa5c4c2142","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 09:49:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"12b508e7c3a184a298fd1cf98260173f9e878981","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:42:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fb4a6ac39c54d90bcc6fe61faf42b26\n\n- grenade https://zuul.opendev.org/t/openstack/build/3f080912301248ddac2b3ecdacb3d6c9 : SUCCESS in 57m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d5c2c7ef2325471e8d9fb713f7d80694 : SUCCESS in 1h 45m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccf9484c58264688b95c4c6d4de02ac0 : SUCCESS in 10m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a8d039ff4744fd82a19b32e2fa1f8a : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b164c60b7e64f3ea855cffcb2e48511 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8f507afe56d94ca3b2f8d83bc2d61455 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd5eda29795a40dc930d7193fca4bcab : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5699436eb22449ae9a04093b9724858b : SUCCESS in 4m 00s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/5b0742f1c1404d49b992a14ecca95d54 : SUCCESS in 1h 24m 41s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/ff1d9a2a12da4a88b6eecf286aacddb7 : SUCCESS in 1h 11m 47s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/936a0d983af6493d9629a0c54e521e9b : SUCCESS in 52m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6b9824d663704d4f91413709463694d9 : SUCCESS in 59m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d221e63fad2c40d18d2e26ed804e6e8e : SUCCESS in 1h 12m 37s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c1cd4052e4fa40468e492ab3609e0b67 : SUCCESS in 1h 23m 09s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/df41f867be514773ab62be105693bcc8 : SUCCESS in 1h 30m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/d118a3d7b83947ff9c212eaee6a9ee4a : SUCCESS in 47m 17s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/49334a8cbe6c4637984fa480117f541f : SUCCESS in 1h 08m 11s","accounts_in_message":[],"_revision_number":1},{"id":"38ce3b2f65f5fda619cc8ef72c2b3a9eead60556","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:42:30.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3e5be37dbcf14e6a7070d05803bf3429478d7a99","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-30 11:43:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87b4128d6e304ef5be2b60f0d4cb44b3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1aac0213439a4a54ae37f1a9ba632877 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b42fc4905892ab6b139a68d535e4a55a2dd5716b","revisions":{"b42fc4905892ab6b139a68d535e4a55a2dd5716b":{"kind":"REWORK","_number":1,"created":"2024-04-18 19:00:56.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/30/916330/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/30/916330/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/916330/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/916330/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/30/916330/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/30/916330/1"}}},"commit":{"parents":[{"commit":"c0da6e843a74c2392c8e87e8ff36d2fea12949c4","subject":"Merge \"[codespell] fix typos\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/c0da6e843a74c2392c8e87e8ff36d2fea12949c4"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-04-18 18:59:28.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-04-18 18:59:28.000000000","tz":-420},"subject":"Fix constraint usage for unmaintained/victoria|wallaby","message":"Fix constraint usage for unmaintained/victoria|wallaby\n\nvistoria and wallaby branches are renamed to unmaintained/\nnamespace. Fixing run-tempest role to adopt those new name\nfor those branches so that it can set the correct constraints\nwhile installing the tempest.\n\nChange-Id: I595c3530461173fe3c9ba258aeb1ae90314874bf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b42fc4905892ab6b139a68d535e4a55a2dd5716b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b42fc4905892ab6b139a68d535e4a55a2dd5716b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
