)]}'
{"id":"openstack%2Ftempest~920011","triplet_id":"openstack%2Ftempest~master~Id96bd9b30598844457963b842f741e5f4bad7c74","project":"openstack/tempest","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-05-20 08:46:44.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Id96bd9b30598844457963b842f741e5f4bad7c74","subject":"Increase swap size of tempest-integrated-storage","status":"NEW","created":"2024-05-20 06:11:39.000000000","updated":"2024-05-20 10:54:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8c06e8d0707dd20a06826ab468e62bc0e55ed875","_number":920011,"virtual_id_number":920011,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-05-20 10:54:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-05-20 08:21:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"08a997c1795b81ed84186b2bacad8a14ff8bd2c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-20 06:11:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b48da46c1672c05f1c25a9e9dca896b5f6df1a12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 08:21:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8c22dc866254d248bb0cc51aa639f2a\n\n- grenade https://zuul.opendev.org/t/openstack/build/68826b83750d46f39dd1b03f92966866 : SUCCESS in 1h 05m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1c23e94eb39544c8a07d0b4b4951c227 : SUCCESS in 1h 41m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9f591e0a59f480e8800880e7248746a : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ac543e0b03f4b6d99785d1acfdaff61 : SUCCESS in 12m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/560bfc420f044602b88d028b9ff11c67 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/febe457ec4ef4099b3ed8afbfab38793 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7abcdf7f8f1f4a69a339c8db404986ee : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21c24ec23f9148a1b637e1b26f264636 : SUCCESS in 3m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5c22c6d3f414436b3d2b74efd252461 : SUCCESS in 3m 41s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/456386dea66f4cd18b18006d1a199c2d : SUCCESS in 53m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a44620268cc64ac296bf9036b62cbf58 : SUCCESS in 1h 50m 11s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/75c86cee8e7744e0b11a750512e97d01 : SUCCESS in 1h 29m 55s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/3e87e3cbc02d4d96b202d3c1e1e3f8fb : SUCCESS in 1h 00m 31s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/3878da2cc62345dcb013f9c330de0551 : FAILURE in 58m 12s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ad4a8eb1f15e4742b043c5efb90553b5 : SUCCESS in 24m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc439de8a8494a82a108f102e424b709 : SUCCESS in 55m 21s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e6fcdcb127bc4ec2a9e0f8b53e0285a8 : SUCCESS in 1h 45m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0929fbb62b954f1fbc9f2715ef232d69 : SUCCESS in 50m 23s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7dedd41073464e0a813a56a28e212d6d : SUCCESS in 1h 21m 35s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/54456b06eb3e45379e344b73b77963c8 : SUCCESS in 1h 35m 38s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/b157b7fcfb7341dc966b2b77ef64729b : SUCCESS in 59m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e97fd51aceea4ce6ba5ea13b09a270b5 : SUCCESS in 1h 00m 35s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/c5682075ba1f478bb67b61e18f76cf3c : SUCCESS in 1h 11m 50s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/94b4e19e6aef44d08107f7e393f52235 : SUCCESS in 3m 41s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/1b56978216c44ab5a87363393f25f8f9 : FAILURE in 1h 08m 42s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/658a6c3382e3467e995363423ef136ad : SUCCESS in 1h 08m 16s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/5fe68d5827fa4a4cac941a9782728552 : SUCCESS in 2m 41s","accounts_in_message":[],"_revision_number":1},{"id":"d276c786329e817168cdbddb4a3a4c0cfffdcbfc","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-05-20 08:46:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c06e8d0707dd20a06826ab468e62bc0e55ed875","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-20 10:54:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcb7878580174eaa8f1db9e22f4b4b95\n\n- grenade https://zuul.opendev.org/t/openstack/build/25a9cae316d54dc393a813b5fdf2a47b : SUCCESS in 44m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b6a89680651f461cb27b31c144795ce0 : SUCCESS in 1h 13m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26311d4486724b72b8852ec932023837 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72baa20f9bf447c3806b6a8ef378829f : SUCCESS in 12m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89a7e62c7f114671bcf817d7cecfdb3f : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4123a67fbc9a4ec1a43a1625bfa9c887 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba8a62db4c7447c2898104d9ecd29efc : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2954302a696245a6a33130d7756e2ef2 : SUCCESS in 3m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6a91c7a84c7b413c9be837088f1705a6 : SUCCESS in 3m 35s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/ec4a9af154fb4a54a74c05d81e19f8c6 : SUCCESS in 56m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/269c36730d3648e2afcd2c5003c33381 : FAILURE in 1h 58m 06s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/48bc1535fe614ce4bc0dad9419ad91e1 : SUCCESS in 2h 00m 14s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/0c422679593848dd97318a532dde34c2 : SUCCESS in 1h 21m 17s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/52e48747d0f94b86942d81684e3da25e : SUCCESS in 1h 15m 55s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/9eb1f89711b04b72a090e772a708e66f : SUCCESS in 23m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ca7f5988d7604b7caad49e368ee5d282 : SUCCESS in 53m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/0d85d3c6517443f78b23c01e74b8f53b : SUCCESS in 1h 38m 03s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/ba987c028b4e4bfeae6b7f1bea3d4821 : SUCCESS in 56m 45s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1c20e36066a44c13a70dfc14613eb381 : SUCCESS in 1h 34m 31s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/0673d4092a074f60acc998739b282904 : SUCCESS in 1h 55m 15s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/1a309c7a420f45329b95cd2f2effcc97 : SUCCESS in 1h 04m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/180547992fc84739896c2802c703abb2 : SUCCESS in 1h 04m 52s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/f498198750f649599170605617692e54 : SUCCESS in 1h 12m 11s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/f7d860dc6a3c4d93b9f527344ec415ec : SUCCESS in 3m 41s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/6aed79086f8346069c7fd2165a03c620 : FAILURE in 1h 24m 48s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/90617781a9484db48c7e1b2fcb4bf273 : SUCCESS in 1h 00m 11s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/0a9a30d029084f4e91d9db9841ffbda9 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1eca493977d6300caa18274f35732f95653ab3ab","revisions":{"1eca493977d6300caa18274f35732f95653ab3ab":{"kind":"REWORK","_number":1,"created":"2024-05-20 06:11:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/920011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/11/920011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/920011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/920011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/11/920011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/11/920011/1"}}},"commit":{"parents":[{"commit":"aec7f39cbe29fef41f64001cb2abbbdef35b3500","subject":"Add servers tests with LUN direct volume","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/aec7f39cbe29fef41f64001cb2abbbdef35b3500"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-20 06:10:59.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-05-20 06:10:59.000000000","tz":540},"subject":"Increase swap size of tempest-integrated-storage","message":"Increase swap size of tempest-integrated-storage\n\nThe tempest-integrated-storage job occasionally fails because of OOM\nduring tempest tests. Increase swap size from 4096 to 8192 to avoid\nfrequent memory shortage.\n\nChange-Id: Id96bd9b30598844457963b842f741e5f4bad7c74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1eca493977d6300caa18274f35732f95653ab3ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1eca493977d6300caa18274f35732f95653ab3ab"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
