)]}'
{"id":"openstack%2Ftempest~922379","triplet_id":"openstack%2Ftempest~master~I0c4523c6916821781c3a67b01ed2e0091407734e","project":"openstack/tempest","branch":"master","topic":"placement-api-calls","attention_set":{},"removed_from_attention_set":{"34860":{"account":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"last_update":"2024-07-02 09:15:20.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-06-26 19:08:36.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I0c4523c6916821781c3a67b01ed2e0091407734e","subject":"Adds placement trait api calls","status":"MERGED","created":"2024-06-20 09:50:17.000000000","updated":"2024-07-02 09:16:31.000000000","submitted":"2024-07-02 09:15:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":17,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"922379-placement-api-calls","meta_rev_id":"79e55a11374f0858a59df7995576f09a8b14e886","_number":922379,"virtual_id_number":922379,"owner":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-02 09:15:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":2,"date":"2024-07-02 07:22:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":2,"date":"2024-06-26 19:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":1,"date":"2024-07-02 07:22:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-06-20 12:14:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-06-20 19:10:53.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-07-02 07:22:56.000000000","updated_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"}],"messages":[{"id":"96d7a73ef72e7e9dd66db44f5086b0fa660fdf3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-06-20 09:50:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74fe0699a67a7f4b89f838bf072627a73d1691f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 12:14:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e3b0655e4f142dabce02d5688528c01\n\n- grenade https://zuul.opendev.org/t/openstack/build/ab9662e6487e4e628187ed416a02314b : SUCCESS in 1h 04m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/85fd13359252434d816dbd0ed27d5f4a : SUCCESS in 1h 41m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f59f709ef2c4433a102fba3eb38a27d : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fb89a0d823e491688a3e49708e56a12 : SUCCESS in 18m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/246c3a7ce47046d4a2edbeb244077bc9 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18db708de8124c24bf3b1b197b058b01 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ee82373169bb4bfcbc99886d3f2595fd : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a568a44cc6a34e17a19371c5c40ffb6e : SUCCESS in 4m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fc88fe9368d14e7cb4c707994e36a2c1 : SUCCESS in 4m 18s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/4796c53b14824e149a9a6a71f09ef966 : SUCCESS in 55m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d9eeabd65f25438ab458b69b5df8b230 : SUCCESS in 2h 06m 12s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/c56ea2e848dd455eaf4ca1d632ee3938 : SUCCESS in 1h 40m 56s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/8fdb5408b6524194930c975e3faf875f : SUCCESS in 1h 24m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/745013c6913042a2a3a579a4c0bcf026 : SUCCESS in 1h 09m 19s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3853c740995c4930b1a95aea652ec5c4 : SUCCESS in 23m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7454970b6f894193bd2c61e75c2fbced : SUCCESS in 1h 03m 29s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/498f3ef899b7442f9c92fc2d5fc1627b : SUCCESS in 1h 37m 35s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0e2dda4082b84d1ea54c2e364044d8d7 : SUCCESS in 49m 27s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/edc61b507e9b480280432e347669d607 : SUCCESS in 1h 38m 46s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/770dc8beb2504975ae12fd6428454702 : SUCCESS in 2h 04m 58s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/377cf63448a643659857a123d3a5204b : SUCCESS in 51m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a4eb586882ac456da0195676dbb72768 : SUCCESS in 51m 21s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/96c2d01adc0a42b2adf881ad59314c2a : SUCCESS in 1h 19m 17s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/ea6dc20efb3f4657a087639cfc010122 : SUCCESS in 3m 43s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/115dd5d5e6f84b429a50649eade30f40 : FAILURE in 1h 17m 27s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/a7163635d9a843bfab17d44187189fdb : POST_FAILURE in 1h 06m 34s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/8ceb0bee16c440e3bc4195ae9c512fe4 : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1},{"id":"24aa2680a43000dcd19418987ce98f0baff6c2ae","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-06-20 12:21:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9b33333d157b26446792730d1783e6843732f3a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 14:51:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1bc94a4c91f0424790ac3901884a8d8f\n\n- grenade https://zuul.opendev.org/t/openstack/build/06a92bf0ff6e46ea82b478782a6037b0 : SUCCESS in 56m 24s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f72e3e4226d433b8e95ecec86bec298 : FAILURE in 1h 14m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fbf3fe2f244468daadcab7818c248b1 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06445f5144a64c379f0fdf24b6c82056 : SUCCESS in 12m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ced7381168b2489aa3ea7cd5b5988694 : SUCCESS in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e3d87b73939544efac9c626e8038cb20 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6a4e7a1284114603a3951a942cad369c : SUCCESS in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f7515409d9e0489db50fc4c74cfde652 : SUCCESS in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a7705d92ac1241f9a4b9695d18d4a17a : SUCCESS in 3m 29s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/13a58b23dd4c49b3af2f9f80876dd190 : SUCCESS in 1h 08m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9d76eac7c19e40b0ad50cb7004e0eaac : SUCCESS in 2h 18m 43s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/aa662595f07a49fd8dd7ff96d285c3fc : SUCCESS in 1h 36m 24s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/3a0475e3caca40f080784d6184ae692b : SUCCESS in 1h 19m 45s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e1b5a8396ce64beea727717627e9177c : SUCCESS in 1h 09m 32s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/3ef81e114d864149be5570c6fff6e09f : SUCCESS in 28m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d630e1abf36c4f7c9d99c1684e8b8a0b : SUCCESS in 52m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9f00f33cfcef4b64b5ee7eadb5d42d4b : SUCCESS in 1h 52m 21s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/328da9eed81748fdbacd6fd5b010b5f6 : SUCCESS in 49m 44s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5c7b1ec12e4e43b4acd113cf0067c7b2 : SUCCESS in 1h 38m 04s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1f091e8c18a64effbf1f7fc9999533c5 : SUCCESS in 2h 01m 08s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/64ff1b0a1bc9400b839023ad5ea546c0 : SUCCESS in 56m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/878eb7e155174e389f28ac0fae36f9a0 : SUCCESS in 1h 04m 41s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/50380f978c704f4ab1519e8cffd3d87a : SUCCESS in 1h 12m 45s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/c2b010889c1f4686910c1bc25b5c1492 : SUCCESS in 3m 31s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c056df47e0d847f08f5ccb6a954645fc : FAILURE in 1h 28m 40s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/6cb6e8d74647401d823332ba30541033 : SUCCESS in 52m 00s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e88111e3d51a41c9844843adedb02145 : SUCCESS in 2m 42s","accounts_in_message":[],"_revision_number":1},{"id":"ae916368dbb2c8bc9a9b876b6961b0551f94b2f9","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-06-20 19:02:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b1fadfde6a5ed39144543e923113a025803876cd","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-06-20 19:10:53.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"17c31351205efc8dff5608a9143280c7a08ed9d2","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-06-20 19:13:10.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"624b98d8946c4abaf357fc9c84eae11130d44062","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-20 21:11:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9260e7a42fbb4dbba24ad25dc149e032\n\n- grenade https://zuul.opendev.org/t/openstack/build/1d8f2f0527a942029a405f6cc67312ce : SUCCESS in 1h 15m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fc9cabd5423e483991ad9dbd3d3a0b06 : SUCCESS in 1h 49m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23fb6aa0bc18461bbd9bf8584632fe9b : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/958afd1b2bc344af938318bdaa686570 : SUCCESS in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/08fefe675ff54916afb714e48ce2a4f9 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95545db4ce5f4ef5ad413336b6aea0ec : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cbcadcc055ec4241bfb8a3bb67824552 : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec3b28fa991842608122bbc2722443b3 : SUCCESS in 3m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/212699e424fb478f99d4f18adf482609 : SUCCESS in 3m 17s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/896ec2d14949443bb19f56ddfb6c0718 : SUCCESS in 1h 00m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/808ddce42dfd42299d2208848a957755 : SUCCESS in 2h 06m 45s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/9d79203369b4412b9e063c88acd71473 : SUCCESS in 1h 45m 09s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/fd40431d3f27443cb4d8570ba92bf72d : SUCCESS in 1h 23m 20s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c0ae5d3981e3491aa2c70179d2f530e6 : SUCCESS in 1h 20m 40s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/2fbf74bb84f64eb88c6e35e074651ee7 : SUCCESS in 24m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af14785903594f8595eb7b86ff7aae2f : SUCCESS in 1h 03m 36s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/1c9e78b71f0e407f9fec2f6b88c1296b : SUCCESS in 1h 50m 36s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/0e4c1da5248f49af8a6b40e5341b7068 : SUCCESS in 54m 40s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/10a79ac0b34946839189e00554867220 : SUCCESS in 1h 41m 10s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/97928390f13f466b80c7688a2410e0b8 : SUCCESS in 1h 56m 22s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2e5e1b8072464808bd3237a7849fc2d7 : SUCCESS in 1h 00m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/728d1cdf162f4bdf825b549112230a30 : SUCCESS in 45m 08s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/24fab738ceca42398c103babe10d04b4 : SUCCESS in 1h 22m 02s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/332dd23ab87646e7950e3552a8087fcc : SUCCESS in 3m 34s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/d71855d91c7f42269766587feff1dc44 : FAILURE in 1h 24m 09s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/618dca012a3a4627a10cfb72fc79c7b9 : SUCCESS in 48m 21s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/9f6ebca5330d451ea05d476be4e53929 : SUCCESS in 2m 39s","accounts_in_message":[],"_revision_number":1},{"id":"b0c19e3904cfab6f46d35a74f0b077ca5ec591ac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-06-21 13:52:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2cce889e9df69d85b5b1af4489625d32c02b0689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-21 16:01:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/265a4c7884b94c8095af1a92d2e5659d\n\n- grenade https://zuul.opendev.org/t/openstack/build/4e08a3bef158469a8eb6be9d47b237f1 : SUCCESS in 59m 29s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/93bcb15559dd4738bee724cd44c9cc44 : SUCCESS in 1h 30m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f18fcad6e9454fd3b4ffaeab37810927 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db1c55c4a2748788053dbe1b8e261c2 : SUCCESS in 10m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a14dc10b070848fab9aad9f1a0159927 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef64b0968e02422bbd16c49166b3cac9 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d02f5acb20454eaeaae4a0026b543eec : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/496e77a719be4fdd87d53bbd9557b51e : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b7a60cc282dd425ab1bf2d4b6338dbc2 : SUCCESS in 3m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d4ababf581342499ee051ce42cd72e1 : SUCCESS in 4m 49s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/914ddf410c7c4c85ba32eb9ce0aa314d : SUCCESS in 45m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/57981845ff5c4183ae24a94c5676d32d : SUCCESS in 2h 03m 21s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/96710e06e0294d25947b88e3228668e0 : SUCCESS in 1h 30m 20s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/520d6b4047c345bfb0a57206d86fd5e7 : SUCCESS in 1h 29m 40s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d10ea21ecbe045059e5fd059f3ecd308 : SUCCESS in 1h 20m 04s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/519fb9092dfc4d6c8df6319a1070f624 : SUCCESS in 25m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22a4a7eaadd44e7e9827ca14b78dcfb5 : SUCCESS in 50m 22s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/e9e2738b5e8040c594c0c2aad9e7745c : FAILURE in 1h 35m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/bc1021e5fe114858b7e82f7aa4cf4f9e : SUCCESS in 42m 10s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/025c1ba51491454c95ea435df30c8409 : FAILURE in 1h 14m 12s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6b438d98e5024f8b8b7264e4b29bd32d : SUCCESS in 1h 58m 14s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/fcde1b5d6bfc4f8189929a7327ad9d32 : SUCCESS in 52m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9f0e1e2ba1a14691b1a0aac4a76bf565 : SUCCESS in 55m 49s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/34397bcd4d2641f4952578acceb02eef : SUCCESS in 1h 12m 32s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/1812612b66a84b3c9273c2c92f11ff69 : SUCCESS in 6m 09s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/b3ae7e15db1a4df39739d2ca1cad6a03 : FAILURE in 1h 06m 47s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/df3ed655296a4dbfa611c40e19505088 : SUCCESS in 1h 02m 47s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/e24feb02d06348f8bde31022c2e173ae : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":2},{"id":"3dfd271d53727ef6645e950f28d8b7acb9f8d660","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-06-21 16:58:09.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b96af7e723718391036bf3e7ec5734e43cd5b98d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-06-23 18:07:25.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"77fb72d08aa7234415b911e8259bac7aefa24f2a","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-06-23 18:23:33.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"69fa8eb7e4aad5bad86c012b1e7b33511e64ab3e","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-06-23 18:29:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"15a583c67103ef654d84f151ff32452050433b54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-23 20:12:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3aed52bd933e4d208496b8759d40514c\n\n- grenade https://zuul.opendev.org/t/openstack/build/b3098f63026b4b27a9e4eded898d7bf4 : SUCCESS in 1h 03m 20s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c0489d9fbb36459194c3c111ec7ccf57 : SUCCESS in 1h 55m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8c936b7894048dfbe4f84514911ce37 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e741e55679d46cb851666ab704c681d : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/502395e80b7043caa750c643454650a1 : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d61cd6e70eb4f45885618cf2c58dde6 : SUCCESS in 3m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fe38bf1ca934fa3b701ab42d90fdd11 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0442ed3ef034d438696012fe8ca5e93 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/08dd958392c14abd8a3ffdfea7d218be : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b96ce6a7eeff469986fbb5e3de8dfed0 : SUCCESS in 3m 55s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/bb21c2933f2144e1baa986612cefc8d0 : SUCCESS in 1h 07m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c3f678909e204b7680be288776ed3484 : SUCCESS in 1h 33m 52s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/00e48f372a024235be2444cad34a4aea : SUCCESS in 1h 12m 01s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/e8c7c93ea51f4345b4f48c6d173cef74 : SUCCESS in 1h 00m 59s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/76acba793486487d9171c08f79259eb5 : SUCCESS in 51m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c253d57f36fa4140ae1e6230f3c13e8b : SUCCESS in 21m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/88bc494499b14ba1b9c25fc339c13041 : SUCCESS in 43m 44s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/86db2ae0b95d4d51a28ed697089cd9ea : SUCCESS in 1h 54m 58s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/48a347fb00e6443a90e88fbfbb2e67af : SUCCESS in 49m 11s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ba5f68ad141c47a29247155b9791deae : SUCCESS in 1h 38m 02s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a83411acaf5240028df7f7ff89d095a1 : SUCCESS in 1h 50m 18s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/d8305425353241e792ffb1658a4618c8 : SUCCESS in 1h 06m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6f3e718e3465485895283b556ff0193a : SUCCESS in 1h 03m 31s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/b11069cf38c846089427b84344845fb2 : SUCCESS in 1h 19m 24s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/d3119f7b7a9740329b18083d49ce55c3 : SUCCESS in 3m 36s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/61eaa141517342d68da5f987e4d6b32d : FAILURE in 1h 16m 04s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/71abc301f14d4986914cfd6b6d6f841b : SUCCESS in 1h 00m 14s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/06a8e700406949ebb9cb41b4128f1305 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":3},{"id":"3bcc18bde12312d14307d8897327fef78ddae23b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-06-24 18:35:56.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7b7424b648852e4b25a97fe79fb898c42c6e7259","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-06-26 07:08:39.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"1aa02728d2550b6202a7340fcbc30fade37a9dc6","author":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"date":"2024-06-26 07:10:13.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9408d1cb1f48f7e1cb5d5b4677d3acf1acd16756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-26 10:19:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d495f6fb43b64914ba4a832580a80d0f\n\n- grenade https://zuul.opendev.org/t/openstack/build/fd5970cede574d08ae9a3c707003839f : SUCCESS in 56m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c3ff7c90a31843bb9a8bf59d89a083a5 : SUCCESS in 1h 47m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3e96984eb214f45af1d18fced26883f : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/996254336d9b405d99189e9aa27aef20 : SUCCESS in 14m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84f6c26677eb4aeaa351c69a454bf49d : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb76c1afb97c40e69f11879a0671137b : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3064885d1af4bfda96f3befecd81df3 : SUCCESS in 3m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2b1cda6dcfc1483ba15bfecd3414344c : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0a8689d212a747da92bbbabdf43a6b56 : SUCCESS in 3m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8bd0c4435ee4d979a245f3e4109a283 : SUCCESS in 4m 48s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/6f559ec9d2ed450e923a2f0e319804e9 : SUCCESS in 47m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ecc5b00c52a4415bb7cc318985df9165 : FAILURE in 2h 26m 46s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/ed243129e99243ef9c0321cae5a19c42 : SUCCESS in 1h 33m 22s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/b4967207d70849409f5cfdbee571a2e3 : SUCCESS in 1h 27m 44s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a774634c0ab446de849524c9be568ad7 : SUCCESS in 1h 04m 59s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/a6fe85f1a63b448aa4e60b5f4668da28 : SUCCESS in 29m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd690b46704a408da7b38b167cb7cd9b : SUCCESS in 52m 03s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/9883b2400e6b47ccbae12cb0170c4dd5 : SUCCESS in 1h 30m 44s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4180c4ceed9c48afac0facffa5313daa : SUCCESS in 51m 47s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1a943ab4d78c4a2eb9f3a669b179274d : SUCCESS in 1h 46m 52s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/be70a13c41d7447c83976df7f1620295 : SUCCESS in 1h 38m 42s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0ab205e943f24812832459f5256725a8 : SUCCESS in 1h 01m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94bf3f330e5e4fda915db8d8487da6c3 : SUCCESS in 57m 34s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/30a22e328afb49bcbb5a2fe95bdb8287 : SUCCESS in 1h 17m 28s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/ce75634766d044cf961068db0187feef : SUCCESS in 3m 33s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/7c177b481c9649b7bbd2c86a1fd6d02a : FAILURE in 1h 25m 48s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/833113875155452ca9c5307407312df4 : SUCCESS in 58m 35s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/25fde810fc3640918d9c6d4dcb3d8850 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":4},{"id":"232e246bf4a58737964d646c383f8182ae03f720","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-06-26 19:08:36.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ec6796062f12a7ddb01979b0de3a0917d5d83c55","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-07-02 07:22:56.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3977c5be6fa5f773758ce6d8939a7ab2ad367574","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 07:23:30.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"facf13f9711930fc761b2c946b4d53a132624d5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 09:15:20.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b490faa438ce484786fd02be1af0f1af\n\n- grenade https://zuul.opendev.org/t/openstack/build/7f8c672b510b42e3ac8dbf6a86917b86 : SUCCESS in 49m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/841825b9e0464596989c332554b8bd83 : SUCCESS in 1h 29m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/573d4e72082f44f8b642e76460530687 : SUCCESS in 15m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be95253849a349ed85e43ee6d695c930 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3880326f049d491f802a0296772a6b55 : SUCCESS in 3m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8842f9a03fcd4335a75becbe0bf1dae1 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e98d287c47a41fc82ae03493a9fb922 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/16fd6d4349de4c2cbd34d77a4cce857d : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5ea5bd2c53da4fafa1a7e1fe202f610b : SUCCESS in 3m 50s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c54bd196ea7d4c4a920f5dce2d1da34d : SUCCESS in 1h 30m 05s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/47603771e2264f55be81c1e8a4326a65 : SUCCESS in 1h 06m 58s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/65e6cab51d3d4e7dadf8eb4480784a56 : SUCCESS in 56m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8f2ec1fc16d14a0f8ac55cb20ff6cd50 : SUCCESS in 1h 04m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a1e3de7db9c44fba82882350f5fed6e : SUCCESS in 1h 05m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/6f79161cfa214a208d6f779dae275872 : SUCCESS in 1h 23m 30s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c9f8cb5316e44063ac30887e3973e6b9 : SUCCESS in 1h 46m 23s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/9611e46a748642d68da5e79299759552 : SUCCESS in 52m 16s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/750a18e570db43868a992dabed826848 : SUCCESS in 59m 52s","accounts_in_message":[],"_revision_number":4},{"id":"b535de0c2f50ab3c99d71576c662456fa15bce0e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 09:15:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"79e55a11374f0858a59df7995576f09a8b14e886","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 09:16:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01e1801c48734295ac03be629c739a4c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd266391318045b581a3f94b840f2a07 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1841ab652a834809b2d28c895eab490c : SUCCESS in 36s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3c3985c2ec5ad3a25de0fb644cd66cef71511b6f","revisions":{"0c4ad93c95c26c0c999a51414069a638b667a47f":{"kind":"REWORK","_number":1,"created":"2024-06-20 09:50:17.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/79/922379/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/79/922379/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/922379/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/922379/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/922379/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/79/922379/1"}}},"commit":{"parents":[{"commit":"3619d297d57c719d2f297f16e8867b5b4f7abf20","subject":"Merge \"Create independent mount path for each device\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3619d297d57c719d2f297f16e8867b5b4f7abf20"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-06-20 09:48:44.000000000","tz":-240},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-06-20 09:49:39.000000000","tz":-240},"subject":"Adds placement trait api calls","message":"Adds placement trait api calls\n\nChange-Id: I0c4523c6916821781c3a67b01ed2e0091407734e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0c4ad93c95c26c0c999a51414069a638b667a47f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/0c4ad93c95c26c0c999a51414069a638b667a47f"}]},"branch":"refs/heads/master"},"9446af18dae0af0d1ff0818d0eb35dd7ece6e3e4":{"kind":"REWORK","_number":2,"created":"2024-06-21 13:52:10.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/79/922379/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/79/922379/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/922379/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/922379/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/922379/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/79/922379/2"}}},"commit":{"parents":[{"commit":"3619d297d57c719d2f297f16e8867b5b4f7abf20","subject":"Merge \"Create independent mount path for each device\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3619d297d57c719d2f297f16e8867b5b4f7abf20"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-06-20 09:48:44.000000000","tz":-240},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-06-21 13:51:24.000000000","tz":-240},"subject":"Adds placement trait api calls","message":"Adds placement trait api calls\n\nChange-Id: I0c4523c6916821781c3a67b01ed2e0091407734e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9446af18dae0af0d1ff0818d0eb35dd7ece6e3e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/9446af18dae0af0d1ff0818d0eb35dd7ece6e3e4"}]},"branch":"refs/heads/master"},"e0b46988c751991438e3b586957a0a94616b35cb":{"kind":"REWORK","_number":3,"created":"2024-06-23 18:07:25.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/79/922379/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/79/922379/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/922379/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/922379/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/922379/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/79/922379/3"}}},"commit":{"parents":[{"commit":"3619d297d57c719d2f297f16e8867b5b4f7abf20","subject":"Merge \"Create independent mount path for each device\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3619d297d57c719d2f297f16e8867b5b4f7abf20"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-06-20 09:48:44.000000000","tz":-240},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-06-23 18:01:22.000000000","tz":-240},"subject":"Adds placement trait api calls","message":"Adds placement trait api calls\n\nChange-Id: I0c4523c6916821781c3a67b01ed2e0091407734e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e0b46988c751991438e3b586957a0a94616b35cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e0b46988c751991438e3b586957a0a94616b35cb"}]},"branch":"refs/heads/master"},"3c3985c2ec5ad3a25de0fb644cd66cef71511b6f":{"kind":"REWORK","_number":4,"created":"2024-06-26 07:08:39.000000000","uploader":{"_account_id":34860,"name":"Amit Uniyal","email":"auniyal@redhat.com","username":"auniyal"},"ref":"refs/changes/79/922379/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/79/922379/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/922379/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/922379/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/79/922379/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/79/922379/4"}}},"commit":{"parents":[{"commit":"3619d297d57c719d2f297f16e8867b5b4f7abf20","subject":"Merge \"Create independent mount path for each device\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3619d297d57c719d2f297f16e8867b5b4f7abf20"}]}],"author":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-06-20 09:48:44.000000000","tz":-240},"committer":{"name":"Amit Uniyal","email":"auniyal@redhat.com","date":"2024-06-26 07:07:40.000000000","tz":-240},"subject":"Adds placement trait api calls","message":"Adds placement trait api calls\n\nChange-Id: I0c4523c6916821781c3a67b01ed2e0091407734e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3c3985c2ec5ad3a25de0fb644cd66cef71511b6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3c3985c2ec5ad3a25de0fb644cd66cef71511b6f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
