)]}'
{"id":"openstack%2Ftempest~924127","triplet_id":"openstack%2Ftempest~master~Ibde7620c9388fdbc93f741e38c9d63fca44a1bbb","project":"openstack/tempest","branch":"master","topic":"new-location-api-support","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-07-19 07:39:05.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-16 06:45:55.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2024-07-19 05:47:14.000000000","reason":"\u003cGERRIT_ACCOUNT_22873\u003e replied on the change","reason_account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-07-18 17:57:41.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"Ibde7620c9388fdbc93f741e38c9d63fca44a1bbb","subject":"Add wait for location import task","status":"MERGED","created":"2024-07-15 07:32:07.000000000","updated":"2024-07-19 07:40:37.000000000","submitted":"2024-07-19 07:39:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"924127-new-location-api-support","meta_rev_id":"9492a5458bfc80685a68c0e19c03cfd9e3b29675","_number":924127,"virtual_id_number":924127,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-19 07:39:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-16 06:45:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-07-19 05:47:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-07-18 17:57:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-07-19 05:47:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-15 08:24:23.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-07-15 10:30:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-16 07:27:08.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"state":"REVIEWER"},{"updated":"2024-07-16 07:27:08.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"}],"messages":[{"id":"e6aa710a9a5aa302d7aaf7957f2164f8101ba114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-15 07:32:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a7e37b3ad6e6fec6f564d2cdf4b5f3b12a0d6f5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-15 08:24:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"810a9aea121dc2e4f76a882cf61fe7c778bd18b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 10:30:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e1e1dc5e279441b92aee5484ef97644\n\n- grenade https://zuul.opendev.org/t/openstack/build/fce1c4fc57af4267bc2bd3c43cab03c7 : SUCCESS in 43m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87338cc4b6194c4cb87e5c4cd4db88bf : SUCCESS in 1h 35m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f879959800274d56b67bbb6758e686ec : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cca0b7973c0c4c11bd0b30cac6529cee : SUCCESS in 10m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92405dd7bde04104ba96c03bfde022db : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19ba4c1bed6c45c79bf980ae55947aff : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3557f3eab371419b92e4b98a44151606 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c02e30194ec4bd48b9aacc100dc0952 : SUCCESS in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/127287fe1c04455595198954fc7a7971 : SUCCESS in 3m 50s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/c67fb84e22c04d689668c30fd5a51c44 : SUCCESS in 41m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0e78bc3bd7dc41b68a8d93e0439377de : SUCCESS in 1h 51m 34s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/d8692fa6018f48a6865ed87ac16c9800 : SUCCESS in 1h 25m 48s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/012074a17dac42a1829c56d0746b7997 : SUCCESS in 1h 23m 10s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/493022bc51d443d2a8f17192a0a5a88d : SUCCESS in 51m 12s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/0760f913023647e286e3ed692d4774f8 : SUCCESS in 19m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bc9e29881394eeabf80bf64e2015ee9 : SUCCESS in 56m 16s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/67d5cf322eec45b39914928da9fd00d2 : SUCCESS in 1h 45m 12s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/4197808ccf914dfcade298c097a87901 : SUCCESS in 52m 12s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/279d75e2f6914dd5be477a1af49e450d : SUCCESS in 1h 17m 02s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/a8893d467a9843babe265b53517d6021 : SUCCESS in 1h 41m 25s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/6530f2913f9544d9bbc656eb9b23fd76 : SUCCESS in 42m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ffd7df2b0b06491ba41e2db05ac37ff3 : SUCCESS in 47m 23s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/8f1124c095264a83a24014322289993a : SUCCESS in 47m 32s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/629896af4cd84e3c8a84ffee3b08154d : SUCCESS in 3m 57s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/38f59295f24e407eb578f7adcb76b0ef : FAILURE in 1h 19m 40s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/501165bf3ee44ccdad29bdb881ed3d0f : SUCCESS in 59m 44s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/b7747f07a1e740fbbf73cbb80ac578de : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"158a09b97d060ad2849d2244582a4a8d27b01bb0","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-15 10:41:01.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"023213034b6deb06517c42cf42906f0daba70dbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-15 13:48:03.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5c31530a31b647009ce3429eabdc29e37d098d06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-15 14:55:26.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"f389f6c3077bb6a0baa449646cac20af78e3aac7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 17:37:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cdbe7c34e854603bb0111d61358b4e1\n\n- grenade https://zuul.opendev.org/t/openstack/build/3a7aefefa5ae4f6c8b6f5b0afbd0b446 : SUCCESS in 1h 06m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fd3a92a8bf2431ebc859776bad59bf6 : SUCCESS in 1h 38m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17672f5cacdc47a89a366c30b14849da : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a72302e76b6a4d51abb3c997bd666a59 : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704a1b3c31314ac092a6d075dcd59bb2 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b7d5896e516453a9a2ca7d966dad4d2 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a54a49e04db4edaaaea7171bde7331e : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67a4a3b4465b473c8d7bbec5de30c9c0 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e954c93f06784a149ed3862bc5229297 : SUCCESS in 3m 31s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/f900839550504f24a13b9cfb16cdd783 : SUCCESS in 46m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/169d728705b4447d86d8cacba862df00 : SUCCESS in 1h 51m 13s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/dea871b466d342ff975c83c7169e9e87 : SUCCESS in 1h 54m 12s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/513b7464cbaa4edb98c8c74c6c5e08e0 : SUCCESS in 1h 18m 57s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/aa2e69ed119440ac8d96dd70c3d38ebf : SUCCESS in 56m 37s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/ea2ffe7ae019416089e744bb377c41ee : SUCCESS in 27m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba9a47500bac4ceda9a01b6325c5f8ec : SUCCESS in 46m 46s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/93a5f0419d7b483f895e0193b56b0ea2 : SUCCESS in 1h 23m 09s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/fc53da201e8a4a77ab08934bf44b831e : SUCCESS in 53m 21s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9acdcbb1cbf447a6af5d9a6d8710aa27 : SUCCESS in 1h 26m 26s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/481148189459498c8562c592c9395ffa : SUCCESS in 1h 39m 02s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/79230d2e29ae4097b78516e666f6e424 : SUCCESS in 50m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/57e3d06e77aa4cd6a2a90d2f50e4fb3a : SUCCESS in 57m 18s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/0ac3df9c7e87491594ee7d86fd5b1960 : SUCCESS in 1h 14m 07s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/3d40bd9ae5744acf95b0998373421d61 : SUCCESS in 4m 01s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/05444c7ede49435794b4d614edd9efd7 : FAILURE in 1h 19m 46s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/c2194873c94043e1b42def9f190b4a87 : SUCCESS in 1h 03m 28s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7d2bc6573d5349f8869c49a44cc586c4 : SUCCESS in 2m 52s","accounts_in_message":[],"_revision_number":3},{"id":"26b4f96a2403a017c8973dca8099ef422e650d10","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-16 06:45:55.000000000","message":"Patch Set 3: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"06b8f4f4ee9492e9da319e7a74547f47f3745a4b","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-07-18 17:57:41.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1b9809edb160249e4e183778b06371570c661345","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-07-19 05:47:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f7cee7814fe3e51a373fb9c6bfb43072141f0801","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 05:47:54.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ee4a526671d39c2552b09e499e71a0c60c6c3dce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 07:39:05.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0332a50fea2a484d931b3d03941bbedc\n\n- grenade https://zuul.opendev.org/t/openstack/build/463914d5dc624bea95f7f147cb4aa1c4 : SUCCESS in 44m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ec8898611a3244b6b78eb173b92a2f48 : SUCCESS in 1h 29m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18a8d5ca520f4e03aeff61892fbf1884 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e61f5e2bbb494bd8867b4ea19a78697e : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e7991adfdab14870873869fb8d230592 : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d9adbd346a643038106c721d236fda9 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3036fd3af2c2457d89a67eed784ba361 : SUCCESS in 4m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/981323088a95482b8ad7963ec7844dc7 : SUCCESS in 3m 39s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/15aea485603040099f2db2c4f9aaf94d : SUCCESS in 1h 40m 18s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c73299331e604f2085e29d531691fd5e : SUCCESS in 1h 01m 48s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/8d1287c7edf743618703d1db1ebbdc0e : SUCCESS in 38m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e849ce45467e4f6ebd10b34641ffd968 : SUCCESS in 52m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ed08cdbd89d433abd8a22fcafa6844f : SUCCESS in 56m 39s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/a1b327c411214eb99061d6f5b78ef50e : SUCCESS in 1h 06m 26s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/19b5ac4ae2244487b3474a9011a2ee91 : SUCCESS in 1h 42m 27s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/49ded4c4948144ada7a4ad4aab7288f5 : SUCCESS in 40m 41s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/29a802c5f1a3452c8d6905fad1824e08 : SUCCESS in 1h 11m 00s","accounts_in_message":[],"_revision_number":3},{"id":"fe83bfb762a1d5daee04d114d4b8bdcb255a0900","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 07:39:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9492a5458bfc80685a68c0e19c03cfd9e3b29675","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 07:40:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eff2c364843842ffa93ad1771212f65e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0332b9a6b4a94f28990d5b29c7df8b06 : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"953940288c51e2c035a21aee518a925caabe7549","revisions":{"93ff2bb11fc3b72ff06e07fe367354f16c01b297":{"kind":"REWORK","_number":1,"created":"2024-07-15 07:32:07.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/27/924127/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/27/924127/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/27/924127/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/27/924127/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/27/924127/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/27/924127/1"}}},"commit":{"parents":[{"commit":"e47f4457cbe5991c2670ad6a5902f3a056d19f6c","subject":"Merge \"Drop identity v2 api tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e47f4457cbe5991c2670ad6a5902f3a056d19f6c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-15 07:13:56.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-15 07:31:06.000000000","tz":0},"subject":"Add wait for location import task","message":"Add wait for location import task\n\nWe are working to get support of new location APIs in glance,\nwhere we are adding hash calculation mechanism once the image\nlocation is added. Since new location api is async,\nwe need to wait for the task to finish before performing the\ncleanup else we will fail due to the existing bug [1].\n\n[1]: https://bugs.launchpad.net/glance/+bug/2045769\n\nChange-Id: Ibde7620c9388fdbc93f741e38c9d63fca44a1bbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/93ff2bb11fc3b72ff06e07fe367354f16c01b297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/93ff2bb11fc3b72ff06e07fe367354f16c01b297"}]},"branch":"refs/heads/master"},"6854c76581dcb2a702956e6ebf2dc59763912328":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-07-15 13:48:03.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/27/924127/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/27/924127/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/27/924127/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/27/924127/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/27/924127/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/27/924127/2"}}},"commit":{"parents":[{"commit":"e47f4457cbe5991c2670ad6a5902f3a056d19f6c","subject":"Merge \"Drop identity v2 api tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e47f4457cbe5991c2670ad6a5902f3a056d19f6c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-15 07:13:56.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-15 13:48:02.000000000","tz":0},"subject":"Add wait for location import task","message":"Add wait for location import task\n\nWe are working to get support of new location APIs in glance,\nwhere we are adding hash calculation mechanism once the image\nlocation is added. Since new location api is async,\nwe need to wait for the task to finish before performing the\ncleanup else we will fail due to the existing bug [1].\n\n[1]: https://bugs.launchpad.net/glance/+bug/2045769\n\nDepends-On: https://review.opendev.org/c/openstack/nova/+/891207/15\nChange-Id: Ibde7620c9388fdbc93f741e38c9d63fca44a1bbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6854c76581dcb2a702956e6ebf2dc59763912328"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/6854c76581dcb2a702956e6ebf2dc59763912328"}]},"branch":"refs/heads/master","description":"Edit commit message"},"953940288c51e2c035a21aee518a925caabe7549":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-07-15 14:55:26.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/27/924127/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/27/924127/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/27/924127/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/27/924127/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/27/924127/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/27/924127/3"}}},"commit":{"parents":[{"commit":"e47f4457cbe5991c2670ad6a5902f3a056d19f6c","subject":"Merge \"Drop identity v2 api tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e47f4457cbe5991c2670ad6a5902f3a056d19f6c"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-15 07:13:56.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-15 14:55:26.000000000","tz":0},"subject":"Add wait for location import task","message":"Add wait for location import task\n\nWe are working to get support of new location APIs in glance,\nwhere we are adding hash calculation mechanism once the image\nlocation is added. Since new location api is async,\nwe need to wait for the task to finish before performing the\ncleanup else we will fail due to the existing bug [1].\n\n[1]: https://bugs.launchpad.net/glance/+bug/2045769\n\nChange-Id: Ibde7620c9388fdbc93f741e38c9d63fca44a1bbb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/953940288c51e2c035a21aee518a925caabe7549"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/953940288c51e2c035a21aee518a925caabe7549"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
