)]}'
{"id":"openstack%2Ftempest~924169","triplet_id":"openstack%2Ftempest~master~I25d27b3bbb9640f35f0748de0200e1175eb90007","project":"openstack/tempest","branch":"master","topic":"bug/1835186","attention_set":{"30674":{"account":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"last_update":"2024-07-16 02:55:15.000000000","reason":"\u003cGERRIT_ACCOUNT_37102\u003e replied on the change","reason_account":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I25d27b3bbb9640f35f0748de0200e1175eb90007","subject":"WIP: Replace provider with encryption_id in update/delete endpoints","status":"NEW","created":"2024-07-15 16:21:54.000000000","updated":"2024-07-16 05:00:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"d7fa82cd90bd7643d9f86a48660b88ed657fe52f","_number":924169,"virtual_id_number":924169,"owner":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-16 05:00:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"}]},"reviewer_updates":[{"updated":"2024-07-15 18:29:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-16 02:55:15.000000000","updated_by":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"reviewer":{"_account_id":30674,"name":"Lukas Piwowarski","email":"lpiwowar@redhat.com","username":"lukas-piwowarski"},"state":"REVIEWER"}],"messages":[{"id":"78578508d45defc952d86227cec79c00b71e5b5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"date":"2024-07-15 16:21:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d340654d6595fb3726c8209a8cd30b3984389c6a","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"date":"2024-07-15 16:22:34.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"d79a0d1452c4f4e89876166b0a385d8e749b1180","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"date":"2024-07-15 16:24:21.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2ec3c95477c2d55912bf591defb9531d206fcdbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 18:29:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/50f60adfa3a74cab82ec17279635a311\n\n- grenade https://zuul.opendev.org/t/openstack/build/8f873be9b33842d2a8ddec0868d77879 : SUCCESS in 57m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e36cb32dbb664c9ea0009a9f8e38a26d : SUCCESS in 1h 20m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65d8a3b910674c9a986a0abd2dcd9392 : FAILURE in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98051ec63b3046e3b6e69903737d67a5 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ee77f01f834a23bcfeb3a98e1e5c38 : FAILURE in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42a039429f684c8a853c26987523e9fa : FAILURE in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbeb7a841129433d96f698faa71362ab : FAILURE in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/234025a212ae4f269a3867a0dcf183bf : FAILURE in 3m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b2bafc88ecc54e898f8af87d9895eb79 : FAILURE in 4m 27s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/1ab1cc067a8146428f42614b918de86e : FAILURE in 48m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1d906ecab5d24ec5a392f40545964a6a : FAILURE in 1h 48m 54s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/7f489a49dc2b454c8ecfe6016a55414b : SUCCESS in 1h 33m 39s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/3b86c46fd54b44e5b353777f1bd7ab24 : SUCCESS in 1h 28m 23s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d6d68611a35a458380bf49cec6c18e60 : SUCCESS in 1h 06m 09s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/c0c02ca0b3db41f2b4b5844521012b4d : SUCCESS in 22m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab42d19605214245bad8b771fe6875d6 : SUCCESS in 54m 14s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/c8bd6e9489bc45e0a9d0fdac97741153 : SUCCESS in 1h 18m 34s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/97894aeb171245a4ac0827792025e97a : SUCCESS in 44m 56s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7595168f273b4ea58927cb3f2e38697c : SUCCESS in 1h 21m 15s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1427387a390b49bdaf6c6bd6fdd4ab7c : FAILURE in 1h 42m 19s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f22eaaa2b724412585ac5cba7cca4583 : SUCCESS in 48m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7718acacdebd44c792733e54f583f7d8 : SUCCESS in 57m 01s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/c154fa9b11504fda809d5e542f1d7f0e : SUCCESS in 1h 05m 26s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/6ccde0034d4b4e2ca8f3c37bbf8a3996 : SUCCESS in 3m 16s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/347813447d18476982adafe1dc3f1644 : FAILURE in 1h 02m 15s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/7751ba71e59440cb8470a1639b853539 : SUCCESS in 50m 32s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/da31fac685644c9d8674dcb9b8d59b61 : SUCCESS in 3m 10s","accounts_in_message":[],"_revision_number":2},{"id":"492e73ff88964cab62b88de64253fa080c7d365c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"date":"2024-07-16 02:55:30.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d7fa82cd90bd7643d9f86a48660b88ed657fe52f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 05:00:25.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bfd3df6173d7413b9f6103c1c71d397b\n\n- grenade https://zuul.opendev.org/t/openstack/build/475ff0c15784484bb284a84f2331401d : SUCCESS in 57m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bfd8a7e653dc45c68a03dfce0f231866 : SUCCESS in 1h 35m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5df1023d723a45469afed77f94d7765d : FAILURE in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/518c20e862f94676a2e7909892df0bfc : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f04bc0169dc443f78b9476e1ff37cbc7 : FAILURE in 3m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c853a5246bb439aab7c57430e68b1a3 : FAILURE in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e5076501c7744d19aacea7c27cc4c3e : FAILURE in 4m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/572bfe07f0b24b53a0c7650499cbf29e : FAILURE in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e232c4d98294d45a730a029c29e05d9 : FAILURE in 6m 20s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/c1d01496506143f6ab0e700c53cd306b : FAILURE in 43m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ec546306fa8f4315981189b75f422a40 : FAILURE in 1h 57m 51s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/a857c90c548a4d61981a45c3e270571f : SUCCESS in 1h 29m 09s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/ef503559717d45c6b21c23288a242394 : SUCCESS in 1h 10m 00s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/c9ceecc8e56e46ae957ab46413e8e7a9 : SUCCESS in 56m 47s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/30c30490ed9048b491c6bc56d622a6ca : SUCCESS in 27m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac8e78b12206494ebf97dca0634a0723 : SUCCESS in 54m 17s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/45cfbb4a2e734dcd8370dad82a93512a : SUCCESS in 1h 46m 07s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/650a3beaad4d4799835547aaf9030e44 : SUCCESS in 40m 44s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d042d18c91db42849e89f85094423c7b : SUCCESS in 1h 27m 22s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/c8a27319bb4b428aaad032915a67bed0 : FAILURE in 1h 31m 17s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/0d018ba486c24892a92d45f3f56bee8a : SUCCESS in 55m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dbe62268fd084ed3a8c563fb6f4dcb07 : SUCCESS in 1h 06m 33s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/baa35c95ea104b4fac2f18cc20d1a063 : SUCCESS in 1h 22m 14s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/2043e2c66ebb4d12af892f447e7b3571 : SUCCESS in 3m 40s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/8231738a8b124c13a62b38ca33334ff0 : FAILURE in 1h 02m 31s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/35d250a7753f4f6688765e6e430849b3 : SUCCESS in 58m 45s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/7c8990a738ca42cab904c9e9d773a41d : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"86e8cdefb0ee053fe0bfaac0f473102d0f29915a","revisions":{"bd35c381a17bd821d596e2c221615f9a8b1cd049":{"kind":"REWORK","_number":1,"created":"2024-07-15 16:21:54.000000000","uploader":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"ref":"refs/changes/69/924169/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/69/924169/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/924169/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/924169/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/924169/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/69/924169/1"}}},"commit":{"parents":[{"commit":"e47f4457cbe5991c2670ad6a5902f3a056d19f6c","subject":"Merge \"Drop identity v2 api tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e47f4457cbe5991c2670ad6a5902f3a056d19f6c"}]}],"author":{"name":"Evan Wever","email":"evan.wever@gmail.com","date":"2024-07-15 15:52:37.000000000","tz":-240},"committer":{"name":"Evan Wever","email":"evan.wever@gmail.com","date":"2024-07-15 16:21:44.000000000","tz":-240},"subject":"WIP: Replace provider with encryption_id in update/delete endpoints","message":"WIP: Replace provider with encryption_id in update/delete endpoints\n\nThis patch modifies the update_encryption_type and delete_encryption_type methods in encryption_types_client.py to use the correct API endpoints as per the OpenStack documentation (https://docs.openstack.org/api-ref/block-storage/v3/index.html?expanded\u003dupdate-an-encryption-type-detail#update-an-encryption-type). Specifically, the endpoint now correctly includes encryption_id instead of the placeholder provider.\n\nSome Tempest test cases fail when calling the updated methods because they do not pass in `encryption_id` and still need to be modified.\n\nRelated-Bug: #1835186\nChange-Id: I25d27b3bbb9640f35f0748de0200e1175eb90007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bd35c381a17bd821d596e2c221615f9a8b1cd049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/bd35c381a17bd821d596e2c221615f9a8b1cd049"}]},"branch":"refs/heads/master"},"250eb6a4da4a24b23a3e70a2cd3b224082c3e9c1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-07-15 16:24:21.000000000","uploader":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"ref":"refs/changes/69/924169/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/69/924169/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/924169/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/924169/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/924169/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/69/924169/2"}}},"commit":{"parents":[{"commit":"e47f4457cbe5991c2670ad6a5902f3a056d19f6c","subject":"Merge \"Drop identity v2 api tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e47f4457cbe5991c2670ad6a5902f3a056d19f6c"}]}],"author":{"name":"Evan Wever","email":"evan.wever@gmail.com","date":"2024-07-15 15:52:37.000000000","tz":-240},"committer":{"name":"Evan Wever","email":"evan.wever@gmail.com","date":"2024-07-15 16:24:21.000000000","tz":0},"subject":"WIP: Replace provider with encryption_id in update/delete endpoints","message":"WIP: Replace provider with encryption_id in update/delete endpoints\n\nThis patch modifies the update_encryption_type and delete_encryption_type methods in encryption_types_client.py to use the correct API endpoints as per the OpenStack documentation[1]. Specifically, the endpoint now correctly includes encryption_id instead of the placeholder provider.\n\nSome Tempest test cases fail when calling the updated methods because they do not pass in `encryption_id` and still need to be modified.\n\n[1] https://docs.openstack.org/api-ref/block-storage/v3/index.html\n\nRelated-Bug: #1835186\nChange-Id: I25d27b3bbb9640f35f0748de0200e1175eb90007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/250eb6a4da4a24b23a3e70a2cd3b224082c3e9c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/250eb6a4da4a24b23a3e70a2cd3b224082c3e9c1"}]},"branch":"refs/heads/master","description":"Edit commit message"},"86e8cdefb0ee053fe0bfaac0f473102d0f29915a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-07-16 02:55:30.000000000","uploader":{"_account_id":37102,"name":"Evan Wever","display_name":"Evan Wever","email":"evan.wever@gmail.com","username":"room"},"ref":"refs/changes/69/924169/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/69/924169/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/924169/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/924169/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/69/924169/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/69/924169/3"}}},"commit":{"parents":[{"commit":"e47f4457cbe5991c2670ad6a5902f3a056d19f6c","subject":"Merge \"Drop identity v2 api tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/e47f4457cbe5991c2670ad6a5902f3a056d19f6c"}]}],"author":{"name":"Evan Wever","email":"evan.wever@gmail.com","date":"2024-07-15 15:52:37.000000000","tz":-240},"committer":{"name":"Evan Wever","email":"evan.wever@gmail.com","date":"2024-07-16 02:55:30.000000000","tz":0},"subject":"WIP: Replace provider with encryption_id in update/delete endpoints","message":"WIP: Replace provider with encryption_id in update/delete endpoints\n\nThis patch modifies the update_encryption_type and delete_encryption_type methods in encryption_types_client.py to use the correct API endpoints as per the OpenStack documentation[1]. Specifically, the endpoint now correctly includes encryption_id instead of the placeholder provider.\n\nSome Tempest test cases fail when calling the updated methods because they do not pass in encryption_id and still need to be modified.\n\n[1] https://docs.openstack.org/api-ref/block-storage/v3/index.html\n\nRelated-Bug: #1835186\nChange-Id: I25d27b3bbb9640f35f0748de0200e1175eb90007\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/86e8cdefb0ee053fe0bfaac0f473102d0f29915a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/86e8cdefb0ee053fe0bfaac0f473102d0f29915a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
