)]}'
{"id":"openstack%2Ftempest~924726","triplet_id":"openstack%2Ftempest~master~Ia3e64078943cd000dba2651f2f81dbb6b4d328c6","project":"openstack/tempest","branch":"master","topic":"user-visible-metadata-in-volume-types","hashtags":[],"change_id":"Ia3e64078943cd000dba2651f2f81dbb6b4d328c6","subject":"Adjustment for Volume Type Metadata","status":"ABANDONED","created":"2024-07-23 11:41:43.000000000","updated":"2024-11-12 12:29:00.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a64cbc4fd322179d2a2dcaf3726b0cb2670f2857","_number":924726,"virtual_id_number":924726,"owner":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-23 13:58:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-23 13:58:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c5f04c023f632fa8a1d4ffa28657beb7edbe4c7f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-07-23 11:41:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c65db86db740031bd57d9fa5b4508eaaadd33b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 13:58:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef7bb78d07fc4e90b13e8088a0c98d3a\n\n- grenade https://zuul.opendev.org/t/openstack/build/9f804da1e67e4943ac1d36cbb5456673 : SUCCESS in 47m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/95afd1de3ed9405aa998f3d8f40c756b : SUCCESS in 1h 37m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bb217aa9571481995d04a0216699532 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04b92e62916b4963a60bbb0bc0566419 : SUCCESS in 11m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b7f0dcb6ad64ab39163e99d5165805c : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49c31d7d60a745788cc2e012a1664752 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b885dd7761ca45168110c3c923bc6ef1 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5397c02f17764eb4be4c333b3c90e8ca : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/59c7a4317acf4e0eaa42e0f0c991e939 : SUCCESS in 4m 03s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/040d17d1203a457f8e2936cf6f58f40f : SUCCESS in 56m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6c6166eb57904e4ca4255ae3474613ac : SUCCESS in 2h 04m 07s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/77edc974229a40acb7b55459a2e02188 : SUCCESS in 1h 27m 13s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/36b7c5d273f64f90b1cb7a539ad3df41 : SUCCESS in 1h 21m 32s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b616c10728e0490d9f51d728045acbd0 : SUCCESS in 56m 48s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/20861f18cff746a0bc14aa0c3f49b9a4 : SUCCESS in 23m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1b71daacfea643af8de2e855f00c3571 : SUCCESS in 58m 26s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/2d850cf015c54239ab7fa394462e0879 : SUCCESS in 1h 31m 57s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/b5f2ef2b2c264640811e1c026ecb5c22 : SUCCESS in 48m 30s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cd0d410e18c94b59bb1bd4551bffd297 : SUCCESS in 1h 49m 08s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/9f05a9517d684af4883034c7e90a825e : SUCCESS in 1h 55m 24s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/7c727cbad89d4968b7a54a3eaacf7e1c : SUCCESS in 56m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8443f06a55af4cf1b9a6333958ee5184 : SUCCESS in 50m 52s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/10275e0025834e939d0f81e340605e07 : SUCCESS in 1h 23m 38s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8cc380532093452b8621a594133ccd7f : SUCCESS in 4m 02s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/a028f1d686d643e49cd675a00f3fc83a : FAILURE in 1h 11m 01s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/31e76ca02d9044fba40a2d34124b5f19 : SUCCESS in 50m 05s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/78d33621b5ad409ba8b66f7f036efc82 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":1},{"id":"a64cbc4fd322179d2a2dcaf3726b0cb2670f2857","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-11-12 12:29:00.000000000","message":"Abandoned\n\nAbandoned because new microversion will need new tests","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8f29d4a28595181cd7f9ed6fc8c80ee4ec461c14","revisions":{"8f29d4a28595181cd7f9ed6fc8c80ee4ec461c14":{"kind":"REWORK","_number":1,"created":"2024-07-23 11:41:43.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/26/924726/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/26/924726/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/924726/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/924726/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/26/924726/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/26/924726/1"}}},"commit":{"parents":[{"commit":"1af21705c53bc9911ea467eaeee2bc12489a43ed","subject":"Merge \"Add wait for location import task\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/1af21705c53bc9911ea467eaeee2bc12489a43ed"}]}],"author":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-07-23 11:36:09.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-07-23 11:41:34.000000000","tz":120},"subject":"Adjustment for Volume Type Metadata","message":"Adjustment for Volume Type Metadata\n\nVolume Types have a new property called metadata. This change adjusts test,\nto include the new property in responses from the volume type API.\n\nDepends-On: I9d1c75d25c2daac74d1e4f75d9e7cc96cbe56f66\nChange-Id: Ia3e64078943cd000dba2651f2f81dbb6b4d328c6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8f29d4a28595181cd7f9ed6fc8c80ee4ec461c14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/8f29d4a28595181cd7f9ed6fc8c80ee4ec461c14"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
