)]}'
{"id":"openstack%2Ftempest~924957","triplet_id":"openstack%2Ftempest~master~I606cc94ab14022d25d5fb1ecafc5e5f5e701696c","project":"openstack/tempest","branch":"master","topic":"early-failure-detection","attention_set":{},"removed_from_attention_set":{"4146":{"account":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"last_update":"2024-07-26 05:09:59.000000000","reason":"Only change owner, uploader, reviewers, and cc can be in the attention set"},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-07-26 08:34:27.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}},"hashtags":[],"change_id":"I606cc94ab14022d25d5fb1ecafc5e5f5e701696c","subject":"DNM: Test early failure detection","status":"ABANDONED","created":"2024-07-25 17:16:16.000000000","updated":"2024-07-26 08:34:35.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2de11e23cd2b941d0497a4bbb7accb01d9f18a69","_number":924957,"virtual_id_number":924957,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-07-26 07:04:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":0,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-25 18:33:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"047298fa6a52471367470cdf5d34eeb57c0dff9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-25 17:16:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"65cd7e48c222bebf24d620f591f686811ea11e35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 18:33:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5efb93b7f5f94495b8dd6490003cbd66\n\n- grenade https://zuul.opendev.org/t/openstack/build/38701c8bb0f0438eaced632e787d352a : SUCCESS in 1h 09m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5ff3397c63a64304b866e4f8b9e71c84 : FAILURE in 6m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/283b245c8ce44116ae8a92d71d3d1d39 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/567a077316a24b3eac1aed463b10d68c : SUCCESS in 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d041c365ef5f425d8f6ad2ea525f6b63 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3fbe5cc805a04347ba1604f5859c52f8 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e005742f28c4964a9b3827042aa1e85 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1cd7ce0c36f42a1a0d9a6fd767f5b5c : SUCCESS in 3m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/555d08c950594b14be50fc450118de09 : SUCCESS in 3m 58s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/b294dc989e024e4cb6d5df7f520d4f72 : FAILURE in 4m 29s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/860a74fa33c34b4a954962b9615d5780 : FAILURE in 16m 20s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/686c672e36434d70a90e9d45e5075c3f : FAILURE in 4m 23s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/45899ab722914e728f7c9f85c37f2640 : FAILURE in 15m 49s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/1042bc43d25c465381ffbb9fa3ac10aa : FAILURE in 5m 57s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/6d32cd6b2f3a489e96b8afa8ba1e55e3 : SUCCESS in 25m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d88d0952b6cb44639c28a83b580bbfe5 : SUCCESS in 59m 34s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/96c531bdfcb542acb5a7150e89a73277 : FAILURE in 18m 26s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/e81388dd26824ea8abb36d93325d81c8 : POST_FAILURE in 16m 29s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bf040270e4364cbeb50b89c3d57c54f9 : FAILURE in 14m 14s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/1006b3d6e43c4565b8d70f4167b5a779 : FAILURE in 5m 00s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/9460310061684e62847bc2dc8c61d688 : SUCCESS in 57m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/08d6b3806bca4aa6860458c5d5b1e59a : SUCCESS in 1h 05m 58s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/2f2064a375434b8aa5c146932bd335db : FAILURE in 15m 17s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/941feaff8e19450e8d2c0723546af82e : SUCCESS in 2m 59s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/06f2eaba60934bfe9a05f6d9965d340f : FAILURE in 6m 28s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/b8797ae6f02b4c5482315475c1686f13 : FAILURE in 4m 54s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/6cd1d1f5b0e5438994588eab6bb763f8 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"34e28a8ec36757d1f88ad559d15a8ffdc4c8902b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-25 18:39:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9014c832dfbd7701dfbc2d2b018bc7089a8c358b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 20:39:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcb43afec4cf43b880c8c731dbc634ea\n\n- grenade https://zuul.opendev.org/t/openstack/build/08a13f32fc8e4eaea19b3b8824059b59 : SUCCESS in 1h 06m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/362ce61889144fdda4180910533b077e : SUCCESS in 1h 40m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d25b011d1332412f9a972f0e73055c87 : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fd187bf8f70468f859cd0e055dc89ab : SUCCESS in 17m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3ed88b769c04ddf8204a8fc98f21476 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/516f26aa018b4a2eb7d1769943af70c6 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dbe39b722b394cbe962e2fa045a39850 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d42b2212fa8f4e2395cabde43f0d74c0 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2e1f7e21da684c859bea6b8caa85e9ae : SUCCESS in 5m 59s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/fccba458e3db4d208ea0053e390a7019 : SUCCESS in 34m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/684d8666cf1b443baa1f41842e0cf0e3 : SUCCESS in 1h 35m 28s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/a8f4a3dba2f14e0cb128344d45d80e61 : SUCCESS in 1h 39m 35s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/24c3b38fbfb042b18efb6f43033a44f6 : SUCCESS in 1h 22m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/e17c22717eaf47e5b3c6d9cb10edb868 : SUCCESS in 1h 07m 55s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/5938ffef69074ef5ac213c2abef56222 : SUCCESS in 25m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c22ac22273f7497bb3d3023f8ea473a9 : SUCCESS in 55m 58s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fc120f658822421eb049ef19fc8f0138 : SUCCESS in 1h 46m 55s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/76e37c914b4b4969a8006dbe769379e1 : SUCCESS in 48m 25s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5d89c05482614f2a8e36c191c538bd26 : SUCCESS in 1h 39m 42s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/98e63b5a2ac8430b9c780914618b0db5 : SUCCESS in 1h 48m 36s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/bda4ff7d6eed4a5e891682b0700dcb5d : SUCCESS in 57m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2da87561c839428cac86b598cc82a2c7 : SUCCESS in 1h 05m 25s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/254085aab3c64b8185f0268df287f219 : SUCCESS in 1h 12m 30s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/131dbd386205421eb83d6a4a220c42c4 : SUCCESS in 3m 19s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/c5d90b4c27ec4bd0b4f988702476ae1a : FAILURE in 1h 12m 10s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/21c14134f7624c00ab126faf0ebf4720 : SUCCESS in 55m 43s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/3ef6758235744cfea38208fff01516d9 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"ddc97060585b1a4d11df7a7db00d4c91093b17ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2024-07-25 21:21:07.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c200303b45305da7c358aab19064b2a5ebe76cdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 23:09:54.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b68f3aba777d423cb1f33da5acaf064e\n\n- grenade https://zuul.opendev.org/t/openstack/build/0e526006e62b4f21abcfbf5053f6279f : SUCCESS in 45m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/274a567ab9f34edf9d1d20d8c9c22b24 : FAILURE in 1h 43m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e992cbec0fce4fbcb7e03158c781e924 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b98f1bd1877a4a87a027fef9e5d1d9ba : SUCCESS in 10m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9d8997ee0a4136812a1b119e7c02d6 : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4afdcdab6be40e1b9f25ed371c0381f : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/55d23c390d7746beab546b99810f4981 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f594d686c2044cf906a6378281e975f : SUCCESS in 3m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51b0d4eaa19f46efad7dd073f3d96a81 : SUCCESS in 4m 00s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/28d3fec8adb74b6d9bd16e22d1178eff : FAILURE in 55m 13s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/083bd97b4aa348e0b8b7aa8559dbc0d8 : FAILURE in 1h 46m 21s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/0427a00b3d5d493a8954ba88dcd8bdcf : FAILURE in 1h 36m 10s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/8c991b969b63493989236e48db5fa477 : FAILURE in 1h 11m 03s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/b539956263674836a53654e1af9fa204 : FAILURE in 58m 57s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/d1d9e91bf86345b1aa543ea7bcca4c1d : SUCCESS in 28m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c1793351f1240e6987de7406b7a9074 : SUCCESS in 45m 25s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/daa7915e739a40269b51b3c6f2f8f87b : FAILURE in 1h 18m 53s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c69c5546f3c448d294f933aa4dae934d : FAILURE in 40m 34s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/df6fa37811ab43c9b318e074d873b57c : FAILURE in 1h 24m 49s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/432c090ee11b4c95b8a904d9e914bcba : FAILURE in 1h 34m 10s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/c882feba47544757ab60838bbcd5e616 : SUCCESS in 43m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/450d2c08d7544ed1bbb102b4d4f4f723 : SUCCESS in 43m 50s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/162ee3e966c4407cb90f68d43594a93d : FAILURE in 1h 01m 34s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/42e90a930b834a18bcd58ccacb28c105 : SUCCESS in 3m 14s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/ae12d7104ab34a52a87185b71e3fd2a7 : FAILURE in 1h 12m 06s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/318f29c2487248e396560490a689f895 : FAILURE in 41m 27s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/a56459de5ba64edab7656dee1465c479 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":3},{"id":"f104eb060d35d64aeeb8699bd05396eb002fd2f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-26 05:06:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3829d7f94a757f7e040f31deac41941dc7d82d8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 07:04:49.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a77b2ed33ec44524bcad62641cd32e87\n\n- grenade https://zuul.opendev.org/t/openstack/build/4fed3ecd84804ca3be9c73a8522814a6 : SUCCESS in 58m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d646347e4ecb43a7a2f1553ec204c9bb : SUCCESS in 1h 43m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/33bb894202b447b2bd34217edc105f45 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f44e8fc6f61412c87066bf675e9cf36 : SUCCESS in 11m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1fbf8ec205244348f6194e8b1f1f5e2 : SUCCESS in 6m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5b37185be3a4d218e0896c59163ba4c : SUCCESS in 7m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7883b2009a2b46c0b894c7d048eb00fb : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f58d43e726874eb8b219251d757567b9 : SUCCESS in 6m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/40590b8be2b14654b424399bbc9b4546 : SUCCESS in 6m 12s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/331bd06c18d746009bd76d9fca4f40d6 : SUCCESS in 37m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/52972911c60345048d2ec2086ba7f5ba : SUCCESS in 1h 55m 58s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/b431b3121cf54e1c81b971c306c2bfd9 : SUCCESS in 1h 14m 59s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/7f3bd80d9c10421c97a84e99ef92ce0e : SUCCESS in 1h 02m 19s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/712ab9e018d84f6799fdbed84d5a7d4f : FAILURE in 1h 06m 58s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/30deeeed74264b7cbb3e72419cb74fe9 : SUCCESS in 21m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a389e3b5de044c6e801599bf5cdb44b4 : SUCCESS in 39m 02s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/39cbb865e8ec42369c7ea69b4c1ea8b7 : SUCCESS in 1h 21m 54s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/55ba6bebacbd4af4bc624cb4c254b936 : SUCCESS in 49m 20s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/54277024c1ad4395bde39d99d034a99e : SUCCESS in 1h 45m 06s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/046f86854dea40e1b8dafee35fa148b8 : SUCCESS in 1h 45m 44s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/f6e50179cf3e4b869656fa9f7c80434c : SUCCESS in 1h 03m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/71131e706d3c4480b6171d7f6b288876 : SUCCESS in 1h 05m 34s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/723f1e00ce084dc2821427d21426ab02 : SUCCESS in 1h 12m 26s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/76e0539aff6b4a7098a7390a763a87b5 : SUCCESS in 3m 33s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/5eb3b0a4916f4b2ab103e97bd90a5f6d : FAILURE in 1h 14m 42s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/30517a5cc7d44fd0830e3feb9c5c1b3c : SUCCESS in 1h 00m 01s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/ace9261af58c43e291872c9a81d04237 : SUCCESS in 2m 38s","accounts_in_message":[],"_revision_number":4},{"id":"c372a665cb3120b48a33931a125e3e3d828ccf4b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-26 08:34:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2de11e23cd2b941d0497a4bbb7accb01d9f18a69","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-26 08:34:35.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3032e72004ffe951547e935dce610cdc5ed9efe4","revisions":{"d84ec014555cd32c046887a08c72bede6ce84df0":{"kind":"REWORK","_number":1,"created":"2024-07-25 17:16:16.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/57/924957/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/57/924957/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/924957/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/924957/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/924957/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/57/924957/1"}}},"commit":{"parents":[{"commit":"4009355e7106e458567b86d4aee76682bcbdf0c8","subject":"Add early failure detection for tempest jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/4009355e7106e458567b86d4aee76682bcbdf0c8"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-25 17:15:51.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-25 17:15:51.000000000","tz":120},"subject":"DNM: Test early failure detection","message":"DNM: Test early failure detection\n\nChange-Id: I606cc94ab14022d25d5fb1ecafc5e5f5e701696c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d84ec014555cd32c046887a08c72bede6ce84df0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d84ec014555cd32c046887a08c72bede6ce84df0"}]},"branch":"refs/heads/master"},"d7d1220a8fa109c7bcd787a9ae6bc864a0f8661c":{"kind":"REWORK","_number":2,"created":"2024-07-25 18:39:20.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/57/924957/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/57/924957/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/924957/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/924957/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/924957/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/57/924957/2"}}},"commit":{"parents":[{"commit":"a4a48b2f299ed9a507fb7f512c6b0a26c77f621d","subject":"Add early failure detection for tempest base job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a4a48b2f299ed9a507fb7f512c6b0a26c77f621d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-25 17:15:51.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-25 18:38:41.000000000","tz":120},"subject":"DNM: Test early failure detection","message":"DNM: Test early failure detection\n\nChange-Id: I606cc94ab14022d25d5fb1ecafc5e5f5e701696c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d7d1220a8fa109c7bcd787a9ae6bc864a0f8661c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/d7d1220a8fa109c7bcd787a9ae6bc864a0f8661c"}]},"branch":"refs/heads/master"},"cc272df3e97eaa4cee6e62aa891b3b964a0c98e5":{"kind":"REWORK","_number":3,"created":"2024-07-25 21:21:07.000000000","uploader":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"ref":"refs/changes/57/924957/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/57/924957/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/924957/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/924957/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/924957/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/57/924957/3"}}},"commit":{"parents":[{"commit":"a4a48b2f299ed9a507fb7f512c6b0a26c77f621d","subject":"Add early failure detection for tempest base job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a4a48b2f299ed9a507fb7f512c6b0a26c77f621d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-25 17:15:51.000000000","tz":120},"committer":{"name":"Clark Boylan","email":"clark.boylan@gmail.com","date":"2024-07-25 21:21:00.000000000","tz":-420},"subject":"DNM: Test early failure detection","message":"DNM: Test early failure detection\n\nChange-Id: I606cc94ab14022d25d5fb1ecafc5e5f5e701696c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cc272df3e97eaa4cee6e62aa891b3b964a0c98e5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/cc272df3e97eaa4cee6e62aa891b3b964a0c98e5"}]},"branch":"refs/heads/master"},"3032e72004ffe951547e935dce610cdc5ed9efe4":{"kind":"REWORK","_number":4,"created":"2024-07-26 05:06:30.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/57/924957/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/57/924957/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/924957/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/924957/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/57/924957/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/57/924957/4"}}},"commit":{"parents":[{"commit":"a4a48b2f299ed9a507fb7f512c6b0a26c77f621d","subject":"Add early failure detection for tempest base job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/a4a48b2f299ed9a507fb7f512c6b0a26c77f621d"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-25 17:15:51.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-26 05:06:13.000000000","tz":120},"subject":"DNM: Test early failure detection","message":"DNM: Test early failure detection\n\nChange-Id: I606cc94ab14022d25d5fb1ecafc5e5f5e701696c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3032e72004ffe951547e935dce610cdc5ed9efe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3032e72004ffe951547e935dce610cdc5ed9efe4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
