)]}'
{"id":"openstack%2Ftempest~927831","triplet_id":"openstack%2Ftempest~master~I4e8fe7650f088ecbbbb279c35f8931c71de4ece1","project":"openstack/tempest","branch":"master","attention_set":{"22873":{"account":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"last_update":"2024-09-04 14:26:01.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}}},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-09-04 00:29:39.000000000","reason":"\u003cGERRIT_ACCOUNT_8556\u003e replied on the change","reason_account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}}},"hashtags":[],"change_id":"I4e8fe7650f088ecbbbb279c35f8931c71de4ece1","subject":"Use stable constraint in tox to release Tempest new tag","status":"MERGED","created":"2024-09-03 14:13:26.000000000","updated":"2024-09-06 19:42:37.000000000","submitted":"2024-09-04 08:20:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"927831","meta_rev_id":"0f808d727c3977f882fc1221bd64db1b18a6e28b","_number":927831,"virtual_id_number":927831,"owner":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-04 08:20:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"recommended":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"all":[{"value":1,"date":"2024-09-06 19:42:37.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":1},"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":2,"date":"2024-09-04 00:29:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"all":[{"value":0,"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":1,"date":"2024-09-04 00:29:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"all":[{"value":1,"date":"2024-09-06 19:42:37.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Release Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}],"REVIEWER":[{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-03 16:43:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-03 17:19:03.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"CC"},{"updated":"2024-09-03 17:19:07.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2024-09-04 14:26:01.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"},{"updated":"2024-09-06 19:42:37.000000000","updated_by":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"reviewer":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"state":"REVIEWER"}],"messages":[{"id":"6791c79764db8a6d170adf4c8d5aa9e0a25c4a34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-09-03 14:13:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"779bf880f9d17f0cde9811bf552654d1e20590f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-03 16:43:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a9985e2417048dda423c8d8eca77338\n\n- grenade https://zuul.opendev.org/t/openstack/build/a5cdf90adb634396ba9e940fa9d01391 : SUCCESS in 1h 01m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7e8b8f4592e04a90960cff9f4d1cb030 : SUCCESS in 1h 37m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cb8c5f417fc4fd98d2fac2ef0f473de : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17cc0d23c93f4908b8b74ec6620b0433 : SUCCESS in 12m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e6fe5ab271f4e0ea0049c7ef9bf3de1 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/907ea3678d244dfdae41f7c399816e56 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c464647fd4e84378983bfa1725d92875 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/292152f7228642acbbdde5df0048d0ec : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/82553a3c1b1e4d79a330602a434b236d : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/633ead4ff3e54a6abf874cc9518f8d3d : SUCCESS in 4m 08s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/61089916cd3d42cd85b6c536b3971908 : SUCCESS in 50m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bc0144ddc408404aa7aeafe4421dbfcb : FAILURE in 2h 06m 19s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/fc159412ba4c41ea93dc1d7c3092c2e1 : SUCCESS in 1h 39m 16s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/f91a845f79e449209ef26b05d873163d : SUCCESS in 1h 48m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/7421ef7b4020438696cf782732f12b1f : SUCCESS in 52m 58s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/f831e9609c3b4b00ba7b254b6d3126c7 : SUCCESS in 27m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af032a58ea024c01ac5590dacd8054e7 : SUCCESS in 1h 05m 33s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/babfd4e8c8c34a04986b88baf8de27e8 : SUCCESS in 1h 36m 38s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/70c366981b224a38a8f901419cfead61 : SUCCESS in 44m 09s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e9acff896e6d422aa715fc06bd579a8b : SUCCESS in 1h 40m 16s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/d54a7bcb328f49f2b354073d95d85c38 : FAILURE in 1h 58m 31s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/2cc59e9d6c5d4f40a0581016216eee02 : SUCCESS in 55m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b150eb8ada2049cfbd8b1bc06d3254b0 : SUCCESS in 1h 09m 54s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/9cf5260391804d4985fb0a542f7c0ad7 : SUCCESS in 1h 05m 34s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/8970bed08ff44fd493db0ab4d632e7fa : SUCCESS in 4m 16s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/328399645c0f47d5ac6be153e351d715 : FAILURE in 1h 20m 37s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/22ad39aee7a8427888ae769151341d66 : FAILURE in 37m 41s\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/54ab490710ea48f7a357befa6c140e6d : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4ff946433c1a0b8151c54f7a535e3f9f9005b977","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-03 17:19:03.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d475ebf09fd39b2ff395e61ba5781d47fa565250","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-03 17:19:07.000000000","message":"Patch Set 1: Code-Review-1","accounts_in_message":[],"_revision_number":1},{"id":"f3f94d2732fc417f1398c1add0cca1723609bc0c","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-09-03 21:45:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f0ca820f5edcb8b3a4da3fabf5cd68340d3b9be","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-09-03 21:45:37.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3cd27cfba67a7dd07f19a899a84dbb7359fab6cf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"date":"2024-09-03 21:47:40.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"396b5092d6bf79a263c633c58c3a42c2d1b0761f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 00:10:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aed57e962c7644c190ca4d250224f627\n\n- grenade https://zuul.opendev.org/t/openstack/build/e4cda3997ea14f32a1e52e20496f56d0 : SUCCESS in 42m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/363b9eb32ee1401688ccd743ea7a8128 : SUCCESS in 1h 20m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/715c80b693f64687bccdfff7d8ddce5d : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b585b5443a0454e8fc1de674ea6170b : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/365eb12575064bf685de9a5e8a4b645c : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ba3e36b59c94d2cbdcb44561b2186c3 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90f61f2065984b398f81650661950857 : SUCCESS in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77802f064eee46f79769ad43373b598e : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/701daf88ff744f358aa0ec11643390af : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/357f57fb271e42f3b0ad13a8c3dc5311 : SUCCESS in 3m 47s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/658462fff005434da39da7cdd72313c3 : SUCCESS in 52m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d666787a58834672947bf9ff16ce6bca : SUCCESS in 2h 21m 20s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/c65fd4eba2a74f3f88258f05d9ef536c : SUCCESS in 1h 41m 00s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/d01b4c741476417da4d9e08f1c3ae46d : SUCCESS in 1h 28m 26s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/45423095c8d9489481e6b1557aed394d : SUCCESS in 57m 16s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/e82cd7fde04e4b419cbe4071748d38f2 : SUCCESS in 30m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2c7fc8623da944af8c1e4925f90c9520 : SUCCESS in 41m 48s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/401095e254de4943aaab66a4bc68e4c4 : SUCCESS in 1h 53m 32s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/2702f12fb23746fa8f7aea2c2d47a031 : SUCCESS in 37m 55s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f3db5688483246f9a7e24ce72915fe1b : SUCCESS in 1h 40m 24s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/6e70d23602484b13b361b679ac9cedad : FAILURE in 1h 39m 44s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/089bb417c21640619f2270e619cebeb4 : SUCCESS in 47m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6805c7e3f06a45d2b3ddc761e49e8ed9 : SUCCESS in 1h 06m 20s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/708f3c9471ea4ba88cfeb15e69fcc1b5 : SUCCESS in 1h 18m 18s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/e6d4906d49f24f2c86023765d1212f0d : SUCCESS in 3m 55s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/b296357f47564139960c0f2c7bc49ad8 : FAILURE in 56m 35s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/4b381a0786b141fbab0cc63ee841cd55 : FAILURE in 34m 31s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1bc98d4a9c2748a8952404c692d79685 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":3},{"id":"94b6d0fea7b51f4dfac17d55fc7bf25713769d72","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-04 00:29:39.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8910208f01454a161e2c30aa61d30317da445a63","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 00:30:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b277cfd7cd361d2f48a67c9949d2277f7de53e41","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 02:15:03.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bda7da6b6bc64396ad68aac825772d3d\n\n- grenade https://zuul.opendev.org/t/openstack/build/d674af06f52643dd9fd16d591a49de61 : SUCCESS in 1h 04m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9bc774c043ba4eff970f861997879a1f : SUCCESS in 1h 29m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77e2da1a525f49d09d4646df381ca1fa : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/efa07e121e1547e6ade8b2aa8d10c5d8 : SUCCESS in 5m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6c3ecf005074b41bd27c113713e53ef : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/461c863975ab4a23a9e9669ee8581134 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54335c09e7b14bc2af7c5e2c9ad6ce77 : SUCCESS in 3m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a5735ea81e74cb185134ddc0aaf12af : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2dbc6af3d974455db90b3cffda3f75d8 : SUCCESS in 4m 51s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/168cf3f59fa148f4ac3fa6c53c0f0589 : FAILURE in 1h 23m 22s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/135e726ab9d344aab56c509a766163fb : SUCCESS in 59m 05s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/2870ea5584f8441da578e3001038cbe7 : SUCCESS in 50m 36s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/000da26f01234d02a0c0a1bf74a3829d : SUCCESS in 1h 01m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fb86722bd8b246799a5d4ac449f2c821 : SUCCESS in 55m 27s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/fffe4f9a01bc43509d520cd8294a4413 : SUCCESS in 53m 06s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/93cd218846e7493c910867920f56ccb1 : SUCCESS in 1h 39m 06s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/254736e5cc4448f2b39050701c97e42b : SUCCESS in 46m 18s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/1466fda67f7c4a0698fbb0f4481a96f0 : FAILURE in 33m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1dd1d623701051f21a51ca6116cbdb4d81e5d54e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-04 03:50:54.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0e3b1325c1264d0609fbc8b841b7b92fd2aa8470","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 06:35:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0d6252d599c4fdb87aa49c0afc2cd66\n\n- grenade https://zuul.opendev.org/t/openstack/build/67951802936a420596694a2887c32726 : SUCCESS in 55m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/375984f07ce346eb9c3b5d8fb21f9596 : SUCCESS in 1h 42m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88279766b4c645b3962d99dd38406397 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28187c164aca4fc2a41bbb2836466ed8 : SUCCESS in 10m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfdfaf48846c4d93b86848d35e705cba : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2e652a5e8ba45bf8ea47a4f96fbda8b : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4d20af238fff4e7eb757420f5dfaaef2 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/008f5c41997c46ef9bfd9c75cdcde9e3 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed9584a9f6394d08a6476fd8e0ba74b7 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/31afb82a30bb4e878923579d9e990cf3 : SUCCESS in 4m 43s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/277881f1494e4780a02efdab09330376 : SUCCESS in 1h 07m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e1853aa7e05b454685b6e87f262a30bb : SUCCESS in 2h 43m 11s (non-voting)\n- tempest-full-2024-1 https://zuul.opendev.org/t/openstack/build/280cf94527744d65bc6e527d44cca3ac : SUCCESS in 1h 30m 02s\n- tempest-full-2023-1 https://zuul.opendev.org/t/openstack/build/88579f6233884729a0ea691eb35ac3d8 : SUCCESS in 1h 22m 21s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/d678017ae8744b11857b1e9e9582ded1 : SUCCESS in 1h 13m 00s\n- tempest-tox-plugin-sanity-check https://zuul.opendev.org/t/openstack/build/818bad9870b34c369b2c0a28e9e1af07 : SUCCESS in 20m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61683846a277408a92aae59d44b6cad5 : SUCCESS in 52m 31s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/54f4ee62bcc94551b4c6de879a1a312a : SUCCESS in 1h 23m 51s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/c27dec215704419a8dad7081955f06bd : SUCCESS in 40m 36s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0de43ab16e4044f0a43dd0d3da56c638 : SUCCESS in 1h 35m 16s\n- devstack-plugin-ceph-tempest-py3 https://zuul.opendev.org/t/openstack/build/eb399f6776a54d81b42aa2b1be1c552c : FAILURE in 1h 43m 36s (non-voting)\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/8cd134906b7045f59e6cce27ce4333bd : SUCCESS in 1h 00m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4020ff4564234fabbb4423505aaa70c1 : SUCCESS in 46m 03s\n- neutron-ovs-tempest-dvr https://zuul.opendev.org/t/openstack/build/627963ad83364db6b904c04553a2ac35 : SUCCESS in 1h 12m 34s (non-voting)\n- interop-tempest-consistency https://zuul.opendev.org/t/openstack/build/a638a1d5da5f43689088dd886876de61 : SUCCESS in 4m 36s\n- tempest-full-test-account-py3 https://zuul.opendev.org/t/openstack/build/276c726f05e145739c45251e810b39a2 : FAILURE in 1h 01m 49s (non-voting)\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/18a9835a63dc48d79d6aba2340f45dc9 : FAILURE in 36m 28s (non-voting)\n- openstack-tox-bashate https://zuul.opendev.org/t/openstack/build/1a847e3ddfc74a9b9dd1433dcc487cf6 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":3},{"id":"30e4d0e2d28ee3ab7390f43b3a32bc8f06662a3a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 06:36:16.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"ce85d2c2f2a6771622fb5f16f1846638ecba1068","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 08:20:36.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db40bac1946845ac9ad500c984c0b1e1\n\n- grenade https://zuul.opendev.org/t/openstack/build/884cb5e5a5c84a5cab01e7783aaaeb1f : SUCCESS in 1h 13m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2cb49d7200c542a59de757832c9690fe : SUCCESS in 1h 39m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4832314fb44640fc9add86cd8353d537 : SUCCESS in 10m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4a50dedea6444998351dab83d15cc6f : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e503b69daec44baaac2490fabb3cdb6b : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1fd9c7522474c9d8799bb33e9066989 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28a2875f6cc84ecb82e3932fbdb1ac0f : SUCCESS in 4m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4c142c528354d9b8a5dce88fd1949f2 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4b0bfaf87a414fb2bf6dd8b00923062f : SUCCESS in 4m 05s\n- tempest-slow-py3 https://zuul.opendev.org/t/openstack/build/fb084ba96c194abb8c700d8b746565bc : SUCCESS in 1h 25m 09s\n- neutron-ovs-grenade-multinode https://zuul.opendev.org/t/openstack/build/72e8c8f2570a4f22820a33d546d2593f : SUCCESS in 54m 27s\n- tempest-extra-tests https://zuul.opendev.org/t/openstack/build/5a26da0f9d1d4c519341fb6e0e7f1400 : SUCCESS in 59m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c6d7bdf06a3642aea5ebf237d6a19e2f : SUCCESS in 57m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a4706535c1394cfb8cd13ed611035d3b : SUCCESS in 59m 46s\n- tempest-multinode-full-py3 https://zuul.opendev.org/t/openstack/build/ab3eada7f845438f9ec928dbeba04510 : SUCCESS in 1h 15m 39s\n- tempest-full-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ad7a779999f141fc889e1123e55763bc : SUCCESS in 1h 33m 50s\n- nova-live-migration https://zuul.opendev.org/t/openstack/build/09c1c50b94164e53a8fffe3cb9b2cd07 : SUCCESS in 45m 09s\n- ironic-tempest-bios-ipmi-direct-tinyipa https://zuul.opendev.org/t/openstack/build/b94bbf3ae1b6448d8420800c5a693d13 : FAILURE in 28m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7987f8311ba21189111d657db9747018e44e03de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 08:20:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"5e1abda21eefe4f089943d5324894daf1d26e25d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 08:23:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c0d99548f2f4868a2ea0cb5623e944d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b410efac5b4439a32ad8ca20e84a7a : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3088a43cbfe0467a8d68e9b3a736dbcf : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3},{"id":"0c3e065d439b85f5fd1794d8c7252b83b1346eb5","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-04 14:26:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"46c3e35a1420c0ce457c2f175b3111d3ee7f99ea","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-04 17:50:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"98d87c86d55688aaf660ade5a6e56b18e4d87e4c","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2024-09-05 06:39:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0f808d727c3977f882fc1221bd64db1b18a6e28b","author":{"_account_id":15527,"name":"foundjem","display_name":"Armstrong","email":"foundjem@ieee.org","username":"foundjem"},"date":"2024-09-06 19:42:37.000000000","message":"Patch Set 3: Code-Review+1 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b7ab42af303ba13a5e3b57a7fd7a6502f69bb01f","revisions":{"b711313a793c0b9726261b3c0d01405c4729207b":{"kind":"REWORK","_number":1,"created":"2024-09-03 14:13:26.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/31/927831/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/927831/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/927831/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/927831/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/927831/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/927831/1"}}},"commit":{"parents":[{"commit":"3ba975fce29779f779b0f48ea1e0c422a35f6821","subject":"Add release notes for 2024.2 cycle intermediate release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3ba975fce29779f779b0f48ea1e0c422a35f6821"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-09-03 14:09:37.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-09-03 14:09:37.000000000","tz":120},"subject":"Use stable constraint in tox to release Tempest new tag","message":"Use stable constraint in tox to release Tempest new tag\n\nWe are going to release Tempest new tag which work with\nthe stable/2024.2 constraint so use that in the tox env.\n\nFor example: if anyone use Tempest new tag in future say\n1 year later then tox env also should use the 2024.2 contstraint for\ncompatibility instead master.\n\nThese need to be move back to master constraint once release is done.\n\nChange-Id: I4e8fe7650f088ecbbbb279c35f8931c71de4ece1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b711313a793c0b9726261b3c0d01405c4729207b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b711313a793c0b9726261b3c0d01405c4729207b"}]},"branch":"refs/heads/master"},"431054d377faa28e11f2e5a2566b959203685585":{"kind":"REWORK","_number":2,"created":"2024-09-03 21:45:37.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/31/927831/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/927831/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/927831/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/927831/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/927831/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/927831/2"}}},"commit":{"parents":[{"commit":"3ba975fce29779f779b0f48ea1e0c422a35f6821","subject":"Add release notes for 2024.2 cycle intermediate release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3ba975fce29779f779b0f48ea1e0c422a35f6821"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-09-03 14:09:37.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-09-03 21:43:50.000000000","tz":120},"subject":"Use stable constraint in tox to release Tempest new tag","message":"Use stable constraint in tox to release Tempest new tag\n\nWe are going to release Tempest new tag which work with\nthe stable/2024.2 constraint so use that in the tox env.\n\nFor example: if anyone use Tempest new tag in future say\n1 year later then tox env also should use the 2024.2 contstraint for\ncompatibility instead master.\n\nThese need to be move back to master constraint once release is done.\n\nChange-Id: I4e8fe7650f088ecbbbb279c35f8931c71de4ece1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/431054d377faa28e11f2e5a2566b959203685585"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/431054d377faa28e11f2e5a2566b959203685585"}]},"branch":"refs/heads/master"},"b7ab42af303ba13a5e3b57a7fd7a6502f69bb01f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-09-03 21:47:40.000000000","uploader":{"_account_id":22873,"name":"Martin Kopec","email":"mkopec@redhat.com","username":"mkopec"},"ref":"refs/changes/31/927831/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/tempest","ref":"refs/changes/31/927831/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/927831/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/927831/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/tempest refs/changes/31/927831/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/tempest refs/changes/31/927831/3"}}},"commit":{"parents":[{"commit":"3ba975fce29779f779b0f48ea1e0c422a35f6821","subject":"Add release notes for 2024.2 cycle intermediate release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/3ba975fce29779f779b0f48ea1e0c422a35f6821"}]}],"author":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-09-03 14:09:37.000000000","tz":120},"committer":{"name":"Martin Kopec","email":"mkopec@redhat.com","date":"2024-09-03 21:46:32.000000000","tz":120},"subject":"Use stable constraint in tox to release Tempest new tag","message":"Use stable constraint in tox to release Tempest new tag\n\nWe are going to release Tempest new tag which work with\nthe stable/2024.1 constraint so use that in the tox env.\n\nFor example: if anyone use Tempest new tag in future say\n1 year later then tox env also should use the 2024.1 contstraint for\ncompatibility instead master.\n\nThese need to be move back to master constraint once release is done.\n\nChange-Id: I4e8fe7650f088ecbbbb279c35f8931c71de4ece1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b7ab42af303ba13a5e3b57a7fd7a6502f69bb01f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/tempest/commit/b7ab42af303ba13a5e3b57a7fd7a6502f69bb01f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
